)]}'
{"id":"openstack%2Fcinder~997591","triplet_id":"openstack%2Fcinder~master~Ie8ee7c1d8c2f0546ae7f33e3b7da7a7d05a6613d","project":"openstack/cinder","branch":"master","topic":"eventlet-removal","attention_set":{"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2026-08-13 16:23:01.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ie8ee7c1d8c2f0546ae7f33e3b7da7a7d05a6613d","subject":"WIP: Enable H905: no-eventlet","status":"NEW","created":"2026-07-16 14:45:51.000000000","updated":"2026-08-13 16:30:46.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f877f5726eebccea869c7d873560ac1117046e79","_number":997591,"virtual_id_number":997591,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-08-13 16:23:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-16 14:56:55.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2026-07-16 15:39:30.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-08-13 14:31:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-08-13 16:23:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-13 16:30:46.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"4f09c9f95086c24635c0ce5134ee56b3130c8abf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-07-16 14:45:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6891fc9b44802833c529bb88c37dbbd94f4e7ba0","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-16 14:56:55.000000000","message":"Patch Set 1:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4814/.","accounts_in_message":[],"_revision_number":1},{"id":"b8057cbb4aa895726ef9833d68107ff3fb410369","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-16 15:39:30.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ie8ee7c1d8c2f0546ae7f33e3b7da7a7d05a6613d .","accounts_in_message":[],"_revision_number":1},{"id":"ef10fb81241e57a6b3cfa49c7370f26f8a70b673","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-18 09:16:32.000000000","message":"Patch Set 1:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4638/.","accounts_in_message":[],"_revision_number":1},{"id":"5017b5de6c688fa99d3c7951ba0890a82ff3628f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-08-13 13:36:37.000000000","message":"Topic set to eventlet-removal","accounts_in_message":[],"_revision_number":1},{"id":"f826d6eb3fe1f344b588e8df589e03376dc3abe2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-08-13 14:02:51.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"703b8e10affc34c400efbe467d7e3861705006e4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 14:31:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d062f94b69f24c0dac11cabe1017fd99\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a5b6b3d6d10e4ea5aef73b764b4cc94d : SUCCESS in 17m 46s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3bffec4a9c334ddcb8b0a54631d346b6 : SUCCESS in 10m 54s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/8426bb4cc0a14569a0249323507fd81b : SUCCESS in 22m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"92714f5b7acad33ed19c33b0bc758c92c972e2bf","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-13 15:01:27.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ie8ee7c1d8c2f0546ae7f33e3b7da7a7d05a6613d .","accounts_in_message":[],"_revision_number":2},{"id":"58ef7e6024af6bc522bd6e175490414b0986c73a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-13 15:59:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4935/.","accounts_in_message":[],"_revision_number":2},{"id":"a9b8f61a10de2bdeef464027f3cfccb116496ef8","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-13 16:06:51.000000000","message":"Patch Set 2:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5111/.","accounts_in_message":[],"_revision_number":2},{"id":"6c9d15bae4d30bb259e46b7a9cea356ef1b09a3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 16:23:01.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1aac540b58d649979af6889561228dcf\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/1429d9ffb36043479d6e9439e366979c : SUCCESS in 5m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/023a9128dea7489da5ba70cafabcf416 : FAILURE in 4m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/71b8f561e2cf4c3c94ad9d5e6fe84622 : SUCCESS in 10m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/390aa040f16f414abde992e8b7124e16 : SUCCESS in 10m 31s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8332ebe12c1048468c09fad75a65ea3b : SUCCESS in 13m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5bfc446568848649e61b5d406d9fc2d : SUCCESS in 14m 23s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/15c0fd58f1884d69b1743838c8e44c98 : SUCCESS in 52m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/83549aa87aad4f6cb66a679cf9b5bdad : SUCCESS in 9m 35s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e7ada1be60ef4cd0a342d88b35e46385 : SUCCESS in 3m 44s\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/4d4c3a3df3ae4bfebeb1b9862653b021 : FAILURE in 8m 55s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/578048f2a9b2420eb009359de593b8dc : SUCCESS in 6m 57s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/945015f32a414b23ba9cdc5f207baee9 : SUCCESS in 5m 31s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/667300a840f44270ac94f168bf01449d : SUCCESS in 9m 41s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f9ec9f7f6fce4f1884e0002a29c0573a : FAILURE in 35m 11s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/cd7ef9ebbb4d4d2f9fbaa7e93a31ba37 : SUCCESS in 4m 01s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/88f0b809286f46ed8a0d01029bb757b7 : SUCCESS in 2h 09m 39s\n\nWarning:\n  Comments left for invalid file cinder/cmd/backup.py\n  Comments left for invalid file cinder/image/image_utils.py\n  Comments left for invalid file cinder/manager.py\n  Comments left for invalid file cinder/ssh_utils.py\n  Comments left for invalid file cinder/tests/unit/fake_utils.py\n  Comments left for invalid file cinder/tests/unit/test.py\n  Comments left for invalid file cinder/tests/unit/volume/drivers/ibm/test_ds8k_proxy.py\n  Comments left for invalid file cinder/tests/unit/volume/test_volume.py\n  Comments left for invalid file cinder/utils.py\n  Comments left for invalid file cinder/volume/drivers/datera/datera_iscsi.py\n  Comments left for invalid file cinder/volume/drivers/hitachi/hbsd_replication.py\n  Comments left for invalid file cinder/volume/drivers/ibm/ibm_storage/ds8k_proxy.py\n  Comments left for invalid file cinder/volume/drivers/netapp/dataontap/client/api.py\n  Comments left for invalid file cinder/volume/volume_utils.py\n  Comments left for invalid file cinder/zonemanager/drivers/brocade/brcd_fc_zone_client_cli.py\n  Comments left for invalid file cinder/zonemanager/drivers/cisco/cisco_fc_san_lookup_service.py\n  Comments left for invalid file cinder/zonemanager/drivers/cisco/cisco_fc_zone_client_cli.py","accounts_in_message":[],"_revision_number":2},{"id":"f877f5726eebccea869c7d873560ac1117046e79","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-13 16:30:46.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/2ca26477600947d58c3b5f1e8d309d92 : SUCCESS in 45m 08s\n- cinder-storpool-tempest-iscsi-multipath build/d9b0fa5dd919477cb39295cd5db75714 : SUCCESS in 55m 46s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e610f2aa11a079c75fea8d4619cccac880ac3076","revisions":{"ac3ba498e846a46e4213094da127f674c5854bd8":{"kind":"REWORK","_number":1,"created":"2026-07-16 14:45:51.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/91/997591/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/91/997591/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/91/997591/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/91/997591/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/91/997591/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/91/997591/1"}}},"commit":{"parents":[{"commit":"0829dd157f0f20c8333b95ef537ac38dffacfb51","subject":"WIP: Run Ceph Backup driver in native threads","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0829dd157f0f20c8333b95ef537ac38dffacfb51"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2026-07-16 14:27:38.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2026-07-16 14:45:40.000000000","tz":-240},"subject":"WIP: Enable H905: no-eventlet","message":"WIP: Enable H905: no-eventlet\n\nPrevent re-addition of eventlet imports.\n\nChange-Id: Ie8ee7c1d8c2f0546ae7f33e3b7da7a7d05a6613d\nSigned-off-by: Eric Harney \u003ceharney@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ac3ba498e846a46e4213094da127f674c5854bd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ac3ba498e846a46e4213094da127f674c5854bd8"}]},"branch":"refs/heads/master"},"e610f2aa11a079c75fea8d4619cccac880ac3076":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-08-13 14:02:51.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/91/997591/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/91/997591/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/91/997591/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/91/997591/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/91/997591/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/91/997591/2"}}},"commit":{"parents":[{"commit":"763f958a3b7b4fb00c3963fbd1e5a7f99fd34520","subject":"Run Ceph Backup driver in native threads","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/763f958a3b7b4fb00c3963fbd1e5a7f99fd34520"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2026-07-16 14:27:38.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2026-08-13 14:02:51.000000000","tz":0},"subject":"WIP: Enable H905: no-eventlet","message":"WIP: Enable H905: no-eventlet\n\nPrevent re-addition of eventlet imports.\n\nChange-Id: Ie8ee7c1d8c2f0546ae7f33e3b7da7a7d05a6613d\nSigned-off-by: Eric Harney \u003ceharney@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e610f2aa11a079c75fea8d4619cccac880ac3076"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e610f2aa11a079c75fea8d4619cccac880ac3076"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"0829dd157f0f20c8333b95ef537ac38dffacfb51","ours":"ac3ba498e846a46e4213094da127f674c5854bd8","theirs":"763f958a3b7b4fb00c3963fbd1e5a7f99fd34520","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
