)]}'
{"id":"openstack%2Fcinder~997765","triplet_id":"openstack%2Fcinder~master~Id9be84b11d53dbdfc0d6150bf4beec64dd494df5","project":"openstack/cinder","branch":"master","attention_set":{},"removed_from_attention_set":{"38081":{"account":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"last_update":"2026-09-04 15:00:14.000000000","reason":"Change was submitted"},"37923":{"account":{"_account_id":37923,"name":"anthony gamboa","display_name":"Anthony Gamboa","email":"anthony.gamboa@hitachivantara.com","username":"anthonygamboa","status":"Hitachi Vantara Software Developer"},"last_update":"2026-07-31 17:11:58.000000000","reason":"Only change owner, uploader, reviewers, and cc can be in the attention set"},"9236":{"account":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"last_update":"2026-08-30 18:19:36.000000000","reason":"\u003cGERRIT_ACCOUNT_9236\u003e replied on the change","reason_account":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},"10058":{"account":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"last_update":"2026-09-04 15:00:14.000000000","reason":"Change was submitted"},"38059":{"account":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"last_update":"2026-09-04 15:00:14.000000000","reason":"Change was submitted"},"36171":{"account":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"last_update":"2026-09-04 15:00:14.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id9be84b11d53dbdfc0d6150bf4beec64dd494df5","subject":"[Hitachi] Update version for Hibiscus release / small doc updates","status":"MERGED","created":"2026-07-17 16:05:45.000000000","updated":"2026-09-09 05:16:40.000000000","submitted":"2026-09-04 15:00:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"997765","meta_rev_id":"d4f4fef3cd3a09d599d610c88c0efa88f29aa67b","_number":997765,"virtual_id_number":997765,"owner":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-04 15:00:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"all":[{"value":2,"date":"2026-09-04 13:03:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"all":[{"value":1,"date":"2026-09-04 13:03:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"}],"CC":[{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},{"_account_id":39391,"name":"Rashi Jain","display_name":"HPE AlletraMP NVMeTCP CI","email":"rashi.jain@hpe.com","username":"rashij"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-17 16:10:50.000000000","updated_by":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"reviewer":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"state":"CC"},{"updated":"2026-07-17 16:10:52.000000000","updated_by":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"reviewer":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"state":"CC"},{"updated":"2026-07-17 16:10:53.000000000","updated_by":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"reviewer":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"state":"CC"},{"updated":"2026-07-17 16:16:59.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2026-07-17 17:00:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-17 17:01:28.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-07-17 17:03:39.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-07-17 18:29:31.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2026-07-17 21:15:25.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"},{"updated":"2026-07-18 08:56:08.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2026-07-21 07:02:50.000000000","updated_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"reviewer":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"state":"REVIEWER"},{"updated":"2026-08-06 02:26:53.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-08-21 14:21:36.000000000","updated_by":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"reviewer":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"state":"REVIEWER"},{"updated":"2026-08-21 15:17:12.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2026-08-28 07:48:07.000000000","updated_by":{"_account_id":39391,"name":"Rashi Jain","display_name":"HPE AlletraMP NVMeTCP CI","email":"rashi.jain@hpe.com","username":"rashij"},"reviewer":{"_account_id":39391,"name":"Rashi Jain","display_name":"HPE AlletraMP NVMeTCP CI","email":"rashi.jain@hpe.com","username":"rashij"},"state":"CC"},{"updated":"2026-08-28 17:37:56.000000000","updated_by":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"reviewer":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"state":"CC"},{"updated":"2026-08-28 22:43:09.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2026-08-28 23:00:00.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"050c4d9f6071df6bdd630c878f00b6e305d29478","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-07-17 16:05:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"13fde2ac6ee5e246226cb9c98cf02ce023a1bc56","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-07-17 16:08:10.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"0b432af452b5541694f5d22a9155d1214f180b1b","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-07-17 16:08:13.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"c409c00fa924d0d4077af08cb26dc138388f1473","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-07-17 16:10:50.000000000","message":"Patch Set 2:\n\nHPE AlletraMP iSCSI CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_iSCSi_CI build/904 : FAILURE in 0h 2m 19s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/997765","accounts_in_message":[],"_revision_number":2},{"id":"2bebae4dd4f824fe2fcfd5992dc9a5ea0f192080","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-07-17 16:10:52.000000000","message":"Patch Set 2:\n\nHPE AlletraMP FC CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_FC_CI build/641 : FAILURE in 0h 2m 18s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/997765","accounts_in_message":[],"_revision_number":2},{"id":"fdaef52b22fb04b871e94e537c9a5edad1d7c377","author":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"date":"2026-07-17 16:10:53.000000000","message":"Patch Set 2:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_NVME_CI build/549 : FAILURE in 0h 2m 23s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/997765","accounts_in_message":[],"_revision_number":2},{"id":"ae2c83f4a4ec77733e9ac3dcf80bfc48064353fd","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-17 16:16:59.000000000","message":"Patch Set 1:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4833/.","accounts_in_message":[],"_revision_number":1},{"id":"6c3eba10a59d5178e8f7cdf85e9fc5828ec10652","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-07-17 16:17:58.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"3edb2148ec45c95ee3ca7da6be32cb587cfc3b3f","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-07-17 16:23:46.000000000","message":"Patch Set 3:\n\nHPE AlletraMP FC CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_FC_CI build/644 : FAILURE in 0h 2m 9s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/997765","accounts_in_message":[],"_revision_number":3},{"id":"a486b75262cf3df95ea6b3d6093823744fd25d06","author":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"date":"2026-07-17 16:24:03.000000000","message":"Patch Set 3:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_NVME_CI build/552 : FAILURE in 0h 2m 27s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/997765","accounts_in_message":[],"_revision_number":3},{"id":"2db3ee71f6a8f07e1beceb514bcecaecdf26080b","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-07-17 16:24:05.000000000","message":"Patch Set 3:\n\nHPE AlletraMP iSCSI CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_iSCSi_CI build/907 : FAILURE in 0h 2m 15s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/997765","accounts_in_message":[],"_revision_number":3},{"id":"6a2a8461b0bbef2a1868811c03c8edaa94d9debc","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-17 16:25:26.000000000","message":"Patch Set 2:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4834/.","accounts_in_message":[],"_revision_number":2},{"id":"ea84815f921848dbc03f38d70cf80641a477ed2c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-07-17 16:36:13.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"201d2c7564f05e2d300e6afc78826c1358aa902a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-17 16:55:38.000000000","message":"Patch Set 3:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4837/.","accounts_in_message":[],"_revision_number":3},{"id":"e836dd0701d7d3ffed3901be9abf161f84d5825d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-17 17:00:09.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/83eaaea98d2b4a979be5d0ddf49630a6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d975b1c691824411b9fcc055b0709b3b : SUCCESS in 6m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b915adc72314dc699694d2237a349ae : SUCCESS in 14m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b718da769b354dc9be66894cc2d961b7 : SUCCESS in 9m 45s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/440791519d1a484f90ce2d52be424307 : SUCCESS in 6m 32s","accounts_in_message":[],"_revision_number":4},{"id":"50a973d6ba703943082fa78096d9bb96729e967d","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-17 17:03:39.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Id9be84b11d53dbdfc0d6150bf4beec64dd494df5 .","accounts_in_message":[],"_revision_number":1},{"id":"d2ad70e94d2980836049bc23a2af986029306af1","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-17 17:06:10.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Id9be84b11d53dbdfc0d6150bf4beec64dd494df5 .","accounts_in_message":[],"_revision_number":2},{"id":"6c8dbc3f7174acf2bcd5e04e27bb6a465ed2f103","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-07-17 17:11:39.000000000","message":"Patch Set 4:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/17_07_2026_17_04_52\nLeave a comment with","accounts_in_message":[],"_revision_number":4},{"id":"63d638f42d40693f3eb4516a713c6af8577385a0","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-17 17:14:15.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Id9be84b11d53dbdfc0d6150bf4beec64dd494df5 .","accounts_in_message":[],"_revision_number":3},{"id":"5a0eabe18bb74d12a6ca9ed1a6591290ebf53459","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-17 17:34:13.000000000","message":"Patch Set 4:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4840/.","accounts_in_message":[],"_revision_number":4},{"id":"394277b3929f15577620532220e85e4be3336dc1","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-17 17:34:21.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Id9be84b11d53dbdfc0d6150bf4beec64dd494df5 .","accounts_in_message":[],"_revision_number":4},{"id":"a2222c6d651b8f4841abb2b0282b91e640ac375e","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-07-17 18:29:31.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/997765 : FAILURE in 30m 16s","accounts_in_message":[],"_revision_number":4},{"id":"2eff6f0d72087df6bc3315e470307028d7362578","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-07-17 21:15:25.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-65-997765-4-fc : FAILURE in 10m\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-65-997765-4-iscsi : SUCCESS in 1h 20m 4s","accounts_in_message":[],"_revision_number":4},{"id":"664ecfd6f72e1c2413a47bb80ab0b9cdcf540a55","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2026-07-18 08:56:08.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-65-997765-3/FC/ : FAILURE in 16h 24m 26s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-65-997765-3/iSCSI/ : SUCCESS in 1h 50m 33s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":3},{"id":"56750983225ae95d747e115541cdba05f185fc9f","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-07-18 12:31:30.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-65-997765-3-fc : FAILURE in 9m 44s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-65-997765-3-iscsi : SUCCESS in 1h 6m 40s","accounts_in_message":[],"_revision_number":3},{"id":"16478820410ca27e7e4a69e9f6f79378089b2793","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-18 12:57:55.000000000","message":"Patch Set 1:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4657/.","accounts_in_message":[],"_revision_number":1},{"id":"8da1ae4c6a3a9e1561fcb31535131e5b6138b9b1","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-18 13:47:15.000000000","message":"Patch Set 3:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4661/.","accounts_in_message":[],"_revision_number":3},{"id":"11d6d75305761852adda6389b847810dd2f2d5b1","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-18 14:16:31.000000000","message":"Patch Set 4:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4664/.","accounts_in_message":[],"_revision_number":4},{"id":"04ce033f493b1f0b555a3dafc479035decd2195c","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-07-21 07:02:50.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"f1c6c7e6898e838a9b50be879fc883ed440eac2d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-07-22 04:20:54.000000000","message":"Patch Set 5: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"86ddbbe3078314c8819084f6010c7866957689fa","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-07-22 04:23:46.000000000","message":"Patch Set 5:\n\nHPE AlletraMP FC CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_FC_CI build/670 : FAILURE in 0h 2m 32s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/997765","accounts_in_message":[],"_revision_number":5},{"id":"ca335c24a19d09a955c709cb17bd7f87db618c90","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-07-22 04:23:50.000000000","message":"Patch Set 5:\n\nHPE AlletraMP iSCSI CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_iSCSi_CI build/933 : FAILURE in 0h 2m 31s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/997765","accounts_in_message":[],"_revision_number":5},{"id":"b2ca771ff70cffad07857dc5528005f0a6f5a0e9","author":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"date":"2026-07-22 04:24:08.000000000","message":"Patch Set 5:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_NVME_CI build/578 : FAILURE in 0h 2m 53s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/997765","accounts_in_message":[],"_revision_number":5},{"id":"fdd232dbf29fc3e1b955ad7b0916e97b620c5782","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-22 04:33:44.000000000","message":"Patch Set 5:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4875/.","accounts_in_message":[],"_revision_number":5},{"id":"138ada76e0246932f73f2ec2eaff6e1d515ee5c6","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-22 04:35:22.000000000","message":"Patch Set 5:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4699/.","accounts_in_message":[],"_revision_number":5},{"id":"0ab3ba2ac8496c9f2526b80bf8806b61b0421089","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-07-22 04:39:05.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"9be70258ea9a89f3d3e7ee20856d8714aba9aa12","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-07-22 04:40:07.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"38384ea96e38e1e5336de2a4df01693fe33540db","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-22 04:47:33.000000000","message":"Patch Set 6:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4876/.","accounts_in_message":[],"_revision_number":6},{"id":"39ba6188b77bf27ea169db4a3a1c3b29628135c7","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-07-22 04:52:10.000000000","message":"Patch Set 7:\n\nHPE AlletraMP iSCSI CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_iSCSi_CI build/937 : FAILURE in 0h 2m 12s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/997765","accounts_in_message":[],"_revision_number":7},{"id":"b3dcd468760afe2f35e4f99db61bdb55662c0573","author":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"date":"2026-07-22 04:52:28.000000000","message":"Patch Set 7:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_NVME_CI build/582 : FAILURE in 0h 2m 26s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/997765","accounts_in_message":[],"_revision_number":7},{"id":"c5bbee0162c6afff58c1a4c5b5859f85822c9532","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-07-22 04:52:32.000000000","message":"Patch Set 7:\n\nHPE AlletraMP FC CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_FC_CI build/674 : FAILURE in 0h 2m 29s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/997765","accounts_in_message":[],"_revision_number":7},{"id":"4ba5d5ab8eb1abc8c9b71cd5dc1204fd4db6fd6f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-22 04:53:25.000000000","message":"Patch Set 6:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4700/.","accounts_in_message":[],"_revision_number":6},{"id":"b480caf4fed1e5a1a6969c4a2781628bc72a50fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 05:05:45.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c89cba5898d447828c5dec37bc32deb5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98946f26cff84e40ada7ebd5a9321c1f : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa0ae20d4664406fba4d6d460a340d7c : SUCCESS in 9m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bc34f692b08749cea26768db1320acf1 : SUCCESS in 5m 30s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c20d40b0f4074b5d9d198b6f0f01f74c : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":7},{"id":"35ea9b36f247f2d4ede2def232ae5fea21701de9","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-22 05:15:12.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Id9be84b11d53dbdfc0d6150bf4beec64dd494df5 .","accounts_in_message":[],"_revision_number":5},{"id":"47edea872f05e0c6400901fd153352a49a5d4690","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-07-22 05:16:15.000000000","message":"Patch Set 7:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/22_07_2026_05_09_18\nLeave a comment with","accounts_in_message":[],"_revision_number":7},{"id":"8ee9a2f12f79ceb1d13137a36db71f0a9bf46d2c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-22 05:18:59.000000000","message":"Patch Set 7:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4703/.","accounts_in_message":[],"_revision_number":7},{"id":"d9921ff8b2f1f8dc1d59495c0303b2c318352e4a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-22 05:20:37.000000000","message":"Patch Set 7:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4879/.","accounts_in_message":[],"_revision_number":7},{"id":"9b872fcd61fdd120abb4dd342c94b6bcabce7a1c","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-22 05:36:13.000000000","message":"Patch Set 6:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Id9be84b11d53dbdfc0d6150bf4beec64dd494df5 .","accounts_in_message":[],"_revision_number":6},{"id":"d73942692dd18a035a26de8f97d982333bb74cf7","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-22 05:36:29.000000000","message":"Patch Set 7:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Id9be84b11d53dbdfc0d6150bf4beec64dd494df5 .","accounts_in_message":[],"_revision_number":7},{"id":"5bd5c86a62e8bbbefeba771181ff55aa8954a4a9","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-07-22 06:34:30.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/997765 : FAILURE in 30m 08s","accounts_in_message":[],"_revision_number":7},{"id":"fcc1160b648cb7d2481c1d9489ca2ada2690bbc4","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-07-22 07:30:51.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-65-997765-7-fc : FAILURE in 11m\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-65-997765-7-iscsi : SUCCESS in 59m 19s","accounts_in_message":[],"_revision_number":7},{"id":"630497464c877514c495cad42e114859b5b73f32","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2026-07-22 19:55:02.000000000","message":"Patch Set 5:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-65-997765-5/FC/ : FAILURE in 15h 33m 32s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-65-997765-5/iSCSI/ : UNSTABLE in 18s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":5},{"id":"6282cba56db5a1fab360a7b1f1d0a05016fbf810","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-07-22 23:57:17.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-65-997765-5-fc : FAILURE in 9m 54s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-65-997765-5-iscsi : SUCCESS in 1h 4m 28s","accounts_in_message":[],"_revision_number":5},{"id":"b7b6289e44af0683b3b3c9a330ffc5b0570c0e2b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37923,"name":"anthony gamboa","display_name":"Anthony Gamboa","email":"anthony.gamboa@hitachivantara.com","username":"anthonygamboa","status":"Hitachi Vantara Software Developer"},"date":"2026-07-31 16:48:40.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"151c81052888900cab4502e9bf30d61f309471bc","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-07-31 16:52:10.000000000","message":"Patch Set 8:\n\nHPE AlletraMP FC CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_FC_CI build/775 : FAILURE in 0h 2m 29s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/997765","accounts_in_message":[],"_revision_number":8},{"id":"601ed383c6ad44fc311d299722c7926c649c7763","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-07-31 16:57:08.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"555b7a0ff9c96ab990455756de1606781ce55023","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-31 17:11:58.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b1706e6a59e426abfc3899106701597\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e0ce1cb1df343ab96340ebf6deeb079 : SUCCESS in 6m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edca2bb680464bb2b17dc04581c66590 : SUCCESS in 7m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/94a0388003094f92b1c28999d354201f : SUCCESS in 9m 08s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2373f1e9a6b84e7d9df0da1039021792 : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":9},{"id":"4070fb5d35aedb4cc6da2b9478c44933fa085557","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-07-31 17:28:17.000000000","message":"Patch Set 9:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/31_07_2026_17_21_27\nLeave a comment with","accounts_in_message":[],"_revision_number":9},{"id":"a88c71ca455e55678bde0b15e66602c89f189951","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-31 18:04:34.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/Id9be84b11d53dbdfc0d6150bf4beec64dd494df5 .","accounts_in_message":[],"_revision_number":8},{"id":"28f0950715129872d6c276445dc3b2d7a1e75fa1","author":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"date":"2026-07-31 18:04:52.000000000","message":"Patch Set 9:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_NVME_CI build/690 : FAILURE in 0h 1m 56s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/997765","accounts_in_message":[],"_revision_number":9},{"id":"b6fdd760a607b8ec8d70c6718b1bb09761b24d26","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-07-31 18:05:51.000000000","message":"Patch Set 9:\n\nHPE AlletraMP iSCSI CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1051 : FAILURE in 0h 1m 50s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/997765","accounts_in_message":[],"_revision_number":9},{"id":"cd5c91b87f36f3ab000163b37466db86d3944d23","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-07-31 18:08:24.000000000","message":"Patch Set 9:\n\nHPE AlletraMP FC CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_FC_CI build/781 : FAILURE in 0h 1m 53s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/997765","accounts_in_message":[],"_revision_number":9},{"id":"4bb21a3dab0f99e59f3c8fc3e163669cc4e2a49c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-31 18:31:05.000000000","message":"Patch Set 8:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4815/.","accounts_in_message":[],"_revision_number":8},{"id":"eb4e62cca2f2c959e4c0b5604a23e94cf6585ead","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-31 18:33:39.000000000","message":"Patch Set 8:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4991/.","accounts_in_message":[],"_revision_number":8},{"id":"376935c4591fc7c8273c4c7da379075c8da9652d","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-31 18:44:58.000000000","message":"Patch Set 9:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4992/.","accounts_in_message":[],"_revision_number":9},{"id":"7983a47a7a25239412e584f8344cb9ab6b57323e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-31 18:45:42.000000000","message":"Patch Set 9:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4816/.","accounts_in_message":[],"_revision_number":9},{"id":"d5b55ff189d9ec348451c055fcc1f8405e70c73c","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-31 19:04:25.000000000","message":"Patch Set 9:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/Id9be84b11d53dbdfc0d6150bf4beec64dd494df5 .","accounts_in_message":[],"_revision_number":9},{"id":"02cf0dce343a78494d38d0fd1206e85581721ab9","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2026-08-06 02:26:53.000000000","message":"Patch Set 9:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/65/997765/9/20260806022726/NEC-FC : FAILURE in 6h 18m 03s\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/65/997765/9/20260806022726/NEC-ISCSI : FAILURE in 8h 59m 24s","accounts_in_message":[],"_revision_number":9},{"id":"488524a816ad398ff695a403768bc593d2828653","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-08-11 16:42:44.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-65-997765-9-fc : SUCCESS in 47m 46s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-65-997765-9-iscsi : SUCCESS in 1h 6m 11s","accounts_in_message":[],"_revision_number":9},{"id":"9abd9ab424c27ae71e4bde6a1727e03b7f9609c4","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-08-21 14:21:36.000000000","message":"Patch Set 9: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"5b545446f5612b827f817d23325964ca8ce0f871","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2026-08-21 15:17:12.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"469622e53e2d81ce8d73ea065450f01f58150fc2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-08-28 05:23:46.000000000","message":"Patch Set 10: Patch Set 9 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"1e7c9c001207dbad94e2b2cf0e8608f580d69d50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 05:40:58.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d75ced2c19174a9185ae594aebe936c1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f9d1b40362245f1ac3efb19d4603772 : FAILURE in 3m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29cec2932ebc4e2a80a21a783abbac87 : SUCCESS in 15m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1fa9a3c88c53496a84c4a4b13ea1b68e : SUCCESS in 10m 53s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/79b10b4ece2749568299fe3ba478e382 : SUCCESS in 4m 56s\n\nWarning:\n  Comments left for invalid file cinder/volume/drivers/hitachi/hbsd_rest.py","accounts_in_message":[],"_revision_number":10},{"id":"07fd4a2f9052ac0971b7c5d4eb0b55c290903570","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-28 07:04:20.000000000","message":"Patch Set 10:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5292/.","accounts_in_message":[],"_revision_number":10},{"id":"0e0d99b4429fc01ce076079684269ecd6c5b1280","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-28 07:08:05.000000000","message":"Patch Set 10:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Id9be84b11d53dbdfc0d6150bf4beec64dd494df5 .","accounts_in_message":[],"_revision_number":10},{"id":"5235c2dc2e1c1331937b184c402d3ce2e02ae8a2","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-08-28 07:45:43.000000000","message":"Patch Set 10:\n\nHPE AlletraMP iSCSI CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1340 : FAILURE in 0h 1m 45s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/997765","accounts_in_message":[],"_revision_number":10},{"id":"22ff3c4f22272dd193ab5e6e98fb52713b99ce38","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-08-28 07:46:48.000000000","message":"Patch Set 10:\n\nHPE AlletraMP FC CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_FC_CI build/1071 : FAILURE in 0h 1m 39s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/997765","accounts_in_message":[],"_revision_number":10},{"id":"ede651c17e0be73b548cdebe3cf220214cef55fc","author":{"_account_id":39391,"name":"Rashi Jain","display_name":"HPE AlletraMP NVMeTCP CI","email":"rashi.jain@hpe.com","username":"rashij"},"date":"2026-08-28 07:48:07.000000000","message":"Patch Set 10:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_NVME_CI build/978 : FAILURE in 0h 2m 0s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/997765","accounts_in_message":[],"_revision_number":10},{"id":"a6ab2f54c467c6a65b1de87dc8f09b3b2706d49f","author":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"date":"2026-08-28 17:37:56.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"566b01ee64aaf9e2d5eacb559fbeb8635763990d","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-08-28 18:32:25.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"7552cdc0ae7775bf308a5414990dc726899d65cf","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-08-28 18:33:24.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"7c51bd7e05b3e64f9b8e4110ff60a17b7b1c77a5","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-08-28 20:22:53.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-65-997765-10-fc : NODE_FAILURE\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-65-997765-10-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":10},{"id":"8699d918054d680f01d9994e020a26c0255bdb65","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-08-28 20:33:36.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-65-997765-10-fc : NODE_FAILURE\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-65-997765-10-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":10},{"id":"0937dcff899aede658b91f350a142fa338279563","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-08-28 21:54:28.000000000","message":"Patch Set 11: Patch Set 10 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":11},{"id":"4f33b1a4d142e88002ef6ea4675dcc16388e5b47","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-08-28 21:55:46.000000000","message":"Patch Set 12: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":12},{"id":"060e4e9a3100ea2ae8ae390c774d5d1d1bd1fc7a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-28 22:04:15.000000000","message":"Patch Set 11:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5306/.","accounts_in_message":[],"_revision_number":11},{"id":"8df759ac8334f1de59eadb9577bbcdcf6ea75456","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-08-28 22:09:01.000000000","message":"Patch Set 13: Published edit on patch set 12.","accounts_in_message":[],"_revision_number":13},{"id":"a873badd07a8712d9898185f5726d2d7db527b2c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-28 22:13:35.000000000","message":"Patch Set 12:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5307/.","accounts_in_message":[],"_revision_number":12},{"id":"1ea4473efdc2a977ca0da10e3fd58f921be92b76","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-28 22:22:23.000000000","message":"Patch Set 13:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5308/.","accounts_in_message":[],"_revision_number":13},{"id":"255298e46448783829dbda2785df0f01547c5115","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 22:37:13.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d54688baa5840098a2dc8fc5eb9ec75\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/97208b7aed92469ea68bdaa72680b36c : SUCCESS in 12m 57s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/51c9abafd5ec45a9844bdc878819d4a1 : SUCCESS in 15m 51s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/fd5de4effa1c4ccfae15f65746fd3873 : SUCCESS in 19m 25s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"17941b0787aa0e2833d3c673b74e75b335dad725","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-28 22:43:09.000000000","message":"Patch Set 13:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/108c3c3102634906a0d39a089ac4902d/","accounts_in_message":[],"_revision_number":13},{"id":"c6bda176bef382cb739803b6631b253fb78465ad","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-28 22:50:56.000000000","message":"Patch Set 11:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Id9be84b11d53dbdfc0d6150bf4beec64dd494df5 .","accounts_in_message":[],"_revision_number":11},{"id":"ce65aba7369c1e53e81d398e5fc36cbeb5cfacf6","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-28 22:51:06.000000000","message":"Patch Set 12:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Id9be84b11d53dbdfc0d6150bf4beec64dd494df5 .","accounts_in_message":[],"_revision_number":12},{"id":"0fbe530e54fe271eb5d1bfb0a7988b4f01832f14","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-28 22:57:47.000000000","message":"Patch Set 13:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/108c3c3102634906a0d39a089ac4902d\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e1aefbedc1e44c48bdfcfede590eb8e3 : SUCCESS in 41m 00s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/cee275fda553460f9896d7d35a28659f : SUCCESS in 33m 38s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/23a3e3a78bf542d698fe6f4ebbb0072f : SUCCESS in 42m 50s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/669e29a836274e409d0da05220d6466b : SUCCESS in 36m 53s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d5649914cd48476f935bbb084a3b36a7 : FAILURE in 14m 24s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"3c28de7452cc95ad5528000dbc4433e2446595a2","author":{"_account_id":39391,"name":"Rashi Jain","display_name":"HPE AlletraMP NVMeTCP CI","email":"rashi.jain@hpe.com","username":"rashij"},"date":"2026-08-28 22:59:51.000000000","message":"Patch Set 13:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/992 : SUCCESS in 0h 47m 27s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/997765","accounts_in_message":[],"_revision_number":13},{"id":"666f8c1d95c274958231143b285edaa0e0a14344","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-28 23:00:00.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/53e8b8e74c4747d48d3d741766c02e55 : SUCCESS in 43m 36s\n- cinder-storpool-tempest-iscsi-multipath build/187f3b1c31694172889c61fe3d34eecb : SUCCESS in 49m 30s","accounts_in_message":[],"_revision_number":13},{"id":"7a625ded3b4994d1b392791b195706975dcff810","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-08-28 23:00:23.000000000","message":"Patch Set 13:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/1084 : SUCCESS in 0h 47m 57s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/997765","accounts_in_message":[],"_revision_number":13},{"id":"962969d06b8e3b4bdb9efe3daac3fca94df43a32","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-08-28 23:01:07.000000000","message":"Patch Set 13:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1353 : SUCCESS in 0h 48m 42s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/997765","accounts_in_message":[],"_revision_number":13},{"id":"e992479163c316ac9e5adee74339092f6106d4fd","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-28 23:05:19.000000000","message":"Patch Set 13:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Id9be84b11d53dbdfc0d6150bf4beec64dd494df5 .","accounts_in_message":[],"_revision_number":13},{"id":"dd338cfe6eedc8f2699aa49489b4071f2374126d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-29 00:17:03.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97f8bd9fb37748b3b098df9519952668\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37266e54a1f14ce1b45a48141a4f032e : SUCCESS in 6m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b87500fdfc8b4b0aa1f756af1ec95f1a : SUCCESS in 8m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fdf949581eb44c75b06382126a839f24 : SUCCESS in 10m 00s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b68dd359eb704152a243e14ea15869c3 : SUCCESS in 13m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a95c7656741446748574954c1ed51cef : SUCCESS in 12m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/bebe9cd9119f4a52b2418be9a2fde6a5 : SUCCESS in 44m 13s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/43f0f2f795fd4bc9940fb7d337cf59f6 : SUCCESS in 47m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/155eec17bfe74e7897183b3171b8d1e3 : SUCCESS in 1h 29m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4ca9a5b783a641b9b4ad895997b933ea : SUCCESS in 1h 05m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7fb3013ce05b41d6a77dd18ebe5c302e : SUCCESS in 9m 51s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b3a82cebada24ca694eed44184a28037 : SUCCESS in 16m 22s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e1c2c7c19f8c4db8a6dc6b447eb32136 : SUCCESS in 4m 24s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/dfa5b01161914e56b7aca03a16a9600e : SUCCESS in 5m 23s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/7e1750540535490e87b5d0f244150ee9 : SUCCESS in 12m 56s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/75c787fb1f2545cbb669722b1c2d4f1d : FAILURE in 10m 03s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/3154a3a52d634196858bfa6669b3d014 : SUCCESS in 9m 24s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/1941fe804e0b489a8cade4dc3c0508bb : SUCCESS in 5m 10s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/6abc5c2be7954b5da86dacf7badde85b : FAILURE in 4m 59s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/9a6bf0a2f86744dc85b20ee95d7d462e : SUCCESS in 8m 35s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a49427adb4044161aaf3b8b707830d60 : SUCCESS in 36m 40s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/aa287a353cd24b41bd1bbe552c9c45b8 : FAILURE in 3m 12s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/05e0e481add64ea18cbaece0d530eb71 : SUCCESS in 1h 33m 26s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0fc6972fec534578909ee69d66e06f0c : SUCCESS in 2h 01m 35s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3016732b718d48409e3d9058bc1bebff : SUCCESS in 1h 28m 49s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/2ed260395f29460bbae235d6d907ddfa : FAILURE in 2h 02m 31s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/31ce88fbfbdb4494b39da0c1cd4c4faf : SUCCESS in 1h 34m 07s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/39d942c0d8ee41b299c6e87c861642ff : SUCCESS in 27m 00s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ffd430d2c7cc44718524d197f61b49b8 : SUCCESS in 50m 27s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8c12a82967094f4c88f8d084b81da29c : SUCCESS in 1h 33m 21s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/2940b01c036a4516aa532c0ceb1b5a17 : SUCCESS in 33m 08s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ceaea7df7ef54d63b7e847aabd521458 : SUCCESS in 1h 44m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dc3639fa10594087bc322acc65daf9bd : SUCCESS in 58m 12s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":13},{"id":"9fa518318508e6e47bbdf796c60acaab5b0d3f7d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-29 00:27:30.000000000","message":"Patch Set 13:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/29_08_2026_00_20_38\nLeave a comment with","accounts_in_message":[],"_revision_number":13},{"id":"a0743f541be888190bfc482742ee0ea8c9ade255","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-08-29 02:27:14.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-65-997765-13-fc : NODE_FAILURE\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-65-997765-13-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":13},{"id":"3a4945e24357865f52d10034acc48bdbdb1f2aa9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-08-29 03:47:23.000000000","message":"Patch Set 14: Commit message was updated.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":14},{"id":"bc5181120330bbc3f80a512d448ad970e67d4a4a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-29 03:58:03.000000000","message":"Patch Set 14:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5313/.","accounts_in_message":[],"_revision_number":14},{"id":"34d0680c9ae101b4224688b72a264c7092307a6c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-29 04:12:10.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c30ce22f712b40f2a5fc40217265f203\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0c1691e317aa47588a705d9669dc941f : SUCCESS in 13m 21s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/131294e88ba14db897a645b03911d204 : SUCCESS in 14m 31s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/ffedd65b83a1403882b6c94c02fbac9c : SUCCESS in 21m 49s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"606b5e7febdfc967f05ba0f52d51b7b2f047c0ac","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-29 04:12:29.000000000","message":"Patch Set 14:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/4f8935e875474f28872e36bdba6f8a75/","accounts_in_message":[],"_revision_number":14},{"id":"142dffe0fc5c5b4880d723a54a4f05e55b00242d","author":{"_account_id":39391,"name":"Rashi Jain","display_name":"HPE AlletraMP NVMeTCP CI","email":"rashi.jain@hpe.com","username":"rashij"},"date":"2026-08-29 04:35:47.000000000","message":"Patch Set 14:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/997 : SUCCESS in 0h 47m 58s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/997765","accounts_in_message":[],"_revision_number":14},{"id":"2a8ffd23819b98b03ff06b1c54cce9e555919827","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-08-29 04:35:52.000000000","message":"Patch Set 14:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1358 : SUCCESS in 0h 48m 4s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/997765","accounts_in_message":[],"_revision_number":14},{"id":"06e3b402ed65b49af101757d7de353803ca1202d","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-29 04:36:21.000000000","message":"Patch Set 14:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/4f8935e875474f28872e36bdba6f8a75\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/de0969cf293741409d4fc7d756878086 : SUCCESS in 44m 36s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4ae34d9f74034b789a87fac313c17f4a : SUCCESS in 35m 25s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7b48ec1ebdf04892ae34360d87cd28fd : SUCCESS in 48m 26s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a3cb912d7aae47348f9cd1166b314da1 : SUCCESS in 32m 05s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d6b5e91c35984295bffbc522992bd3cd : FAILURE in 21m 22s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"7f36f0e38de8535ada151e94e4d45c323c69b8d5","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-29 04:38:39.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/d505f550a703440bb0c6d28d6947dd4d : SUCCESS in 49m 58s\n- cinder-storpool-tempest-iscsi-multipath build/3e403ecd34444b3f8014f2d0327cb3ab : SUCCESS in 49m 28s","accounts_in_message":[],"_revision_number":14},{"id":"cd706888b3a36fd30aeda4a47590a46ceaea4df2","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-08-29 04:39:18.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-65-997765-13-fc : SUCCESS in 1h 8m 12s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-65-997765-13-iscsi : SUCCESS in 1h 18m 54s","accounts_in_message":[],"_revision_number":13},{"id":"5c5222e2b43f9e7311159fceb67a56764c2fa564","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-08-29 04:39:25.000000000","message":"Patch Set 14:\n\nHPE AlletraMP FC CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_FC_CI build/1089 : FAILURE in 0h 51m 37s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/997765","accounts_in_message":[],"_revision_number":14},{"id":"0f5d95952803a6885f937f5544a749a9b85d173c","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-29 04:39:40.000000000","message":"Patch Set 14:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Id9be84b11d53dbdfc0d6150bf4beec64dd494df5 .","accounts_in_message":[],"_revision_number":14},{"id":"086cde2a1fe3f1309ce15a9e25c809abae9fcef6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-29 06:17:58.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4c09a55968b498a8e5310a4f7ea6baf\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92247e387baa4ec9875c6c794641fbdb : SUCCESS in 6m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6b80bf565bdb419d84ed1d87cce017ed : SUCCESS in 9m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bae29314047b463e844720705dbff369 : SUCCESS in 10m 40s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/cf4805268d5c457db43992355655940c : SUCCESS in 16m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ae48da63198405a88efb1e6b139ddbe : SUCCESS in 15m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/30ab033c6b0e46508241adfb3d2cc570 : SUCCESS in 29m 50s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/847c9b3f801f46a28a16102d633a70ad : SUCCESS in 1h 05m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f085616368cf416bb9b582b0a4cbea5a : SUCCESS in 59m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a07c97e7c1644dc0900d23bd1e752735 : SUCCESS in 45m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2967b69b1a7e4cf39cc933f996b87d41 : SUCCESS in 5m 18s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/fadc5906c2c6468baf0bda63ac69a3ab : SUCCESS in 19m 31s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/11a5343d5a5a4638a2bae6b998a404c7 : SUCCESS in 3m 58s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d55c41e9411a49548cd0a554ea303047 : SUCCESS in 3m 25s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/b8a233bd82cf410caaf4fed970ae1c72 : SUCCESS in 12m 09s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/cd853b696b0748cfa432ab6264e0739c : SUCCESS in 8m 42s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/16c6642914c844a7bb518ea18db391fe : SUCCESS in 6m 04s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/6b79973745044467bef7f2b03f446d30 : SUCCESS in 6m 09s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/386f7a435a5946078d4029b2ae6eb9b5 : FAILURE in 3m 57s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/a569f7f1c2dc4de9aa969149c2ac243e : SUCCESS in 7m 40s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/6fdb2557918445d8b0810bb6399d3904 : SUCCESS in 20m 00s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/96d429895db34bdb8e0692d52e6c525c : FAILURE in 4m 41s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6fa36401a663433ea7b446693f2c77eb : SUCCESS in 50m 53s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b8e7393a7b6d4737b311a1fea3fdcb84 : FAILURE in 2h 14m 32s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7e8e71e3d2ba470c97317c92bd1a7ce3 : SUCCESS in 1h 22m 55s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/f08a9252283c4ccfb82de8187ea0ebdc : FAILURE in 2h 25m 02s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/88195b49e6ab46edb8154dc6c774ac09 : SUCCESS in 1h 39m 20s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/07d3eaa8cfc0469e954bfabc14730a0c : SUCCESS in 34m 49s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/8642dff0ece5480fb71a3206da0b55b5 : SUCCESS in 1h 19m 28s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a9d0e0b132cc43d8ba191a5c5873dd12 : SUCCESS in 1h 09m 29s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/5b02a61fa0824e66be3bf3e024a573fe : SUCCESS in 45m 50s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/46d9783706384c20af8ae435045e04cd : SUCCESS in 1h 26m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d817944b62c147239b3f06662bcace4d : SUCCESS in 25m 21s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":14},{"id":"91178bdd98d26d34913081e8edbbb63e34844879","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-29 06:28:28.000000000","message":"Patch Set 14:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/29_08_2026_06_21_27\nLeave a comment with","accounts_in_message":[],"_revision_number":14},{"id":"3f0d2b7a4b91ec1ed7452ac414f3d19c63aea015","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-08-29 08:02:08.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-65-997765-14-fc : NODE_FAILURE\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-65-997765-14-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":14},{"id":"69357978bab879bb35300b6d70cda78820881c2a","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2026-08-29 16:28:09.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"dcf44c08ac01215f3b17ec8b5f4123fb86e0267d","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-08-30 00:39:28.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"94377f4c361db9247585ba12628a57c4f3730c47","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-08-30 14:46:55.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-65-997765-14-fc : SUCCESS in 1h 4m 55s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-65-997765-14-iscsi : SUCCESS in 1h 20m 29s","accounts_in_message":[],"_revision_number":14},{"id":"ff3dca30f3e11b13a6f3ce64f742437b99ffbc47","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2026-08-30 18:19:36.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"63185512f6535f3738ef1d92711435d817655410","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-01 16:16:46.000000000","message":"Patch Set 10:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5116/.","accounts_in_message":[],"_revision_number":10},{"id":"3029ec97ab19956ef08aef6ac13fd7443fcf4c24","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-02 01:21:38.000000000","message":"Patch Set 11:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5130/.","accounts_in_message":[],"_revision_number":11},{"id":"fb0eb88755ad88092378dfb4bfafde0e98329af5","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-02 01:59:24.000000000","message":"Patch Set 12:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5131/.","accounts_in_message":[],"_revision_number":12},{"id":"7ee1cb0fec24c6d92058526eefc2c42728ab8b84","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-02 02:37:40.000000000","message":"Patch Set 13:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5132/.","accounts_in_message":[],"_revision_number":13},{"id":"9e8c87a74ad92e88e8963b3dd573d40f65428af5","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-02 05:47:05.000000000","message":"Patch Set 14:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5137/.","accounts_in_message":[],"_revision_number":14},{"id":"80a96502ff3f99e628218ef03353be1e34c21852","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2026-09-04 13:03:42.000000000","message":"Patch Set 14: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"5cd968d8ea8ce0d55067869d95e1a79f0184f665","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 13:04:20.000000000","message":"Patch Set 14: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":14},{"id":"2811464df7c6078fe73a9cc36feca3ea11dd65c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 15:00:13.000000000","message":"Patch Set 14: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9fe8f5af96f44c57aa83299d0463e6c3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4cfb91cfc6bb4545a90b9bb6ec396efa : SUCCESS in 7m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/51a078565f1145f68bfa58d0801708ae : SUCCESS in 8m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e2332bf59fb9425daae9662f4885070f : SUCCESS in 9m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5631080dd8a74ddfa103b7d755ed4b4d : SUCCESS in 15m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/69354c75491f4f49acbc141f81797af5 : SUCCESS in 1h 07m 00s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2b6a190cad414baa88c42db3dec00628 : SUCCESS in 1h 08m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1b39ee1e0989459abbc1f2771d46f3fd : SUCCESS in 1h 33m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c5dcfc6b67a04ae4b0e30f1fc50f7811 : SUCCESS in 42m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3e8fd1b3c2674bea94871f6ee99f42cc : SUCCESS in 11m 02s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a2f092548aa149f5928dd9fad40dea22 : SUCCESS in 1h 00m 43s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c6ae1e7a558341778ed5d05e165dc2f6 : SUCCESS in 1h 44m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/78e3fe97ecf3456abb8e7d44dbba9979 : SUCCESS in 1h 03m 58s","accounts_in_message":[],"_revision_number":14},{"id":"5fe1aec37506c685e829c77da7d6d12ae334e78b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 15:00:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":14},{"id":"3c4db65240a4b6ca9f5d50df3d39a08b9af97365","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 15:01:35.000000000","message":"Patch Set 14:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92ae57607e8440f9aafb635334ef7ce7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68d7f3781de24c6892412390c3188025 : SUCCESS in 1m 04s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bde19328fa114de9aedff2e5f67a3926 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":14},{"id":"1b8a9539110d4fd54cb8b7259625700dc70659ce","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-09-04 17:02:52.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-65-997765-14-fc : NODE_FAILURE\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-65-997765-14-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":14},{"id":"f681e72478f5f9431145bebb0fb8841a5a4ceb69","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-09-04 22:37:28.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-65-997765-14-fc : SUCCESS in 1h 1m 52s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-65-997765-14-iscsi : FAILURE in 58m 21s","accounts_in_message":[],"_revision_number":14},{"id":"d4f4fef3cd3a09d599d610c88c0efa88f29aa67b","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-09-09 05:16:40.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14}],"current_revision_number":14,"current_revision":"54c3352c32b4b55aeb17b9aefeb892a90acecfbc","revisions":{"7ae8237655146eb96bef34355f1e552f2dfce1b1":{"kind":"REWORK","_number":1,"created":"2026-07-17 16:05:45.000000000","uploader":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"ref":"refs/changes/65/997765/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/997765/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/997765/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/997765/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/997765/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/997765/1"}}},"commit":{"parents":[{"commit":"a2fbb270741870f7e7be40521f48f5e1d2051107","subject":"[Hitachi] Add support for compression capacity saving value.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a2fbb270741870f7e7be40521f48f5e1d2051107"}]}],"author":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-07-17 16:05:45.000000000","tz":0},"committer":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-07-17 16:05:45.000000000","tz":0},"subject":"[Hitachi] Small documentation update for better cut+paste preservation.","message":"[Hitachi] Small documentation update for better cut+paste preservation.\n\nChange-Id: Id9be84b11d53dbdfc0d6150bf4beec64dd494df5\nSigned-off-by: Anthony Galica \u003canthony.galica@hitachivantara.com\u003e","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7ae8237655146eb96bef34355f1e552f2dfce1b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7ae8237655146eb96bef34355f1e552f2dfce1b1"}]},"branch":"refs/heads/master","conflicts":{"no_base_reason":"NO_MERGE_PERFORMED","contains_conflicts":false}},"22927e81a00c685a88c142f9749039cba74f6250":{"kind":"REWORK","_number":2,"created":"2026-07-17 16:08:10.000000000","uploader":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"ref":"refs/changes/65/997765/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/997765/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/997765/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/997765/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/997765/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/997765/2"}}},"commit":{"parents":[{"commit":"a2fbb270741870f7e7be40521f48f5e1d2051107","subject":"[Hitachi] Add support for compression capacity saving value.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a2fbb270741870f7e7be40521f48f5e1d2051107"}]}],"author":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-07-17 16:05:45.000000000","tz":0},"committer":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-07-17 16:08:10.000000000","tz":0},"subject":"[Hitachi] Small documentation update for better cut+paste preservation.","message":"[Hitachi] Small documentation update for better cut+paste preservation.\n\nChange-Id: Id9be84b11d53dbdfc0d6150bf4beec64dd494df5\nSigned-off-by: Anthony Galica \u003canthony.galica@hitachivantara.com\u003e","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/22927e81a00c685a88c142f9749039cba74f6250"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/22927e81a00c685a88c142f9749039cba74f6250"}]},"branch":"refs/heads/master"},"02161b7b5de802428b0174a751de2caf3ce80cce":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-07-17 16:17:58.000000000","uploader":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"ref":"refs/changes/65/997765/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/997765/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/997765/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/997765/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/997765/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/997765/3"}}},"commit":{"parents":[{"commit":"f8eea9c74b8017bc06c49bb81468aff9556db7a1","subject":"[Hitachi] Add support for compression capacity saving value.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f8eea9c74b8017bc06c49bb81468aff9556db7a1"}]}],"author":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-07-17 16:05:45.000000000","tz":0},"committer":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-07-17 16:17:58.000000000","tz":0},"subject":"[Hitachi] Small documentation update for better cut+paste preservation.","message":"[Hitachi] Small documentation update for better cut+paste preservation.\n\nChange-Id: Id9be84b11d53dbdfc0d6150bf4beec64dd494df5\nSigned-off-by: Anthony Galica \u003canthony.galica@hitachivantara.com\u003e","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/02161b7b5de802428b0174a751de2caf3ce80cce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/02161b7b5de802428b0174a751de2caf3ce80cce"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"a2fbb270741870f7e7be40521f48f5e1d2051107","ours":"22927e81a00c685a88c142f9749039cba74f6250","theirs":"f8eea9c74b8017bc06c49bb81468aff9556db7a1","merge_strategy":"recursive","contains_conflicts":false}},"941a057d244e52702ff701461e6eaada6652bfd6":{"kind":"REWORK","_number":4,"created":"2026-07-17 16:36:13.000000000","uploader":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"ref":"refs/changes/65/997765/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/997765/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/997765/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/997765/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/997765/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/997765/4"}}},"commit":{"parents":[{"commit":"f8eea9c74b8017bc06c49bb81468aff9556db7a1","subject":"[Hitachi] Add support for compression capacity saving value.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f8eea9c74b8017bc06c49bb81468aff9556db7a1"}]}],"author":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-07-17 16:05:45.000000000","tz":0},"committer":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-07-17 16:36:13.000000000","tz":0},"subject":"[Hitachi] Small documentation update for better cut+paste preservation.","message":"[Hitachi] Small documentation update for better cut+paste preservation.\n\nChange-Id: Id9be84b11d53dbdfc0d6150bf4beec64dd494df5\nSigned-off-by: Anthony Galica \u003canthony.galica@hitachivantara.com\u003e","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/941a057d244e52702ff701461e6eaada6652bfd6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/941a057d244e52702ff701461e6eaada6652bfd6"}]},"branch":"refs/heads/master"},"e4da673e8e7c420b136639a1d2d833eb6e5f521a":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2026-07-22 04:20:54.000000000","uploader":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"ref":"refs/changes/65/997765/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/997765/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/997765/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/997765/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/997765/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/997765/5"}}},"commit":{"parents":[{"commit":"f8eea9c74b8017bc06c49bb81468aff9556db7a1","subject":"[Hitachi] Add support for compression capacity saving value.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f8eea9c74b8017bc06c49bb81468aff9556db7a1"}]}],"author":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-07-17 16:05:45.000000000","tz":0},"committer":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-07-22 04:20:54.000000000","tz":0},"subject":"[Hitachi] Small documentation updates","message":"[Hitachi] Small documentation updates\n\n- Update for better cut+paste preservation.\n- Update release note typo.\n- Add warning about shared ports with CSI (OpenShift) plugin.\n\nChange-Id: Id9be84b11d53dbdfc0d6150bf4beec64dd494df5\nSigned-off-by: Anthony Galica \u003canthony.galica@hitachivantara.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e4da673e8e7c420b136639a1d2d833eb6e5f521a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e4da673e8e7c420b136639a1d2d833eb6e5f521a"}]},"branch":"refs/heads/master","description":"Edit commit message"},"42a179cb53982ed9c45ded730a0ea59be6817224":{"kind":"REWORK","_number":6,"created":"2026-07-22 04:39:05.000000000","uploader":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"ref":"refs/changes/65/997765/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/997765/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/997765/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/997765/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/997765/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/997765/6"}}},"commit":{"parents":[{"commit":"f8eea9c74b8017bc06c49bb81468aff9556db7a1","subject":"[Hitachi] Add support for compression capacity saving value.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f8eea9c74b8017bc06c49bb81468aff9556db7a1"}]}],"author":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-07-17 16:05:45.000000000","tz":0},"committer":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-07-22 04:39:05.000000000","tz":0},"subject":"[Hitachi] Small documentation updates","message":"[Hitachi] Small documentation updates\n\n- Update for better cut+paste preservation.\n- Update release note typo.\n- Add warning about shared ports with CSI (OpenShift) plugin.\n\nChange-Id: Id9be84b11d53dbdfc0d6150bf4beec64dd494df5\nSigned-off-by: Anthony Galica \u003canthony.galica@hitachivantara.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/42a179cb53982ed9c45ded730a0ea59be6817224"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/42a179cb53982ed9c45ded730a0ea59be6817224"}]},"branch":"refs/heads/master"},"2ab16823f475320ec90a2a42465ace83cdec7ac8":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2026-07-22 04:40:07.000000000","uploader":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"ref":"refs/changes/65/997765/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/997765/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/997765/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/997765/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/997765/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/997765/7"}}},"commit":{"parents":[{"commit":"debd48225805adb537d08ebfd579155b159b55d9","subject":"[Hitachi] Add support for compression capacity saving value.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/debd48225805adb537d08ebfd579155b159b55d9"}]}],"author":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-07-17 16:05:45.000000000","tz":0},"committer":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-07-22 04:40:07.000000000","tz":0},"subject":"[Hitachi] Small documentation updates","message":"[Hitachi] Small documentation updates\n\n- Update for better cut+paste preservation.\n- Update release note typo.\n- Add warning about shared ports with CSI (OpenShift) plugin.\n\nChange-Id: Id9be84b11d53dbdfc0d6150bf4beec64dd494df5\nSigned-off-by: Anthony Galica \u003canthony.galica@hitachivantara.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2ab16823f475320ec90a2a42465ace83cdec7ac8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2ab16823f475320ec90a2a42465ace83cdec7ac8"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"f8eea9c74b8017bc06c49bb81468aff9556db7a1","ours":"42a179cb53982ed9c45ded730a0ea59be6817224","theirs":"debd48225805adb537d08ebfd579155b159b55d9","merge_strategy":"recursive","contains_conflicts":false}},"1433b896f57ce4c31c927e693cc81ed3065c94b6":{"kind":"REWORK","_number":8,"created":"2026-07-31 16:48:40.000000000","uploader":{"_account_id":37923,"name":"anthony gamboa","display_name":"Anthony Gamboa","email":"anthony.gamboa@hitachivantara.com","username":"anthonygamboa","status":"Hitachi Vantara Software Developer"},"ref":"refs/changes/65/997765/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/997765/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/997765/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/997765/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/997765/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/997765/8"}}},"commit":{"parents":[{"commit":"0bf09755a9145bee1b0ee1d9fd3f9eb508d9144a","subject":"[Hitachi] Add support for compression capacity saving value.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0bf09755a9145bee1b0ee1d9fd3f9eb508d9144a"}]}],"author":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-07-17 16:05:45.000000000","tz":0},"committer":{"name":"anthonygamboa","email":"anthony.gamboa@hitachivantara.com","date":"2026-07-31 16:48:14.000000000","tz":-420},"subject":"[Hitachi] Small documentation updates","message":"[Hitachi] Small documentation updates\n\n- Update for better cut+paste preservation.\n- Update release note typo.\n- Add warning about shared ports with CSI (OpenShift) plugin.\n\nChange-Id: Id9be84b11d53dbdfc0d6150bf4beec64dd494df5\nSigned-off-by: Anthony Galica \u003canthony.galica@hitachivantara.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1433b896f57ce4c31c927e693cc81ed3065c94b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1433b896f57ce4c31c927e693cc81ed3065c94b6"}]},"branch":"refs/heads/master"},"dae90eec954411bae3689e029e9460a020bc50de":{"kind":"REWORK","_number":9,"created":"2026-07-31 16:57:08.000000000","uploader":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"ref":"refs/changes/65/997765/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/997765/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/997765/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/997765/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/997765/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/997765/9"}}},"commit":{"parents":[{"commit":"0bf09755a9145bee1b0ee1d9fd3f9eb508d9144a","subject":"[Hitachi] Add support for compression capacity saving value.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0bf09755a9145bee1b0ee1d9fd3f9eb508d9144a"}]}],"author":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-07-17 16:05:45.000000000","tz":0},"committer":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-07-31 16:57:04.000000000","tz":0},"subject":"[Hitachi] Small documentation updates","message":"[Hitachi] Small documentation updates\n\n- Update for better cut+paste preservation.\n- Update release note typo.\n- Add warning about shared ports with CSI (OpenShift) plugin.\n\nChange-Id: Id9be84b11d53dbdfc0d6150bf4beec64dd494df5\nSigned-off-by: Anthony Galica \u003canthony.galica@hitachivantara.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dae90eec954411bae3689e029e9460a020bc50de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dae90eec954411bae3689e029e9460a020bc50de"}]},"branch":"refs/heads/master"},"67b2dc77ed0da071b600ed7061ce7d059cac1eac":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2026-08-28 05:23:46.000000000","uploader":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"ref":"refs/changes/65/997765/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/997765/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/997765/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/997765/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/997765/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/997765/10"}}},"commit":{"parents":[{"commit":"49bd2a8c7fc7f8547883aa82f38b182eaf63422f","subject":"[Hitachi] Add support for compression capacity saving value.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/49bd2a8c7fc7f8547883aa82f38b182eaf63422f"}]}],"author":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-07-17 16:05:45.000000000","tz":0},"committer":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-08-28 05:23:46.000000000","tz":0},"subject":"[Hitachi] Small documentation updates","message":"[Hitachi] Small documentation updates\n\n- Update for better cut+paste preservation.\n- Update release note typo.\n- Add warning about shared ports with CSI (OpenShift) plugin.\n\nChange-Id: Id9be84b11d53dbdfc0d6150bf4beec64dd494df5\nSigned-off-by: Anthony Galica \u003canthony.galica@hitachivantara.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/67b2dc77ed0da071b600ed7061ce7d059cac1eac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/67b2dc77ed0da071b600ed7061ce7d059cac1eac"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"0bf09755a9145bee1b0ee1d9fd3f9eb508d9144a","ours":"dae90eec954411bae3689e029e9460a020bc50de","theirs":"49bd2a8c7fc7f8547883aa82f38b182eaf63422f","merge_strategy":"recursive","contains_conflicts":false}},"d0ba6fd986165e7a380ede6f97968b6ecbde8e3c":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2026-08-28 21:54:28.000000000","uploader":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"ref":"refs/changes/65/997765/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/997765/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/997765/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/997765/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/997765/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/997765/11"}}},"commit":{"parents":[{"commit":"dbb4e275c8547b9a714fc828b2a3752a77ec21fa","subject":"Merge \"NetApp ONTAP REST: restore snapshot-dir fallback for CG clones\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dbb4e275c8547b9a714fc828b2a3752a77ec21fa"}]}],"author":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-07-17 16:05:45.000000000","tz":0},"committer":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-08-28 21:54:28.000000000","tz":0},"subject":"[Hitachi] Small documentation updates","message":"[Hitachi] Small documentation updates\n\n- Update for better cut+paste preservation.\n- Update release note typo.\n- Add warning about shared ports with CSI (OpenShift) plugin.\n\nChange-Id: Id9be84b11d53dbdfc0d6150bf4beec64dd494df5\nSigned-off-by: Anthony Galica \u003canthony.galica@hitachivantara.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d0ba6fd986165e7a380ede6f97968b6ecbde8e3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d0ba6fd986165e7a380ede6f97968b6ecbde8e3c"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"49bd2a8c7fc7f8547883aa82f38b182eaf63422f","ours":"67b2dc77ed0da071b600ed7061ce7d059cac1eac","theirs":"dbb4e275c8547b9a714fc828b2a3752a77ec21fa","merge_strategy":"recursive","contains_conflicts":false}},"3eec06c697411309e0ce42c57f0972ba4015c33a":{"kind":"NO_CODE_CHANGE","_number":12,"created":"2026-08-28 21:55:46.000000000","uploader":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"ref":"refs/changes/65/997765/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/997765/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/997765/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/997765/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/997765/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/997765/12"}}},"commit":{"parents":[{"commit":"dbb4e275c8547b9a714fc828b2a3752a77ec21fa","subject":"Merge \"NetApp ONTAP REST: restore snapshot-dir fallback for CG clones\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dbb4e275c8547b9a714fc828b2a3752a77ec21fa"}]}],"author":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-07-17 16:05:45.000000000","tz":0},"committer":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-08-28 21:55:46.000000000","tz":0},"subject":"[Hitachi] Update version for Hibiscus release / small updates","message":"[Hitachi] Update version for Hibiscus release / small updates\n\n- Update versioning for Hibiscus release.\n- Update for better cut+paste preservation.\n- Update release note typo.\n- Add warning about shared ports with CSI (OpenShift) plugin.\n\nChange-Id: Id9be84b11d53dbdfc0d6150bf4beec64dd494df5\nSigned-off-by: Anthony Galica \u003canthony.galica@hitachivantara.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3eec06c697411309e0ce42c57f0972ba4015c33a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3eec06c697411309e0ce42c57f0972ba4015c33a"}]},"branch":"refs/heads/master","description":"Edit commit message","conflicts":{"base":"49bd2a8c7fc7f8547883aa82f38b182eaf63422f","ours":"67b2dc77ed0da071b600ed7061ce7d059cac1eac","theirs":"dbb4e275c8547b9a714fc828b2a3752a77ec21fa","merge_strategy":"recursive","contains_conflicts":false}},"d709905f09a863df21ce1dfd70f079636cf817fc":{"kind":"REWORK","_number":13,"created":"2026-08-28 22:09:01.000000000","uploader":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"ref":"refs/changes/65/997765/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/997765/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/997765/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/997765/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/997765/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/997765/13"}}},"commit":{"parents":[{"commit":"dbb4e275c8547b9a714fc828b2a3752a77ec21fa","subject":"Merge \"NetApp ONTAP REST: restore snapshot-dir fallback for CG clones\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dbb4e275c8547b9a714fc828b2a3752a77ec21fa"}]}],"author":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-07-17 16:05:45.000000000","tz":0},"committer":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-08-28 22:08:56.000000000","tz":0},"subject":"[Hitachi] Update version for Hibiscus release / small updates","message":"[Hitachi] Update version for Hibiscus release / small updates\n\n- Update versioning for Hibiscus release.\n- Update for better cut+paste preservation.\n- Update release note typo.\n- Add warning about shared ports with CSI (OpenShift) plugin.\n\nChange-Id: Id9be84b11d53dbdfc0d6150bf4beec64dd494df5\nSigned-off-by: Anthony Galica \u003canthony.galica@hitachivantara.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d709905f09a863df21ce1dfd70f079636cf817fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d709905f09a863df21ce1dfd70f079636cf817fc"}]},"branch":"refs/heads/master"},"54c3352c32b4b55aeb17b9aefeb892a90acecfbc":{"kind":"NO_CODE_CHANGE","_number":14,"created":"2026-08-29 03:47:23.000000000","uploader":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"ref":"refs/changes/65/997765/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/65/997765/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/997765/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/997765/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/65/997765/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/65/997765/14"}}},"commit":{"parents":[{"commit":"dbb4e275c8547b9a714fc828b2a3752a77ec21fa","subject":"Merge \"NetApp ONTAP REST: restore snapshot-dir fallback for CG clones\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dbb4e275c8547b9a714fc828b2a3752a77ec21fa"}]}],"author":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-07-17 16:05:45.000000000","tz":0},"committer":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-08-29 03:47:23.000000000","tz":0},"subject":"[Hitachi] Update version for Hibiscus release / small doc updates","message":"[Hitachi] Update version for Hibiscus release / small doc updates\n\n- Update versioning for Hibiscus release.\n- Update for better cut+paste preservation.\n- Update release note typo.\n- Add warning about shared ports with CSI (OpenShift) plugin.\n\nChange-Id: Id9be84b11d53dbdfc0d6150bf4beec64dd494df5\nSigned-off-by: Anthony Galica \u003canthony.galica@hitachivantara.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/54c3352c32b4b55aeb17b9aefeb892a90acecfbc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/54c3352c32b4b55aeb17b9aefeb892a90acecfbc"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
