)]}'
{"id":"openstack%2Fcinder~999273","triplet_id":"openstack%2Fcinder~stable%2F2026.1~I790b5b1a2d2991694a33bb2b88f45e1a0ce528fe","project":"openstack/cinder","branch":"stable/2026.1","topic":"bug_2074377_qb-stable/2026.1","attention_set":{"13915":{"account":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"last_update":"2026-07-31 12:06:44.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I790b5b1a2d2991694a33bb2b88f45e1a0ce528fe","subject":"Quobyte: Fix metadata for volume from snapshot #2142477","status":"NEW","created":"2026-07-30 12:08:40.000000000","updated":"2026-08-16 16:30:11.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":978564,"cherry_pick_of_patch_set":9,"meta_rev_id":"6374b1f8a4f2ca39d22021e54f347d937da51eee","_number":999273,"virtual_id_number":999273,"owner":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-07-31 12:06:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"all":[{"value":1,"date":"2026-08-16 16:30:11.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-30 12:38:20.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2026-07-30 13:03:17.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-07-30 13:13:13.000000000","updated_by":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"reviewer":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"state":"CC"},{"updated":"2026-07-30 13:18:38.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2026-07-30 13:33:30.000000000","updated_by":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"reviewer":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"state":"CC"},{"updated":"2026-07-30 13:40:35.000000000","updated_by":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"reviewer":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"state":"CC"},{"updated":"2026-07-30 14:57:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-31 12:06:44.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-08-16 16:30:11.000000000","updated_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"reviewer":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"state":"REVIEWER"}],"messages":[{"id":"1d9335755b04838fccb42ab0aece44f636fef210","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2026-07-30 12:08:40.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"2e98a90af261c75a65314937abab472429b12e90","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-30 12:38:20.000000000","message":"Patch Set 1:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4962/.","accounts_in_message":[],"_revision_number":1},{"id":"5eb1bfa9ed748b374d0dfdfb32b216ff0460076e","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-30 13:03:17.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I790b5b1a2d2991694a33bb2b88f45e1a0ce528fe .","accounts_in_message":[],"_revision_number":1},{"id":"c52d231b1c2f3f45abd4e3a2c230ec6400963d8b","author":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"date":"2026-07-30 13:13:13.000000000","message":"Patch Set 1:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_NVME_CI build/665 : FAILURE in 0h 1m 56s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/999273","accounts_in_message":[],"_revision_number":1},{"id":"b1a6c81af47fe5d1b0927b54db63524f0d9ed53e","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-30 13:18:38.000000000","message":"Patch Set 1:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/05db8fa99bff4250bd5196ee39f1e69c/","accounts_in_message":[],"_revision_number":1},{"id":"b1782c8a8945e1a18475edec5697df7caa2c0ee6","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-07-30 13:33:30.000000000","message":"Patch Set 1:\n\nHPE AlletraMP FC CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_FC_CI build/757 : FAILURE in 0h 2m 7s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/999273","accounts_in_message":[],"_revision_number":1},{"id":"c6fa12510b5177311d07ff5a9ae894ff4274526a","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-07-30 13:40:35.000000000","message":"Patch Set 1:\n\nHPE AlletraMP iSCSI CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1026 : FAILURE in 0h 1m 59s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/999273","accounts_in_message":[],"_revision_number":1},{"id":"096e180519d22176dfb174af1b373aeab9f75ea9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-30 14:57:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9efd813b09da479f8c91a94fc2d49eb1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28c6c7f3a1c74ebab4a7f7183abd2b74 : SUCCESS in 6m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0b46d7cba14246ed96ee7704347bb4cd : FAILURE in 12m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/01b9ae9d7d4048b2a08f1b1fbb80d456 : SUCCESS in 12m 11s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/bd8a7b8dce6c4601af67f114bee4648d : FAILURE in 16m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/221348aa1d304efeb680a3c45c437d65 : SUCCESS in 14m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/3b57ce0eddc946c39c57a88cef9c0d0b : SUCCESS in 1h 01m 36s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f290347e08dc41aaa29c8f1ff22f4991 : SUCCESS in 33m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ce38a884ebc4447692044f6605c74aee : SUCCESS in 1h 39m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/50defc158dab413298f0be820fa7a73c : SUCCESS in 1h 09m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/57711617c3fe4023aac1d5c6f0e92175 : SUCCESS in 9m 41s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a0ce735872b34a36af5e9aa1a2da7519 : SUCCESS in 31m 00s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ac648660d66646c1af5a9b221bdea462 : SUCCESS in 5m 09s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f11cd42e6df24867bcae8bf2966baecc : SUCCESS in 7m 09s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7ad4b864b4e242ac8544cd434f8b8db3 : SUCCESS in 6m 18s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/87d061125ed8442aafef1706fe5684d1 : SUCCESS in 6m 20s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/6a3ac533c9ab4cb48568b8b32cef2c4a : SUCCESS in 42m 59s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/55bf05a71d224901a42a320cd0418b30 : FAILURE in 5m 53s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1672a4295c754b469cea63247acc6862 : SUCCESS in 1h 58m 43s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/506ec2ba20bb467dba8e73b46674ece9 : FAILURE in 2h 35m 02s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/66f231330c454d38b43c7b61baaa5b12 : SUCCESS in 1h 58m 55s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/9a37cd7eeeca4e3c97982f157d63b900 : FAILURE in 8m 56s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/3b744ef0536e4904943bd5611d98f471 : SUCCESS in 41m 44s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d8df038e1a224c19870673165002920e : SUCCESS in 1h 13m 01s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6710fca84710483d83ecbb9e997e7490 : SUCCESS in 1h 57m 37s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/4e67ef17fc2f4d6b8d3b2030ee653722 : SUCCESS in 42m 51s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ae643f9fae564642a6a927e55f71f014 : SUCCESS in 1h 57m 06s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/902e384a068941e5a9cfe9835518da92 : FAILURE in 8m 48s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/cab0968c5f6a4dbdbafdc4a2b6803d94 : SUCCESS in 1h 49m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b22e7cf64f5841c6831be3de38f1cf67 : SUCCESS in 57m 14s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job devstack-plugin-nfs-tempest-full-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job devstack-plugin-nfs-tempest-full-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"cfad56cbba66489ec4b9febd11e6cec0be802fee","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-30 15:00:40.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/05db8fa99bff4250bd5196ee39f1e69c\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3243bdea05364c9099b96dc2e164143a : FAILURE in 1h 01m 08s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/04ac2e9329dd4f10ab02f1b6ba2af202 : FAILURE in 34m 28s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1054d3cbbbfe47ac9850211f8f7e6788 : FAILURE in 49m 10s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/39d25147f86f427196e2b8df3d1366f1 : FAILURE in 35m 46s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b1f6afb44bd64d18a32047ead3e388ec : FAILURE in 17m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c337d038c4b2f6cc0db7fd54794d8e87a5af18d7","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2026-07-31 10:10:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7de8258a4b28fda2256a5ba15ce708c8ef326601","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-31 12:06:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5cbf9d87f37f4bc1ac9f06bd9085a6cf\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0fa6fb58a98d4d1fb0b570f047bd73bd : SUCCESS in 3m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f139d2d45e584e4793a526d4a2392662 : SUCCESS in 11m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/78f36e1fd9ee4fda898f42bf676c9fe3 : SUCCESS in 8m 33s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/54ac88944d8a446db9561da2b1170b18 : FAILURE in 8m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13b3c62c0e844e5686a601701d861a7c : SUCCESS in 13m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/ee09c50d8acf433b8e0ad901fa98aa62 : SUCCESS in 36m 33s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c3d74774149541b8891a26ab88d13316 : SUCCESS in 35m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/624ee7b7c33f412aa4880569c17b1b2a : SUCCESS in 1h 24m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/afa54088bfdc4b02807da197c8f3824c : SUCCESS in 1h 05m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8b67357492674de38214974ea7b970cf : SUCCESS in 6m 37s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d1cfbfcbac1e4984a30a8045f9890944 : SUCCESS in 20m 39s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/16741869524a49e2adb3ade35e444a38 : SUCCESS in 4m 28s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c520bf0e5d20409fb28e55a657c9927b : SUCCESS in 4m 04s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e48fd5c1c5004b71ac8d6b6e7a7d04aa : SUCCESS in 7m 39s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/8faef30655c740889aaebbfbc26245b9 : SUCCESS in 5m 30s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b19bee59e25b442fb8d0d8dbedf602be : SUCCESS in 35m 22s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1c971d4c43d24ab988b4f753a0560071 : FAILURE in 5m 49s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/51f269e7b8864bbaad88853a9d8588f2 : SUCCESS in 1h 49m 55s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/c1c15ec0b0c34e8aadd9e037014079b4 : SUCCESS in 1h 43m 54s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/66069c866a6b4766ad61f83d3389a579 : SUCCESS in 1h 48m 26s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/bccb5e09c8964723989f647897757e28 : FAILURE in 9m 07s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/cd9706a2b5554562a42213fed0879bce : SUCCESS in 35m 49s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/34070ef20b554cc3be6e60acbec9999b : SUCCESS in 1h 13m 40s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/bac33b35d97843059cc8696550653bdd : SUCCESS in 1h 14m 17s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/87664b95ec6d4871b208d8ab0b96ed44 : SUCCESS in 46m 01s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/3337bf1e97744638a7774e3cdb614515 : SUCCESS in 1h 13m 41s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/6114ea71ad1944d1b95d2b3290c8e7d1 : FAILURE in 8m 43s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5aa50b419d9b436fb238d8f1541d7f93 : SUCCESS in 1h 43m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1f9b7dbd1de2412b8e827ca148028629 : SUCCESS in 59m 09s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job devstack-plugin-nfs-tempest-full-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job devstack-plugin-nfs-tempest-full-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"e601534dffaf6f7b43cb259a804239369f02fc84","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-07-31 12:17:05.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/31_07_2026_12_10_17\nLeave a comment with","accounts_in_message":[],"_revision_number":1},{"id":"6374b1f8a4f2ca39d22021e54f347d937da51eee","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-08-16 16:30:11.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1f33a97ae71d6ad95824e98269874a13212ff63c","revisions":{"1f33a97ae71d6ad95824e98269874a13212ff63c":{"kind":"REWORK","_number":1,"created":"2026-07-30 12:08:40.000000000","uploader":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"ref":"refs/changes/73/999273/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/73/999273/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/73/999273/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/73/999273/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/73/999273/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/73/999273/1"}}},"commit":{"parents":[{"commit":"a8945538cffde0cce3b491320757bd7abc574bfe","subject":"Quobyte: adapt to image_utils fixes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a8945538cffde0cce3b491320757bd7abc574bfe"}]}],"author":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2026-03-02 16:42:42.000000000","tz":60},"committer":{"name":"Silvan Kaiser","email":"silvan@quobyte.com","date":"2026-07-30 12:08:40.000000000","tz":0},"subject":"Quobyte: Fix metadata for volume from snapshot #2142477","message":"Quobyte: Fix metadata for volume from snapshot #2142477\n\nFixes missing format metadata in the Quobyte volume driver when creating\na volume from a snapshot.\n\nCloses-Bug: #2142477\nChange-Id: I790b5b1a2d2991694a33bb2b88f45e1a0ce528fe\nSigned-off-by: Silvan Kaiser \u003csilvan@quobyte.com\u003e\n(cherry picked from commit 7884bcd49ec2efa1b939f732e015227d16b3668f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1f33a97ae71d6ad95824e98269874a13212ff63c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1f33a97ae71d6ad95824e98269874a13212ff63c"}]},"branch":"refs/heads/stable/2026.1","conflicts":{"base":"57dc9ece9652af87535c533360acb2f07a8d5e0d","ours":"a8945538cffde0cce3b491320757bd7abc574bfe","theirs":"7884bcd49ec2efa1b939f732e015227d16b3668f","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
