)]}'
{"id":"openstack%2Fcinder~999310","triplet_id":"openstack%2Fcinder~master~Ia94ddfacac494a301866eab565f04c921f6c2ea5","project":"openstack/cinder","branch":"master","topic":"bug-2157914-review","attention_set":{},"removed_from_attention_set":{"13425":{"account":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"last_update":"2026-08-21 18:05:18.000000000","reason":"Change was submitted"},"38371":{"account":{"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},"last_update":"2026-08-21 18:05:18.000000000","reason":"Change was submitted"},"36180":{"account":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"last_update":"2026-08-21 18:05:18.000000000","reason":"Change was submitted"},"38059":{"account":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"last_update":"2026-08-21 18:05:18.000000000","reason":"Change was submitted"},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2026-08-21 18:05:18.000000000","reason":"Change was submitted"},"36171":{"account":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"last_update":"2026-08-19 12:49:22.000000000","reason":"\u003cGERRIT_ACCOUNT_36171\u003e replied on the change","reason_account":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"}},"38623":{"account":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"last_update":"2026-08-21 18:05:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia94ddfacac494a301866eab565f04c921f6c2ea5","subject":"NetApp: default HTTPS transport and TLS verify","status":"MERGED","created":"2026-07-30 15:50:04.000000000","updated":"2026-08-21 18:06:36.000000000","submitted":"2026-08-21 18:05:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":71,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"999310-bug-2157914-review","meta_rev_id":"8c5fd2ee4a2eb7b8ac7f6ba25ff70a030925d521","_number":999310,"virtual_id_number":999310,"owner":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},{"value":0,"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-21 18:05:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"recommended":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"all":[{"value":1,"date":"2026-08-18 14:52:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":2,"date":"2026-08-21 16:16:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},{"value":0,"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"value":2,"date":"2026-08-21 15:31:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":1,"date":"2026-08-19 12:49:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":1,"date":"2026-08-21 16:16:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},{"value":0,"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},{"value":0,"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},{"_account_id":39225,"name":"Prudhvi","display_name":"Prudhvi","email":"prudhvi.openstack@gmail.com","username":"Prudhvi"},{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-30 15:55:37.000000000","updated_by":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"reviewer":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"state":"CC"},{"updated":"2026-07-30 15:55:58.000000000","updated_by":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"reviewer":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"state":"CC"},{"updated":"2026-07-30 16:06:19.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2026-07-30 16:26:02.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2026-07-30 16:37:03.000000000","updated_by":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"reviewer":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"state":"CC"},{"updated":"2026-07-30 16:45:40.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-07-30 18:15:46.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-07-30 18:41:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-07-30 18:52:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-30 18:52:47.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-07-31 03:53:21.000000000","updated_by":{"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},"reviewer":{"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},"state":"REVIEWER"},{"updated":"2026-07-31 05:42:21.000000000","updated_by":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"reviewer":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"state":"REVIEWER"},{"updated":"2026-07-31 11:33:12.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-08-03 14:31:56.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2026-08-06 04:05:55.000000000","updated_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"reviewer":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"state":"REVIEWER"},{"updated":"2026-08-06 12:55:29.000000000","updated_by":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"reviewer":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"state":"CC"},{"updated":"2026-08-06 12:55:34.000000000","updated_by":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"reviewer":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"state":"REVIEWER"},{"updated":"2026-08-10 11:44:54.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"},{"updated":"2026-08-12 15:46:44.000000000","updated_by":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"reviewer":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"state":"REVIEWER"},{"updated":"2026-08-17 10:52:01.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2026-08-18 10:25:05.000000000","updated_by":{"_account_id":39225,"name":"Prudhvi","display_name":"Prudhvi","email":"prudhvi.openstack@gmail.com","username":"Prudhvi"},"reviewer":{"_account_id":39225,"name":"Prudhvi","display_name":"Prudhvi","email":"prudhvi.openstack@gmail.com","username":"Prudhvi"},"state":"CC"},{"updated":"2026-08-21 15:31:18.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2026-08-21 16:16:37.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"}],"messages":[{"id":"03abcb83f1e592fb1cfd6d1cafc2b7c1b636192b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-07-30 15:50:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c7bf96dead32da803cb2e80dd3dffea06cce3577","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-07-30 15:55:37.000000000","message":"Patch Set 1:\n\nHPE AlletraMP FC CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_FC_CI build/761 : FAILURE in 0h 0m 11s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/999310","accounts_in_message":[],"_revision_number":1},{"id":"5dd1f51f3ee399f98ac722b60c357fea1aca5870","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-07-30 15:55:58.000000000","message":"Patch Set 1:\n\nHPE AlletraMP iSCSI CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1030 : FAILURE in 0h 0m 11s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/999310","accounts_in_message":[],"_revision_number":1},{"id":"4e1051c0531acac79319db7704399e7044d8fbf7","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-30 16:06:19.000000000","message":"Patch Set 1:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4793/.","accounts_in_message":[],"_revision_number":1},{"id":"1fd0db6b97986ccbfd9584a8f6ea806635eab80e","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-30 16:26:02.000000000","message":"Patch Set 1:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/4e3df88132fb474e9b07c639e196ab5b/","accounts_in_message":[],"_revision_number":1},{"id":"e55215dca5f0d7e090f21cfae8de59afc31a7f6f","author":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"date":"2026-07-30 16:37:03.000000000","message":"Patch Set 1:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/669 : SUCCESS in 0h 46m 41s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/999310","accounts_in_message":[],"_revision_number":1},{"id":"8fc6b82c76f896a2bbdeaf137446d783e9b23515","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-30 16:45:40.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ia94ddfacac494a301866eab565f04c921f6c2ea5 .","accounts_in_message":[],"_revision_number":1},{"id":"8fcc94a0b34d05600d80a152105b7ea6f66bfaf9","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-30 17:31:08.000000000","message":"Patch Set 1:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/4e3df88132fb474e9b07c639e196ab5b\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5c0b373454cc45b0894cb406e6e7f70e : SUCCESS in 38m 40s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/86d8adfa4f624c2b8b81ca4118a4f70c : SUCCESS in 35m 29s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/79f9e4b9901a4f21a362b5f527f15838 : SUCCESS in 50m 21s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5a8787849ba549ad8054188142d09bcd : SUCCESS in 35m 47s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/839a77795da04eaea2a3a18b24606ca3 : FAILURE in 16m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3483ea6257f6cc4cebe150953d1678362d9956d9","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-07-30 18:15:46.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/4e28efd10a19449391a967758657c44e : SUCCESS in 43m 06s\n- cinder-storpool-tempest-iscsi-multipath build/46c355b09ce1413bac990365fefc45fe : SUCCESS in 47m 55s","accounts_in_message":[],"_revision_number":1},{"id":"25c967f58247d0f1e1bc2c36b1774400731abba2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-30 18:41:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d5cdd6768a14d748f44f377c028c4df\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/28bb806cd6be4fe18c75906d630d11ba : SUCCESS in 28m 21s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c9e690b310da40939e9a86c7191a34c1 : SUCCESS in 27m 46s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/857c8bd7dfd848f6a3c55466870c807c : SUCCESS in 40m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7c3130ad1d529f886b40530482934255fd0bfb0a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-30 18:52:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca3723af71aa4b5488107384312e6cbb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5dc319cb8cb6420d892089283d36f83b : SUCCESS in 5m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/30aec437491744149d8e3bca6a5d471d : SUCCESS in 12m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ec02364b40ee4378870ab63921672a6d : SUCCESS in 11m 13s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ee358a74319f4d729609f22716d85759 : SUCCESS in 10m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a233dc1d2c454be689c20757fb33edb2 : SUCCESS in 17m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/d91d84241bfb4ecfa1b9db1329b145db : SUCCESS in 44m 39s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b920a8d7b12f40ab9d81b85aa9f6221c : SUCCESS in 43m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e6cf4e23113c46fe8bd20196171c941c : SUCCESS in 1h 24m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5162e236bdaa4063bf49f761c8e45568 : SUCCESS in 31m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/34a787e202d7424587999e4295aea767 : SUCCESS in 8m 53s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4721e816ba894c19b3c70fc7e434378d : SUCCESS in 27m 03s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b5b43772d5114a3691f834be28d57206 : SUCCESS in 4m 35s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d4c82eb45cac421ea5bb1661db39cc77 : SUCCESS in 3m 48s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/df5c6c95908943c0b37082a5a9b6f1b5 : FAILURE in 14m 27s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/868ab5ded9df4acabb25f3f4bd7634d7 : SUCCESS in 5m 16s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/c1a02db2692a4e69b30137120eaba99d : SUCCESS in 5m 54s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/ed75db829dba48868775d6faa29c3f45 : SUCCESS in 9m 07s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d25282e779e548ff8e006ca9b2600ce3 : SUCCESS in 41m 38s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9bf8186c954a4dfdaf3a7c47b50e393e : SUCCESS in 6m 12s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/93b265b4f629424ea91eeb5bc453c9cd : SUCCESS in 1h 42m 33s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3b5218bcc58d4363aaad8e52663e1c71 : FAILURE in 1h 51m 28s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2b2e8d9b9dc3401f8b30aa2695a75c0e : SUCCESS in 1h 57m 05s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/628924c9e9d24256a1eee589398a408e : FAILURE in 2h 56m 57s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/f07518e43f8345ddbfefb48b2031e6b7 : FAILURE in 1h 56m 27s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/bf46f0e6c7a04007a218d3519ac1eb12 : SUCCESS in 25m 49s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4fdd9c68fbf24255949a1a846755f90a : SUCCESS in 1h 23m 14s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5cf9d1fec5274addb3d179fca6d6c571 : SUCCESS in 1h 36m 40s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/2192ff600ea24e75aa6091e4ddaf0d1a : SUCCESS in 50m 47s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/929ef04489044e29bb6c618c3fa32118 : SUCCESS in 1h 33m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/947372492feb4f5681f517ee303094b0 : SUCCESS in 1h 05m 06s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"90ad795d1b202464f457d2bfffbc371965bf715a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-07-30 19:03:13.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/30_07_2026_18_56_11\nLeave a comment with","accounts_in_message":[],"_revision_number":1},{"id":"8e7cf5a3601e54cfaaa2175e54d74645bb95bb2b","author":{"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},"date":"2026-07-31 03:53:21.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"99f170bf6e45373f368e43f8b8fdbb730fd5d437","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-07-31 03:58:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f3248dd76a32ee5b4f91aea31d70cdd00a0436c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-07-31 04:04:52.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"6b0a3134b88aaaca29a940024f6001b22f6afa05","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-07-31 04:05:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cb99ed3371d18459d69009da69e87c6fa1c4712b","author":{"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},"date":"2026-07-31 04:15:54.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"78f5b3c0c32faa05d80a87dd0fcd1baefeee78e8","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-31 04:19:51.000000000","message":"Patch Set 2:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4797/.","accounts_in_message":[],"_revision_number":2},{"id":"d2430b4af1926d7ce657c5e58adb5c8ad280f857","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-31 04:29:06.000000000","message":"Patch Set 2:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4973/.","accounts_in_message":[],"_revision_number":2},{"id":"cae400fee15c42bc098ee32881ca08c3da0f89bd","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-31 04:40:24.000000000","message":"Patch Set 2:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/9c76d3b83cd3422bb7bb5cb70806cce1/","accounts_in_message":[],"_revision_number":2},{"id":"525caaad14adb4bb6a02f3ae2eb06b7def29ec21","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-07-31 04:55:40.000000000","message":"Patch Set 2:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1033 : SUCCESS in 0h 50m 26s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/999310","accounts_in_message":[],"_revision_number":2},{"id":"326d48bcd5e86f5eea8d544bb306255836736603","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-07-31 04:55:53.000000000","message":"Patch Set 2:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/764 : SUCCESS in 0h 50m 41s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/999310","accounts_in_message":[],"_revision_number":2},{"id":"7672d75551ecc4e59f51c31e373a1850a038345a","author":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"date":"2026-07-31 04:56:54.000000000","message":"Patch Set 2:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/672 : SUCCESS in 0h 51m 43s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/999310","accounts_in_message":[],"_revision_number":2},{"id":"ad7b7c642e8d166aea88ac7b92bfc6579e94d60b","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-07-31 04:58:13.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/1115d590ad1f48c3a3023e22a4b2af50 : SUCCESS in 45m 48s\n- cinder-storpool-tempest-iscsi-multipath build/7661a12286524a96b5552384bcfa8430 : SUCCESS in 51m 57s","accounts_in_message":[],"_revision_number":2},{"id":"3730b21432dbd1155e20c3e4ca9a1d350418bf99","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-31 05:03:21.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ia94ddfacac494a301866eab565f04c921f6c2ea5 .","accounts_in_message":[],"_revision_number":2},{"id":"1a84b1b1e7babaf88d546826a8a58e933b378612","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-31 05:07:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ecbaca85e1940bb89dfe2e6e98288d1\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/64e6f7cf0bb7402bb38261080e386100 : SUCCESS in 25m 07s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3246076b6c1144abb9d0bbbe7dca93a9 : SUCCESS in 16m 31s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/14e690c0f649416bba7f066a37ebd60f : SUCCESS in 34m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2d82611c95226087e0d09b5334c84594d488495f","author":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"date":"2026-07-31 05:42:21.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"9139344dbb442025e9d2246143459f1f38779cf4","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-31 06:24:50.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/9c76d3b83cd3422bb7bb5cb70806cce1\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/bf13dab7fe1e4529838454484ba96b97 : SUCCESS in 40m 40s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4c1289d582394b20b196edc052e45274 : SUCCESS in 38m 26s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9880ec4a48a64363afade20b130d84b9 : FAILURE in 2h 19m 26s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0bc33fc6f568489b80b40b939c0b8a9d : SUCCESS in 35m 03s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7a6390cba94948b89aade7fcc93893df : FAILURE in 14m 02s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"52996ee16bc4932a06e776ff09564a916da1f496","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-31 06:32:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4068b65a42624747b2307a898e12127c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f8939e3c5bc40bd9f15099af274266e : SUCCESS in 3m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/51401fb2f70e4b0dac4ded10817e83bd : SUCCESS in 9m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/be2c6886f24d413493fc36a743bf7428 : TIMED_OUT in 41m 14s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3b951598401947eaa635fd61ea7105e1 : SUCCESS in 15m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ed73d5cf2ae419a9c007cdfa4591063 : SUCCESS in 8m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/fca1bb4e022749789db35fbb3fc5a149 : SUCCESS in 32m 31s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4599e4d251f04208b7d2a2f9b3c720cd : SUCCESS in 30m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7d76c3a2d4914e29bb8f5eef4ebcb72e : SUCCESS in 53m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/141b6a001f414a238865f88dfd109cc0 : SUCCESS in 32m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/726abb57d19e426cade1f6fdc96402f2 : SUCCESS in 9m 45s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/1043acc713ff4d318360cd076ab3b9dd : TIMED_OUT in 40m 58s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/874a915fda3046b3a5f974d9de4afe6b : SUCCESS in 6m 20s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d8f7276555b945dfa4d968b12e520bb7 : SUCCESS in 7m 38s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/3dd71e8562f94a90b35d6244a9484d77 : FAILURE in 10m 08s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/6397b071038d4dd09e902f5a4422a2f1 : SUCCESS in 6m 33s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/d34372c90dd34cd78bfcb1f6e7ea54fe : SUCCESS in 4m 39s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/67c1fe9273444e539878803ce32fb3ef : SUCCESS in 6m 14s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/138a2e3b0fe34724b99dd313c7694c5b : SUCCESS in 36m 24s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1ca0d4faffbb4e94831b41eaf773fd2f : SUCCESS in 3m 47s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ed048a107cc3415f8d009ee11629c311 : SUCCESS in 1h 33m 34s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d736dbf4fe874add9927291183876cc7 : FAILURE in 1h 05m 59s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2f5bc2110e4e4a73a3deaaf92a8ace30 : SUCCESS in 1h 58m 53s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/59c4e388e28e48318b284b53b3fcd36b : FAILURE in 2h 24m 14s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/b17d9a18320e4d49934f833731dcc517 : FAILURE in 1h 31m 48s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/31b4aec846034529a875a50445ed764c : SUCCESS in 33m 40s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/489eb12e7a924b949140e6383eae23a3 : SUCCESS in 1h 19m 32s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/57eed711aaf84a08abe0afc65c9e7248 : SUCCESS in 1h 31m 54s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a9a5c9e1077749088e3841818b535138 : SUCCESS in 38m 46s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2c5946a7d0db4ac49dd2f5a2cb3ef1a8 : SUCCESS in 1h 42m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/06cf84a15ba34e168d2a9797b9372034 : SUCCESS in 48m 17s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":2},{"id":"06bc1c3a487ee6d151f53b817b224329d1a569d5","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2026-07-31 11:33:12.000000000","message":"Patch Set 2:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/10/999310/2/20260731153151/NEC-ISCSI : FAILURE in 4h 24m 43s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/10/999310/2/20260731153151/NEC-FC : FAILURE in 5h 01m 19s","accounts_in_message":[],"_revision_number":2},{"id":"6969edd722573c1ea95e7a2ee4b835a1d10aa0be","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-08-03 14:31:56.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8ca313751f7c18ac0bc02f166459655141181d72","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-08-03 14:32:04.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d088919e8d5d50e5d37a12c317ca013e79d4986d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-03 15:25:53.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/81f6b0a9339c4f368cf1a9fa87ccd04a : SUCCESS in 46m 56s\n- cinder-storpool-tempest-iscsi-multipath build/a8e457529ef6473cab3e9b4887894f2a : SUCCESS in 52m 24s","accounts_in_message":[],"_revision_number":2},{"id":"018837d2c3458f466632b9201135309c7d5d9075","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-03 16:02:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0454d211f6fc4e098f8367c2ca70f042\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/44f5941e92304508947c2efc14dfd19c : SUCCESS in 17m 48s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6f9cc663c7d74dee9171765504f4ed8c : SUCCESS in 20m 46s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/60b633dec57542d4adb480b08e9bafd9 : SUCCESS in 27m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9f8872ec12e7649a5adbfb358838524a75cea135","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-03 17:03:21.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a361dba8798c4c7797a1ab2f06128c0d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2ec7c60a08a42109e1654e055bc7539 : SUCCESS in 6m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/067f0c27d2334dd484622488afe67419 : SUCCESS in 11m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b76f5965c0bc40c58bf52306077bcb73 : SUCCESS in 8m 41s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5938a950c78c42f2a2a1f552b45b8684 : SUCCESS in 12m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2d16d6e8994466baddd723260ea8a56 : SUCCESS in 12m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/0d2e923c3ccb46b69d1fa151de59c9f5 : SUCCESS in 48m 06s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2f222b92ecc84fbcaf5b8b1998a1cabd : FAILURE in 16m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3237fef7f78d495aa6adf15431b9b9cb : SUCCESS in 1h 00m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/59525a860286446b93a52dd277a4424c : SUCCESS in 59m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/08ff62cc79324580a991fa88d60a43dc : SUCCESS in 9m 21s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/828b9eff306a46929495d6cf8fb1f418 : SUCCESS in 24m 43s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/0adc2eaddcab4d81a5fc710dd16f4af3 : SUCCESS in 4m 33s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2a4339ed0e13434083a2efa5a5766b9c : SUCCESS in 6m 52s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/f2c13c91b9414e419e43c89bb9e0000a : FAILURE in 10m 43s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/9dd4d3e1bbd24b7ba5c9f952c53e123d : SUCCESS in 4m 39s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/d52c88e1f9084405a3d1ae79fc905213 : SUCCESS in 5m 48s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/2cb1b2ad6c0d4b7ea68e3b0f9b38bda8 : SUCCESS in 8m 58s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3689dc39fe7640a3b289ec1613e317f8 : SUCCESS in 38m 16s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b1c59ee86cac4dee9be270cf0ea7fddf : SUCCESS in 3m 59s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/323a4f0bcbab43e98d9bd5a06c3bb045 : SUCCESS in 1h 12m 04s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/beb259495bd044a68811fa01b80a63fe : FAILURE in 2h 00m 11s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/93a4aff5c3d148b2a3d1d1e1e4078be2 : SUCCESS in 1h 39m 04s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/d6644ef3623646d9a9e673249ee4754c : FAILURE in 2h 25m 36s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/5e4e596517574f93b112f6e0231e14a2 : FAILURE in 1h 20m 10s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/d99558c8bcca4acc9b39d80c3baabbae : SUCCESS in 36m 14s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b682971b67c04fa795000cddb9da39ca : SUCCESS in 1h 15m 51s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0b60fb00d287440db1624add1f3db93c : SUCCESS in 1h 03m 29s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/b5294a5ce38d4fec93732221f77ee88e : SUCCESS in 34m 38s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/38b383fd47ab4a888086ef0ab7bd58fa : SUCCESS in 1h 18m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f064c7a273414add89b40fade25dd9b6 : SUCCESS in 58m 46s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":2},{"id":"79953b2cec659e540cdb80f4dd5cca7d450ad6e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-08-05 03:56:00.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"b4b1d2ced2f799fce34f9b94f5ad87c1f4ed2420","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-05 04:09:27.000000000","message":"Patch Set 3:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5027/.","accounts_in_message":[],"_revision_number":3},{"id":"fe69e22f2acad2c77d402bdc27acc3dd09030ee6","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-05 04:15:14.000000000","message":"Patch Set 3:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/7ead0471f4c44ee5ab897a1e3d218fb4/","accounts_in_message":[],"_revision_number":3},{"id":"8481fc513f87f296652c0bb131cd664bc2e041a3","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-05 04:18:03.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-storpool-tempest build/ca5765d16ff9460e9eb8335d7bfacaef : RETRY_LIMIT in 20m 42s\n- cinder-storpool-tempest-iscsi-multipath build/16500cbc06d1425ea125d08dcddd4395 : RETRY_LIMIT in 18m 26s","accounts_in_message":[],"_revision_number":3},{"id":"189554b2095e874b76c797ccb2d2116335d3fbe2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-08-05 04:25:20.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"3f81ed7b5d68726f22e0e586718250daceea9ddb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-08-05 04:27:27.000000000","message":"Uploaded patch set 5: New patch set was added with same tree, parent tree, and commit message as Patch Set 4.","accounts_in_message":[],"_revision_number":5},{"id":"561e1da44d3c0ffc152679c7ccef151ed6c97098","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-08-05 04:34:18.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"68b9c635cc3a79e6eb649a8c1d932be6e9a21a5b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-05 04:34:18.000000000","message":"Patch Set 4:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4852/.","accounts_in_message":[],"_revision_number":4},{"id":"7ea4f6e8afa45dac26e1bc13c810a608fb3cee75","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-08-05 04:34:31.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"8f987f3cbdef3c86b2bfb4fca1d488ba2a2ce107","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-05 04:38:28.000000000","message":"Patch Set 4:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5028/.","accounts_in_message":[],"_revision_number":4},{"id":"884d0435e39a072cada101cffab9bac74db7ad4c","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-08-05 04:40:12.000000000","message":"Patch Set 6:\n\nHPE AlletraMP FC CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_FC_CI build/819 : FAILURE in 0h 3m 8s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/999310","accounts_in_message":[],"_revision_number":6},{"id":"d399f13633b0957deca22dea4081abed9b233c22","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-05 04:48:46.000000000","message":"Patch Set 5:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5029/.","accounts_in_message":[],"_revision_number":5},{"id":"f2b99f1cb6a55de434a9d9e8e26e93843394a174","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-05 04:52:58.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eaef20f4457444a9b80d5b50ab1d2bde\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5563134c1f9f4f9896e12c4dbf3afa5a : FAILURE in 10m 47s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0dad085cafed471a917fdfe0e0cfd171 : FAILURE in 10m 46s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/9ad4fd68a7af4384b35e99eb5de98194 : FAILURE in 16m 56s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"f0f309e8b97e2c1646519f19ea2a995ed34a2ed1","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-05 04:53:38.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ia94ddfacac494a301866eab565f04c921f6c2ea5 .","accounts_in_message":[],"_revision_number":3},{"id":"b56e2195d57a71415ce33fa4526ea5fbb26f0db3","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-05 04:54:47.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-storpool-tempest build/68d288d189044acc9e8f569f70529dc3 : RETRY_LIMIT in 16m 57s\n- cinder-storpool-tempest-iscsi-multipath build/33fa19b59413407da7188d4019b45171 : RETRY_LIMIT in 17m 51s","accounts_in_message":[],"_revision_number":6},{"id":"514a396bf044b355ae12db4a266ac3e194979341","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-05 04:57:54.000000000","message":"Patch Set 6:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4854/.","accounts_in_message":[],"_revision_number":6},{"id":"0263ccee37409dbbbf5f3d29710272b57ec5623d","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-05 04:59:09.000000000","message":"Patch Set 6:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5030/.","accounts_in_message":[],"_revision_number":6},{"id":"892296e43831ef24068e9dd4f7d52c60e4a6f8a1","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-05 05:11:33.000000000","message":"Patch Set 6:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/5b1294326dc644609428b66dbf993060/","accounts_in_message":[],"_revision_number":6},{"id":"7187ea51f2fbddb4c76515081de9e5f5b868a9c7","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-05 05:23:55.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ia94ddfacac494a301866eab565f04c921f6c2ea5 .","accounts_in_message":[],"_revision_number":4},{"id":"9e97ce329528087ea316b157aec067a9ada7756a","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-05 05:25:09.000000000","message":"Patch Set 6:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/5b1294326dc644609428b66dbf993060\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3248627a57a14f21aa3d404defd3d280 : SUCCESS in 39m 48s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/234a392b3efd4a2fba140fb425cc4de2 : SUCCESS in 38m 30s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/410940e32fe44a2ea2e0c4966798acb3 : SUCCESS in 50m 07s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/98901c9927e04ff09642f4d952705335 : SUCCESS in 36m 35s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6aa9d9698eee4eceb5496248c23e97d9 : FAILURE in 16m 39s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"0edca618ddcfa33f11d2fe0e152bc00c6b9de2c4","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-08-05 05:26:22.000000000","message":"Patch Set 6:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1087 : SUCCESS in 0h 48m 48s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/999310","accounts_in_message":[],"_revision_number":6},{"id":"5fddb02373bedfd9466d6ca3202058cde4bbd58b","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-05 05:26:22.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ia94ddfacac494a301866eab565f04c921f6c2ea5 .","accounts_in_message":[],"_revision_number":5},{"id":"b235c74e64d9155c75296861ed5b5abd749fe77a","author":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"date":"2026-08-05 05:26:41.000000000","message":"Patch Set 6:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/726 : SUCCESS in 0h 49m 8s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/999310","accounts_in_message":[],"_revision_number":6},{"id":"17efdeb484a898befc5c93adeadffe6c62ab50c1","author":{"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},"date":"2026-08-05 05:27:51.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"87c2005b705e419d232cd4b626afc8c5af64866e","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-05 05:32:43.000000000","message":"Patch Set 6:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ia94ddfacac494a301866eab565f04c921f6c2ea5 .","accounts_in_message":[],"_revision_number":6},{"id":"894909b78119af3a9ce64c9d26f382fa884092b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-05 06:59:22.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0a08162941ef44249702bb3ccee23029\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f28ccbdc43a2469f8d5d3f084cca1f13 : SUCCESS in 3m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2a96e4c13a4d409ab4d76b9ddc0ad6d2 : FAILURE in 9m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d12597d9af044c7c8906a9c1692389a2 : FAILURE in 6m 10s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/fe94819729344c9fb36b936d0677c769 : FAILURE in 13m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/245cc2799c1e4ef09402ad94a848f8bd : SUCCESS in 8m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/8823a60a123b4f3e9951039345cbff0f : SUCCESS in 28m 53s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0902a61268c4430fb983467690f91f2c : SUCCESS in 59m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bae50023ce3244f2a56d58b1b83c92c5 : SUCCESS in 55m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c3422ce7e9ab4d10a6aa831988e70bcc : SUCCESS in 41m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0a6656ea6b3944e4814b0a9c03d0db91 : SUCCESS in 8m 02s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/5d06450f06ae4e6e95146dc827a29557 : FAILURE in 18m 14s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2b7704b25f4f4f269b6f97478fff7dc7 : SUCCESS in 3m 13s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/49ff4879aa0040d68308a332e756e76a : SUCCESS in 7m 19s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/192863aeecd7487d922de000b83ca0c6 : FAILURE in 10m 45s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/e9a55958d7cf471995454915b1964290 : SUCCESS in 7m 09s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/170c7796180f441988aacb52feb1cc23 : SUCCESS in 6m 28s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/63d9b373e1764519b8d4d94df965dfcd : SUCCESS in 5m 56s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/6f82b4e0834846e2b3299385e966c719 : SUCCESS in 27m 11s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6fb872182c334cc598a09beecaa4d2ca : SUCCESS in 5m 05s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8a4cbfd1d04b48058b62ec26c01236f6 : SUCCESS in 1h 30m 15s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0fa588dfc31745e48208cf48ee3e7119 : FAILURE in 1h 37m 49s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/83e4f6d7a2b345c39b720fb01b1dd02a : SUCCESS in 1h 17m 27s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/aea1cc055cd343a1884782bbc200470b : FAILURE in 2h 12m 21s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/1b3ad64b70314cc9800376906f7302af : FAILURE in 1h 54m 05s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/e2cdc896629942a28d88c278fbcdd2d8 : SUCCESS in 21m 39s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/dd2bc0ab4dfd4d3f8c2dc352e17add74 : SUCCESS in 1h 13m 22s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8d7bbaa62d2046a98af62a6497ae414a : SUCCESS in 1h 21m 39s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/e9612134cf5e4ecaabc0d1fd56c09492 : SUCCESS in 28m 51s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0c497f0e176b44e792d34c6a1cfbae48 : SUCCESS in 51m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9363796cdcfb465d8201494a4239dea5 : SUCCESS in 33m 28s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":6},{"id":"598a4d86e53e8bb30f04d656e2de1c65693f2b4b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-08-05 10:09:29.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"58c00043676287f5c81e4eefaae33605642df5bb","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-08-05 10:13:13.000000000","message":"Patch Set 7:\n\nHPE AlletraMP FC CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_FC_CI build/822 : FAILURE in 0h 3m 26s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/999310","accounts_in_message":[],"_revision_number":7},{"id":"42a13337fe0b956cab89720e4543540e977f567f","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-08-05 10:15:38.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"e48d4210b12d57030a59cfbd1e13f91017aa88be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-08-05 10:18:19.000000000","message":"Uploaded patch set 8: Commit message was updated.","accounts_in_message":[],"_revision_number":8},{"id":"096d8f0f37df547d7a81ff2f3a4f4b30d25da9b5","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-05 10:20:52.000000000","message":"Patch Set 7:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5033/.","accounts_in_message":[],"_revision_number":7},{"id":"ecf5e4b6f2e19f7114a79d98b5a3a88153ca08d1","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-08-05 10:21:43.000000000","message":"Patch Set 8:\n\nHPE AlletraMP FC CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_FC_CI build/823 : FAILURE in 0h 3m 3s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/999310","accounts_in_message":[],"_revision_number":8},{"id":"3436c146e2ba12eb32cd4c9a5955c068228e10c2","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-05 10:22:03.000000000","message":"Patch Set 7:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4857/.","accounts_in_message":[],"_revision_number":7},{"id":"a5da5740e980f9a3c7c1b8e3556f77ab2896a166","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-05 10:33:07.000000000","message":"Patch Set 8:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4858/.","accounts_in_message":[],"_revision_number":8},{"id":"de25bc687942cd05f1fe2998339a9f9e6e814737","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-05 10:34:03.000000000","message":"Patch Set 8:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5034/.","accounts_in_message":[],"_revision_number":8},{"id":"d9de24a7d983c17ef4635a7ef34123743993287a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-05 10:38:24.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-storpool-tempest build/843ebb5d73f243e1877606de56d54cc8 : RETRY_LIMIT in 16m 58s\n- cinder-storpool-tempest-iscsi-multipath build/ab19f7ccc8b84fdeb66454868f19c186 : RETRY_LIMIT in 17m 26s","accounts_in_message":[],"_revision_number":8},{"id":"b5a436b124991c09618a99de4af12c0bc52e72ce","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-05 10:39:28.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9fd3c1edbdb4d3f89a25e5e611e9af5\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/514c0e70c5914bf09ea12206d1a7ca34 : SUCCESS in 12m 25s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1056ff04b80f425db4bcee0f7f647e5a : SUCCESS in 12m 43s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/42aa1204260646d4af59bacd00967a67 : SUCCESS in 18m 51s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"d414ae675a36f5873e89d713fcf1777049442b1e","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-05 10:43:14.000000000","message":"Patch Set 8:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/d080c14200e049b68fb84610222f6364/","accounts_in_message":[],"_revision_number":8},{"id":"405102cfba2eee0a8021760c5095d8349e789ad3","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-05 11:06:34.000000000","message":"Patch Set 7:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/Ia94ddfacac494a301866eab565f04c921f6c2ea5 .","accounts_in_message":[],"_revision_number":7},{"id":"16447c85f290d81b26a5846852d945a1a7916ddc","author":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"date":"2026-08-05 11:08:52.000000000","message":"Patch Set 8:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/730 : SUCCESS in 0h 47m 6s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/999310","accounts_in_message":[],"_revision_number":8},{"id":"990ad53a4e254c9c3067875b34c1c2e0bbb7a5ae","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-08-05 11:09:32.000000000","message":"Patch Set 8:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1091 : SUCCESS in 0h 47m 53s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/999310","accounts_in_message":[],"_revision_number":8},{"id":"f76b47a2cfe1144bd4de24ed0278d966365199a5","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-05 11:15:08.000000000","message":"Patch Set 8:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ia94ddfacac494a301866eab565f04c921f6c2ea5 .","accounts_in_message":[],"_revision_number":8},{"id":"f09197f52f7f4197a1db816ab3a4089a26391a98","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-05 11:32:12.000000000","message":"Patch Set 8:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/d080c14200e049b68fb84610222f6364\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/86fe84935a7d495292af28bdc9251913 : SUCCESS in 40m 27s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/56a90294bb0b497dbff05ce6af938bc5 : SUCCESS in 40m 49s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/18c4ce1f88504fefbf727dc2b6e625c3 : SUCCESS in 49m 40s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d88da4488b6b4224b9845a524bdc8f68 : SUCCESS in 39m 31s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/31f28ff93fe64de096c6c19add6271c3 : FAILURE in 21m 31s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"fec41a5a5d102966353383fa7a86970e8c796859","author":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"date":"2026-08-05 12:51:19.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"23fea3ad02e6c4c883e72df970850a3baf582124","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-05 13:26:47.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8629b316b054423887b5b6a5eb81f800\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89dd2a76d6a24be69848c4a57e2dd4bc : SUCCESS in 3m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c8389605ba3c42b9b2fe8244480e9ddd : SUCCESS in 10m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/46b097389cdc4a5294bfac0fac05534c : SUCCESS in 9m 24s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2d5d6391b3894e4cb6fb9a96f4a32efa : SUCCESS in 12m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4bd0dede55a40cb805a9af0e16610df : SUCCESS in 8m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/7653f3e243ca4b95a7074c4227e2d5ef : SUCCESS in 48m 56s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9422473e92644416a76f5fa2c1ce0ac7 : SUCCESS in 51m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/79bb6b4a21fd4e6fac4996c2289ec5b8 : SUCCESS in 53m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d6ff4fc4ede24748b3c042d48ddff198 : SUCCESS in 32m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e8a206c0855449d3a51eca633acd564f : SUCCESS in 10m 17s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4e6164733ce34fe1a72395e0eb363b10 : SUCCESS in 17m 33s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/55e3b58404df4e98a90a1af3fd0efc14 : SUCCESS in 4m 30s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8073a971b7f64e1bab785fef68a8ec85 : SUCCESS in 6m 07s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/5e60600b4c874d1291161881d186902a : SUCCESS in 12m 16s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/e2dba2ed55d94cf5abedbb3e7928d5d0 : SUCCESS in 6m 26s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/e2d928fc225f44f1a39d1439738cecfc : SUCCESS in 8m 20s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/01ae6f72175d4be4891080bfa916683e : SUCCESS in 6m 37s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2c388087b4604bc895045c91b22b58fb : SUCCESS in 39m 17s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/02a59d81a9e8407e9f3988f064e9c31b : SUCCESS in 6m 28s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2124744689884498b07277d8e2792a62 : SUCCESS in 1h 33m 07s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/521512f6ab3e438fb6e1a74ef2d508c8 : FAILURE in 2h 01m 51s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/861af87a9f924cb7ada48e4999bea97d : SUCCESS in 1h 51m 40s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/31fe19771efc406c9833e205a5c630b8 : TIMED_OUT in 3h 06m 11s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/edd6a1e8ac6b42afaa0115ec72b162fb : SUCCESS in 1h 54m 46s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/e722b50244e84614bb0f49101eb28059 : SUCCESS in 20m 37s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/de01920ddb9547f1921c8aece589220e : SUCCESS in 37m 10s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/544e4366f3cc4dbfba0638318ed494bf : SUCCESS in 1h 14m 11s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/34b84300d6fb4803afed9205a5268221 : SUCCESS in 37m 06s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/85803b49db274322919906cda6a3ef80 : FAILURE in 1h 37m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/59b06014bd2c4a55910233fd0a7f826e : SUCCESS in 54m 30s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":8},{"id":"7aafca763e7f52bd4242999b550c6cb2aa5bdcb5","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-08-06 04:05:55.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"583848b588e637001b78d4ca5c935a7550a91acf","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-08-06 12:20:32.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"abc0c588084b0bb625442dd8bf0c99eef1cdba91","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-08-06 12:22:46.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"f93590d5c2ef902185df97eed0918429a89506d6","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-08-06 12:37:33.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"d7ed85dcab4351d1e4168988ab962273a23c1978","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-06 12:50:49.000000000","message":"Patch Set 8:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/a36d68121b10444fb22d0a2f015ba86d/","accounts_in_message":[],"_revision_number":8},{"id":"fbea1f2a282f57a18460171d37e35b39e0e9e9c1","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2026-08-06 12:55:29.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"6d2ed0220c174b9e3e6e12e58abd3a756bbbe69e","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2026-08-06 12:55:34.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"9e8b2c6a2bac73b1780e067be263a19951377175","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-06 13:21:47.000000000","message":"Patch Set 8:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/a36d68121b10444fb22d0a2f015ba86d\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/425c3c4557d04541bccd0e7d3408309f : SUCCESS in 54m 41s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ed75b549c64d4382a8d828b4bb2e755d : SUCCESS in 44m 03s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/645c772bd34e456c89f67abb26e5c3ee : SUCCESS in 58m 29s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c757300265214284a883bdcafe1da95c : SUCCESS in 35m 53s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d12b50d7ba4748719a4b40b5ff4e684e : FAILURE in 24m 37s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"21a273057ddb4602aae7712f68b52e6edbdebf78","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-06 13:27:07.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a87043df23664e1994f7842d85eab3c8\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ab7b7dda4aba4e3ebabff1003fce6703 : SUCCESS in 12m 46s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/15656cfb4b4c4636bec316678849ec83 : SUCCESS in 11m 53s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/5630e0d94e704a748e72931363adb221 : TIMED_OUT in 1h 01m 48s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"a3435955ef33eb593d40ddd5d13baa64463a6e39","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-06 13:37:02.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/faa70308cb314d0888439425b525806e : SUCCESS in 1h 12m 56s\n- cinder-storpool-tempest-iscsi-multipath build/8f3ea32e1f2145c587a78c5625532c2b : SUCCESS in 1h 06m 46s","accounts_in_message":[],"_revision_number":8},{"id":"62c5ea8847d686f4b56c0a9543b20e073cc6a7dd","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-08-06 14:18:05.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"3bb19c44781d1319b0039f830718a357104ef945","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-06 15:03:25.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/37ca81ea26074674b6b0f27e74a1f218\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe97421e6fec4b5d99f8db2ed555513e : SUCCESS in 4m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ce0cc549398b42c99d320c695f70c180 : SUCCESS in 11m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1d4c7f41878343b8aed428b818a71e95 : TIMED_OUT in 40m 57s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/64ded6e096794c48b11aaef8a93e2115 : SUCCESS in 11m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/523b036f0ee84fc0b33620b4b6e0482a : SUCCESS in 8m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/25e2b28b44484aad9b8ab4344ca4bb18 : SUCCESS in 1h 04m 59s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d47924d9b78e4d56878d89247f62c320 : FAILURE in 31m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0dfa1a9d41a24044acd997a53db91b2d : SUCCESS in 1h 42m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/28c7f9425c694d17a3f41c26931624c1 : SUCCESS in 1h 10m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/280a93f69da641859ddfc9fe55aeebe3 : SUCCESS in 5m 04s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ea0620397a5948b1bb211b4763c2e18b : SUCCESS in 26m 12s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b6e1586514fa4f62b794c9b0d6eedd29 : SUCCESS in 2m 45s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/70e8ef02ae72445192da0921e12721df : SUCCESS in 6m 36s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/ef7b941e3b9c466d80cfab472b3390c4 : SUCCESS in 11m 30s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/fd79708d8bca411a9c08b304f1a33459 : POST_FAILURE in 6m 33s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/d877f14b99394f03bcf76b0af394145d : SUCCESS in 7m 30s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/8cbe47bc7b674ef68d290693fc9e20f1 : SUCCESS in 9m 31s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/22b0228222da4ba8bd7e4847ad80d95d : SUCCESS in 42m 54s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c4efb9d9a2d643d78c1709b46703c60c : SUCCESS in 5m 41s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e58409fff85248f289391bd1533a933d : SUCCESS in 1h 31m 04s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/8fd854a76db64f90b15bea0f3cdd191d : SUCCESS in 1h 56m 24s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ffce7491878f4384b075cb425e5885a3 : SUCCESS in 2h 21m 38s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/f01c50a17a3b4afc99468a783e722c8f : FAILURE in 2h 39m 19s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/a529ff61d6664cc2b5f07b36ae690839 : SUCCESS in 1h 53m 32s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/fd276271e26d4df491dbb9ca24f87e67 : SUCCESS in 35m 47s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e7f2e2a98e36463f9b14bcc6077b3e6c : SUCCESS in 38m 57s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/bdbd80acbf3f4785aad880754636cd2a : FAILURE in 31m 25s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/8d819297f0fd4fc083fcc1d1a4996ceb : SUCCESS in 51m 02s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/488327c99c624f14b9bcd39dd11929a2 : SUCCESS in 1h 37m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b2c2b98fc2854dae9e2f51ff497b2f95 : SUCCESS in 55m 47s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":8},{"id":"5c0028b63bf67b3086a796c202782328d4e3850e","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-08-10 04:04:23.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"a8c70c72c34a8b0fd935a8fdcb32f056fc496ea0","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-10 04:23:18.000000000","message":"Patch Set 8:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/958ccd1df9c845258fd7e0f1fe813daf/","accounts_in_message":[],"_revision_number":8},{"id":"25db6520bbb0da2ecf0230b3fad1ac55aea7d41e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-10 04:41:37.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5025164d278a467293dfb830d95affab\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9ce98d8d93e44fe0b0dd05ab3bb9a42b : SUCCESS in 18m 15s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0d9915e6ce574f4690f3ab0213672b57 : SUCCESS in 17m 26s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/6382a282adc3442390e7a18b579e2109 : SUCCESS in 22m 51s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"3ac9a78a898e6938011ccf61a3ba19c23d21f5a5","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-10 04:47:18.000000000","message":"Patch Set 8:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/958ccd1df9c845258fd7e0f1fe813daf\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/aa9fe0b517ca4c369bcc7a0d65abac32 : SUCCESS in 42m 24s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/18606b52b47c44fbb85ab63e574e8dd9 : SUCCESS in 39m 21s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b37aaa3b92cb4fc58916f0c61bef9976 : SUCCESS in 42m 01s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e9a141a29f574b3cbdd0d1a24ff12858 : SUCCESS in 36m 37s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0c43739d48f444d78cbdba47e0622ddc : FAILURE in 15m 42s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"00625dca7fbe908d9e4b6361f7aab16b0f863d63","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-10 04:57:56.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/198dc78f32154fa6acc0dbb1dfd2c4fc : SUCCESS in 47m 56s\n- cinder-storpool-tempest-iscsi-multipath build/781015584e47473d9b2693fa7583ec3e : SUCCESS in 52m 03s","accounts_in_message":[],"_revision_number":8},{"id":"48535ed78ea4fe5d859cbb4e4e7a5552f0925871","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-10 07:12:34.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2baccb39aa9d4b2badbe342e01139e37\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/296a9a1764154185841d6171cf3f2944 : SUCCESS in 5m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/be2ddf5321a8404c90fa634aa4643012 : SUCCESS in 7m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c882b3ff457b44e8bc96882614bbe09d : SUCCESS in 8m 14s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1ad345e20a684ce88e35fc78f5907ff0 : TIMED_OUT in 41m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a2d17e3db4240b69b9c2eacdf8e2ea4 : SUCCESS in 14m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/66dd527b3d114722b068bd8e7ffa291f : SUCCESS in 41m 34s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6a858218d8a94f1db858da8e940bf4f1 : SUCCESS in 49m 45s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/181527ec86064367b4cf0efc319e738a : SUCCESS in 1h 25m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6831cae482044bf8b7e73fe49248ff3c : FAILURE in 54m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e912cf6dd4f547d29d95d947a9c018d7 : SUCCESS in 5m 18s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ae5cee43649f42228b86cc0df4142561 : SUCCESS in 16m 35s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/039e336e25854028a39a886363b4d3b8 : SUCCESS in 2m 45s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/fb54e9ebb01f4207b35dc33037dc0805 : SUCCESS in 4m 12s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/0a2034ca6e394e8bb77e8fa6cfb4002b : SUCCESS in 9m 20s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/9a36e74717844db3873842ef22863729 : SUCCESS in 6m 45s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/fcecbad2ea804610a2988c1ea2dfbf41 : SUCCESS in 5m 25s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/11978711148d451394f8928892f97150 : SUCCESS in 6m 24s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f9503e0e76f64303a8149ec2d0b594d6 : SUCCESS in 18m 05s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/beffc181d60f4edda10af8b7ed2d5fff : SUCCESS in 3m 23s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4638404bfafe4398a798d815e7e75962 : SUCCESS in 1h 32m 35s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/260a35953e6e4923b3d74d7417a0e6dd : SUCCESS in 1h 06m 52s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/18906b1c4f194fd99459dd959e000599 : SUCCESS in 2h 03m 57s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/78b5c87e7c48462f8d94f665659bb53a : TIMED_OUT in 3h 06m 28s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/b074fb7341624898940473ce3a5d373d : SUCCESS in 1h 41m 21s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/f572d733ed394d0fb817089bebfc1859 : SUCCESS in 33m 23s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/37a5968e0a6c4217b708338c0d1f7f0f : SUCCESS in 37m 46s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/03944d26a9164934b5a6fb10936ee4ca : SUCCESS in 1h 37m 05s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/4caf4289b75948408af1b4f9b9ace33e : SUCCESS in 36m 06s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/cf39f6e707ec4aa5b92644a781a73301 : SUCCESS in 55m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2d4334ad5a274eb792dacbccf4247219 : SUCCESS in 54m 02s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":8},{"id":"e850a84709360b51e42ee3003cfe6ee5fd46efae","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-08-10 11:44:54.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-10-999310-8-fc : FAILURE\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-10-999310-8-iscsi : FAILURE in 52m 46s","accounts_in_message":[],"_revision_number":8},{"id":"c6dafd5b510d9801773ff55b4c2f744f703652d2","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-08-12 14:17:58.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"f79b40d406e8e3dcd52f45a92230cca65ff8e893","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-12 14:56:01.000000000","message":"Patch Set 8:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/e4b2a14bd5d24d0d9af39e35d5996e13/","accounts_in_message":[],"_revision_number":8},{"id":"28da0de2fa9fe901e034c13b25a6506749e45553","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-12 15:12:15.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/f1f9096a4acd484e997978b080ce345b : SUCCESS in 46m 13s\n- cinder-storpool-tempest-iscsi-multipath build/1c51751a74854b479164711039c9ac45 : SUCCESS in 53m 01s","accounts_in_message":[],"_revision_number":8},{"id":"6f0e834b0299ac7e63554cc9d4982f57c8d77a33","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-12 15:37:19.000000000","message":"Patch Set 8:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/e4b2a14bd5d24d0d9af39e35d5996e13\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0c424eba4c0f4a2bb6b4d8d32f4880f3 : SUCCESS in 45m 25s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/297371dc1aea4701b5307ba1aace74eb : SUCCESS in 37m 35s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1c933337f15f4fa4a9e58b7f3ab83276 : SUCCESS in 49m 48s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5e53901fa1d44edd9a90f2a4e90029f6 : SUCCESS in 38m 49s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2ad57dcb0aa74a78916d04900ea348fb : FAILURE in 14m 31s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"11f99347090ea93d6750b83e468f34907432f036","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2026-08-12 15:46:44.000000000","message":"Patch Set 8: Code-Review-1\n\n(9 comments)","accounts_in_message":[],"_revision_number":8},{"id":"7a645407272c0f017d254bff946f32fe090f549c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-12 16:56:17.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9315274e32b402fb6e94ff2bf1f4a7f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd4835d8caa64830a894598233c6c4d3 : SUCCESS in 4m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/eaa4d9b735594b2d9611d9af83e1775e : SUCCESS in 11m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3cb2064836d94d42b47bf5e26f9a17ab : SUCCESS in 9m 14s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/374f276f03d44e1895fb6964aebc0588 : SUCCESS in 10m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a890e2856f743c5927bab4d93f9169f : SUCCESS in 14m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/fbb36949c71c473bac40e562bea894b8 : SUCCESS in 30m 48s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c7786bb777854846afcc1a6060f195c0 : SUCCESS in 1h 08m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8b527b0a5b4a4ef48a346573206fa5b7 : SUCCESS in 1h 06m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/45b6633e3aa94ffeb74065c006455fc8 : SUCCESS in 31m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ab35037538f6422e894c4802a9758d0e : SUCCESS in 6m 57s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7a5ffc8857974a6f893167a5ed7f23a6 : SUCCESS in 21m 44s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/d4c00dfa0f094bfa960e5e84ae569baa : SUCCESS in 3m 31s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/230db4cf6c364cc28b0b22dbea9fb18d : SUCCESS in 6m 09s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/b7026a8414734e3494a58c77c4d8c272 : FAILURE in 27m 01s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/efcf7f9b11bc430eacd6cf3acbd090f3 : SUCCESS in 6m 18s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/3792497b1a4749f89cf42e3f8b3d7ff8 : SUCCESS in 6m 02s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/964a5420c33b4098ad29f2d6573200a4 : SUCCESS in 9m 21s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c70415069bc6461cb146509a39445332 : SUCCESS in 20m 42s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8a64b6c18f3848abb3ee09ee9b529941 : SUCCESS in 5m 10s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/28bfa4e4c5874a4ebaa1bc2f98a93c22 : SUCCESS in 1h 35m 29s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/5d3afb44d47a4d3c973552b3b2f54865 : FAILURE in 1h 05m 19s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1ff4285242054d56a7a249b3aea7eab7 : SUCCESS in 1h 45m 10s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/0cafdad0305a415ea48b174787058bf6 : FAILURE in 2h 32m 32s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/cac2d9174d924f35b1fb1bde8c8d2dd0 : SUCCESS in 1h 51m 02s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/83b96dc260994e5dbfab7c0c3e11e51e : SUCCESS in 18m 21s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/36a0dd4a004f49f698d9264f713e561b : SUCCESS in 1h 15m 24s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/654a2919fa5d47c39a1a589cdab56ee8 : SUCCESS in 1h 04m 54s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/5733a517b41645449b857ab89702ff98 : SUCCESS in 33m 09s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/cc0cf29dd0554eb2bd8791d7fef8e8e8 : SUCCESS in 1h 31m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f00247e39015414c8a1335a111ff46a3 : SUCCESS in 1h 04m 07s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":8},{"id":"4282c5bf7f11822ab280226e7d74c8e1d3f13f60","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-12 17:06:39.000000000","message":"Patch Set 8:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/12_08_2026_17_00_00\nLeave a comment with","accounts_in_message":[],"_revision_number":8},{"id":"a5c0753988d280345af399a917d9ff8f524ed5ca","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-12 17:57:25.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/226f12c77e194a7888476588a1b109da\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/af71be927ebb4d8dba14bad5a5c2cded : SUCCESS in 19m 07s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/af19e7a67cc0404e8d8e0ef9bf96862a : SUCCESS in 17m 06s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/047383823d2044e7904076caefc1ea00 : SUCCESS in 25m 03s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"99f64e01b2d3e0db6fcc1731d3f3352fd5320c39","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-08-13 13:50:49.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":9},{"id":"197097938b49b7723b7006959b8d5b38c2f7fa3c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-13 13:59:58.000000000","message":"Patch Set 9:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5109/.","accounts_in_message":[],"_revision_number":9},{"id":"712a0233ebe96ab544f4ea703b01c67af49467c3","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-13 13:59:58.000000000","message":"Patch Set 9:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4933/.","accounts_in_message":[],"_revision_number":9},{"id":"f21d7b2633902619ac7638f705f79ecd6ae1ae21","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-13 14:11:51.000000000","message":"Patch Set 9:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/cb1bf81304674cd7989ed2e193261e7e/","accounts_in_message":[],"_revision_number":9},{"id":"a351b590a121a9e3c4f359af31709971788a7b6b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 14:21:55.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f8c7f8dc21ad497cadd999b3db5798c9\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/02c4d4d7572140f390e398467fab5d8d : FAILURE in 16m 03s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0882c44a138c499c833697d071614539 : SUCCESS in 17m 06s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/2e6fbb2756b44085a914b7b9b4625de4 : SUCCESS in 27m 49s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"6507210d2dc4978a89f64b93ba4908d4e51742a3","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-13 14:43:02.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/bdcdad313bed45e397889f6f8dadcbdc : SUCCESS in 45m 06s\n- cinder-storpool-tempest-iscsi-multipath build/1424d65364974608a81d3452972ff3ac : SUCCESS in 50m 52s","accounts_in_message":[],"_revision_number":9},{"id":"10ab1e8c5444c3614b6a64dd0701fb15ac71c44b","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-13 14:46:08.000000000","message":"Patch Set 9:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ia94ddfacac494a301866eab565f04c921f6c2ea5 .","accounts_in_message":[],"_revision_number":9},{"id":"30a6a589c18332d50425cd78f8b831a5e8310e24","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-08-13 14:48:50.000000000","message":"Patch Set 9:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/897 : SUCCESS in 0h 48m 52s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/999310","accounts_in_message":[],"_revision_number":9},{"id":"28cf3ef929d7e79ffd47eff8de048847cd313a7f","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-08-13 14:49:39.000000000","message":"Patch Set 9:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1164 : SUCCESS in 0h 49m 56s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/999310","accounts_in_message":[],"_revision_number":9},{"id":"4170bfe70d2c56bd2cbc6f6387284292a803fd32","author":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"date":"2026-08-13 14:49:46.000000000","message":"Patch Set 9:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/803 : SUCCESS in 0h 49m 51s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/999310","accounts_in_message":[],"_revision_number":9},{"id":"d0975fa667cd6424a0cb8c4eb3ac86f606abb721","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2026-08-13 14:51:13.000000000","message":"Patch Set 9:\n\n(5 comments)","accounts_in_message":[],"_revision_number":9},{"id":"acf7001118b2fdb565033cf6bd9154dba125b994","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2026-08-13 14:53:36.000000000","message":"Patch Set 9:\n\n(10 comments)","accounts_in_message":[],"_revision_number":9},{"id":"d1e7bee1e2047d9d45c03e419f18c8c33da66b22","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2026-08-13 14:54:12.000000000","message":"Patch Set 9: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"e9b78981db837a7524e4faf67b1704331e1ea96b","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-13 16:23:24.000000000","message":"Patch Set 9:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/cb1bf81304674cd7989ed2e193261e7e\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c910a049fd7441de8e31eb9539621db9 : SUCCESS in 45m 33s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/82329a9160304ec1808e68d576f24380 : SUCCESS in 37m 47s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1bf050cfc72541f7b7cb9a0c4c5d18f7 : FAILURE in 2h 32m 02s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/27f57966883349698a2927ce2ece8969 : SUCCESS in 38m 09s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/36f4de284d1a4c71afc77ea6af6689e9 : FAILURE in 17m 49s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"b6a537c0e417147b1b57cf52b042d56424c457ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 17:34:58.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ad58495ee484e43bb81caa4565ddf4e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d318b7e529e4473b9042f9a6e877571c : SUCCESS in 6m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b8e991d6f3d04e43867c95a877753939 : SUCCESS in 8m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b2ca239225404421b2bb6f38cdc7f8f7 : SUCCESS in 11m 12s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a6271ccf0d7f46cdaa15f8932589ef6d : SUCCESS in 8m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3593af98c0c34fb6819d2e533a1b8336 : SUCCESS in 14m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/b9d41cc35bba41b895c598f9ea41b812 : SUCCESS in 1h 09m 03s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/45d09fbb60d24e35a3157d538cb992c4 : SUCCESS in 1h 09m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6acbab322f274cb9971d0d41e1fa071e : SUCCESS in 1h 17m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c0872ba6d364498da3da7b1f0ea6d7c2 : SUCCESS in 48m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9b29958cc9b141f58fc9afb13b3816e9 : SUCCESS in 9m 40s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e5952b7b12184fde82c013cf07b0bd4f : SUCCESS in 17m 38s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f834527f181e4d54b8c67c08e5165149 : SUCCESS in 3m 05s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3a84b49ed1a2482490e05c565227b2b5 : SUCCESS in 3m 20s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/06ff59a991e9403ab9e3aec89513bf12 : FAILURE in 10m 49s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/ddad5b78a8be4c008f4f387b298dc214 : SUCCESS in 6m 00s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/bb6b38d9930e467f843eb48de546f08d : SUCCESS in 6m 10s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/3e31f2728dc744bfbcf2fc7e5af74fd5 : SUCCESS in 8m 38s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/17abe2f8710446628fc019e18a275da7 : FAILURE in 7m 12s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/42f60cf1201f4ed28bf9074f7cad84d3 : SUCCESS in 5m 53s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/450618fd49354440b6b4e0cb38e5b07a : SUCCESS in 1h 52m 44s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/00e41fb9f6604fad9a1168d3c8cdd882 : TIMED_OUT Host unreachable in 3h 38m 31s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/108e1a12a38940d5826ca37c7ca1fb1a : SUCCESS in 2h 04m 29s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/1034de3606b54849acc337f4bc97fb2a : FAILURE in 2h 40m 35s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/f0a02d7a69ee401fb47cb8ce8357a602 : SUCCESS in 2h 01m 58s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/541d815c762f4cc0908ffe77269573bc : SUCCESS in 38m 33s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ef6fd75c129543bcb3570b5661e1e20e : SUCCESS in 1h 08m 52s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4b71e70875784f54be7c458bc5b274d1 : SUCCESS in 1h 40m 11s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a02f5a25923a4b74bc64d43d728dc8a0 : SUCCESS in 49m 56s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/96600d319b4e4bfc94b14ab40da4407d : SUCCESS in 1h 55m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c381d9f7a9bc433d933d608263f5fe62 : SUCCESS in 1h 07m 16s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":9},{"id":"21338b19fdfbfc930ed202828c1d98eac1001b7e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-13 18:23:33.000000000","message":"Patch Set 9:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/13_08_2026_18_16_18\nLeave a comment with","accounts_in_message":[],"_revision_number":9},{"id":"3e23ed26a0093d66b354df22450251f7907c404d","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-08-14 01:43:40.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-10-999310-9-fc : FAILURE in 11m 19s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-10-999310-9-iscsi : SUCCESS in 1h 10m 35s","accounts_in_message":[],"_revision_number":9},{"id":"839163f5867927e63397b6a1b51662631ea7f949","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-08-14 06:27:51.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"415ff00cabeea9c74a234c93b46460c695681fe8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-14 06:28:54.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91280cd8873e4f39bbf7c0a1ce338dc1\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6373e65985b04acbae6402d201faffab : NODE_FAILURE Node(set) request 75d270bb5be64e39bff0c9a25e2f3c81 failed in 0s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e3fab20365964f47b7cab45f617d97ad : NODE_FAILURE Node(set) request 09aa1737bc224ecd8d122e0cb74e70c0 failed in 0s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/95f92fde80674ac298d547ef5ee05917 : NODE_FAILURE Node(set) request 7f0ec245151444cb8e9dd84c2e652973 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"3beb0e57202a41fed584fa37484d8d736755cc31","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-08-14 06:32:20.000000000","message":"Patch Set 10:\n\n(5 comments)","accounts_in_message":[],"_revision_number":10},{"id":"b4bd6e30e2b49ae8c69f4584cf726af7842cdddd","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-14 07:05:02.000000000","message":"Patch Set 10:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/8b76f98b412b4146a0dca1b71e00ec9d/","accounts_in_message":[],"_revision_number":10},{"id":"6e5caebd243220858712dd73594e6b498def5a78","author":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"date":"2026-08-14 07:09:40.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"850bbaa9530d7667b134d7db4035f1044c81ad2d","author":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"date":"2026-08-14 07:18:03.000000000","message":"Patch Set 10:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/817 : SUCCESS in 0h 49m 48s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/999310","accounts_in_message":[],"_revision_number":10},{"id":"a1bee613f4add65543a759982ee3c97c89b332ce","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-08-14 07:18:16.000000000","message":"Patch Set 10:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/911 : SUCCESS in 0h 50m 4s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/999310","accounts_in_message":[],"_revision_number":10},{"id":"6526dd130233ad2ab4c343de312dbc40225e07c0","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-08-14 07:18:30.000000000","message":"Patch Set 10:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1178 : SUCCESS in 0h 50m 15s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/999310","accounts_in_message":[],"_revision_number":10},{"id":"3b2c9e356feb91b04e9a5d60cf116e817ef07dd2","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-14 07:18:58.000000000","message":"Patch Set 10:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/8b76f98b412b4146a0dca1b71e00ec9d\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c473f93426ba44da8ace472072ef22ce : SUCCESS in 38m 16s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8b406379463046ca97c94f00e9544cf9 : SUCCESS in 36m 42s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c38c4e4c36ab4d7f84bb965d41d757b2 : SUCCESS in 45m 04s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2cd86141bd284124aa72cf99c675b75c : SUCCESS in 40m 49s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1733666c23d64786a5f650ef9795b829 : FAILURE in 15m 34s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"615386727d289bb3af13f357a45a90a874a801fe","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-14 07:22:30.000000000","message":"Patch Set 10:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ia94ddfacac494a301866eab565f04c921f6c2ea5 .","accounts_in_message":[],"_revision_number":10},{"id":"3616febd6cd6c20052aa4ebb4d3fd64e6f2f81e4","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-14 07:31:53.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/70ac98e9d52a4f6a9aa8ca3fa5944831 : SUCCESS in 46m 04s\n- cinder-storpool-tempest-iscsi-multipath build/00f6fcd4a8564e35984c0c8a12d958a2 : SUCCESS in 1h 00m 59s","accounts_in_message":[],"_revision_number":10},{"id":"1ea005fb7570d0055f21cbdf0cbb3ac1e41bfd9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-14 08:34:04.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f82ba9707ade470bb77a463e7abe071f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80f3f378ed134f928f1e6794cfc66b1e : SUCCESS in 3m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1f92e1cede6d415da7cd369defa07d9b : SUCCESS in 7m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e0f0a2a0b9514e4f91fb6f649748d481 : SUCCESS in 10m 01s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ec09dc679fbf42e1bf46a224a92b4861 : SUCCESS in 14m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20157e462e69435ebb23bdf3bccb2ae7 : SUCCESS in 14m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/9c69458dda3e46c8a8669469d28b83b5 : SUCCESS in 46m 35s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b0e482c344d84beb8aad2fc5414c958e : SUCCESS in 33m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3bd1a12b64b2409b83b564f48e3c698a : SUCCESS in 1h 30m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5a5e7bdbf7114c2d9125ca3463975e03 : SUCCESS in 1h 01m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b8da6512b092473ea614d71658108d65 : SUCCESS in 9m 31s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b0c5e1b1d7ba41eb839e52d651ba7938 : TIMED_OUT in 40m 56s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e9d72952b57847d8822f253c5f917a0b : SUCCESS in 4m 27s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e24e1ef7b12f45e486969d486f673488 : SUCCESS in 5m 06s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/727f00b21ae94428a3a0a139db0ef9ca : SUCCESS in 10m 39s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/4ceace5445a342d79c1746f61428a277 : SUCCESS in 6m 04s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/0b9946dcbde94e6d8bf4363195481ddd : SUCCESS in 5m 36s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/b774ef1928a741a39eb4d1762e8524b2 : SUCCESS in 6m 03s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d065a945f5ce487384dccdf1ef8064e2 : SUCCESS in 16m 52s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7291a785280749fc9e4e027b4b1f6267 : SUCCESS in 5m 41s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1ddedf46c1c54139ac4abb702e1166f5 : SUCCESS in 1h 37m 29s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/7e6623766bbf4136b92fc1f13e4a9134 : FAILURE in 1h 03m 48s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/90e9b1988f164da7b5b26712d62f13b0 : SUCCESS in 1h 40m 47s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/7aa36df28d274771a3fa347428147970 : FAILURE in 1h 32m 24s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/cfaa9a038a164bb59d32ddc02b555abd : FAILURE in 1h 59m 29s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/9f2ea3fc01e8444aa6ff643757a6ec7e : SUCCESS in 23m 58s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/729e0f37cabf4bd49385ed3ac829a295 : SUCCESS in 1h 20m 41s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e0b7ef40e39b40b8bb6a781f9fe7d272 : SUCCESS in 1h 06m 29s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/dbeb649eeeae480bb84e8a0057577181 : SUCCESS in 42m 15s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4f6975a88e6c4ed7a2f5666e9154ac6e : SUCCESS in 1h 25m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/913ccb86df2b423bafbddbf7bbc92c32 : SUCCESS in 36m 18s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":10},{"id":"f5eb200a334f4acc7eee7cc8d7c98f75c7bbefda","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-14 08:44:46.000000000","message":"Patch Set 10:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/14_08_2026_08_37_36\nLeave a comment with","accounts_in_message":[],"_revision_number":10},{"id":"c0f795c1ebb06773abc4917b61318cc6951ede97","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-14 11:58:29.000000000","message":"Patch Set 10:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4948/.","accounts_in_message":[],"_revision_number":10},{"id":"c1c32d5043525d6ebd6f8c4c7e1af3171afdaa6c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-14 12:58:18.000000000","message":"Patch Set 10:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5124/.","accounts_in_message":[],"_revision_number":10},{"id":"0660893cef40c6ec36c4dfadec5af2f4622b8806","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2026-08-14 13:48:53.000000000","message":"Patch Set 10:\n\n(5 comments)","accounts_in_message":[],"_revision_number":10},{"id":"a96f1198f70597267376eea0d48296bd6d73bb6f","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-08-14 18:05:13.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-10-999310-10-fc : FAILURE in 0s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-10-999310-10-iscsi : FAILURE in 0s","accounts_in_message":[],"_revision_number":10},{"id":"7450092275bf78fd566a71e5369fc412d497c5a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-08-17 04:58:42.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":11},{"id":"d5ed62b38d7473e0eadd797c0f1013c651984bab","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-17 05:37:55.000000000","message":"Patch Set 11:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/1c2bd88c746e4b9abe1f44133f49159d/","accounts_in_message":[],"_revision_number":11},{"id":"0422682e09ad9d0e34896699bc03469a918e4005","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-17 05:53:27.000000000","message":"Patch Set 11:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ia94ddfacac494a301866eab565f04c921f6c2ea5 .","accounts_in_message":[],"_revision_number":11},{"id":"17efecee73e9828f0ebae7e7f3a21cbc60369d05","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-17 05:54:42.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/965b1d5eea2a42b592d588040363396f : SUCCESS in 52m 26s\n- cinder-storpool-tempest-iscsi-multipath build/6e5f1a9164244147a4bddb9316d11d04 : SUCCESS in 54m 36s","accounts_in_message":[],"_revision_number":11},{"id":"eea8f8ea1064a8d5637b57ed535c2b52876ee26e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-17 06:01:45.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f4af6ab31d14731977f2bbb1becf86b\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1bdd2b695857444db1b5a5d79bbd9f5d : SUCCESS in 11m 47s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4d3c15d64bbd46ebb6e28d2a0a95fb95 : SUCCESS in 16m 53s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/c9f935035e8a42feb4318332797b7011 : TIMED_OUT in 1h 00m 54s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"aac0bd78bede924e7dee9501256259b68581d228","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-08-17 06:26:09.000000000","message":"Patch Set 11:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1198 : SUCCESS in 1h 27m 5s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/999310","accounts_in_message":[],"_revision_number":11},{"id":"52b13fdf8fb3e33e65269c5d460bfaec8b6cdd90","author":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"date":"2026-08-17 06:29:01.000000000","message":"Patch Set 11:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/837 : SUCCESS in 1h 29m 58s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/999310","accounts_in_message":[],"_revision_number":11},{"id":"155c51e4a014d86bdddc4ec34982c68ed7bc411f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-17 06:32:21.000000000","message":"Patch Set 11:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/1c2bd88c746e4b9abe1f44133f49159d\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3a04a4b07f75461c8dd226ca573440d4 : SUCCESS in 40m 35s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/01f6cdff1f864ca381937fc6d0c238b4 : FAILURE in 1h 33m 07s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/cf1a3cc9e8bd4fe5a7f39fa6afb42164 : SUCCESS in 47m 19s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f49e975ef92249309afaa07afb9eb6e0 : SUCCESS in 38m 44s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/520867a47e784234bb31579012f64f76 : FAILURE in 13m 14s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"b1dc628330c454f8ce9e6b45b683cc3e882a2e5f","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-08-17 06:34:13.000000000","message":"Patch Set 11:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/931 : SUCCESS in 1h 35m 5s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/999310","accounts_in_message":[],"_revision_number":11},{"id":"aae308fd9fb2d405c27e0eca0857de89033ce771","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-17 06:42:41.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73f82ecba3f449c0b0fab385b1d57df3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5df27be93cfb4f619d49538c8bbc27ef : SUCCESS in 4m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e8e6fc4fb70f4ee79fe7ffdf424e104e : SUCCESS in 7m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8373c6f1ddf1456982d959d9b6181ac6 : SUCCESS in 8m 35s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7090646a3d144f639c13131305c78341 : SUCCESS in 9m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7994955530934196b4fae44384cc796b : SUCCESS in 16m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/afa53a224ba74eba879a7d00e0ba54d5 : SUCCESS in 34m 50s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2f3168242e3a40e7b442fd3684389fff : SUCCESS in 33m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ff39757d6f1946f39b1da29cfe845343 : SUCCESS in 54m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7dca3958545f42d3b13e945c5a88f500 : SUCCESS in 31m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/090805e6588b45e4a16c58db023c8481 : SUCCESS in 9m 48s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4dbe1b8a6af843bda107ec047f1c6432 : SUCCESS in 20m 44s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/7cb61ba0c50f441d990bbc506317b607 : SUCCESS in 4m 25s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5b7d0f93ee644f97b56a72173f4e112c : SUCCESS in 4m 05s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/68adf5bd7570429ea43d90eaef9f773c : SUCCESS in 7m 17s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/d7850012ac6f4278abb9510868aa3139 : SUCCESS in 5m 41s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/06f4946fa6834b3eaa348d19ab432214 : SUCCESS in 5m 10s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/b894f96bd3c048e48c3063f1630ed2a3 : SUCCESS in 6m 49s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/249ccb7145f04edeb912c26cbe733df3 : SUCCESS in 37m 11s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ff9542e075274022bf220d075d85314b : SUCCESS in 5m 43s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/392876a58f504d90b76bb5c1d3c3316e : SUCCESS in 46m 55s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/4828738e0560406aa56e3889b4063780 : FAILURE in 1h 27m 36s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9643fd1a54024df8959ee516d1e85043 : SUCCESS in 1h 21m 02s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/edb29105472044cfab4fc74bf4b9e185 : FAILURE in 1h 36m 37s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/66a9da7447ba46cca5b2dc0094ff326e : SUCCESS in 1h 20m 55s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/6a93f8ac4eb04c3c94d40f6cc6020205 : SUCCESS in 20m 18s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d4fe0f50e66545f286ecaef4f750ad4c : SUCCESS in 52m 27s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b504e6774deb44b292bada4877ee920c : SUCCESS in 1h 32m 04s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/75ba276c2f444de2bb330f7c34912704 : SUCCESS in 44m 22s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3628d7e607334e49984f31620f1fd864 : SUCCESS in 1h 02m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/15c6466deb634747a5d24eea48694d8c : SUCCESS in 29m 31s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":11},{"id":"4754615d23693f5a8a793bf34e271ce3340d61f9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-17 06:53:32.000000000","message":"Patch Set 11:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/17_08_2026_06_46_21\nLeave a comment with","accounts_in_message":[],"_revision_number":11},{"id":"7210de46872582237c9d84743acb21966a761397","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-08-17 08:31:59.000000000","message":"Patch Set 11:\n\n(5 comments)","accounts_in_message":[],"_revision_number":11},{"id":"9c9d1593c046ee74ae615deb7c3f11093c4a2216","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-08-17 10:29:43.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"22fcda1641ee8e3ff5aae692a898b415fa5d19c3","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-08-17 10:52:01.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/10/999310/11/upstream-check/cinder-tempest-plugin-ontap-fcp/aefdd4b/ : FAILURE in 17m 51s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/10/999310/11/upstream-check/cinder-tempest-plugin-ontap-iscsi/31b6690/ : FAILURE in 18m 16s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/10/999310/11/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/632b054/ : FAILURE in 18m 16s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":11},{"id":"0cd364708d4461f129921714c34ffed6eba20328","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-08-17 13:44:10.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-10-999310-11-fc : FAILURE in 12m 6s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-10-999310-11-iscsi : SUCCESS in 55m 48s","accounts_in_message":[],"_revision_number":11},{"id":"55c8cf46452686fc576bb51827a3a4de4264f995","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-17 16:16:37.000000000","message":"Patch Set 11:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4968/.","accounts_in_message":[],"_revision_number":11},{"id":"86f8fa72f3b4604444ac302e11aeb5a980948c12","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2026-08-17 17:11:43.000000000","message":"Patch Set 11: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"daa9cde503704db7cab502ac1b66823328270ef6","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-17 17:43:17.000000000","message":"Patch Set 11:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5144/.","accounts_in_message":[],"_revision_number":11},{"id":"3f05086e90b01a74545ecebec425f82d4ae967d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-08-18 09:36:40.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":12},{"id":"62dd5cfc9274cd6b2e951109d6108dbf09f2d3c4","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-08-18 09:37:54.000000000","message":"Patch Set 12:\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"a42e5e2363caa90f72a239c2c854e57c283e7fbb","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-08-18 09:38:13.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"a68176757c2babc7adfa86ccbfbd4b543dfeb051","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-08-18 09:48:11.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"b9d813e2c55ca943503257c07284da71ea71cee8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-18 10:01:59.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a97e5b54e0c44bd5bee73ccf3865e762\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/17de43c81fec479e94b14b8f39ca81ea : SUCCESS in 16m 44s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2d4e73cf59784f1690c2fb655ebe959c : SUCCESS in 17m 45s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b8e87ccde87c4400b18c2436e238dc96 : SUCCESS in 18m 25s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"6e44aa784938646e3e8f526d672ff09963a4993d","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-18 10:03:48.000000000","message":"Patch Set 12:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/ff25dfe577194f6bb0a2cb53ee2b0356/","accounts_in_message":[],"_revision_number":12},{"id":"33d35bbb4be23e27339d86290cfc2aa343f3b771","author":{"_account_id":39225,"name":"Prudhvi","display_name":"Prudhvi","email":"prudhvi.openstack@gmail.com","username":"Prudhvi"},"date":"2026-08-18 10:25:05.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"d8aadcef1fab0c362e0384ef31aace5028099aa5","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-18 10:34:57.000000000","message":"Patch Set 12:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/Ia94ddfacac494a301866eab565f04c921f6c2ea5 .","accounts_in_message":[],"_revision_number":12},{"id":"cff9696a7b5e8fdffd85dc08cfc44acfcdc0e255","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-18 10:49:39.000000000","message":"Patch Set 12:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/ff25dfe577194f6bb0a2cb53ee2b0356\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e946fbaa0c654f37aed94b391007e860 : SUCCESS in 40m 15s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4ffcec19980743c59ecc89aa9aaa7644 : SUCCESS in 30m 45s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7905be5fc8e94285bf2de1c14e69a97b : SUCCESS in 44m 08s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2a9b2e437fd24e9994c2f5272c2d5176 : SUCCESS in 33m 54s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/396e7157bd0441b5879b4e63a11d4d37 : FAILURE in 11m 19s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"e6b7542533bbe35b40d5061f78c4de13a87aef5b","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-08-18 10:54:42.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/10/999310/12/upstream-check/cinder-tempest-plugin-ontap-fcp/127bc26/ : FAILURE in 26m 47s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/10/999310/12/upstream-check/cinder-tempest-plugin-ontap-iscsi/39aa1c5/ : SUCCESS in 27m 37s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/10/999310/12/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/3079e4f/ : SUCCESS in 27m 47s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":12},{"id":"c782ca70afcf9fcf0795baf312c11a86cc0f20eb","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-08-18 11:57:08.000000000","message":"Patch Set 12:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1235 : SUCCESS in 1h 1m 0s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/999310","accounts_in_message":[],"_revision_number":12},{"id":"a609822b38d5549e8992a1ac3f839a9988ce73c7","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-08-18 11:58:06.000000000","message":"Patch Set 12:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/966 : SUCCESS in 0h 59m 54s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/999310","accounts_in_message":[],"_revision_number":12},{"id":"42c200b082ad1cf4c739dc4ab65f863970758cbe","author":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"date":"2026-08-18 11:58:08.000000000","message":"Patch Set 12:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/872 : SUCCESS in 1h 2m 42s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/999310","accounts_in_message":[],"_revision_number":12},{"id":"e887f0d0fa35e0771b6b1143c9741da7cd8d2899","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-18 11:59:34.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b729d54de1b64dbd934f87d6ce3f0302\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4dd945ac07e74137835c666016a2682b : SUCCESS in 6m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/427733bdf3e64793b8e2a7feb04de977 : SUCCESS in 10m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0deef2879d104ddca2e4b8468f6df406 : SUCCESS in 9m 01s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c7341c90ca0e4fc08063f833cfaf40a8 : SUCCESS in 13m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71d783fdba3c4ca5b3e3ea1780ba475d : SUCCESS in 14m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/5d66b168bd694218ae64301711fa88da : SUCCESS in 1h 08m 06s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6e5a3d40a28244148e365ecbcf90e832 : SUCCESS in 32m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7e0e71908f4842278893b986a6b9ba41 : SUCCESS in 1h 38m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/88d6ed1764a749c18f082304d0f2f4fe : SUCCESS in 55m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c72f837c236e4c17b9f103dc4abd0138 : SUCCESS in 9m 20s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/3b92366de87c4b68a89c81c03475c25e : SUCCESS in 22m 39s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/5cc40451b0dc4b40ae2d797f0952cbd9 : SUCCESS in 4m 34s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9944499d45a94aecb15be56b986a77fd : SUCCESS in 6m 26s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/429b42d522f54fd6aad3308bb3256ab1 : SUCCESS in 8m 07s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/926bb8dc80704fe1b85089986f22b5b5 : SUCCESS in 7m 21s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/6b80ef153a7a4ede9fffa368cdf3f1f4 : SUCCESS in 5m 52s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/cad6aaa25f0d45cdaabaadd696f8e360 : SUCCESS in 10m 11s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/cf2c1963276747b9a2d5da03734e8c20 : SUCCESS in 35m 38s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/706b0bbb3e3c454b88fe223f9c6c0008 : SUCCESS in 3m 51s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6ce6232c27e3405e81ed1939600ec9a1 : SUCCESS in 1h 46m 44s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ab401313651f4693b3211671b8f81d50 : FAILURE in 2h 08m 39s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/4d17c111be854edca911d1d1e36ca188 : SUCCESS in 1h 27m 33s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/54491e85b47d4a2a82780ec27de16736 : FAILURE in 2h 12m 50s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/5f1f22c45f3a425d90ea827cb9ea6f48 : SUCCESS in 1h 49m 36s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/235cac64472c4fb381b42a414a785152 : SUCCESS in 17m 42s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/dc54d473c4c44975bd715b5012a249ae : SUCCESS in 1h 12m 37s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e2d7d37fc56d4a9ab0415315de4c1144 : SUCCESS in 1h 45m 20s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/b854017482ca4dc99827d40ae9683b5e : SUCCESS in 43m 09s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/01b27fa81cbc4fdda6b321b42225c040 : SUCCESS in 44m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/13fb804acee94872ad15f880f9a7d1e2 : SUCCESS in 1h 01m 16s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":12},{"id":"065e4d746b67108591ace738a1b171dcc988f88d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-18 12:10:39.000000000","message":"Patch Set 12:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/18_08_2026_12_03_38\nLeave a comment with","accounts_in_message":[],"_revision_number":12},{"id":"77c2dc7bf4156dbe2fdec7349efa84a7c75575bb","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-18 12:17:53.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/257f6b735a82442a889280490190b2a1 : SUCCESS in 46m 42s\n- cinder-storpool-tempest-iscsi-multipath build/ca6b7611964d4a50a9895155fc30f955 : SUCCESS in 51m 49s","accounts_in_message":[],"_revision_number":12},{"id":"f4d57a455c1ebc15106395c557f69e8f4c99f156","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2026-08-18 14:52:17.000000000","message":"Patch Set 12: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"7b0abce69d03bf0b84c15e40d6fb460e5f593de9","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-08-18 19:15:16.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-10-999310-12-fc : FAILURE in 13m 27s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-10-999310-12-iscsi : SUCCESS in 1h 4m 34s","accounts_in_message":[],"_revision_number":12},{"id":"3bf1b8eda74dd7c143d972b92a50ee1302b7a173","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-08-19 11:42:17.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"2a7b54d4a7bc5513841cbb9a18a50c1eb15a860b","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-08-19 12:49:22.000000000","message":"Patch Set 12: Code-Review+1","accounts_in_message":[],"_revision_number":12},{"id":"d157934f764253c510a729af44a20e99c7b65010","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-20 06:31:36.000000000","message":"Patch Set 12:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5179/.","accounts_in_message":[],"_revision_number":12},{"id":"bebc82eed5b99b68d3dcca3f30d3a7fd93a81f7f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-20 11:01:24.000000000","message":"Patch Set 12:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5003/.","accounts_in_message":[],"_revision_number":12},{"id":"e90b866643225dcb75852b1314eb407869c19201","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2026-08-21 15:31:18.000000000","message":"Patch Set 12: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"0cd2739db7ab2ebadf96ffba94f39e15eb551e12","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-08-21 16:16:37.000000000","message":"Patch Set 12: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"51c973698566f2fb0b84447dc6e86f8dbf2acb74","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-21 16:17:16.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"f5ae74f7dc9ac1b07e7fe3b44617853b7402c94f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-21 18:05:16.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a90e28146fc4a7da9ad5e9872ac104d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5373223d97f44f408d86d057763276c6 : SUCCESS in 9m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e81282439ca3472185c5f5458a5594ca : SUCCESS in 11m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4514586b2fe6490488ad1f14ed5f6618 : SUCCESS in 6m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77c10d941ec6477c9ed8cf8a6520eb17 : SUCCESS in 18m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/ec16fb061e504b61914925b9cbe86666 : SUCCESS in 1h 05m 14s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6a90736cf29e4d38bed1c18ce6aace9b : SUCCESS in 1h 03m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/cd07fafb96bc426a9d1573a4fc70ddd6 : SUCCESS in 52m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7ba25060164c466ca2d190ed5592c823 : SUCCESS in 1h 02m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/09452b6f3e78439ab67d0960b08946f9 : SUCCESS in 11m 29s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/cf71cd10401741e1bab52b0048a809f7 : SUCCESS in 1h 17m 35s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2d5f309d5bf040e1a2b58effac72a8a8 : SUCCESS in 1h 33m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c668b1da9f9a49c197886851b0e7cde2 : SUCCESS in 1h 01m 07s","accounts_in_message":[],"_revision_number":12},{"id":"adc10a4e025ed6f72bede8558a9a7a085fdb8a56","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-21 18:05:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":12},{"id":"8c5fd2ee4a2eb7b8ac7f6ba25ff70a030925d521","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-21 18:06:36.000000000","message":"Patch Set 12:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d03a0dac7964f838ebba8c74368c757\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/703decadc6754abaa564f278a5b67862 : SUCCESS in 53s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6c739179a6144d46bf4a6610f9c630ca : SUCCESS in 47s","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"e07c074df9819ced54008f730f5e53881e568437","revisions":{"0d94d698d6f3a723b9fb9d1d618196783ad5fc0c":{"kind":"REWORK","_number":1,"created":"2026-07-30 15:50:04.000000000","uploader":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"ref":"refs/changes/10/999310/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/10/999310/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/10/999310/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/10/999310/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/10/999310/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/10/999310/1"}}},"commit":{"parents":[{"commit":"8bb9a7f7d1fb0fad7cb26d2aece58014d60e6cd4","subject":"Merge \"Remove unnecessary direct dependency on Pygments\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8bb9a7f7d1fb0fad7cb26d2aece58014d60e6cd4"}]}],"author":{"name":"anoop2","email":"anoop.shukla@netapp.com","date":"2026-07-30 15:47:49.000000000","tz":-240},"committer":{"name":"anoop2","email":"anoop.shukla@netapp.com","date":"2026-07-30 15:49:58.000000000","tz":-240},"subject":"NetApp: default transport to HTTPS and warn on HTTP Change NetApp Cinder driver default netapp_transport_type from http to https for secure-by-default behavior. Also add an initialization warning when operators explicitly configure http transport, indicating that it is insecure and not recommended for production. Includes unit coverage for warning/no-warning behavior and a regression test for the HTTPS default. Closes-Bug: #2157914","message":"NetApp: default transport to HTTPS and warn on HTTP\nChange NetApp Cinder driver default netapp_transport_type from http to\nhttps for secure-by-default behavior.\nAlso add an initialization warning when operators explicitly configure\nhttp transport, indicating that it is insecure and not recommended for\nproduction.\nIncludes unit coverage for warning/no-warning behavior and a regression\ntest for the HTTPS default.\nCloses-Bug: #2157914\n\nChange-Id: Ia94ddfacac494a301866eab565f04c921f6c2ea5\nSigned-off-by: anoop2 \u003canoop.shukla@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0d94d698d6f3a723b9fb9d1d618196783ad5fc0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0d94d698d6f3a723b9fb9d1d618196783ad5fc0c"}]},"branch":"refs/heads/master"},"711c2a48764b2539ff0c6e799ae6183a5e375dae":{"kind":"REWORK","_number":2,"created":"2026-07-31 04:04:52.000000000","uploader":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"ref":"refs/changes/10/999310/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/10/999310/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/10/999310/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/10/999310/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/10/999310/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/10/999310/2"}}},"commit":{"parents":[{"commit":"8bb9a7f7d1fb0fad7cb26d2aece58014d60e6cd4","subject":"Merge \"Remove unnecessary direct dependency on Pygments\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8bb9a7f7d1fb0fad7cb26d2aece58014d60e6cd4"}]}],"author":{"name":"anoop2","email":"anoop.shukla@netapp.com","date":"2026-07-30 15:47:49.000000000","tz":-240},"committer":{"name":"anoop2","email":"anoop.shukla@netapp.com","date":"2026-07-31 04:04:36.000000000","tz":-240},"subject":"NetApp: default transport to HTTPS and warn on HTTP","message":"NetApp: default transport to HTTPS and warn on HTTP\n\nChange NetApp Cinder driver default netapp_transport_type from http to\nhttps for secure-by-default behavior.\n\nAlso add an initialization warning when operators explicitly configure\nhttp transport, indicating that it is insecure and not recommended for\nproduction.\n\nIncludes unit coverage for warning/no-warning behavior.\n\nCloses-Bug: #2157914\nChange-Id: Ia94ddfacac494a301866eab565f04c921f6c2ea5\nSigned-off-by: anoop2 \u003canoop.shukla@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/711c2a48764b2539ff0c6e799ae6183a5e375dae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/711c2a48764b2539ff0c6e799ae6183a5e375dae"}]},"branch":"refs/heads/master"},"3d3beea475c22ecb47bdce620d7681993f0b0a8f":{"kind":"REWORK","_number":3,"created":"2026-08-05 03:56:00.000000000","uploader":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"ref":"refs/changes/10/999310/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/10/999310/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/10/999310/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/10/999310/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/10/999310/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/10/999310/3"}}},"commit":{"parents":[{"commit":"8bb9a7f7d1fb0fad7cb26d2aece58014d60e6cd4","subject":"Merge \"Remove unnecessary direct dependency on Pygments\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8bb9a7f7d1fb0fad7cb26d2aece58014d60e6cd4"}]}],"author":{"name":"anoop2","email":"anoop.shukla@netapp.com","date":"2026-07-30 15:47:49.000000000","tz":-240},"committer":{"name":"anoop2","email":"anoop.shukla@netapp.com","date":"2026-08-05 03:55:48.000000000","tz":-240},"subject":"NetApp: default transport HTTPS and clarify TLS verify options Switch NetApp ONTAP client defaults to HTTPS and add explicit TLS certificate verification controls for both REST and legacy ZAPI paths. Changes include: - Default transport type in NaServer/RestNaServer constructors to HTTPS. - Introduce and plumb netapp_ssl_cert_verify through client factory and   constructors. - Keep support for netapp_ssl_cert_path as trust-anchor path and update   option help text to correctly apply to both REST and legacy ZAPI clients. - Update unit tests for default transport, verify semantics, and URL behavior. - Add release notes describing verification precedence and operator guidance   for self-signed certificate deployments. Verification precedence: * ssl_cert_verify\u003dfalse -\u003e disable certificate verification * ssl_cert_verify\u003dtrue + ssl_cert_path set -\u003e verify against provided path * ssl_cert_verify\u003dtrue + no ssl_cert_path -\u003e verify against default CA roots Related pattern: LP#2157914 (Manila analogue).","message":"NetApp: default transport HTTPS and clarify TLS verify options\nSwitch NetApp ONTAP client defaults to HTTPS and add explicit TLS\ncertificate verification controls for both REST and legacy ZAPI paths.\nChanges include:\n- Default transport type in NaServer/RestNaServer constructors to HTTPS.\n- Introduce and plumb netapp_ssl_cert_verify through client factory and\n  constructors.\n- Keep support for netapp_ssl_cert_path as trust-anchor path and update\n  option help text to correctly apply to both REST and legacy ZAPI clients.\n- Update unit tests for default transport, verify semantics, and URL behavior.\n- Add release notes describing verification precedence and operator guidance\n  for self-signed certificate deployments.\nVerification precedence:\n* ssl_cert_verify\u003dfalse -\u003e disable certificate verification\n* ssl_cert_verify\u003dtrue + ssl_cert_path set -\u003e verify against provided path\n* ssl_cert_verify\u003dtrue + no ssl_cert_path -\u003e verify against default CA roots\nRelated pattern: LP#2157914 (Manila analogue).\n\nChange-Id: Ia94ddfacac494a301866eab565f04c921f6c2ea5\nSigned-off-by: anoop2 \u003canoop.shukla@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3d3beea475c22ecb47bdce620d7681993f0b0a8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3d3beea475c22ecb47bdce620d7681993f0b0a8f"}]},"branch":"refs/heads/master"},"0272a9116025fd9dfec4b9da0ffc251c2998d49d":{"kind":"REWORK","_number":4,"created":"2026-08-05 04:25:20.000000000","uploader":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"ref":"refs/changes/10/999310/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/10/999310/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/10/999310/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/10/999310/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/10/999310/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/10/999310/4"}}},"commit":{"parents":[{"commit":"8bb9a7f7d1fb0fad7cb26d2aece58014d60e6cd4","subject":"Merge \"Remove unnecessary direct dependency on Pygments\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8bb9a7f7d1fb0fad7cb26d2aece58014d60e6cd4"}]}],"author":{"name":"anoop2","email":"anoop.shukla@netapp.com","date":"2026-07-30 15:47:49.000000000","tz":-240},"committer":{"name":"anoop2","email":"anoop.shukla@netapp.com","date":"2026-08-05 04:25:10.000000000","tz":-240},"subject":"NetApp: default transport HTTPS and clarify TLS verify options Switch NetApp ONTAP client defaults to HTTPS and add explicit TLS certificate verification controls for both REST and legacy ZAPI paths. Changes include: - Default transport type in NaServer/RestNaServer constructors to HTTPS. - Introduce and plumb netapp_ssl_cert_verify through client factory and   constructors. - Keep support for netapp_ssl_cert_path as trust-anchor path and update   option help text to correctly apply to both REST and legacy ZAPI clients. - Update unit tests for default transport, verify semantics, and URL behavior. - Add release notes describing verification precedence and operator guidance   for self-signed certificate deployments. Verification precedence: * ssl_cert_verify\u003dfalse -\u003e disable certificate verification * ssl_cert_verify\u003dtrue + ssl_cert_path set -\u003e verify against provided path * ssl_cert_verify\u003dtrue + no ssl_cert_path -\u003e verify against default CA roots Related pattern: LP#2157914 (Manila analogue).","message":"NetApp: default transport HTTPS and clarify TLS verify options\nSwitch NetApp ONTAP client defaults to HTTPS and add explicit TLS\ncertificate verification controls for both REST and legacy ZAPI paths.\nChanges include:\n- Default transport type in NaServer/RestNaServer constructors to HTTPS.\n- Introduce and plumb netapp_ssl_cert_verify through client factory and\n  constructors.\n- Keep support for netapp_ssl_cert_path as trust-anchor path and update\n  option help text to correctly apply to both REST and legacy ZAPI clients.\n- Update unit tests for default transport, verify semantics, and URL behavior.\n- Add release notes describing verification precedence and operator guidance\n  for self-signed certificate deployments.\nVerification precedence:\n* ssl_cert_verify\u003dfalse -\u003e disable certificate verification\n* ssl_cert_verify\u003dtrue + ssl_cert_path set -\u003e verify against provided path\n* ssl_cert_verify\u003dtrue + no ssl_cert_path -\u003e verify against default CA roots\nRelated pattern: LP#2157914 (Manila analogue).\n\nChange-Id: Ia94ddfacac494a301866eab565f04c921f6c2ea5\nSigned-off-by: anoop2 \u003canoop.shukla@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0272a9116025fd9dfec4b9da0ffc251c2998d49d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0272a9116025fd9dfec4b9da0ffc251c2998d49d"}]},"branch":"refs/heads/master"},"42bdd763bec5a7ed653a4a0dfa27c587551bc571":{"kind":"NO_CHANGE","_number":5,"created":"2026-08-05 04:27:27.000000000","uploader":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"ref":"refs/changes/10/999310/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/10/999310/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/10/999310/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/10/999310/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/10/999310/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/10/999310/5"}}},"commit":{"parents":[{"commit":"8bb9a7f7d1fb0fad7cb26d2aece58014d60e6cd4","subject":"Merge \"Remove unnecessary direct dependency on Pygments\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8bb9a7f7d1fb0fad7cb26d2aece58014d60e6cd4"}]}],"author":{"name":"anoop2","email":"anoop.shukla@netapp.com","date":"2026-07-30 15:47:49.000000000","tz":-240},"committer":{"name":"anoop2","email":"anoop.shukla@netapp.com","date":"2026-08-05 04:27:17.000000000","tz":-240},"subject":"NetApp: default transport HTTPS and clarify TLS verify options Switch NetApp ONTAP client defaults to HTTPS and add explicit TLS certificate verification controls for both REST and legacy ZAPI paths. Changes include: - Default transport type in NaServer/RestNaServer constructors to HTTPS. - Introduce and plumb netapp_ssl_cert_verify through client factory and   constructors. - Keep support for netapp_ssl_cert_path as trust-anchor path and update   option help text to correctly apply to both REST and legacy ZAPI clients. - Update unit tests for default transport, verify semantics, and URL behavior. - Add release notes describing verification precedence and operator guidance   for self-signed certificate deployments. Verification precedence: * ssl_cert_verify\u003dfalse -\u003e disable certificate verification * ssl_cert_verify\u003dtrue + ssl_cert_path set -\u003e verify against provided path * ssl_cert_verify\u003dtrue + no ssl_cert_path -\u003e verify against default CA roots Related pattern: LP#2157914 (Manila analogue).","message":"NetApp: default transport HTTPS and clarify TLS verify options\nSwitch NetApp ONTAP client defaults to HTTPS and add explicit TLS\ncertificate verification controls for both REST and legacy ZAPI paths.\nChanges include:\n- Default transport type in NaServer/RestNaServer constructors to HTTPS.\n- Introduce and plumb netapp_ssl_cert_verify through client factory and\n  constructors.\n- Keep support for netapp_ssl_cert_path as trust-anchor path and update\n  option help text to correctly apply to both REST and legacy ZAPI clients.\n- Update unit tests for default transport, verify semantics, and URL behavior.\n- Add release notes describing verification precedence and operator guidance\n  for self-signed certificate deployments.\nVerification precedence:\n* ssl_cert_verify\u003dfalse -\u003e disable certificate verification\n* ssl_cert_verify\u003dtrue + ssl_cert_path set -\u003e verify against provided path\n* ssl_cert_verify\u003dtrue + no ssl_cert_path -\u003e verify against default CA roots\nRelated pattern: LP#2157914 (Manila analogue).\n\nChange-Id: Ia94ddfacac494a301866eab565f04c921f6c2ea5\nSigned-off-by: anoop2 \u003canoop.shukla@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/42bdd763bec5a7ed653a4a0dfa27c587551bc571"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/42bdd763bec5a7ed653a4a0dfa27c587551bc571"}]},"branch":"refs/heads/master"},"762f3a21db94362764b4c57a8d626d9e45941364":{"kind":"REWORK","_number":6,"created":"2026-08-05 04:34:18.000000000","uploader":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"ref":"refs/changes/10/999310/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/10/999310/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/10/999310/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/10/999310/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/10/999310/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/10/999310/6"}}},"commit":{"parents":[{"commit":"8bb9a7f7d1fb0fad7cb26d2aece58014d60e6cd4","subject":"Merge \"Remove unnecessary direct dependency on Pygments\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8bb9a7f7d1fb0fad7cb26d2aece58014d60e6cd4"}]}],"author":{"name":"anoop2","email":"anoop.shukla@netapp.com","date":"2026-07-30 15:47:49.000000000","tz":-240},"committer":{"name":"anoop2","email":"anoop.shukla@netapp.com","date":"2026-08-05 04:33:21.000000000","tz":-240},"subject":"NetApp: default transport HTTPS and clarify TLS verify options Switch NetApp ONTAP client defaults to HTTPS and add explicit TLS certificate verification controls for both REST and legacy ZAPI paths. Changes include: - Default transport type in NaServer/RestNaServer constructors to HTTPS. - Introduce and plumb netapp_ssl_cert_verify through client factory and   constructors. - Keep support for netapp_ssl_cert_path as trust-anchor path and update   option help text to correctly apply to both REST and legacy ZAPI clients. - Update unit tests for default transport, verify semantics, and URL behavior. - Add release notes describing verification precedence and operator guidance   for self-signed certificate deployments. Verification precedence: * ssl_cert_verify\u003dfalse -\u003e disable certificate verification * ssl_cert_verify\u003dtrue + ssl_cert_path set -\u003e verify against provided path * ssl_cert_verify\u003dtrue + no ssl_cert_path -\u003e verify against default CA roots Related pattern: LP#2157914 (Manila analogue).","message":"NetApp: default transport HTTPS and clarify TLS verify options\nSwitch NetApp ONTAP client defaults to HTTPS and add explicit TLS\ncertificate verification controls for both REST and legacy ZAPI paths.\nChanges include:\n- Default transport type in NaServer/RestNaServer constructors to HTTPS.\n- Introduce and plumb netapp_ssl_cert_verify through client factory and\n  constructors.\n- Keep support for netapp_ssl_cert_path as trust-anchor path and update\n  option help text to correctly apply to both REST and legacy ZAPI clients.\n- Update unit tests for default transport, verify semantics, and URL behavior.\n- Add release notes describing verification precedence and operator guidance\n  for self-signed certificate deployments.\nVerification precedence:\n* ssl_cert_verify\u003dfalse -\u003e disable certificate verification\n* ssl_cert_verify\u003dtrue + ssl_cert_path set -\u003e verify against provided path\n* ssl_cert_verify\u003dtrue + no ssl_cert_path -\u003e verify against default CA roots\nRelated pattern: LP#2157914 (Manila analogue).\n\nChange-Id: Ia94ddfacac494a301866eab565f04c921f6c2ea5\nSigned-off-by: anoop2 \u003canoop.shukla@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/762f3a21db94362764b4c57a8d626d9e45941364"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/762f3a21db94362764b4c57a8d626d9e45941364"}]},"branch":"refs/heads/master"},"6176ce494aa3cd67ecc86d580338d103041d0d6f":{"kind":"REWORK","_number":7,"created":"2026-08-05 10:09:29.000000000","uploader":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"ref":"refs/changes/10/999310/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/10/999310/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/10/999310/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/10/999310/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/10/999310/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/10/999310/7"}}},"commit":{"parents":[{"commit":"8bb9a7f7d1fb0fad7cb26d2aece58014d60e6cd4","subject":"Merge \"Remove unnecessary direct dependency on Pygments\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8bb9a7f7d1fb0fad7cb26d2aece58014d60e6cd4"}]}],"author":{"name":"anoop2","email":"anoop.shukla@netapp.com","date":"2026-07-30 15:47:49.000000000","tz":-240},"committer":{"name":"anoop2","email":"anoop.shukla@netapp.com","date":"2026-08-05 10:09:20.000000000","tz":-240},"subject":"NetApp: default transport HTTPS and clarify TLS verify options Switch NetApp ONTAP client defaults to HTTPS and add explicit TLS certificate verification controls for both REST and legacy ZAPI paths. Changes include: - Default transport type in NaServer/RestNaServer constructors to HTTPS. - Introduce and plumb netapp_ssl_cert_verify through client factory and   constructors. - Keep support for netapp_ssl_cert_path as trust-anchor path and update   option help text to correctly apply to both REST and legacy ZAPI clients. - Update unit tests for default transport, verify semantics, and URL behavior. - Add release notes describing verification precedence and operator guidance   for self-signed certificate deployments. Verification precedence: * ssl_cert_verify\u003dfalse -\u003e disable certificate verification * ssl_cert_verify\u003dtrue + ssl_cert_path set -\u003e verify against provided path * ssl_cert_verify\u003dtrue + no ssl_cert_path -\u003e verify against default CA roots Related pattern: LP#2157914 (Manila analogue).","message":"NetApp: default transport HTTPS and clarify TLS verify options\nSwitch NetApp ONTAP client defaults to HTTPS and add explicit TLS\ncertificate verification controls for both REST and legacy ZAPI paths.\nChanges include:\n- Default transport type in NaServer/RestNaServer constructors to HTTPS.\n- Introduce and plumb netapp_ssl_cert_verify through client factory and\n  constructors.\n- Keep support for netapp_ssl_cert_path as trust-anchor path and update\n  option help text to correctly apply to both REST and legacy ZAPI clients.\n- Update unit tests for default transport, verify semantics, and URL behavior.\n- Add release notes describing verification precedence and operator guidance\n  for self-signed certificate deployments.\nVerification precedence:\n* ssl_cert_verify\u003dfalse -\u003e disable certificate verification\n* ssl_cert_verify\u003dtrue + ssl_cert_path set -\u003e verify against provided path\n* ssl_cert_verify\u003dtrue + no ssl_cert_path -\u003e verify against default CA roots\nRelated pattern: LP#2157914 (Manila analogue).\n\nChange-Id: Ia94ddfacac494a301866eab565f04c921f6c2ea5\nSigned-off-by: anoop2 \u003canoop.shukla@netapp.com\u003e\nCo-authored-by: Cursor \u003ccursoragent@cursor.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6176ce494aa3cd67ecc86d580338d103041d0d6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6176ce494aa3cd67ecc86d580338d103041d0d6f"}]},"branch":"refs/heads/master"},"9688c220979de542ff9183b8aeaf5c72769e015e":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2026-08-05 10:18:19.000000000","uploader":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"ref":"refs/changes/10/999310/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/10/999310/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/10/999310/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/10/999310/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/10/999310/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/10/999310/8"}}},"commit":{"parents":[{"commit":"8bb9a7f7d1fb0fad7cb26d2aece58014d60e6cd4","subject":"Merge \"Remove unnecessary direct dependency on Pygments\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8bb9a7f7d1fb0fad7cb26d2aece58014d60e6cd4"}]}],"author":{"name":"anoop2","email":"anoop.shukla@netapp.com","date":"2026-07-30 15:47:49.000000000","tz":-240},"committer":{"name":"anoop2","email":"anoop.shukla@netapp.com","date":"2026-08-05 10:18:09.000000000","tz":-240},"subject":"NetApp: default HTTPS transport and TLS verify","message":"NetApp: default HTTPS transport and TLS verify\n\nSwitch NetApp ONTAP client defaults to HTTPS and add explicit TLS\ncertificate verification controls for both REST and legacy ZAPI paths.\n\nChanges include:\n- Default transport type in NaServer/RestNaServer constructors to HTTPS.\n- Introduce and plumb netapp_ssl_cert_verify through client factory and\n  constructors.\n- Keep support for netapp_ssl_cert_path as trust-anchor path and update\n  option help text to correctly apply to both REST and legacy ZAPI clients.\n- Update unit tests for default transport, verify semantics, and URL behavior.\n- Add release notes describing verification precedence and operator guidance\n  for self-signed certificate deployments.\n\nVerification precedence:\n* ssl_cert_verify\u003dfalse -\u003e disable certificate verification\n* ssl_cert_verify\u003dtrue + ssl_cert_path set -\u003e verify against provided path\n* ssl_cert_verify\u003dtrue + no ssl_cert_path -\u003e verify against default CA roots\n\nRelated pattern: LP#2157914 (Manila analogue).\n\nChange-Id: Ia94ddfacac494a301866eab565f04c921f6c2ea5\nSigned-off-by: anoop2 \u003canoop.shukla@netapp.com\u003e\nAssisted-By: Cursor\nCo-authored-by: Cursor \u003ccursoragent@cursor.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9688c220979de542ff9183b8aeaf5c72769e015e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9688c220979de542ff9183b8aeaf5c72769e015e"}]},"branch":"refs/heads/master"},"40dd40672f5d9af0bb889ea5c55050e8f76cdb2a":{"kind":"REWORK","_number":9,"created":"2026-08-13 13:50:49.000000000","uploader":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"ref":"refs/changes/10/999310/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/10/999310/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/10/999310/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/10/999310/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/10/999310/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/10/999310/9"}}},"commit":{"parents":[{"commit":"911497601b299bcb679f454fe9f3fb60e66ada94","subject":"Merge \"Run Cinder\u0027s Backup service in native threads\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/911497601b299bcb679f454fe9f3fb60e66ada94"}]}],"author":{"name":"anoop2","email":"anoop.shukla@netapp.com","date":"2026-07-30 15:47:49.000000000","tz":-240},"committer":{"name":"anoop2","email":"anoop.shukla@netapp.com","date":"2026-08-13 13:49:49.000000000","tz":-240},"subject":"NetApp: default HTTPS transport and TLS verify","message":"NetApp: default HTTPS transport and TLS verify\n\nSwitch NetApp ONTAP client defaults to HTTPS and add explicit TLS\ncertificate verification controls for both REST and legacy ZAPI paths.\n\nChanges include:\n- Default transport type in NaServer/RestNaServer constructors to HTTPS.\n- Introduce and plumb netapp_ssl_cert_verify through client factory and\n  constructors.\n- Keep support for netapp_ssl_cert_path as trust-anchor path and update\n  option help text to correctly apply to both REST and legacy ZAPI clients.\n- Update unit tests for default transport, verify semantics, and URL behavior.\n- Add release notes describing verification precedence and operator guidance\n  for self-signed certificate deployments.\n\nVerification precedence:\n* ssl_cert_verify\u003dfalse -\u003e disable certificate verification\n* ssl_cert_verify\u003dtrue + ssl_cert_path set -\u003e verify against provided path\n* ssl_cert_verify\u003dtrue + no ssl_cert_path -\u003e verify against default CA roots\n\nRelated-Bug: #2157914\n\nChange-Id: Ia94ddfacac494a301866eab565f04c921f6c2ea5\nSigned-off-by: anoop2 \u003canoop.shukla@netapp.com\u003e\nAssisted-By: Cursor\nCo-authored-by: Cursor \u003ccursoragent@cursor.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/40dd40672f5d9af0bb889ea5c55050e8f76cdb2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/40dd40672f5d9af0bb889ea5c55050e8f76cdb2a"}]},"branch":"refs/heads/master"},"e4b6a1fc0beac9614d042887282270e4be0944a3":{"kind":"REWORK","_number":10,"created":"2026-08-14 06:27:51.000000000","uploader":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"ref":"refs/changes/10/999310/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/10/999310/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/10/999310/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/10/999310/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/10/999310/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/10/999310/10"}}},"commit":{"parents":[{"commit":"911497601b299bcb679f454fe9f3fb60e66ada94","subject":"Merge \"Run Cinder\u0027s Backup service in native threads\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/911497601b299bcb679f454fe9f3fb60e66ada94"}]}],"author":{"name":"anoop2","email":"anoop.shukla@netapp.com","date":"2026-07-30 15:47:49.000000000","tz":-240},"committer":{"name":"anoop2","email":"anoop.shukla@netapp.com","date":"2026-08-14 05:12:33.000000000","tz":-240},"subject":"NetApp: default HTTPS transport and TLS verify","message":"NetApp: default HTTPS transport and TLS verify\n\nSwitch NetApp ONTAP client defaults to HTTPS and add explicit TLS\ncertificate verification controls for both REST and legacy ZAPI paths.\n\nChanges include:\n- Default transport type in NaServer/RestNaServer constructors to HTTPS.\n- Introduce and plumb netapp_ssl_cert_verify through client factory and\n  constructors.\n- Keep support for netapp_ssl_cert_path as trust-anchor path and update\n  option help text to correctly apply to both REST and legacy ZAPI clients.\n- Log startup warnings for insecure transport options from driver do_setup\n  paths only (block_base, nfs_base, nvme_library), not NetAppDriver.__new__.\n- Update unit tests for default transport, verify semantics, and URL behavior.\n- Add release notes describing verification precedence and operator guidance\n  for self-signed certificate deployments.\n\nVerification precedence:\n* ssl_cert_verify\u003dfalse -\u003e disable certificate verification\n* ssl_cert_verify\u003dtrue + ssl_cert_path set -\u003e verify against provided path\n* ssl_cert_verify\u003dtrue + no ssl_cert_path -\u003e verify against default CA roots\n\nRelated-Bug: #2157914\n\nChange-Id: Ia94ddfacac494a301866eab565f04c921f6c2ea5\nSigned-off-by: anoop2 \u003canoop.shukla@netapp.com\u003e\nAssisted-By: Cursor\nCo-authored-by: Cursor \u003ccursoragent@cursor.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e4b6a1fc0beac9614d042887282270e4be0944a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e4b6a1fc0beac9614d042887282270e4be0944a3"}]},"branch":"refs/heads/master"},"ba818871aacaccc83fc03fa6bf1a965425dfdfee":{"kind":"REWORK","_number":11,"created":"2026-08-17 04:58:42.000000000","uploader":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"ref":"refs/changes/10/999310/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/10/999310/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/10/999310/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/10/999310/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/10/999310/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/10/999310/11"}}},"commit":{"parents":[{"commit":"911497601b299bcb679f454fe9f3fb60e66ada94","subject":"Merge \"Run Cinder\u0027s Backup service in native threads\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/911497601b299bcb679f454fe9f3fb60e66ada94"}]}],"author":{"name":"anoop2","email":"anoop.shukla@netapp.com","date":"2026-07-30 15:47:49.000000000","tz":-240},"committer":{"name":"anoop2","email":"anoop.shukla@netapp.com","date":"2026-08-17 04:58:37.000000000","tz":-240},"subject":"NetApp: default HTTPS transport and TLS verify","message":"NetApp: default HTTPS transport and TLS verify\n\nSwitch NetApp ONTAP client defaults to HTTPS and add explicit TLS\ncertificate verification controls for both REST and legacy ZAPI paths.\n\nChanges include:\n- Default transport type in NaServer/RestNaServer constructors to HTTPS.\n- Introduce and plumb netapp_ssl_cert_verify through client factory and\n  constructors.\n- Keep support for netapp_ssl_cert_path as trust-anchor path and update\n  option help text to correctly apply to both REST and legacy ZAPI clients.\n- Log startup warnings for insecure transport options from driver do_setup\n  paths only (block_base, nfs_base, nvme_library), not NetAppDriver.__new__.\n- Assert do_setup wiring for insecure-transport warnings in block/nfs/nvme\n  unit tests.\n- Fix .inc config table ordering and add netapp_ssl_cert_path entries.\n- Update unit tests for default transport, verify semantics, and URL behavior.\n- Add release notes describing verification precedence and operator guidance\n  for self-signed certificate deployments.\n\nVerification precedence:\n* ssl_cert_verify\u003dfalse -\u003e disable certificate verification\n* ssl_cert_verify\u003dtrue + ssl_cert_path set -\u003e verify against provided path\n* ssl_cert_verify\u003dtrue + no ssl_cert_path -\u003e verify against default CA roots\n\nRelated-Bug: #2157914\n\nChange-Id: Ia94ddfacac494a301866eab565f04c921f6c2ea5\nSigned-off-by: anoop2 \u003canoop.shukla@netapp.com\u003e\nAssisted-By: Cursor\nCo-authored-by: Cursor \u003ccursoragent@cursor.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ba818871aacaccc83fc03fa6bf1a965425dfdfee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ba818871aacaccc83fc03fa6bf1a965425dfdfee"}]},"branch":"refs/heads/master"},"e07c074df9819ced54008f730f5e53881e568437":{"kind":"REWORK","_number":12,"created":"2026-08-18 09:36:40.000000000","uploader":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"ref":"refs/changes/10/999310/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/10/999310/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/10/999310/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/10/999310/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/10/999310/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/10/999310/12"}}},"commit":{"parents":[{"commit":"911497601b299bcb679f454fe9f3fb60e66ada94","subject":"Merge \"Run Cinder\u0027s Backup service in native threads\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/911497601b299bcb679f454fe9f3fb60e66ada94"}]}],"author":{"name":"anoop2","email":"anoop.shukla@netapp.com","date":"2026-07-30 15:47:49.000000000","tz":-240},"committer":{"name":"anoop2","email":"anoop.shukla@netapp.com","date":"2026-08-18 09:36:33.000000000","tz":-240},"subject":"NetApp: default HTTPS transport and TLS verify","message":"NetApp: default HTTPS transport and TLS verify\n\nSwitch NetApp ONTAP client defaults to HTTPS and add explicit TLS\ncertificate verification controls for both REST and legacy ZAPI paths.\n\nChanges include:\n- Default transport type in NaServer/RestNaServer constructors to HTTPS.\n- Introduce and plumb netapp_ssl_cert_verify through client factory and\n  constructors.\n- Keep support for netapp_ssl_cert_path as trust-anchor path and update\n  option help text to correctly apply to both REST and legacy ZAPI clients.\n- Log startup warnings for insecure transport options from driver do_setup\n  paths only (block_base, nfs_base, nvme_library), not NetAppDriver.__new__.\n- Assert do_setup wiring for insecure-transport warnings in block/nfs/nvme\n  unit tests.\n- Fix .inc config table ordering and add netapp_ssl_cert_path entries.\n- Scope netapp_ssl_cert_path help text to basic-auth; cert-auth uses\n  netapp_ca_certificate_file (options.py and .inc tables).\n- Update unit tests for default transport, verify semantics, and URL behavior.\n- Add release notes describing verification precedence and operator guidance\n  for self-signed certificate deployments.\n\nVerification precedence:\n* ssl_cert_verify\u003dfalse -\u003e disable certificate verification\n* ssl_cert_verify\u003dtrue + ssl_cert_path set -\u003e verify against provided path\n* ssl_cert_verify\u003dtrue + no ssl_cert_path -\u003e verify against default CA roots\n\nRelated-Bug: #2157914\n\nChange-Id: Ia94ddfacac494a301866eab565f04c921f6c2ea5\nSigned-off-by: anoop2 \u003canoop.shukla@netapp.com\u003e\nAssisted-By: Cursor\nCo-authored-by: Cursor \u003ccursoragent@cursor.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e07c074df9819ced54008f730f5e53881e568437"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e07c074df9819ced54008f730f5e53881e568437"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
