)]}'
{"id":"openstack%2Fcinder~999463","triplet_id":"openstack%2Fcinder~master~I9a2ee22fbed0dc0cbbe04214213d0363457d744f","project":"openstack/cinder","branch":"master","topic":"rbd-group-replication","attention_set":{"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2026-08-03 19:35:44.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I9a2ee22fbed0dc0cbbe04214213d0363457d744f","subject":"WIP: Group failover: Update host and AZ","status":"NEW","created":"2026-07-31 21:44:14.000000000","updated":"2026-09-15 23:50:33.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a95914c63f719602d693861583755b9a9bc52040","_number":999463,"virtual_id_number":999463,"owner":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-15 12:04:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-31 21:47:11.000000000","updated_by":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"reviewer":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"state":"CC"},{"updated":"2026-07-31 21:47:16.000000000","updated_by":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"reviewer":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"state":"CC"},{"updated":"2026-07-31 21:47:30.000000000","updated_by":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"reviewer":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"state":"CC"},{"updated":"2026-07-31 21:58:27.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2026-07-31 22:13:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-07-31 22:17:15.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2026-07-31 22:49:57.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-08-01 00:12:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-01 00:33:20.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-08-03 22:15:53.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-09-15 11:27:56.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-15 12:14:49.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"}],"messages":[{"id":"2bd92675db2944b8aab8a927e2af09a8c75f354d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2026-07-31 21:44:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7a08058aaf6280e4b25580d477da948e5bb0cb0b","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-07-31 21:47:11.000000000","message":"Patch Set 1:\n\nHPE AlletraMP iSCSI CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1053 : FAILURE in 0h 2m 37s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/999463","accounts_in_message":[],"_revision_number":1},{"id":"8dac4a1466712df4ccfd53f7c66505178a67b482","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-07-31 21:47:16.000000000","message":"Patch Set 1:\n\nHPE AlletraMP FC CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_FC_CI build/783 : FAILURE in 0h 2m 40s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/999463","accounts_in_message":[],"_revision_number":1},{"id":"2f3f7e4358be3c6a567a65d944d8cc887290daa5","author":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"date":"2026-07-31 21:47:30.000000000","message":"Patch Set 1:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_NVME_CI build/692 : FAILURE in 0h 2m 56s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/999463","accounts_in_message":[],"_revision_number":1},{"id":"a7348c8e6d5f6b742a1fe1bb69ca5245cf0753a2","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-31 21:58:27.000000000","message":"Patch Set 1:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4818/.","accounts_in_message":[],"_revision_number":1},{"id":"2b64656d08577a37937feb92e2d24f7bc5c852c7","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-31 21:59:46.000000000","message":"Patch Set 1:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4994/.","accounts_in_message":[],"_revision_number":1},{"id":"b1c6d75cc6612b1b4ee0cc21658b11214116f8bd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-31 22:13:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fccfc663d9e54962bf65c058e3c25f9f\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5be6e93ceb3f43e7945e856662e1af21 : SUCCESS in 20m 18s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1ae36883ba2b4cf4a7f297ec500abfe7 : SUCCESS in 19m 02s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/1cb6d9aeb8794feeb420be1970568c52 : SUCCESS in 26m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"696c28a7eff37382bd86f4b933ee1215a0c3ab2b","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-31 22:17:15.000000000","message":"Patch Set 1:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/d0bc17b02375482ea17b17a120386c57/","accounts_in_message":[],"_revision_number":1},{"id":"9cb4fb77e42a2c34ba224f2ce49135204937d27e","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-31 22:47:46.000000000","message":"Patch Set 1:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/d0bc17b02375482ea17b17a120386c57\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9b8e6d7747b845adb4c3333d85b52674 : SUCCESS in 43m 13s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/21619d0deabb45c7a0341af73c9ebbf4 : SUCCESS in 40m 34s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ff736b6f283d44289d8492c1f4255f32 : SUCCESS in 50m 29s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2e8835b365544437a795faf3352e8fba : SUCCESS in 32m 16s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e8a743c67fc44caaa5fa9e23f50ce6df : FAILURE in 14m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4aec8c7b5f4e729271ac97db6ba846c7bcf0d8ed","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-31 22:49:57.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I9a2ee22fbed0dc0cbbe04214213d0363457d744f .","accounts_in_message":[],"_revision_number":1},{"id":"4ed25e1d15252889691522ae03af233919efb6d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-01 00:12:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d18d05ba2abc4b6891238f6f5ca6bce0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/883488beb559450991ffc3449f8152ea : SUCCESS in 4m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bca85be1571b48c58912a4e0a83986c3 : SUCCESS in 7m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9139df13af264a038ed6e03deaa6a36b : SUCCESS in 8m 25s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4fe08180646a4dc5a82905c96c6ddad8 : SUCCESS in 12m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d75d20717fb441dcabdb88d8c0502402 : SUCCESS in 9m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/16b19167dc1c414b88dc886c5668afd0 : SUCCESS in 32m 21s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c8d7a82a421b4b61a3cbb131f73e7967 : SUCCESS in 36m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/97811fded05548fea928a9faaf9f4417 : SUCCESS in 54m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1b226fac7e054154a4bae158f3740484 : SUCCESS in 30m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/79c5ca9d3f504e9ca0a19cd0c4e4c4c6 : SUCCESS in 5m 20s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/99f6331cf6334828b42c4ec517885768 : SUCCESS in 21m 29s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9cc7de3e36ad46958e0bd71f442f0941 : SUCCESS in 2m 55s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/cb6ff2f151144d61949d2f093b891eab : SUCCESS in 3m 34s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/d06f52e8d8624c8bb46bbac83caa57e9 : FAILURE in 10m 38s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/7fd1c2beb921416aab2061b5c6eae530 : FAILURE in 6m 49s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/8db78b83f50b40efa557eea24129aad5 : FAILURE in 6m 09s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/48d5dbf5c689430d8eb41138d7215d46 : FAILURE in 10m 45s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8730bfe143cd4e448775c152df1bee06 : SUCCESS in 38m 16s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7017e5d9e2334c50aa3a4c421be1c3db : SUCCESS in 5m 46s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b5b339f602c14b8bbb6c1456208d4dbc : SUCCESS in 1h 34m 03s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/a53ad3e22ad7442e8ded754f2a3a6969 : FAILURE in 2h 08m 50s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/73b0dc11b8ec4f93bd8c03483c674e47 : SUCCESS in 1h 50m 54s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/448d3835879b48c59c1b6c64ab6911a8 : FAILURE in 2h 23m 19s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/55e8a4f2cdbc4a71a10a9790e53f8853 : SUCCESS in 1h 56m 31s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/c0c2e4c4335a4e2fac1c96635b68ddc1 : SUCCESS in 18m 18s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b4e9f7df270e48d99b5e769e431b3270 : SUCCESS in 46m 17s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f29ac6ddbc4b4b66b67df12d22603b15 : SUCCESS in 1h 25m 33s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/7b8cf1e46cec4733a01205001b531693 : SUCCESS in 34m 48s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/7e3769892ebd426cb9e225f45a5bb9c6 : SUCCESS in 2h 22m 35s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c1c793a4544e4e9e952bba9ec24bc0af : SUCCESS in 53m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2974c750ce294c93b6841379c92a9ce0 : SUCCESS in 49m 25s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"e2adcd4c921c893bb8fb6fb16f7ddbf6d8b2747a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-01 00:33:20.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/d1930f19989f4bb6b3af6c431e539172 : SUCCESS in 42m 59s\n- cinder-storpool-tempest-iscsi-multipath build/fee3f5cbc7e5444ea3531f97169b330d : SUCCESS in 43m 56s","accounts_in_message":[],"_revision_number":1},{"id":"74bf808b42273832574f24d8426445a0d6ab0d0e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2026-08-03 18:56:28.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"7d98c89842036fcd435c1ce0aa422822ea805ad4","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-03 19:35:44.000000000","message":"Patch Set 2:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5007/.","accounts_in_message":[],"_revision_number":2},{"id":"e13ebace461d8163baa1e25e01dbea5611ac0d5e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-03 19:39:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79cb737a401b4403a44bbd28422ce9c0\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9cdc3eac384d46cab2c83f7a4de6ec2e : SUCCESS in 14m 05s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/69c015f7a7604864b2b4ccac4c39ec16 : SUCCESS in 11m 37s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/7a335698867d4c0fbb7ddbafc220e724 : SUCCESS in 23m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"13d4abe9d9b47f04fdd9c8923b6c7a7f6aecc4af","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-03 19:46:31.000000000","message":"Patch Set 2:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4831/.","accounts_in_message":[],"_revision_number":2},{"id":"e6ae24a104421b6aa8ee4f5366f5202f8f2629c2","author":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"date":"2026-08-03 19:55:42.000000000","message":"Patch Set 2:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_NVME_CI build/701 : FAILURE in 0h 1m 50s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/999463","accounts_in_message":[],"_revision_number":2},{"id":"9fe39c2e98b00d285aaef78918f8a371283b85f8","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-03 19:56:57.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I9a2ee22fbed0dc0cbbe04214213d0363457d744f .","accounts_in_message":[],"_revision_number":2},{"id":"8d210ee7ce8491caedbc8d90db79a29ed62e47c7","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-08-03 19:57:04.000000000","message":"Patch Set 2:\n\nHPE AlletraMP iSCSI CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1062 : FAILURE in 0h 2m 3s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/999463","accounts_in_message":[],"_revision_number":2},{"id":"1158eeafa12e88cdc93962f94c4b7e8a6bb4fe8a","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-08-03 19:57:08.000000000","message":"Patch Set 2:\n\nHPE AlletraMP FC CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_FC_CI build/794 : FAILURE in 0h 2m 8s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/999463","accounts_in_message":[],"_revision_number":2},{"id":"154cf0cddc881829542f3ac5c4d356b626ec0613","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-03 20:11:21.000000000","message":"Patch Set 2:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/c3f4a23bd8cd41629633c31b27ca1ee2/","accounts_in_message":[],"_revision_number":2},{"id":"d8430e2e52e6b00a986ffa9c174633bf91e0dc93","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-03 21:57:48.000000000","message":"Patch Set 2:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/c3f4a23bd8cd41629633c31b27ca1ee2\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/dd0ef0c7ef5441b6aab5a784404fa91c : SUCCESS in 39m 42s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8e5af9be4aef45159986c7db4f1b13f7 : SUCCESS in 36m 40s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/afad0397cf8648cf9b282b0086eb7e7a : SUCCESS in 53m 04s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/91c5651a8c9c4b2ba600390460cdffde : SUCCESS in 35m 01s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/892d5304d83f4893ae65afbb6ead5f99 : FAILURE in 17m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"06cea3bbf7203b3c83ef0daaf405473bb513107f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-03 22:15:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7440650104884159b3d090b78d3a7fa8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a0ca4ecb6be4526a6a18c01981fef81 : SUCCESS in 6m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/76d3c79b456e4c5791314458fcf66326 : SUCCESS in 12m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/96b8ba30577e42f996aadd470e3e4c6f : SUCCESS in 8m 04s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/dc2fd10534fb491194f89def1bcf3983 : SUCCESS in 11m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aead0ebca4a54e5eae0bea9c0bb5beec : SUCCESS in 16m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/a328bb9ab1cc4404a77dda194ab57915 : SUCCESS in 36m 21s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6eb58e49dbc24da1b2064e08ce46fa34 : SUCCESS in 1h 07m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/046586f68bd84d7a8f0479f77bf93471 : SUCCESS in 1h 19m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f19fade14a874a2086318c44ecde7348 : SUCCESS in 35m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ddad190ef103419d875e2353101113e9 : SUCCESS in 7m 57s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/66dae2aee2d946c483f7e8682e5968dc : TIMED_OUT in 41m 19s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/5c256e5b083543688b92430c8f1a576f : SUCCESS in 3m 40s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ec6b931fa53f4c778b7b0a8b54d5e543 : SUCCESS in 6m 57s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/ca09348bf9f84ed7be4af08a2e5f3915 : FAILURE in 24m 53s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/d1947658c78642b186a7101092465a73 : SUCCESS in 6m 23s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/9379dc0a554b4e8bb5427a5b9f912907 : SUCCESS in 5m 52s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/3dccb92512e1410986a2e49681596d1b : SUCCESS in 7m 35s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ce99cf8d8fc44db99f71f15c35839df8 : SUCCESS in 36m 54s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/197917527c2b47e19bd710244a1767db : SUCCESS in 4m 27s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b87ba5dcfcc94a3e9fd5058b40d5e117 : SUCCESS in 48m 32s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d25670f382bb43dc92b9004d5d361532 : FAILURE in 2h 03m 59s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6c14740c91d642fa9314d2fa588a7a8e : SUCCESS in 1h 18m 23s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/a2e1fedef1c5421b894d4542910d63cd : TIMED_OUT in 3h 07m 22s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/3781130710024a199e7c5330b806471b : FAILURE in 1h 47m 20s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/afe5c763dcc1404580f94a580895e29e : SUCCESS in 22m 22s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b4240d9241994246ae2152c9d0a42579 : SUCCESS in 1h 13m 54s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/680cc872359a498c9932a73de4cdb18d : SUCCESS in 1h 27m 04s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/74478fec753c4a3d92dcf21e8e7f735b : SUCCESS in 51m 57s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a30281d7a94240a880eed01ad94b3811 : SUCCESS in 1h 46m 39s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2334fee92ac2413a865810c44cc25cfb : SUCCESS in 1h 42m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/94ca34a051664051905b5361e0330a3e : SUCCESS in 51m 43s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":2},{"id":"9b029787c50a2e1fdd6de222a346b83cb2a075b0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-03 22:26:12.000000000","message":"Patch Set 2:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/03_08_2026_22_19_19\nLeave a comment with","accounts_in_message":[],"_revision_number":2},{"id":"61f1ff1fa40af46850235dcc39d4e0c83611f60f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-03 22:33:57.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/4cf5eff22fb646aeada27aba0bb8d4ac : SUCCESS in 43m 11s\n- cinder-storpool-tempest-iscsi-multipath build/519300af473f416893793649c014b414 : SUCCESS in 50m 32s","accounts_in_message":[],"_revision_number":2},{"id":"25af2ba96e2dd7f0d1689d0996e18eb273da135e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2026-09-15 09:25:56.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"24af333f9cc05c7c771af236c6ef0e5a83c90cae","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 10:13:55.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c91bfde40044c91b95407d16417ab62\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/cf51d113311e496489db49062c732f78 : SUCCESS in 18m 30s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/acfec45b62dd47c5ac15cbc8214591d3 : SUCCESS in 17m 41s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/7b8c6555cf61464380773e67edbdde8b : SUCCESS in 26m 45s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d99d1a2958cc2c47862fa355ebc6d3f1225c4646","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-15 10:14:17.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I9a2ee22fbed0dc0cbbe04214213d0363457d744f .","accounts_in_message":[],"_revision_number":3},{"id":"9b43c41aad1e0323c23a9968d48709fd6153df73","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-09-15 11:27:56.000000000","message":"Patch Set 3:\n\nHPE AlletraMP CI\n\nBuild failed.\n\n- iSCSI    : FAILURE in 0h 1m 10s\n- FC       : FAILURE in 0h 1m 19s\n- NVMe/TCP : FAILURE in 0h 1m 39s\n\nLogs: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/999463","accounts_in_message":[],"_revision_number":3},{"id":"14ab450303cc22c98e70b94c662f382374c987b1","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-15 11:39:45.000000000","message":"Patch Set 3:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/73465653940e47da811c8152af1f6503/","accounts_in_message":[],"_revision_number":3},{"id":"aba3bf40e57914f6d4fd3ff865da99d7072bbec5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 12:04:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd2b8d09631b4b0aadd14dd746879a77\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d162f8ccb73340189f827fc82a3a38c6 : SUCCESS in 6m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/19f97fe8608f44509a697cbfb7662b55 : SUCCESS in 11m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/232629a5d7d344f1bb38abe89da3d0e3 : SUCCESS in 12m 36s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6e56f0c0bd84418eb989f8dcedb78f29 : SUCCESS in 15m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a597465510eb4d668752ae6df3b9c53a : SUCCESS in 15m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/cdd788c5a6634462b9bc58b8efb57955 : SUCCESS in 1h 00m 37s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5124d66faf134d039277530f904f8eea : SUCCESS in 47m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/99768285583b4e51b940c5a5143fb188 : SUCCESS in 1h 35m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/11f115a0b4924124bbfa64e9a7c77e5c : SUCCESS in 30m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/310aa7943508404fbea5df202676e863 : SUCCESS in 6m 43s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/283ce1e67ee84680ab689ef7a516788e : TIMED_OUT in 41m 49s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9bf399eea5454e59a9c4ee5e9bfcdc48 : SUCCESS in 3m 28s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/139ebda8b522429c81b4a6e03c3dcfc1 : SUCCESS in 7m 03s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/c193c6d675334a0fa95adc24dc0976f3 : FAILURE in 12m 43s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/05de08e323d246f6b2c1d03c4642ade7 : SUCCESS in 11m 56s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/763db3ea3f1a4080a2ae9edecc4567af : SUCCESS in 6m 32s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/6bc81f567c2e4202aff3b704c4861b6c : SUCCESS in 6m 27s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/3a6f74e05c1c44cbb36d141d1ccbf235 : FAILURE in 5m 52s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/7aa7f7ad2f5440f0af5c8d0bf4cc8a65 : SUCCESS in 9m 06s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/77599385d71c4f848c78bd81f9574e8e : SUCCESS in 20m 15s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/107f80b659714e61b0da488d2a0c605c : FAILURE in 5m 57s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/86be22cf827d4b79bd26a65d484a9ff3 : SUCCESS in 48m 48s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b9590947d4124d8e8eb95f5fdeba9b2c : SUCCESS in 2h 07m 45s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/4d25a2ff18ec4f7683cf24ce6111f63c : SUCCESS in 1h 50m 45s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/1170500c58ba4f8fbf280db2a6a84092 : FAILURE in 2h 04m 52s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/d89ea603435643bbb83bcb4608db2130 : SUCCESS in 2h 25m 37s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/4e0b564e788846c48117f5030404d875 : SUCCESS in 38m 47s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b6d22935905745ab99dfd858483d5ca9 : SUCCESS in 1h 09m 03s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c677d7aea49944108d401bee5170941c : SUCCESS in 1h 37m 40s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/4f2bc221f583458e840650dd746b90fc : SUCCESS in 34m 27s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b5c9f51c86ea4643b3b60182be4ace99 : SUCCESS in 1h 56m 28s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/624c17eb37994555bfc82ef58a3d47eb : SUCCESS in 1h 36m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/de3db462cc6442329ae154a8c6ca07b6 : SUCCESS in 54m 35s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":3},{"id":"39dab242f52a7b59444f20a134387765f7b2d6c1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-09-15 12:13:41.000000000","message":"Patch Set 3:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/15_09_2026_12_08_08\nLeave a comment with","accounts_in_message":[],"_revision_number":3},{"id":"8c74cd1748ec5112504594b148979e47dbbdfede","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-09-15 12:14:49.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-63-999463-3-fc : NODE_FAILURE\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-63-999463-3-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"c5ab9597a1a8c1b3b1c16ff2ae666530c60bc84c","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-15 12:53:33.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/eef8a5ccb2f341d785511a08594ff337 : SUCCESS in 50m 21s\n- cinder-storpool-tempest-iscsi-multipath build/eaafa1343c8d4eb9b7143210d07d4e17 : SUCCESS in 48m 42s","accounts_in_message":[],"_revision_number":3},{"id":"cc682295010ffa6d47a52f1c668f54b9fa79afb4","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-15 13:02:39.000000000","message":"Patch Set 3:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/73465653940e47da811c8152af1f6503\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b9c66ba1172e41f99ccf685cf3f9b73c : SUCCESS in 51m 51s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0cd9a0cc41ca4103ac5697efd8dd67f6 : NODE_FAILURE Node request 200-0001284088 failed in 0s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/be6bb959752c4cea983cbf390e352db0 : SUCCESS in 48m 02s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/881edeb5bece457b85f77de84b18bcc1 : SUCCESS in 36m 02s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d194cf9efb9b442cb5e01390ae9dc244 : FAILURE in 15m 49s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"02c82df8a1aacd5f6fed0cece3e10c31a15d9f1e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-15 13:46:25.000000000","message":"Patch Set 3:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5549/.","accounts_in_message":[],"_revision_number":3},{"id":"0f249e1b60b54847f2605c09d783d595b958c919","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-15 15:19:26.000000000","message":"Patch Set 3:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5372/.","accounts_in_message":[],"_revision_number":3},{"id":"a95914c63f719602d693861583755b9a9bc52040","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-09-15 23:50:33.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/hitachi-vantara/hbsd-tempest-result-files/tree/main/refs-changes-63-999463-3-fc : SUCCESS in 50m 43s\n- hitachi_ci_iscsi https://github.com/hitachi-vantara/hbsd-tempest-result-files/tree/main/refs-changes-63-999463-3-iscsi : FAILURE","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"0a1e4581bf8821b040dc5e6dc8c53161c02218d5","revisions":{"c27066451487ed0236b1aa59f8db133f5c02517b":{"kind":"REWORK","_number":1,"created":"2026-07-31 21:44:14.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/63/999463/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/63/999463/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/999463/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/999463/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/999463/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/63/999463/1"}}},"commit":{"parents":[{"commit":"4664ecec712777769c8cf192c262c1184cd202eb","subject":"WIP: Add group replication support in RBD","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4664ecec712777769c8cf192c262c1184cd202eb"}]}],"author":{"name":"rajathere","email":"rajatdhasmana@gmail.com","date":"2026-07-31 21:43:12.000000000","tz":330},"committer":{"name":"rajathere","email":"rajatdhasmana@gmail.com","date":"2026-07-31 21:43:33.000000000","tz":330},"subject":"WIP: Group failover: Update host and AZ","message":"WIP: Group failover: Update host and AZ\n\nAssisted-By: Claude (Opus 4.6)\n\nChange-Id: I9a2ee22fbed0dc0cbbe04214213d0363457d744f\nSigned-off-by: rajathere \u003crajatdhasmana@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c27066451487ed0236b1aa59f8db133f5c02517b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c27066451487ed0236b1aa59f8db133f5c02517b"}]},"branch":"refs/heads/master"},"1257986934a75b2595ceadaa416794556519b347":{"kind":"REWORK","_number":2,"created":"2026-08-03 18:56:28.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/63/999463/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/63/999463/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/999463/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/999463/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/999463/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/63/999463/2"}}},"commit":{"parents":[{"commit":"710b0033be8ce96d7e3e4c7ac7c4fc12d5f03446","subject":"WIP: Add group replication support in RBD","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/710b0033be8ce96d7e3e4c7ac7c4fc12d5f03446"}]}],"author":{"name":"rajathere","email":"rajatdhasmana@gmail.com","date":"2026-07-31 21:43:12.000000000","tz":330},"committer":{"name":"rajathere","email":"rajatdhasmana@gmail.com","date":"2026-08-03 18:50:37.000000000","tz":330},"subject":"WIP: Group failover: Update host and AZ","message":"WIP: Group failover: Update host and AZ\n\nAssisted-By: Claude (Opus 4.6)\n\nChange-Id: I9a2ee22fbed0dc0cbbe04214213d0363457d744f\nSigned-off-by: rajathere \u003crajatdhasmana@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1257986934a75b2595ceadaa416794556519b347"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1257986934a75b2595ceadaa416794556519b347"}]},"branch":"refs/heads/master"},"0a1e4581bf8821b040dc5e6dc8c53161c02218d5":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-09-15 09:25:56.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/63/999463/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/63/999463/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/999463/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/999463/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/999463/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/63/999463/3"}}},"commit":{"parents":[{"commit":"469ea8179f8a4599e88cb25956fc35c7a66bf83f","subject":"WIP: Add group replication support in RBD","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/469ea8179f8a4599e88cb25956fc35c7a66bf83f"}]}],"author":{"name":"rajathere","email":"rajatdhasmana@gmail.com","date":"2026-07-31 21:43:12.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2026-09-15 09:24:16.000000000","tz":330},"subject":"WIP: Group failover: Update host and AZ","message":"WIP: Group failover: Update host and AZ\n\nAssisted-By: Claude (Opus 4.6)\n\nChange-Id: I9a2ee22fbed0dc0cbbe04214213d0363457d744f\nSigned-off-by: rajathere \u003crajatdhasmana@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0a1e4581bf8821b040dc5e6dc8c53161c02218d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0a1e4581bf8821b040dc5e6dc8c53161c02218d5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
