)]}'
{"id":"openstack%2Fcinder~999688","triplet_id":"openstack%2Fcinder~master~I2725b5b02d047e561b5168c64f357d9c17947744","project":"openstack/cinder","branch":"master","topic":"bug/2160669","attention_set":{},"removed_from_attention_set":{"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2026-08-12 09:05:22.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I2725b5b02d047e561b5168c64f357d9c17947744","subject":"Fix service enable/disable not updating database","status":"ABANDONED","created":"2026-08-04 06:45:42.000000000","updated":"2026-08-12 09:05:22.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"94062854610be7bcabf454dd5cb0978645939b1c","_number":999688,"virtual_id_number":999688,"owner":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":34260,"name":"Masayoshi Mizuma","display_name":"Masayoshi Mizuma","email":"m.mizuma@fujitsu.com","username":"mmizuma"},{"_account_id":36151,"name":"Cyprian Kleist","display_name":"Cyprian","email":"Cyp2505@wp.pl","username":"gerard25"},{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-08-04 09:51:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"all":[{"value":1,"date":"2026-08-12 01:03:59.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":34260,"name":"Masayoshi Mizuma","display_name":"Masayoshi Mizuma","email":"m.mizuma@fujitsu.com","username":"mmizuma"},{"value":1,"date":"2026-08-11 08:46:47.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":36151,"name":"Cyprian Kleist","display_name":"Cyprian","email":"Cyp2505@wp.pl","username":"gerard25"},{"value":2,"date":"2026-08-08 00:28:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":34260,"name":"Masayoshi Mizuma","display_name":"Masayoshi Mizuma","email":"m.mizuma@fujitsu.com","username":"mmizuma"},{"_account_id":36151,"name":"Cyprian Kleist","display_name":"Cyprian","email":"Cyp2505@wp.pl","username":"gerard25"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":34260,"name":"Masayoshi Mizuma","display_name":"Masayoshi Mizuma","email":"m.mizuma@fujitsu.com","username":"mmizuma"},{"_account_id":36151,"name":"Cyprian Kleist","display_name":"Cyprian","email":"Cyp2505@wp.pl","username":"gerard25"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34260,"name":"Masayoshi Mizuma","display_name":"Masayoshi Mizuma","email":"m.mizuma@fujitsu.com","username":"mmizuma"},{"_account_id":36151,"name":"Cyprian Kleist","display_name":"Cyprian","email":"Cyp2505@wp.pl","username":"gerard25"}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-04 06:55:45.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2026-08-04 07:22:53.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2026-08-04 07:36:32.000000000","updated_by":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"reviewer":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"state":"CC"},{"updated":"2026-08-04 07:37:29.000000000","updated_by":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"reviewer":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"state":"CC"},{"updated":"2026-08-04 07:37:51.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-08-04 07:38:03.000000000","updated_by":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"reviewer":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"state":"CC"},{"updated":"2026-08-04 07:41:14.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-08-04 07:48:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-08-04 09:51:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-04 09:56:42.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-08-08 00:28:03.000000000","updated_by":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"reviewer":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"state":"REVIEWER"},{"updated":"2026-08-11 08:46:47.000000000","updated_by":{"_account_id":36151,"name":"Cyprian Kleist","display_name":"Cyprian","email":"Cyp2505@wp.pl","username":"gerard25"},"reviewer":{"_account_id":36151,"name":"Cyprian Kleist","display_name":"Cyprian","email":"Cyp2505@wp.pl","username":"gerard25"},"state":"REVIEWER"},{"updated":"2026-08-12 01:03:59.000000000","updated_by":{"_account_id":34260,"name":"Masayoshi Mizuma","display_name":"Masayoshi Mizuma","email":"m.mizuma@fujitsu.com","username":"mmizuma"},"reviewer":{"_account_id":34260,"name":"Masayoshi Mizuma","display_name":"Masayoshi Mizuma","email":"m.mizuma@fujitsu.com","username":"mmizuma"},"state":"REVIEWER"}],"messages":[{"id":"6516387ff7ed21392aadb960a0204cb85fe5e3e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2026-08-04 06:45:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b3e76fb7e36f169c6e9e1166bd408c68f23571e2","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-04 06:55:45.000000000","message":"Patch Set 1:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4832/.","accounts_in_message":[],"_revision_number":1},{"id":"feb4150eee136a9f65547f6754fbbbb4eace359e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-04 06:59:55.000000000","message":"Patch Set 1:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5008/.","accounts_in_message":[],"_revision_number":1},{"id":"b4ed7e6f3148404a8fa41d0424b10e60e8d6288a","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-04 07:22:53.000000000","message":"Patch Set 1:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/dc44399698e14ced95a6094398c8930e/","accounts_in_message":[],"_revision_number":1},{"id":"c80a50e32271a4a73311be64b1ab01da17c99cc2","author":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"date":"2026-08-04 07:36:32.000000000","message":"Patch Set 1:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/704 : SUCCESS in 0h 50m 24s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/999688","accounts_in_message":[],"_revision_number":1},{"id":"a9fd4d143052be05ec805d581cab95e17172e581","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-08-04 07:37:29.000000000","message":"Patch Set 1:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1065 : SUCCESS in 0h 51m 24s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/999688","accounts_in_message":[],"_revision_number":1},{"id":"44ce4d09b06cf91e9507bf39f67ee1ee0cc001bd","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-04 07:37:51.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/f2ee89eeb9f14aeda0a6939a66ea0643 : SUCCESS in 45m 06s\n- cinder-storpool-tempest-iscsi-multipath build/62605a5876fb45658bb15989b8b8b42a : SUCCESS in 50m 43s","accounts_in_message":[],"_revision_number":1},{"id":"18be8aa32b01c07b9073422c6ef3ff4b70755b74","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-08-04 07:38:03.000000000","message":"Patch Set 1:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/797 : SUCCESS in 0h 51m 59s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/999688","accounts_in_message":[],"_revision_number":1},{"id":"26726f29482c4439f000af735068c71f47f2e689","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-04 07:41:14.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I2725b5b02d047e561b5168c64f357d9c17947744 .","accounts_in_message":[],"_revision_number":1},{"id":"99494ab8403016580e338df70cd10383b5b4aeb6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-04 07:48:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ee565f3e73c4591bbc7b7bc21ddfb33\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d74c594c2a804d20b1ad9694ac5be50a : TIMED_OUT in 1h 01m 16s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4309d421f3c2433ebe0c547472e631f6 : FAILURE in 15m 50s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/6efcdc0ccf9e446a929d97fed2bd8012 : SUCCESS in 21m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"65c2950757b18a4818225616447b44c80e3f39d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-04 09:51:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c6b061590614a568c363b57500b3a10\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43c8fdf759694012b67e54c3148396a3 : SUCCESS in 4m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e4b3ec021069457e83b8591d52f8d457 : SUCCESS in 11m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7bf50f780fe447cd8e0d7a7a3d39e251 : SUCCESS in 8m 37s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c39317398f4f4e01a0549c291ac58c8e : SUCCESS in 16m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35be5b380952477d93f62b36bfc948b7 : SUCCESS in 7m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/04824cf1df6d4a369f094d5f3303f8b1 : SUCCESS in 35m 13s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1c9738e9a95c4cb6af0558695cb9e388 : SUCCESS in 1h 01m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c5e6dc94f451425996858aee40d22dfb : SUCCESS in 1h 48m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/69d77d6416d540ea81fe4d88d4e1f9cb : SUCCESS in 54m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6d94eae2fc204f968e3af47f37ab4d86 : SUCCESS in 8m 57s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d8517508482a475db5077ec79f4cfb2b : SUCCESS in 27m 50s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/50420743a7c74dc58c8a78c5020ed871 : SUCCESS in 4m 51s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/dba4216f657d47219e9bc2a89f6505ff : SUCCESS in 6m 14s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/41edcc6888ef4167885b829a5bad9b56 : FAILURE in 19m 39s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/cac6624da04b4491b4b01f788b91e174 : SUCCESS in 8m 41s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/cb0ed9e21ec94d109f92c51050ed68d9 : SUCCESS in 5m 47s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/18e275b0cc174eac9ee813065b8b30b5 : SUCCESS in 9m 18s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/87d4954e6bd44d2f99c7394726647edd : SUCCESS in 29m 45s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6acea6ab29f740a3b47081345574bbec : SUCCESS in 5m 43s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/18f299247a6945e4b37d050a1165980c : SUCCESS in 1h 08m 34s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ed8f798dfd9c46a9949f774218c90b58 : FAILURE in 1h 57m 43s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b210c59db0fb4098a413430b050d3e7c : SUCCESS in 1h 57m 06s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/968039caf5664d1da025b9f01f1df054 : FAILURE in 2h 59m 28s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/2351f51b681d441fb7598b5739e0244d : FAILURE in 1h 24m 52s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/e9044cdeb5d346e3888647df411f5856 : SUCCESS in 26m 14s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b379c5cac3f94173af4b8fc4d7237e42 : SUCCESS in 1h 25m 16s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8e395bfe3a8f4a1aadb66bda65cde19a : SUCCESS in 1h 21m 30s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/dcfc796aa0d9456488ae2ad30cc1b50c : SUCCESS in 45m 53s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/cf94fc62d7f44e78a3b70e56d8a69bdc : SUCCESS in 1h 29m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/156e67f323324b9f8380b365471188fb : SUCCESS in 43m 44s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"cdc527c0a0678a49afff169ac30e6c206804e2b6","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-04 09:52:32.000000000","message":"Patch Set 1:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/dc44399698e14ced95a6094398c8930e\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e2864b8f905f4be0bdeb913d6b77b71e : SUCCESS in 50m 08s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/bc06f87319264503bacd01bca8869461 : SUCCESS in 36m 42s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/973249cfc503495987d5846bffc9da3b : SUCCESS in 45m 10s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f81c7d500709414aa3e5dce7eaef74ad : SUCCESS in 37m 10s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d5f640139d6648229f61014b0babad2a : FAILURE in 12m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bc6b04d6de7a94e8729c3ea8d34115e28163f208","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-04 10:06:49.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/04_08_2026_10_00_00\nLeave a comment with","accounts_in_message":[],"_revision_number":1},{"id":"8ca0e015093057dfebba78b8c870b88d978eab25","author":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"date":"2026-08-08 00:28:03.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7f6c64277700b3a5a479553c9b0b0068c89d5e7e","author":{"_account_id":36151,"name":"Cyprian Kleist","display_name":"Cyprian","email":"Cyp2505@wp.pl","username":"gerard25"},"date":"2026-08-11 08:46:47.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"2eb6b72ff4c5baa749f35a0c25a308fbee667851","author":{"_account_id":34260,"name":"Masayoshi Mizuma","display_name":"Masayoshi Mizuma","email":"m.mizuma@fujitsu.com","username":"mmizuma"},"date":"2026-08-12 01:03:59.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"94062854610be7bcabf454dd5cb0978645939b1c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2026-08-12 09:05:22.000000000","message":"Abandoned\n\nin favor of https://review.opendev.org/c/openstack/cinder/+/997333","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8866939d3e0674e02a5efda3ca427b78fb9ebe30","revisions":{"8866939d3e0674e02a5efda3ca427b78fb9ebe30":{"kind":"REWORK","_number":1,"created":"2026-08-04 06:45:42.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/88/999688/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/88/999688/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/999688/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/999688/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/999688/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/88/999688/1"}}},"commit":{"parents":[{"commit":"e3b9582a60a58d6ea630e845bbe5d96c8780fd8c","subject":"Merge \"mypy: annotate lvm\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e3b9582a60a58d6ea630e845bbe5d96c8780fd8c"}]}],"author":{"name":"rajathere","email":"rajatdhasmana@gmail.com","date":"2026-08-03 20:06:24.000000000","tz":330},"committer":{"name":"rajathere","email":"rajatdhasmana@gmail.com","date":"2026-08-04 06:39:37.000000000","tz":330},"subject":"Fix service enable/disable not updating database","message":"Fix service enable/disable not updating database\n\nThe enable and disable API endpoints return success but do not\nactually update the service\u0027s disabled flag in the database. This\nis a regression from commit 74b312d554 which split the monolithic\nupdate() method into separate enable() and disable() methods but\nomitted setting svc.disabled before calling svc.save().\n\nSet svc.disabled to the correct value in both methods. Also clear\ndisabled_reason when enabling a service.\n\nCloses-Bug: #2160669\nChange-Id: I2725b5b02d047e561b5168c64f357d9c17947744\nSigned-off-by: rajathere \u003crajatdhasmana@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8866939d3e0674e02a5efda3ca427b78fb9ebe30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8866939d3e0674e02a5efda3ca427b78fb9ebe30"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
