)]}'
{"id":"openstack%2Fcinder~999911","triplet_id":"openstack%2Fcinder~master~Id59d70da2326ce5a34e61b22c833903398ac58c4","project":"openstack/cinder","branch":"master","attention_set":{},"removed_from_attention_set":{"13425":{"account":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"last_update":"2026-09-11 18:15:38.000000000","reason":"Change was submitted"},"36171":{"account":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"last_update":"2026-09-11 18:15:38.000000000","reason":"Change was submitted"},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2026-09-10 17:28:02.000000000","reason":"\u003cGERRIT_ACCOUNT_4523\u003e replied on the change","reason_account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},"34860":{"account":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"last_update":"2026-09-11 13:00:49.000000000","reason":"\u003cGERRIT_ACCOUNT_34860\u003e replied on the change","reason_account":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"}},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2026-09-11 18:15:38.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id59d70da2326ce5a34e61b22c833903398ac58c4","subject":"Fix data-to-zero corruption restoring an incremental backup chain","status":"MERGED","created":"2026-08-05 22:33:36.000000000","updated":"2026-09-11 18:15:38.000000000","submitted":"2026-09-11 18:15:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":38,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"999911","meta_rev_id":"227848f782fc8911942432956c033053fb0e65b3","_number":999911,"virtual_id_number":999911,"owner":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-11 18:15:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2026-09-11 16:06:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2026-09-11 16:19:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-09-11 13:00:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2026-09-11 16:19:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2026-09-11 16:06:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2026-09-11 16:19:32.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},{"_account_id":39391,"name":"Rashi Jain","display_name":"HPE AlletraMP NVMeTCP CI","email":"rashi.jain@hpe.com","username":"rashij"}],"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-05 22:36:29.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-08-05 22:43:18.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2026-08-05 22:52:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-08-05 22:57:59.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2026-08-05 23:49:14.000000000","updated_by":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"reviewer":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"state":"CC"},{"updated":"2026-08-06 00:00:05.000000000","updated_by":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"reviewer":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"state":"CC"},{"updated":"2026-08-06 00:01:37.000000000","updated_by":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"reviewer":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"state":"CC"},{"updated":"2026-08-06 00:03:46.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-08-06 00:57:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-06 12:57:35.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2026-08-07 23:20:55.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-08-14 15:10:10.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"},{"updated":"2026-08-16 09:53:14.000000000","updated_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"reviewer":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"state":"REVIEWER"},{"updated":"2026-08-26 10:31:40.000000000","updated_by":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"reviewer":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"state":"REVIEWER"},{"updated":"2026-08-31 19:27:41.000000000","updated_by":{"_account_id":39391,"name":"Rashi Jain","display_name":"HPE AlletraMP NVMeTCP CI","email":"rashi.jain@hpe.com","username":"rashij"},"reviewer":{"_account_id":39391,"name":"Rashi Jain","display_name":"HPE AlletraMP NVMeTCP CI","email":"rashi.jain@hpe.com","username":"rashij"},"state":"CC"},{"updated":"2026-09-03 10:13:13.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"state":"CC"},{"updated":"2026-09-03 12:57:33.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2026-09-10 15:16:13.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-10 22:24:42.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"}],"messages":[{"id":"75453fbd2c27c36a628f5837b46259189c925129","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2026-08-05 22:33:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bc798b047373ff9f9eae3fcd86ca2f5d872e436a","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-05 22:36:29.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/Id59d70da2326ce5a34e61b22c833903398ac58c4 .","accounts_in_message":[],"_revision_number":1},{"id":"91efe40f0fa2bd57b601681b85735ec000387849","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-05 22:43:18.000000000","message":"Patch Set 1:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5044/.","accounts_in_message":[],"_revision_number":1},{"id":"8b94f107c590f9b10f73c018a84c73ffb4f32596","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-05 22:44:57.000000000","message":"Patch Set 1:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4868/.","accounts_in_message":[],"_revision_number":1},{"id":"9f72afdc119785f3844a206fc4fd46136afa110d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-05 22:52:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e11b03234dd484db573b70a82a3bf92\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/63b7a6ab3c1244f7b4224ac0eb2169c0 : SUCCESS in 13m 02s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9588fe0a44284bbba64886c4f3620d0f : SUCCESS in 12m 21s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/f1086e89e0f04680bda67fd77a7e31a1 : SUCCESS in 16m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ebcf498d719a3078f49dd0f6f85bae8cf6c81bc2","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-05 22:57:59.000000000","message":"Patch Set 1:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/2987b9d7623a4244904828c366e0c453/","accounts_in_message":[],"_revision_number":1},{"id":"c8411679cfb3fdd21487b98076438f4e1413f8ee","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-05 23:33:56.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/2987b9d7623a4244904828c366e0c453\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b1495dae87704a758b194792c6e325d8 : SUCCESS in 44m 47s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3592286ce1034a2382f466bd6517f03f : SUCCESS in 44m 19s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/fb6966908c3747518c890477312d58f0 : SUCCESS in 56m 30s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/34ea7339c77040eaae581e8901041823 : FAILURE in 7m 13s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/692053b8231546cd80a434d9aee39d43 : FAILURE in 10m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"335794bddcf28cb5e02795d35de4c4ab1e9b852f","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-08-05 23:49:14.000000000","message":"Patch Set 1:\n\nHPE AlletraMP iSCSI CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1102 : FAILURE in 0h 35m 34s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/999911","accounts_in_message":[],"_revision_number":1},{"id":"de88537fc5c3ec8345f368ccb65ac0e0538cf187","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-08-06 00:00:05.000000000","message":"Patch Set 1:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/835 : SUCCESS in 0h 47m 36s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/999911","accounts_in_message":[],"_revision_number":1},{"id":"ee10dbcd666a0c9e92eab702f04042ad1b854662","author":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"date":"2026-08-06 00:01:37.000000000","message":"Patch Set 1:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/741 : SUCCESS in 0h 48m 39s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/999911","accounts_in_message":[],"_revision_number":1},{"id":"ae456cd943ff8d72ae977bc9b32645763c2ce72f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-06 00:03:46.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/de0b8da1d0a34eefbd6e5cb10983dbd3 : SUCCESS in 46m 10s\n- cinder-storpool-tempest-iscsi-multipath build/6b44433726d54039a107e14425f39649 : SUCCESS in 53m 59s","accounts_in_message":[],"_revision_number":1},{"id":"a4b06585ea9e90b09300c7466523766c62a9faac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-06 00:57:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0bbbc38cd3ed4359bb567a3b62201727\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6317ecee8b7541d5b9ccab5ffcb80e61 : SUCCESS in 5m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4e0d3adc84d8402abf775cfff37735d7 : SUCCESS in 8m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ca0a549cf1574243b7e218186cb92dc1 : SUCCESS in 9m 51s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6f0b67fff454411eb623772b618af683 : SUCCESS in 13m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/412a76c7ac4a477e8489559b5f82d74d : SUCCESS in 6m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/56d8c18d02be42eda58d915158c52536 : SUCCESS in 1h 05m 49s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5abf10bad82e4ec38abe99c01c652008 : FAILURE in 28m 45s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a607973c3efc44758177ec8e34701bc4 : SUCCESS in 1h 23m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/212b93d09f6845c087e780e5c73e006e : SUCCESS in 56m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eb41548e916047a398d882ad94a06129 : SUCCESS in 8m 45s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/34431d8913e14ad7871fdbaefb80c787 : SUCCESS in 17m 01s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/670745a5f3af4347a1a125ff61d1137c : SUCCESS in 4m 43s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/bb0ee9f5c91c42eeac171b7d743e564b : SUCCESS in 3m 37s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/2bc1f283ceed4a96bbd83f547d1cf203 : SUCCESS in 9m 05s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/569e92b4b27741f6a4c6d732bbd60d77 : SUCCESS in 5m 56s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/2931bba0711841eda127cc0590da6680 : SUCCESS in 5m 45s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/f202e9eb5f4f4061815c86015df3a105 : SUCCESS in 5m 53s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2cdee43dcd5c41c7b868fad63f74a7fb : SUCCESS in 26m 30s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/210fc344114d4ddbaa3b4287a30ee15f : SUCCESS in 5m 34s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c873c8d6abb343858f1e5c9e05006964 : SUCCESS in 1h 38m 52s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f41999a1c61c48c0ba548f525de9356a : FAILURE in 1h 58m 54s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6b231a636a27447d89d1e5ff531c1de2 : SUCCESS in 2h 08m 34s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/2ba9a112389740098faca2dd57967dc2 : FAILURE in 2h 21m 24s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/17a934d0cf604c3f98e220b937c43a5c : FAILURE in 1h 57m 36s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/9706999994ba43df87578bfe0f5e5761 : SUCCESS in 16m 52s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/01f3ce93edb24d4890318b961b565bd3 : SUCCESS in 1h 15m 46s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/7067f6204778403389410ab7a9efb547 : SUCCESS in 1h 10m 17s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/e1803b505be5439c8a5ff89af57f39be : SUCCESS in 45m 49s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2229548ac612461aa7976c26b028943c : SUCCESS in 1h 16m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/81eff2e3457e4947a7e5aede66620a2b : SUCCESS in 36m 48s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"c6859503671372ba2c126ef2d4a5bed4ca80155b","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2026-08-06 03:38:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3f4270c31280cfb836906afe3ea968f46c40717e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-06 04:28:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/962358c4aec549818c5416ffc7f4b631\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b478c3549f454cf3a957e85ec85d4676 : SUCCESS in 18m 14s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/41b548972b2f4c3cadc499eacaf50d5c : SUCCESS in 17m 41s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/554a7ff5dcaa41208f93231e986f3ca9 : SUCCESS in 22m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f87c42c0c68f874ef4324924b2c9488a3749cedf","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-06 04:49:30.000000000","message":"Patch Set 1:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/0985deeba32e4ba5805b09149e599dd9/","accounts_in_message":[],"_revision_number":1},{"id":"e26214ffb2d1ecfa44be36ec19ceba4d99bfe19f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-06 05:47:55.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/b8f692eef02b4163a1fb1a7ccdc7d035 : SUCCESS in 58m 44s\n- cinder-storpool-tempest-iscsi-multipath build/d3563965a2fc41fb87b4c34fb702bb68 : SUCCESS in 1h 12m 13s","accounts_in_message":[],"_revision_number":1},{"id":"af61f9a14e98323444082a9f2eb0687530cbfbe9","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-06 06:02:37.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/0985deeba32e4ba5805b09149e599dd9\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/53d02d033b7949c2bf128bd023a897e7 : SUCCESS in 41m 10s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/276d853ea9fa4194be7774f807d1ac2c : NODE_FAILURE Node request 200-0000689043 failed in 0s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b9c8b34c9fa243da8e0dd9cc1f6d6f58 : SUCCESS in 51m 45s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ad85fcb7a4ba4729b34587f421b69f26 : SUCCESS in 34m 51s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6e204933e85147aeb4c51528e068cf11 : FAILURE in 14m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a8d7977a65824f8ddd17b33ab1230f3fba33c6e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-06 06:47:51.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c8f305b906494fc980d84df0bd68d221\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b8d012c4f9e4b22b316bf1ad08c6cb8 : SUCCESS in 6m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9bfb26ca76b949b48e65e80bdb702476 : SUCCESS in 9m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/23d586bc91364b0285c2f0af5ebacec2 : SUCCESS in 8m 52s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d7d6c09746ca4d729892f69da5fdf9d9 : SUCCESS in 15m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd67e391bad942abb03608f2fa4ec5ad : SUCCESS in 14m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/0b152c67c0764f898074b49d503798dc : SUCCESS in 1h 05m 39s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/47c95a7ac4ef4aaa97a3fa3df7a79686 : FAILURE in 28m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d4774c4cf1234501ab70c2a4e914a5d0 : SUCCESS in 1h 30m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f2ac8e7a75f2477ca3d1c2ea9bfaf8bb : FAILURE in 13m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f4b46bff75a44c83bc1ffe4af36ef113 : SUCCESS in 5m 01s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/8344565e023b4e029cb537a49668e6f1 : SUCCESS in 25m 08s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/1b6bcb8b9fe24031ad5c8d91f1abcd25 : SUCCESS in 3m 07s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ee63b151a73a4451abebb8ac4f06988d : SUCCESS in 5m 48s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/63f3a9157012419f9bc91be07f689783 : FAILURE in 12m 27s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/c31c7989902a4179b616938119085ea4 : SUCCESS in 6m 10s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/01cbe65eef3e49f6ac83cf5f16943bbb : SUCCESS in 5m 54s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/775f151780b74c4897a28dbf2ceabb80 : SUCCESS in 10m 13s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/03d18656d40e41f9a08b7dec0c91a08b : SUCCESS in 37m 06s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/dc563e7cb4b2446bb4350ced27632261 : SUCCESS in 6m 42s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c5c488331423416d91312fb11a1f9a1f : SUCCESS in 1h 44m 07s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/01ef9c3684034a3d850347207ad2cba7 : FAILURE in 1h 04m 52s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a7fe07b2d304400cab26df24258c4697 : FAILURE in 19m 23s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/a510eb07888c46398de49cc7328444bb : TIMED_OUT in 3h 07m 00s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/02ec94553a1946a0862d0f2a59072841 : FAILURE in 30m 22s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/f14cc12cba9949e99b106247fce6f7f7 : SUCCESS in 19m 33s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/05262f563ff94821ac2dcb23b979f482 : SUCCESS in 1h 12m 25s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d5eb31215b004169b69a278de23f08b2 : SUCCESS in 1h 53m 57s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/fda52f8d7517460daf975c19826d81db : FAILURE in 25m 38s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3ee7551e721847599e05e36c33f46737 : SUCCESS in 1h 40m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5c41654b3e9c449190497570af22df5e : FAILURE in 14m 10s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"1b6565b2a5230bf164764ce95a3fcddc2ebc6020","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-08-06 12:57:35.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5a3447f42c236595a73214b4e4dba206073fc794","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2026-08-06 13:53:59.000000000","message":"Uploaded patch set 2: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"287ce3d988c414d8049b4d0d449000535ec3da11","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-06 13:57:24.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/Id59d70da2326ce5a34e61b22c833903398ac58c4 .","accounts_in_message":[],"_revision_number":2},{"id":"6c3c85e87db9194345ffa569b9ad0fb7333574a9","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-06 14:06:44.000000000","message":"Patch Set 2:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5052/.","accounts_in_message":[],"_revision_number":2},{"id":"4d95c660b795e25de5c205aa59710767f22aa50d","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-06 14:07:50.000000000","message":"Patch Set 2:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4876/.","accounts_in_message":[],"_revision_number":2},{"id":"7304bb79c44c120bcf5ccf15bfc275dcb36ff839","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-06 14:20:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e71fa8e04f2c47f2b050cbd19c27c817\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ab8e0121bf034a1f8372b16522336f00 : SUCCESS in 13m 35s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b2f1c6384d904d2f96922d4df1ef7dfd : SUCCESS in 14m 50s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b555a47247e54074a7f742ecce44634a : SUCCESS in 18m 13s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ecac30173e7ca031517da93fd0585c0ddcdb1c9e","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-08-06 14:45:40.000000000","message":"Patch Set 2:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1109 : SUCCESS in 0h 50m 55s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/999911","accounts_in_message":[],"_revision_number":2},{"id":"a1c76be9209a750d64e717f19dd45e196fab9cb9","author":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"date":"2026-08-06 14:46:05.000000000","message":"Patch Set 2:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/748 : SUCCESS in 0h 51m 18s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/999911","accounts_in_message":[],"_revision_number":2},{"id":"a820e7280ec9a07e4fc495d1cca9aa1fa4f44f9b","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-06 14:56:46.000000000","message":"Patch Set 2:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/4c38807c92134a99872340e6e495cca3/","accounts_in_message":[],"_revision_number":2},{"id":"4490a022e917a70526e5172c2a61a68cd6fb87a6","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-08-06 15:39:07.000000000","message":"Patch Set 2:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/842 : SUCCESS in 0h 47m 49s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/999911","accounts_in_message":[],"_revision_number":2},{"id":"77acdb602412884b0bd43a256ddd104197d2876f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-06 16:36:34.000000000","message":"Patch Set 2:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/4c38807c92134a99872340e6e495cca3\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0ce9df6524c244d797381e7b6542a54c : SUCCESS in 48m 18s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a368779220074799acf147ea7ec0f170 : SUCCESS in 46m 19s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c4b7532ebd5446a19806e8ec75afe6cf : SUCCESS in 51m 09s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d0d619b4895d46149dd2b27d84472e26 : SUCCESS in 38m 34s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/80e71172aa5c4179a0c928f95b5379e9 : FAILURE in 18m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"33645ddee34f655ac96ef84049f5a7646c2a3050","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-06 17:26:41.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4d0de217449142a7ab1bbdd30a1dceaa\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42c4f451c37148dcab59cdc267574541 : FAILURE in 6m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cebdfea089c646bfbd120d5136f75bb3 : SUCCESS in 9m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f9e1756c83ba48e1ab9079f2accc7937 : SUCCESS in 10m 04s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/40d0f09a0e2a406fb5a942ba6cc3a345 : SUCCESS in 11m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac80e86b608e406e96feb9507e8a8d8c : SUCCESS in 11m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/d841e7940c4b45e2a569b246ed622552 : SUCCESS in 59m 34s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fa8aeebc88bd46e699682f589f3c1393 : SUCCESS in 33m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4f9b9850b72d4d6297944ecf513a40f4 : SUCCESS in 1h 27m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/19ccf129504f4b56bf8b3cdfe988aa48 : FAILURE in 30m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d2dc6d6f162d4991a3894dcd12edfeb1 : FAILURE in 3m 41s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/98184f7e77ba446b8f9b26dd859c892e : SUCCESS in 19m 16s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/6df8070fda3b456f9028f56e3180c289 : SUCCESS in 4m 52s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/80a779fa147143e28d7200a30ddbbe77 : SUCCESS in 4m 12s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/55e39f6872454aa08a069005cd4c6dbe : SUCCESS in 10m 31s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/730005559212435f830338f704c98bb2 : SUCCESS in 6m 28s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/59d0b333e6824d6aaa7ee3bc5b055321 : SUCCESS in 7m 19s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/86a718258da6458cbd0647ec205d02b8 : SUCCESS in 7m 04s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/28da883d4ab84b57ac6691cb1ca08cb7 : SUCCESS in 24m 00s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/74cfb0aab59d44cfa70a1ca8e67d8d6f : SUCCESS in 4m 08s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/fd48b1948ca543e2b78bf5329674b669 : SUCCESS in 1h 29m 35s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f530f7c7817b4292861b9609f1d87c30 : FAILURE in 32m 17s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a97bed0f605045a1a73d1e1f84d475cb : SUCCESS in 1h 42m 16s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/7fda853138b94049b96b1038234e017c : POST_FAILURE in 3h 22m 04s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/9289e6cb3194490aa685bb86f49fae67 : SUCCESS in 2h 04m 04s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/b54d46876b5b4e328559112a7f718a59 : FAILURE in 25m 42s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/067bdbc64c1e48ddbde1e5dc53d69dff : SUCCESS in 1h 12m 27s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/adc7acdc6e5e458d8d7daba2d564220c : FAILURE in 29m 22s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/3a8f8110359c4617a5edca04228959ef : SUCCESS in 39m 18s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/67112a7adb6447db86c37edb7b91e7ed : FAILURE in 1h 47m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ec7ba8c7955241cfb29892b0f2ed03c8 : SUCCESS in 32m 05s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":2},{"id":"d6225e20b3f34c9c0355e046c74c00b476401543","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-06 17:57:23.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/e121fa5d5c7f4e4f8669c7af1392af83 : SUCCESS in 1h 02m 57s\n- cinder-storpool-tempest-iscsi-multipath build/405bf986f91f40df851424912956d502 : SUCCESS in 50m 09s","accounts_in_message":[],"_revision_number":2},{"id":"35a7d8f515d05f0a89f487ea652e09faff5e21fb","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2026-08-07 20:26:29.000000000","message":"Patch Set 2:\n\nrecheck transient pep8 failure, pre-commit could not clone openstack/hacking due to an opendev git-hosting blip, no files linted and no code change needed","accounts_in_message":[],"_revision_number":2},{"id":"b533a3ac944284695201367c2e3dbf384f7b3a91","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 20:57:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4fa7581b710c4abcb894ddb8149f8c90\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f324d61a22024b25bbfef8c577d6d376 : SUCCESS in 16m 01s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5f88a84daadb4f15a18e96467e9bbe1c : SUCCESS in 19m 26s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/78eedd41d4fd46d789153a76fe869e8f : SUCCESS in 28m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f3af52c44cfc3e7aaf414aecd64a01ba3850cf09","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-07 21:19:53.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest build/599ac5cbc0964aa0a847353ccb696ff2 : SUCCESS in 46m 19s\n- cinder-storpool-tempest-iscsi-multipath build/495e589ca1604c76b84225b7a3dda7cb : FAILURE in 51m 55s","accounts_in_message":[],"_revision_number":2},{"id":"5c40af9f9a2eb0d020f49f9087c5c1065d84923c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 23:20:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/67365b70e8f44c7cadbf12e74efcafb4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d39cdcf7180d4b1bad34392b04e6f598 : SUCCESS in 5m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/74ee83f77aff4c6fa8b11db07a2cb13c : SUCCESS in 10m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7262d7ce75ac477d986f37dc9cbb7645 : SUCCESS in 12m 11s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d523f3e825ce4fbfb9a1b8e3ebc06dbf : SUCCESS in 14m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d03e458e1fa4f41883b9583fbd9a232 : SUCCESS in 13m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/cd61d1fc40214dad91bdb0d78b928b23 : SUCCESS in 32m 23s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/230257d1fa3b40179ffc70c83389fc06 : SUCCESS in 46m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/cff89e12ee8a4af480674a904c08f109 : SUCCESS in 58m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/946a200ee7124121bbb7a68985780a36 : SUCCESS in 1h 00m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fc3f8d57e0444b5e8eb85abf766f483b : SUCCESS in 10m 14s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7385a79e15744faea0f3be790dbdd5b9 : SUCCESS in 16m 52s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/fc2a3de083c34bc796afdaf5d289ba41 : SUCCESS in 4m 30s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/bf642def2f7b4da5a351762c6fd49e3f : SUCCESS in 6m 46s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/e67b7e06eb664832a18d301d48bec7bb : SUCCESS in 11m 09s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/7caa35fc90fc4079a52cda44fb4d83ad : SUCCESS in 6m 34s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/671d028d683b4b56a3e94a90af889954 : SUCCESS in 7m 46s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/0551f5b687904ac9a5eb0a382683b50c : SUCCESS in 7m 21s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/eefd9e530c5746498e765fea5878508f : SUCCESS in 24m 48s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d25fbc820ee64786826104f712001ae6 : SUCCESS in 4m 41s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/839b1ce70666433994805d0ffe38bfb0 : SUCCESS in 1h 30m 23s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e15e5128f0c3467181652c54a6edfa2c : FAILURE in 2h 01m 16s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/28247192697345d6813030b830c77176 : SUCCESS in 1h 23m 41s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/3f0a512a8b8d42ee9d159a744b74406f : FAILURE in 2h 48m 41s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/316eeeee7c3d4862b304d5b963772166 : SUCCESS in 1h 17m 51s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/f32ae4a3195c48b082b26063cc990188 : SUCCESS in 18m 09s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3aef8be25cf0414a8f7a2b2f799572c5 : SUCCESS in 1h 13m 16s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e0788e7f22f4419a82a68ca8a5b45469 : SUCCESS in 1h 27m 45s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/9c7f516b84944233b3832c146a84546d : SUCCESS in 29m 47s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a325a56247b9413c951d494eb6757556 : SUCCESS in 1h 07m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/714c14fc8d1e46bdade30f575081f9a0 : SUCCESS in 57m 22s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":2},{"id":"224a63d2a1796d475fe7e566c56f7c875a5cffaa","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-07 23:31:18.000000000","message":"Patch Set 2:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/07_08_2026_23_24_13\nLeave a comment with","accounts_in_message":[],"_revision_number":2},{"id":"38be935a5904cbf4283035bac8abdae5a2117ba8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2026-08-13 14:12:57.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"919eff9d61dcba06574c546641206a427fcffe25","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 14:38:18.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4dca26537a6430bae13c75f6ae28cf5\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9e97f2de388c48028ad0cf91ceb1fdce : SUCCESS in 19m 43s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2f9f9009efa741e194f5c7d118ef2877 : SUCCESS in 15m 51s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/6f18839137084d63838ad279221bc74c : SUCCESS in 22m 48s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"926a55f576a9c86739f3fa3042288d9d8f0f9a38","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-13 15:02:55.000000000","message":"Patch Set 3:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/9e519c1bfc5f45a6afe36f10e2eda282/","accounts_in_message":[],"_revision_number":3},{"id":"b1745931ddf1d08a221ea7208de456e58408fdda","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-13 15:12:49.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/Id59d70da2326ce5a34e61b22c833903398ac58c4 .","accounts_in_message":[],"_revision_number":3},{"id":"17e818299ed7860a174564bebe2c9a8decd353aa","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-08-13 16:37:27.000000000","message":"Patch Set 3:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/899 : SUCCESS in 0h 47m 56s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/999911","accounts_in_message":[],"_revision_number":3},{"id":"331583d485e90460401ef1a738b081b891e16ddf","author":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"date":"2026-08-13 16:39:43.000000000","message":"Patch Set 3:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/805 : SUCCESS in 0h 47m 58s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/999911","accounts_in_message":[],"_revision_number":3},{"id":"19938bcc8e5ddeb553dcc6df2faa1e155264bf57","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-08-13 16:39:48.000000000","message":"Patch Set 3:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1166 : SUCCESS in 0h 48m 29s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/999911","accounts_in_message":[],"_revision_number":3},{"id":"b69ed95c0feccaab944e42d17478d38d416ca45a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 16:53:22.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d9b0323e67014c3f94668186f260bced\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6c9a060345b46d898d3b364b7eec337 : SUCCESS in 7m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ce93d0eb8037429b90f8f9e47faf67c2 : SUCCESS in 7m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/baf08acce91e4c7fb6378415d6ec0093 : SUCCESS in 10m 31s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1bb1151be1444043b9aabd14fc5686ab : SUCCESS in 16m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a5f5a30158c4fc185f2005fa8370183 : SUCCESS in 11m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/54db2e2fae9740ce9c9335fe2ab3ded1 : SUCCESS in 1h 16m 34s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f1449699b67c47a681b972529cb5fb3a : SUCCESS in 42m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/381cd5b32cf246dab197794462df29f1 : SUCCESS in 53m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0a7c1f41cfdc49a7ad15e04218f5be18 : SUCCESS in 53m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b109172d8d50438697227cf2d6078087 : SUCCESS in 7m 29s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4564cea8a79d48e4af38093c424ea57d : TIMED_OUT in 41m 07s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/dd751ded8bae48a8a8043d53323ddea0 : SUCCESS in 3m 23s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4038e3453a2f410e9601d56d3e1a6964 : SUCCESS in 5m 59s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/23701ceb16d94aba899d22bea060dcae : SUCCESS in 13m 09s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/60421085c9ab4812814724058471705e : SUCCESS in 8m 13s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/dda866c7208c48918c71813829d7ea75 : SUCCESS in 7m 36s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/98786d08432c4be289bf111a6c560f54 : SUCCESS in 9m 18s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/cfe8938c9419479a8a516bac004d2449 : FAILURE in 34m 19s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ab07f46a2f104bf7b2872f8a8e050ad4 : SUCCESS in 7m 39s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/960a4192ddde43ae890d5c02a169051f : SUCCESS in 1h 09m 38s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0d28e57796f44b018484b24cc3a1ea51 : FAILURE in 2h 06m 39s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/205d008d718e4ac1b3c0c5d847b87580 : SUCCESS in 2h 06m 35s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/d89daa3eee424a12abffbfa97468314b : FAILURE in 2h 24m 00s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/2ee243d2c91b421e82bebaa90a9aa5ef : SUCCESS in 1h 45m 29s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/e0ff958e154a40d9bf705c7c19dcba99 : SUCCESS in 29m 16s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/965110bb26f84dae81a89a33a79e4c67 : SUCCESS in 1h 19m 00s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c8b7e2d44d0540b295cb5cf2455c9ef7 : SUCCESS in 1h 33m 21s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/9ac118cddf0f482c921748974765d279 : SUCCESS in 48m 55s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/28ff289b5ae248d8a90aeec0a3d10be6 : SUCCESS in 58m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d8d2b980a20d40c686f3b35534f14bdf : SUCCESS in 1h 08m 49s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":3},{"id":"7d1faa6b1be8cc04427cacb75bf8618abc530b41","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-13 17:20:02.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/67fe4c7d4fa448c997fe1d742f24fb04 : SUCCESS in 44m 26s\n- cinder-storpool-tempest-iscsi-multipath build/ec4896a2ecc641d59c68ff09bc6f9725 : SUCCESS in 48m 19s","accounts_in_message":[],"_revision_number":3},{"id":"975fad0d6b0bbca520e4e26e7fd48e1b2df6100b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-13 17:52:21.000000000","message":"Patch Set 3:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4936/.","accounts_in_message":[],"_revision_number":3},{"id":"956484860aacdb654fdc385dc42ce7e901a0efa3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-13 17:53:16.000000000","message":"Patch Set 3:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/13_08_2026_17_46_47\nLeave a comment with","accounts_in_message":[],"_revision_number":3},{"id":"cd70a8a76a2d303d748fe893d7f5afc2f2bdb01a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-13 18:07:41.000000000","message":"Patch Set 3:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5112/.","accounts_in_message":[],"_revision_number":3},{"id":"25653c34e1384c7ad52fceaada56fa9098ca4f20","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-13 21:53:43.000000000","message":"Patch Set 3:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/9e519c1bfc5f45a6afe36f10e2eda282\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c0a7e68baa274be288738f117b85b1ec : SUCCESS in 38m 51s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1abe6b5acd2341948ba46d662f61adfd : SUCCESS in 35m 51s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9d867ebc560a48b098b915030a8fc18a : SUCCESS in 48m 11s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/26a849c9517449368a44cde6b696bcd6 : SUCCESS in 30m 23s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8dd4a6bfbf5042d0bd0f36d8fca82824 : FAILURE in 15m 07s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d1415b67f5446bc06ad7a416701f95769171be20","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-08-14 15:10:10.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-11-999911-3-fc : FAILURE in 12m 30s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-11-999911-3-iscsi : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"1e6628d1e59f63153ae227431297b5c7239df8f1","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-08-16 09:53:14.000000000","message":"Patch Set 3: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"1dd96420e23e0b6a5adc4e662d7a0f27f149ad62","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2026-08-16 15:22:53.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"10c207c2085295f47b782c8a7b00944790199de4","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2026-08-16 15:23:16.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"90d07cbe163dba3f350917c2f76963117fd0fcde","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2026-08-16 15:26:39.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"92bb3b20a0abf6173020239eb00103e8e17d3d4e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-16 15:52:46.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc3fd49f03794c1b92d4c300abf96692\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/710700b4645c4cea8ab5b4ee86ad2d75 : SUCCESS in 22m 58s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f61754e6fc474526821d7c600dd72f5b : SUCCESS in 12m 10s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/2d197788f61d460db598fa3ea54842a7 : SUCCESS in 24m 02s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"04da26e2205248476169599f94d20ab3b1b159b6","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-16 16:02:26.000000000","message":"Patch Set 5:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/e7ac6b3039284d66810ef9ca98d64a34/","accounts_in_message":[],"_revision_number":5},{"id":"a8b9ce5c6b59bff3afb13033b94c17cb49ab18bd","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-16 16:13:46.000000000","message":"Patch Set 5:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/e7ac6b3039284d66810ef9ca98d64a34\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5c2c3ba4b5ff4139b015a275a81f1939 : SUCCESS in 42m 03s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/127bb2b58fbd4620b57cb0f6eac5d6ce : SUCCESS in 33m 33s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9e893bf70edc4f4eaf34547938d4c019 : SUCCESS in 46m 12s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6d4eeb494d3549b99d244e690370ccbb : SUCCESS in 34m 54s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f1098d8952ea4b94927311e322a1486d : FAILURE in 15m 22s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"1119fae5d06ddd3b00eba1e1bbbb4922f0d3d258","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-08-16 16:22:59.000000000","message":"Patch Set 5:\n\nHPE AlletraMP FC CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_FC_CI build/921 : FAILURE in 0h 22m 2s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/999911","accounts_in_message":[],"_revision_number":5},{"id":"797ffe6bd0168207025ef006a4abbbe7a018d13e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-16 16:23:29.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/7cb7d31cf0174ebbaab5c0ad2c411b44 : SUCCESS in 47m 58s\n- cinder-storpool-tempest-iscsi-multipath build/4575823704294bbd8753d0deac8c8c17 : SUCCESS in 55m 52s","accounts_in_message":[],"_revision_number":5},{"id":"a1a25fb4b07769502b11dc81a70a98269e05fa32","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-16 16:26:46.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Id59d70da2326ce5a34e61b22c833903398ac58c4 .","accounts_in_message":[],"_revision_number":4},{"id":"7704f2e181cad6cb1ebf5e677f18776ecb93d15c","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-16 16:29:44.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Id59d70da2326ce5a34e61b22c833903398ac58c4 .","accounts_in_message":[],"_revision_number":5},{"id":"f4c3814bdc775b82b32f6ccbd8a6f08e599f5dda","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-16 17:10:20.000000000","message":"Patch Set 4:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4953/.","accounts_in_message":[],"_revision_number":4},{"id":"029ff069aae3610f65a64b49ffcf66a62b0947b1","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-16 17:15:12.000000000","message":"Patch Set 4:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5129/.","accounts_in_message":[],"_revision_number":4},{"id":"d5ad4e9380171ce45ff50d0d644e0057244f074c","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-08-16 17:27:34.000000000","message":"Patch Set 5:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1188 : SUCCESS in 1h 27m 13s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/999911","accounts_in_message":[],"_revision_number":5},{"id":"23d0c4fe71fbeb5c6fefb8047d7bcdfeb30c8829","author":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"date":"2026-08-16 17:29:54.000000000","message":"Patch Set 5:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/827 : SUCCESS in 1h 30m 15s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/999911","accounts_in_message":[],"_revision_number":5},{"id":"5da78d416171098866654ddf78db74ae16088e15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-16 17:58:45.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f07fa36fd28a402994505ed632ff558b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ceef4976ff346ed81d07826d46604cf : SUCCESS in 6m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/68291fa349fb45a7abe13f3360804f5e : SUCCESS in 8m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/91acbfda8c004ec383dff07a8fc64c91 : SUCCESS in 9m 39s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/601e306d7cca42f0adfcc2495a5e6237 : SUCCESS in 11m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6621d3cdf194f93a5e6d8aa4703364c : SUCCESS in 15m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/6e53c529ee6646b4825a443cc5dc38c6 : SUCCESS in 1h 03m 31s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/993e835da8f54827843200ad249015ee : SUCCESS in 1h 04m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/df6ad0c2866243218e343150d0f6bb21 : SUCCESS in 1h 36m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5f9a28a680414dd6b684224992bf52f4 : SUCCESS in 1h 08m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/246ca5d5018549b8ae6091dbec5d15aa : SUCCESS in 8m 52s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/69ebea6612c1491ab5bef0eb3ad81a59 : TIMED_OUT in 41m 03s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/892ec702104442a3939764f71ea2f155 : SUCCESS in 4m 22s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/90c22b6557ac4c93bd64df9be4fc92ae : SUCCESS in 6m 26s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/801ccd08f53148faa0243b8c6db95115 : SUCCESS in 12m 17s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/7d2c57fb1c5b478f8be841a18178fa07 : SUCCESS in 6m 32s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/0dd811f9d6df46c286063edb6c3db7a7 : SUCCESS in 5m 50s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/5d6323230589411fbd0d15286db54cfe : SUCCESS in 8m 49s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/546af8928548403984b4f889675645ff : SUCCESS in 36m 40s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a5fbb85b91dd4316a1de336f08789e2d : SUCCESS in 6m 42s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/82b2edb145ac4747b419e8c5c08db3cf : SUCCESS in 1h 39m 02s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/9aa544eb67bd40e2a765fecde6b06b1f : SUCCESS in 2h 25m 39s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5a1dc12a3e2c415390a66ddbf15d086d : SUCCESS in 1h 49m 17s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/1c4e6a4d7a604316ad9ac3d5fe43e20c : FAILURE in 2h 12m 35s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/65a5d5094bbf46f8bba37655e91111f0 : SUCCESS in 1h 43m 05s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/a3f80539a794401b951cc6c747704d6c : SUCCESS in 34m 12s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/243845872b114bac825ef9da747b345a : SUCCESS in 1h 07m 47s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b09ab7e4b554415597640e6c50ad0a6e : SUCCESS in 1h 29m 31s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/5fc716e2e37643a3a234d98ab1b308a3 : SUCCESS in 47m 34s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/86082acef71d4a59af1e8589c3241395 : SUCCESS in 1h 54m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c55765d85be048308d37751b728c5940 : SUCCESS in 54m 47s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":5},{"id":"a1a031168e9d129d9d3eb9602e816c84ef7b9835","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-16 18:09:15.000000000","message":"Patch Set 5:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/16_08_2026_18_02_15\nLeave a comment with","accounts_in_message":[],"_revision_number":5},{"id":"4024eb350e0875fe1489227550b045239cce3a9d","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-08-16 20:42:52.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-11-999911-5-fc : FAILURE in 10m 55s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-11-999911-5-iscsi : SUCCESS in 1h 9m 46s","accounts_in_message":[],"_revision_number":5},{"id":"836b89e9df57ba79b62ef97fb31d887cc142caa5","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-16 22:37:08.000000000","message":"Patch Set 5:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4956/.","accounts_in_message":[],"_revision_number":5},{"id":"ecb093e850ac6761e2e3c950dec6af3115acc485","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-16 22:54:47.000000000","message":"Patch Set 5:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5132/.","accounts_in_message":[],"_revision_number":5},{"id":"16293afdc6002be53511f407fb557b7f62a74cda","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-08-19 14:50:06.000000000","message":"Patch Set 5: Review-Priority+1","accounts_in_message":[],"_revision_number":5},{"id":"05502e45b2d3e70a1d2aeb10c2c57ebb7e92b630","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2026-08-26 10:31:40.000000000","message":"Patch Set 5: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"1e30bd0f388473398e6de491021371988a6290b0","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-08-31 16:51:15.000000000","message":"Patch Set 5: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"aafae185a22ef6352dc9624c15bebb1dbd518020","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2026-08-31 18:38:29.000000000","message":"Uploaded patch set 6.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"f48d5817de3ea6024eea9eaafc28ef9c6c9fcafb","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2026-08-31 18:39:58.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"6ae40ff606b052ff8c43dbf0a73886fbaeff1f7c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-31 18:48:06.000000000","message":"Patch Set 6:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5338/.","accounts_in_message":[],"_revision_number":6},{"id":"2bf8f44101fd550699e9744ddc006dc5ec08bce9","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-31 18:59:21.000000000","message":"Patch Set 6:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/09e94dc8ab9a4ff59d6c041094727587/","accounts_in_message":[],"_revision_number":6},{"id":"e81e70bb0e263aaa3aa5393f3dd553c359971db7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 19:11:24.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6eeaf9ab19a148b7be049c69fee1c68f\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/55ad5a7eaf934c63a5a5ffdefd3a2e19 : SUCCESS in 20m 27s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/493bb41968d24957a3e5ef326ed21e9b : SUCCESS in 12m 15s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/5f1e35bd28814a2e9b8e16d1b80891e6 : SUCCESS in 18m 18s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"4bd048f21dcdd2a1c6b18598b23dec02209b3468","author":{"_account_id":39391,"name":"Rashi Jain","display_name":"HPE AlletraMP NVMeTCP CI","email":"rashi.jain@hpe.com","username":"rashij"},"date":"2026-08-31 19:27:41.000000000","message":"Patch Set 6:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/1022 : SUCCESS in 0h 48m 50s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/999911","accounts_in_message":[],"_revision_number":6},{"id":"8a9d35bbbb9038841aa2604e3d8c5b4a67b63c55","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-08-31 19:28:11.000000000","message":"Patch Set 6:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/1114 : SUCCESS in 0h 49m 19s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/999911","accounts_in_message":[],"_revision_number":6},{"id":"cc5fc7a0bbe542062c9fa2ce6f5788a40d27c7a3","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-08-31 19:28:23.000000000","message":"Patch Set 6:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1383 : SUCCESS in 0h 49m 31s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/999911","accounts_in_message":[],"_revision_number":6},{"id":"c0e9d0e81acf41f7e1d33fa33f91024f36c2585e","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-31 19:32:58.000000000","message":"Patch Set 6:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Id59d70da2326ce5a34e61b22c833903398ac58c4 .","accounts_in_message":[],"_revision_number":6},{"id":"8ee87b97bfd2151c56a2e908c76bfecf390e6c2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 20:46:28.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7bf73aed94a4328bc2c5aee92d18cd9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/457c970959b54bbdb54e65c65c6b5c33 : SUCCESS in 3m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/87d6eb9254e941efb38b1892faf30d90 : SUCCESS in 10m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9db750d3f3f648b28ab6bfae080778e4 : SUCCESS in 9m 29s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7d1600cc81dd4c35b383e47f8d23f3c5 : SUCCESS in 10m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c635c405e2e40308f7d733c3d5ad4dc : SUCCESS in 14m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/74790a7534d94b0d98ed7fd2a42bf715 : SUCCESS in 32m 20s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/bb0f7c5538ab4beeb32cd31bc5705f24 : SUCCESS in 43m 45s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8578bfc638b24bf18b9db1f22d554a7a : SUCCESS in 1h 12m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2581298ade5e413a9d2dde3176db99c0 : SUCCESS in 38m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/35e26d8223d948a7b2c00bf81a7e7d8b : SUCCESS in 8m 59s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/51ec1f2578bd4503949c394b6e08cc32 : SUCCESS in 17m 25s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/cd5db235e8994065a4c39e30dfcf6a15 : SUCCESS in 5m 05s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a93c2e0f256b45bd963c34b412db1e8c : SUCCESS in 7m 36s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/eaf635c71b88408c887cef16f9d173c1 : SUCCESS in 10m 45s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/0270df8b62e1456998279e18cf675d13 : FAILURE in 10m 00s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/512be767e3634c069856a19fb6147e18 : SUCCESS in 7m 02s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/d1dd38d0a56843d9aacc32520539aa7a : SUCCESS in 4m 54s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/261c907542e3437e9ae911dd19767559 : FAILURE in 6m 06s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/6c48efd6ed834fa9854be4ccc2fe527a : SUCCESS in 11m 12s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7f967bcbba4c47898cf0089441d3d708 : SUCCESS in 43m 19s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2d5a09f013d0427788f96568f051637c : FAILURE in 6m 21s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d39511c6abd0452cb1d61c66a17fba47 : SUCCESS in 1h 03m 09s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0e45210855a3459995756066e4e37132 : FAILURE in 1h 32m 31s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c0b91fe734314abbafd253491f0993c7 : SUCCESS in 1h 23m 01s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/e1d79d770f1f427685f2c92c7697cd56 : FAILURE in 1h 38m 26s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/a67d5ef328c947aab37c958850facdc0 : SUCCESS in 1h 54m 35s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/48dd5709e0024d769bfea2fb21532d67 : SUCCESS in 40m 05s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ac70b38d2e2745e4a98e052dd1aefda4 : SUCCESS in 1h 22m 09s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/9e5b6f7489594462bd526c678a47b462 : SUCCESS in 1h 35m 48s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/6df73a886c814d968169057a0360d47e : SUCCESS in 28m 13s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/dc617b73314a4841be1314e41af2dec0 : SUCCESS in 1h 28m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e2e3c3e25ba843719f55f27d6e99d328 : SUCCESS in 1h 00m 30s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":6},{"id":"b25a18244492bcc9a6393b9b7df4d3b30c8987b2","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-31 21:00:49.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/13c863f8b5b64996a81d63f603641483 : SUCCESS in 49m 18s\n- cinder-storpool-tempest-iscsi-multipath build/5fdd17cfb6e34734a1021ba61e85c53d : SUCCESS in 48m 04s","accounts_in_message":[],"_revision_number":6},{"id":"176fef4eb8d786e96cf5b7971ac470ff277e264c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-31 21:05:39.000000000","message":"Patch Set 6:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/31_08_2026_20_58_39\nLeave a comment with","accounts_in_message":[],"_revision_number":6},{"id":"45c8d2e51b46629da5d00b6187ad34b30b32c49f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-31 21:08:57.000000000","message":"Patch Set 6:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/09e94dc8ab9a4ff59d6c041094727587\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ccbf8758e96c4fbf83b6f4e6b4fd9570 : SUCCESS in 40m 27s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8d7e689a439d487da795d0d8daf97d22 : SUCCESS in 36m 57s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4f7e0f1ed5f24d9ba4fb28a5776bad33 : FAILURE in 2h 29m 16s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/45ab6122f09443689f99a332f301eb33 : SUCCESS in 36m 43s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b87f0d5ee92d44f8822c9b0d6caeba67 : FAILURE in 17m 01s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"ea46cfeb572b372d4c73e85ad818bf8b59382ddb","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2026-08-31 21:27:20.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"b7e682a7313cef8dd9cd9c86197ffb30c3ff2590","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-31 22:01:30.000000000","message":"Patch Set 6:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/9768600adf884a649627d9d981f95003/","accounts_in_message":[],"_revision_number":6},{"id":"6cffd12234df68e48792bcb17cf5ec9173fc39b8","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-31 23:00:30.000000000","message":"Patch Set 6:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/9768600adf884a649627d9d981f95003\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6a9d4befe12a405dad4bb757d443f44a : SUCCESS in 42m 06s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/95010dbe99504657a01ce47af33c6639 : SUCCESS in 42m 10s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/83b7a74638614b6199c67daaff84e1da : SUCCESS in 51m 23s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5f6e275fd9314ca4ab6534e379db4032 : SUCCESS in 33m 42s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9d26a21b5da54efbaa621a2d94ea73dc : NODE_FAILURE Node request 200-0001084342 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"9fa2bc5227ff81e00e469e883594969eed880a39","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-09-01 02:01:19.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-11-999911-6-fc : FAILURE in 14m 15s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-11-999911-6-iscsi : SUCCESS in 1h 34m","accounts_in_message":[],"_revision_number":6},{"id":"bab850a36c072e1cfcf23e08d7a7ef120cd20305","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-09-01 15:43:20.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-11-999911-6-fc : SUCCESS in 56m 24s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-11-999911-6-iscsi : FAILURE in 2h 13m 21s","accounts_in_message":[],"_revision_number":6},{"id":"0e5c63f428caa41bef96888bc0ca8d4820f6fd24","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-03 05:29:52.000000000","message":"Patch Set 6:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5162/.","accounts_in_message":[],"_revision_number":6},{"id":"3ed3dc82692e3d192be28571a920841347818798","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-09-03 12:57:33.000000000","message":"Patch Set 6: Review-Priority+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"2b485f2a04e596325e040dec80e04ce81742197d","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2026-09-07 01:12:11.000000000","message":"Patch Set 6: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"b4967fb91d6d51d7ee0ab8eae9d3fc38444d8bd4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2026-09-10 13:48:07.000000000","message":"Uploaded patch set 7.\n\nCopied Votes:\n* Review-Priority+1, Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"c5d4406b3065f27f8ce559176f25ea54178775ea","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2026-09-10 13:51:06.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"618d4ffd15fc09a0d5b4622f1841872073397960","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2026-09-10 13:52:07.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"58163c8411d63954d97e329ce51db8b0aa004b31","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-10 14:17:40.000000000","message":"Patch Set 7:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/bea0aaf9c36c47a095bcb01c5f1f9ec2/","accounts_in_message":[],"_revision_number":7},{"id":"ad8f15efc09d7c012aa08cb05840526d61e215e6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 14:24:12.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02caa33f90e541dca09c478113c9f7b6\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/fe879f2a38d246818ed519fbff26f90c : FAILURE in 12m 44s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/76fa44e98c78494aa181f3de85928427 : SUCCESS in 14m 07s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/3a90baf7f1b840b19da93624ccfca2be : SUCCESS in 16m 50s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"6ab2fd4776685e61e49a1b12031fa85f3dbb5fd9","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-10 14:41:51.000000000","message":"Patch Set 7:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/bea0aaf9c36c47a095bcb01c5f1f9ec2\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/436c1c43d04f44368af0b87b1ce5a0ef : SUCCESS in 45m 09s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/81b112f4d79f458f9e2e3ddcfbd14e79 : SUCCESS in 31m 32s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/02430efac1b444f38864f27eb348c529 : SUCCESS in 46m 23s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/66a24185557b425e8251b158d9ff234f : SUCCESS in 38m 24s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3921915dda8944e79d91a60cd7301128 : FAILURE in 14m 53s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"b1d2434043fde5adec645dff81e428bbfe6a1467","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-10 14:42:24.000000000","message":"Patch Set 7:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Id59d70da2326ce5a34e61b22c833903398ac58c4 .","accounts_in_message":[],"_revision_number":7},{"id":"dd785e9f2dc091af345979ab5995706ee2234c4f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-09-10 15:16:13.000000000","message":"Patch Set 7:\n\nHPE AlletraMP CI\n\nBuild succeeded.\n\n- iSCSI    : SUCCESS in 0h 47m 57s\n- FC       : SUCCESS in 0h 46m 46s\n- NVMe/TCP : SUCCESS in 0h 47m 13s\n\nLogs: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/999911","accounts_in_message":[],"_revision_number":7},{"id":"f8c69f5d8b907a20b406fce70795cc844d3d9007","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-10 15:20:18.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/e28b431feda4466bb146e60d8c590ddf : SUCCESS in 45m 51s\n- cinder-storpool-tempest-iscsi-multipath build/58a990797d7d463da37d29f0c3d82e48 : SUCCESS in 50m 14s","accounts_in_message":[],"_revision_number":7},{"id":"0ea346d7c801759f6c0c38a5e9026f0872d81061","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-10 15:48:09.000000000","message":"Patch Set 7:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5322/.","accounts_in_message":[],"_revision_number":7},{"id":"0cf429e31f35440be4107b745778402f9b84a07b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-10 16:07:55.000000000","message":"Patch Set 7:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5499/.","accounts_in_message":[],"_revision_number":7},{"id":"95823aaf9dd9a1e1ff29f9bb044a2b9d5f24be7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 16:26:12.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3f8a55d0be4434d956fe9f75e1961aa\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2a32c6eee5b42f096f2be34eef1361e : SUCCESS in 6m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e4dc0988bcb14e83b950adcfdb4a14b9 : SUCCESS in 12m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/98a8a001886143c4988f8b0e2c0b2eeb : SUCCESS in 9m 25s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ae70f4f7a7d9423ea2db2340dab44a94 : SUCCESS in 14m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd1ebb573b3e400cb315929e654e5109 : SUCCESS in 7m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/1c76393231e04d339e6b628d0ab52ac8 : SUCCESS in 1h 15m 50s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ab56ce592f1c493f81195fc0bfbb0248 : SUCCESS in 1h 06m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/710658a8aaf94c1bafc04879295f6df8 : SUCCESS in 53m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/339e116c74a145649a0ed7cd1b70c611 : SUCCESS in 1h 07m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ff8ad7bc3beb44f6be781237e1c2e0fa : SUCCESS in 5m 33s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2e604defb99e45cfa634dc88a79070d5 : SUCCESS in 23m 26s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c2fb54d917214dd58535f62fa2b13c16 : SUCCESS in 5m 24s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4766c6af394b4db5a1038c55ecfd859d : SUCCESS in 6m 52s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/e885ca17a7ac4faca60c71dccd5d839c : FAILURE in 8m 02s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/d31d2db7dada4414894f4d14f859edbd : SUCCESS in 10m 30s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/387e50da95044fffb2582f8309a8e4b9 : SUCCESS in 7m 04s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/338a8eaeeff4454989422d13c834abe4 : SUCCESS in 6m 33s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/ae9824d00cd84a7c8c92fc626660304d : FAILURE in 6m 25s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/8d2e46fcb4154feba7d5709cb14edca8 : SUCCESS in 9m 56s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/4580a3380f22435b90084a229164b833 : SUCCESS in 38m 53s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ecea4923f8da467cb2bacdbd3bf8d814 : FAILURE in 4m 10s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c9bb1d749e37442aa3aafd0425b24059 : SUCCESS in 1h 40m 56s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/167b40f781b441e4bbc9381ca89e2dd8 : FAILURE in 2h 13m 38s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2fa4f67b67194d2a930ed8377bdb20c5 : SUCCESS in 1h 46m 12s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/726d1e9f02cc4dd4b0a814a6964eb29f : FAILURE in 2h 34m 32s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/d033db686c204ceca6575c6e8a6cf4d6 : SUCCESS in 1h 54m 18s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/507e28292d444c35b8344e851e59a867 : SUCCESS in 32m 57s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/7fb2c0d06a274e7a9787015ba0804427 : SUCCESS in 1h 13m 37s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/64ddc8a9e6b440989d4f188cf489d46e : SUCCESS in 1h 36m 48s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/89bd1be7e1b64139bf2ab31c1599bc06 : SUCCESS in 52m 10s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d0fe11566a9d46449c260c26b73ad203 : SUCCESS in 1h 57m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e2aa0555f50b4a23a51325ce6cc28dc3 : SUCCESS in 1h 00m 30s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":7},{"id":"2fb93319ef57459c8f37b94f9bf08d7c4938d243","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-09-10 16:37:12.000000000","message":"Patch Set 7:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/10_09_2026_16_31_40\nLeave a comment with","accounts_in_message":[],"_revision_number":7},{"id":"0877d6dde27ed2a4a83e0e98edbf4d29fe0af1d3","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-09-10 17:28:02.000000000","message":"Patch Set 7: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"13298afe3885c5cf2f8cde9b3b0fa916f72af519","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2026-09-10 22:24:42.000000000","message":"Patch Set 7: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":7},{"id":"95b927071c8b497871cd4c6f64891cead11bc902","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-09-10 22:42:42.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-11-999911-7-fc : FAILURE in 17m 36s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-11-999911-7-iscsi : SUCCESS in 1h 49m 58s","accounts_in_message":[],"_revision_number":7},{"id":"ab67737fe14776519b78ddf05e371feff1b307b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2026-09-11 02:20:15.000000000","message":"Uploaded patch set 8.\n\nCopied Votes:\n* Review-Priority+1, Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"49135f6c1490fabf6d7f35512c758f3fa2094b60","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2026-09-11 02:26:41.000000000","message":"Patch Set 8:\n\n(6 comments)","accounts_in_message":[],"_revision_number":8},{"id":"1205887790f6ca14eff0516eacd55610157172c2","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-11 02:53:32.000000000","message":"Patch Set 8:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/24719f0109b246188c7bb4efd7c418dd/","accounts_in_message":[],"_revision_number":8},{"id":"b6bdd5e0fc9d081236d16e1c40fd49f8802ab83f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-09-11 03:13:31.000000000","message":"Patch Set 8:\n\nHPE AlletraMP CI\n\nBuild succeeded.\n\n- iSCSI    : SUCCESS in 0h 48m 18s\n- FC       : SUCCESS in 0h 48m 30s\n- NVMe/TCP : SUCCESS in 0h 48m 2s\n\nLogs: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/999911","accounts_in_message":[],"_revision_number":8},{"id":"5d1b433049cad9baf20a1d12486d26f44269d0ee","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-11 03:16:05.000000000","message":"Patch Set 8:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Id59d70da2326ce5a34e61b22c833903398ac58c4 .","accounts_in_message":[],"_revision_number":8},{"id":"2905ad9be279345deb5ba2ebf9c1971fdcbc6863","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-11 03:16:14.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/4e5a7f78221842a7b04ef70368ace741 : SUCCESS in 50m 14s\n- cinder-storpool-tempest-iscsi-multipath build/d6f2a7e2a38a4f83b4d2b405bb668894 : SUCCESS in 53m 44s","accounts_in_message":[],"_revision_number":8},{"id":"a825576cc1b037ec1cc07f0538fadbdcbda361e9","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-11 03:25:35.000000000","message":"Patch Set 8:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5329/.","accounts_in_message":[],"_revision_number":8},{"id":"edab6dd7bd491107194d2e78d2246a9b0c757aac","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-11 03:29:03.000000000","message":"Patch Set 8:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5506/.","accounts_in_message":[],"_revision_number":8},{"id":"6df1ede0bc873dd40ae727488cb2fe5c8a9e40cf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 03:31:07.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2dd617eee73c40ae97798a4a33cdc671\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6a0a4a509b4a43d8ab90cfbc57cf70ee : SUCCESS in 13m 57s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f871be3012c6403787500085e86699f5 : SUCCESS in 15m 24s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/510a68eb1f4a4d68a68cbaafdf306743 : SUCCESS in 15m 29s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"d7cfca427d1cc50429c9626680b4cf968bc00640","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-11 04:44:20.000000000","message":"Patch Set 8:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/24719f0109b246188c7bb4efd7c418dd\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e94883f4299a4e92a5aa58fe8f90794f : SUCCESS in 38m 47s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c34cd1142d2f4cbb8f9c7f710ee076b2 : SUCCESS in 32m 08s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/cbf0da6b6aef4ddfb6733cf1c770983d : FAILURE in 2h 22m 54s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e67db9c680bd4fe19b17bd7b14e261e6 : SUCCESS in 32m 27s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8b9e363a4ccd4ed8b37e453554752e04 : NODE_FAILURE Node request 200-0001247639 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"f7049381f9257401724c8152a13c1be7c62b3653","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 05:08:00.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/027374ffda2a42379a5a48c66c0a3d84\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d20b07fcbe8c47e699fd4282e10ac323 : SUCCESS in 7m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dc71cc969619406a9162a30814862d34 : SUCCESS in 11m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fd7b8ca4667145cc938c7c3c786c5fbc : SUCCESS in 11m 37s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ba8fb6482d4b44338c7e67bcd58b5054 : SUCCESS in 13m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b50f5892f75b4b0f9eff3ee0597c37e3 : SUCCESS in 11m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/e627789ed6974bbcb09501eff762b0f9 : SUCCESS in 1h 02m 26s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7425fa862f4847f195b4684927eaaa80 : SUCCESS in 1h 00m 54s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6e22da9754fe4e74970c25f2c04dbb22 : SUCCESS in 1h 03m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5a0a97968714443ab1bf9422712e4303 : SUCCESS in 56m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7a40d22c990842a39323cd2483a18210 : SUCCESS in 9m 17s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/5d1cf52497a74e808a08bf13a5f9defd : TIMED_OUT in 41m 25s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c7599114bb5648a9add3ef3d056798d9 : SUCCESS in 5m 25s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d962c338c28a41b0ab42a3a803f8dff0 : SUCCESS in 7m 09s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/4b509448560645dcb49e2f88b6c607d2 : SUCCESS in 11m 58s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/32fd1010267d4fb1bc98d624bdc1dfac : SUCCESS in 9m 37s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/c008c6c4d9094db58d82430ef5077a8b : SUCCESS in 8m 40s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/be0b85bdc4934c99bedfbf04c4364b6a : SUCCESS in 7m 05s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/78b432daaa694ba6bf63fe20c1c715b1 : FAILURE in 6m 04s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/06fe7026675f4b64b549bb3c9c6828c8 : SUCCESS in 11m 28s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/300b0a00a2234052a71c8d2c24d639df : SUCCESS in 24m 22s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ef9e469ad5eb46d3845e646bee503b32 : FAILURE in 6m 09s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e9d868977a624f6e9319f10a448d9916 : SUCCESS in 1h 42m 36s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/7dfe0941f93b4672b7df4d6c1e75dd08 : SUCCESS in 1h 15m 31s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/976e4cbc935245ff90ef47adc4890798 : SUCCESS in 1h 53m 11s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/91045f04311245b3a45a01b35675cc46 : FAILURE in 2h 07m 36s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/94c26b78567a4a95979fa0b9a8a580aa : SUCCESS in 1h 43m 41s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/a97296fc054e480b811ccd0813b1d92b : SUCCESS in 35m 56s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2d077a61f003419697f02e91b3e85baa : SUCCESS in 1h 29m 43s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a3adffd7acd94d239a7c265076e1f379 : SUCCESS in 1h 24m 40s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/cebf24c273c24e4f8990c6662d72e90a : SUCCESS in 46m 48s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/19324b6b62854ff2a4ea605b4b343414 : SUCCESS in 1h 07m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1409a22d46e24a9fbaa6ed3d112562be : SUCCESS in 1h 14m 43s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":8},{"id":"9509d18279e0b89a0d7c1f6940eb16f8d487731f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-09-11 05:17:42.000000000","message":"Patch Set 8:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/11_09_2026_05_11_40\nLeave a comment with","accounts_in_message":[],"_revision_number":8},{"id":"eb3bb9f19e9fed5f2c05d0a85db832504bd74f5a","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2026-09-11 05:27:20.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"e533c018b1a4f920a099d5ad99df6ad439cc1691","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-11 05:57:49.000000000","message":"Patch Set 8:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/cc960861c1934f92ad62a4de0da1a75b/","accounts_in_message":[],"_revision_number":8},{"id":"9995b6eb91b28f87a6c56c94c3e73ee1ae9c2b9c","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-11 08:07:03.000000000","message":"Patch Set 8:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/cc960861c1934f92ad62a4de0da1a75b\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e0dd6baf57494789a28950dab4464f98 : SUCCESS in 40m 55s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8cce33faa78b4de3a98bac7fc9256a1e : SUCCESS in 35m 16s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ec163cfe68354478b1d1044c484d17c9 : SUCCESS in 56m 07s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/06950cfb8ebd497f85165ae978d3cb54 : SUCCESS in 30m 01s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/81a24c2840a14354aded8e083f0f0076 : FAILURE in 44m 58s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"5f369fc0b3da536d2f67e55fa8f5081e69178253","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2026-09-11 13:00:49.000000000","message":"Patch Set 8: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"5c539963491be32186518bc94783c527be99f066","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-09-11 15:58:32.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-11-999911-8-fc : FAILURE in 10m 5s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-11-999911-8-iscsi : SUCCESS in 1h 39m 11s","accounts_in_message":[],"_revision_number":8},{"id":"5297cae48e11eea13fec46b9846c27392f44ff7a","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-09-11 16:06:51.000000000","message":"Patch Set 8: Code-Review+2 Review-Priority+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"589883d5b1b5340d608ea421fbcc12ad62169772","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-09-11 16:19:32.000000000","message":"Patch Set 8: Code-Review+2 Review-Priority+1 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"fa3ba1774b5e0ad77a9d78840692948c315bf26c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 16:20:06.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"348df2d320855f88823754b4574bff3365a283c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 18:15:37.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a295d775ae4448e99c6fe7bbde38e94b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/043f5e492fbd409ab15c18f782205c16 : SUCCESS in 5m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0ec0276a987e4aa0839d9e4c7e0be8c1 : SUCCESS in 9m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/50b3279b90de4d72828cdccf7dad1dac : SUCCESS in 9m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6490910e1ae473da240f1b76c9f7620 : SUCCESS in 11m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/3be7d2409ab1412aa7b4b670a3a91f26 : SUCCESS in 1h 07m 23s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c625be9918b645c38e401440ae721f28 : SUCCESS in 1h 14m 54s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/af71c2fb927e4dcaa8d454e58aae2c05 : SUCCESS in 1h 12m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a424ae288f1642e4bcdce19d9c2b4f19 : SUCCESS in 48m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f7e81d03466a4953a97c4445a2e0afba : SUCCESS in 10m 49s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e20ad133a00d471c80fd00b54014758d : SUCCESS in 1h 17m 02s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4625b6ad0fae4173bd161f65ca505885 : SUCCESS in 1h 27m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3b0e0fb6830844f7b98a60c28660c279 : SUCCESS in 32m 36s","accounts_in_message":[],"_revision_number":8},{"id":"227848f782fc8911942432956c033053fb0e65b3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 18:15:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"c900a0095e79bf04d3b6125cfaffbf63da663341","revisions":{"a6f20e379b9e9726fb88c562de476daf7e28bfa3":{"kind":"REWORK","_number":1,"created":"2026-08-05 22:33:36.000000000","uploader":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"ref":"refs/changes/11/999911/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/11/999911/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/999911/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/999911/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/999911/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/11/999911/1"}}},"commit":{"parents":[{"commit":"c9291171cf1448ce5a887c513418772191f11123","subject":"Merge \"Storwize: stop tests spawning cinder-manage subprocesses\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c9291171cf1448ce5a887c513418772191f11123"}]}],"author":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2026-08-05 22:30:23.000000000","tz":-240},"committer":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2026-08-05 22:30:23.000000000","tz":-240},"subject":"Fix data-to-zero corruption restoring an incremental backup chain","message":"Fix data-to-zero corruption restoring an incremental backup chain\n\nWhen restoring a chunked backup chain (Swift, S3, Google, NFS, Ceph, ...)\nto a new or sparse volume, restore() applied the ``volume_is_new``\n\"skip all-zero chunks\" optimisation to every backup in the chain, not just\nthe base full backup. A region that changed from data to zero in an\nincremental was therefore skipped instead of being zeroed, leaving the base\nbackup\u0027s stale data at that offset in the restored volume.\n\nOnly the base backup, applied first to a pristine (zero) target, may skip\nall-zero chunks. Incrementals layered on top must be written in full so a\ndata-to-zero change overwrites the base\u0027s data. Gate the optimisation on the\nbase backup accordingly.\n\nThis is a pre-existing chunked-driver restore bug independent of Changed\nBlock Tracking; it is split out from the CBT framework change so it can be\nreviewed and backported on its own.\n\nCo-Authored-By: Claude Opus 4.8 \u003cnoreply@anthropic.com\u003e\nChange-Id: Id59d70da2326ce5a34e61b22c833903398ac58c4\nSigned-off-by: Simon Dodsley \u003csimon@purestorage.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a6f20e379b9e9726fb88c562de476daf7e28bfa3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a6f20e379b9e9726fb88c562de476daf7e28bfa3"}]},"branch":"refs/heads/master"},"da9562953a021f0c2077f46510ef496f4e572205":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-08-06 13:53:59.000000000","uploader":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"ref":"refs/changes/11/999911/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/11/999911/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/999911/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/999911/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/999911/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/11/999911/2"}}},"commit":{"parents":[{"commit":"c9291171cf1448ce5a887c513418772191f11123","subject":"Merge \"Storwize: stop tests spawning cinder-manage subprocesses\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c9291171cf1448ce5a887c513418772191f11123"}]}],"author":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2026-08-05 22:30:23.000000000","tz":-240},"committer":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2026-08-06 13:52:39.000000000","tz":-240},"subject":"Fix data-to-zero corruption restoring an incremental backup chain","message":"Fix data-to-zero corruption restoring an incremental backup chain\n\nWhen restoring a chunked backup chain (Swift, S3, Google Cloud Storage,\nPOSIX/NFS) to a new or sparse volume, restore() applied the ``volume_is_new``\n\"skip all-zero chunks\" optimisation to every backup in the chain, not just\nthe base full backup. A region that changed from data to zero in an\nincremental was therefore skipped instead of being zeroed, leaving the base\nbackup\u0027s stale data at that offset in the restored volume.\n\nOnly the base backup, applied first to a pristine (zero) target, may skip\nall-zero chunks. Incrementals layered on top must be written in full so a\ndata-to-zero change overwrites the base\u0027s data. Gate the optimisation on the\nbase backup accordingly.\n\nThis is a pre-existing chunked-driver restore bug independent of Changed\nBlock Tracking; it is split out from the CBT framework change so it can be\nreviewed and backported on its own.\n\nCloses-Bug: #2162961\nCo-Authored-By: Claude Opus 4.8 \u003cnoreply@anthropic.com\u003e\nChange-Id: Id59d70da2326ce5a34e61b22c833903398ac58c4\nSigned-off-by: Simon Dodsley \u003csimon@purestorage.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/da9562953a021f0c2077f46510ef496f4e572205"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/da9562953a021f0c2077f46510ef496f4e572205"}]},"branch":"refs/heads/master"},"ea29d4895197a7be3ebf386e28b6eed15d7fbadf":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-08-13 14:12:57.000000000","uploader":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"ref":"refs/changes/11/999911/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/11/999911/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/999911/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/999911/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/999911/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/11/999911/3"}}},"commit":{"parents":[{"commit":"911497601b299bcb679f454fe9f3fb60e66ada94","subject":"Merge \"Run Cinder\u0027s Backup service in native threads\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/911497601b299bcb679f454fe9f3fb60e66ada94"}]}],"author":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2026-08-05 22:30:23.000000000","tz":-240},"committer":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2026-08-13 14:03:08.000000000","tz":-240},"subject":"Fix data-to-zero corruption restoring an incremental backup chain","message":"Fix data-to-zero corruption restoring an incremental backup chain\n\nWhen restoring a chunked backup chain (Swift, S3, Google Cloud Storage,\nPOSIX/NFS) to a new or sparse volume, restore() applied the ``volume_is_new``\n\"skip all-zero chunks\" optimisation to every backup in the chain, not just\nthe base full backup. A region that changed from data to zero in an\nincremental was therefore skipped instead of being zeroed, leaving the base\nbackup\u0027s stale data at that offset in the restored volume.\n\nOnly the base backup, applied first to a pristine (zero) target, may skip\nall-zero chunks. Incrementals layered on top must be written in full so a\ndata-to-zero change overwrites the base\u0027s data. Gate the optimisation on the\nbase backup accordingly.\n\nThis is a pre-existing chunked-driver restore bug independent of Changed\nBlock Tracking; it is split out from the CBT framework change so it can be\nreviewed and backported on its own.\n\nCloses-Bug: #2162961\nCo-Authored-By: Claude Opus 4.8 \u003cnoreply@anthropic.com\u003e\nChange-Id: Id59d70da2326ce5a34e61b22c833903398ac58c4\nSigned-off-by: Simon Dodsley \u003csimon@purestorage.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ea29d4895197a7be3ebf386e28b6eed15d7fbadf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ea29d4895197a7be3ebf386e28b6eed15d7fbadf"}]},"branch":"refs/heads/master"},"599dd77f086cc4d7320793377e98b6030829c83a":{"kind":"REWORK","_number":4,"created":"2026-08-16 15:22:53.000000000","uploader":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"ref":"refs/changes/11/999911/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/11/999911/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/999911/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/999911/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/999911/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/11/999911/4"}}},"commit":{"parents":[{"commit":"911497601b299bcb679f454fe9f3fb60e66ada94","subject":"Merge \"Run Cinder\u0027s Backup service in native threads\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/911497601b299bcb679f454fe9f3fb60e66ada94"}]}],"author":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2026-08-05 22:30:23.000000000","tz":-240},"committer":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2026-08-16 15:22:29.000000000","tz":-240},"subject":"Fix data-to-zero corruption restoring an incremental backup chain","message":"Fix data-to-zero corruption restoring an incremental backup chain\n\nWhen restoring a chunked backup chain (Swift, S3, Google Cloud Storage,\nPOSIX/NFS) to a new or sparse volume, restore() applied the ``volume_is_new``\n\"skip all-zero chunks\" optimisation to every backup in the chain, not just\nthe base full backup. A region that changed from data to zero in an\nincremental was therefore skipped instead of being zeroed, leaving the base\nbackup\u0027s stale data at that offset in the restored volume.\n\nOnly the base backup, applied first to a pristine (zero) target, may skip\nall-zero chunks. Incrementals layered on top must be written in full so a\ndata-to-zero change overwrites the base\u0027s data. Gate the optimisation on the\nbase backup accordingly.\n\nThis is a pre-existing chunked-driver restore bug independent of Changed\nBlock Tracking; it is split out from the CBT framework change so it can be\nreviewed and backported on its own.\n\nCloses-Bug: #2162961\nCo-Authored-By: Claude Opus 4.8 \u003cnoreply@anthropic.com\u003e\nChange-Id: Id59d70da2326ce5a34e61b22c833903398ac58c4\nSigned-off-by: Simon Dodsley \u003csimon@purestorage.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/599dd77f086cc4d7320793377e98b6030829c83a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/599dd77f086cc4d7320793377e98b6030829c83a"}]},"branch":"refs/heads/master"},"fd0521cf7def0ce1f140365222c663b564d22aea":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-08-16 15:26:39.000000000","uploader":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"ref":"refs/changes/11/999911/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/11/999911/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/999911/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/999911/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/999911/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/11/999911/5"}}},"commit":{"parents":[{"commit":"b5b763129e2bde5077c0cf3a5eb434021abaa6e0","subject":"Merge \"Centralize cooperative yield for eventlet removal\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b5b763129e2bde5077c0cf3a5eb434021abaa6e0"}]}],"author":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2026-08-05 22:30:23.000000000","tz":-240},"committer":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2026-08-16 15:24:56.000000000","tz":-240},"subject":"Fix data-to-zero corruption restoring an incremental backup chain","message":"Fix data-to-zero corruption restoring an incremental backup chain\n\nWhen restoring a chunked backup chain (Swift, S3, Google Cloud Storage,\nPOSIX/NFS) to a new or sparse volume, restore() applied the ``volume_is_new``\n\"skip all-zero chunks\" optimisation to every backup in the chain, not just\nthe base full backup. A region that changed from data to zero in an\nincremental was therefore skipped instead of being zeroed, leaving the base\nbackup\u0027s stale data at that offset in the restored volume.\n\nOnly the base backup, applied first to a pristine (zero) target, may skip\nall-zero chunks. Incrementals layered on top must be written in full so a\ndata-to-zero change overwrites the base\u0027s data. Gate the optimisation on the\nbase backup accordingly.\n\nThis is a pre-existing chunked-driver restore bug independent of Changed\nBlock Tracking; it is split out from the CBT framework change so it can be\nreviewed and backported on its own.\n\nCloses-Bug: #2162961\nCo-Authored-By: Claude Opus 4.8 \u003cnoreply@anthropic.com\u003e\nChange-Id: Id59d70da2326ce5a34e61b22c833903398ac58c4\nSigned-off-by: Simon Dodsley \u003csimon@purestorage.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fd0521cf7def0ce1f140365222c663b564d22aea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fd0521cf7def0ce1f140365222c663b564d22aea"}]},"branch":"refs/heads/master"},"0f6b0d82ac2a33c6ea5aa9bcfa99b2747a22e129":{"kind":"REWORK","_number":6,"created":"2026-08-31 18:38:29.000000000","uploader":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"ref":"refs/changes/11/999911/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/11/999911/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/999911/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/999911/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/999911/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/11/999911/6"}}},"commit":{"parents":[{"commit":"b5b763129e2bde5077c0cf3a5eb434021abaa6e0","subject":"Merge \"Centralize cooperative yield for eventlet removal\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b5b763129e2bde5077c0cf3a5eb434021abaa6e0"}]}],"author":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2026-08-05 22:30:23.000000000","tz":-240},"committer":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2026-08-31 18:35:54.000000000","tz":-240},"subject":"Fix data-to-zero corruption restoring an incremental backup chain","message":"Fix data-to-zero corruption restoring an incremental backup chain\n\nWhen restoring a chunked backup chain (Swift, S3, Google Cloud Storage,\nPOSIX/NFS) to a new or sparse volume, restore() applied the ``volume_is_new``\n\"skip all-zero chunks\" optimisation to every backup in the chain, not just\nthe base full backup. A region that changed from data to zero in an\nincremental was therefore skipped instead of being zeroed, leaving the base\nbackup\u0027s stale data at that offset in the restored volume.\n\nOnly the base backup, applied first to a pristine (zero) target, may skip\nall-zero chunks. Incrementals layered on top must be written in full so a\ndata-to-zero change overwrites the base\u0027s data. Gate the optimisation on the\nbase backup accordingly.\n\nThis is a pre-existing chunked-driver restore bug independent of Changed\nBlock Tracking; it is split out from the CBT framework change so it can be\nreviewed and backported on its own.\n\nCloses-Bug: #2162961\nCo-Authored-By: Claude Opus 4.8 \u003cnoreply@anthropic.com\u003e\nChange-Id: Id59d70da2326ce5a34e61b22c833903398ac58c4\nSigned-off-by: Simon Dodsley \u003csimon@purestorage.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0f6b0d82ac2a33c6ea5aa9bcfa99b2747a22e129"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0f6b0d82ac2a33c6ea5aa9bcfa99b2747a22e129"}]},"branch":"refs/heads/master"},"884d8ff7f7c2179bd4554e6d57bb4b69ed4a5b8a":{"kind":"REWORK","_number":7,"created":"2026-09-10 13:48:07.000000000","uploader":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"ref":"refs/changes/11/999911/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/11/999911/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/999911/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/999911/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/999911/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/11/999911/7"}}},"commit":{"parents":[{"commit":"b5b763129e2bde5077c0cf3a5eb434021abaa6e0","subject":"Merge \"Centralize cooperative yield for eventlet removal\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b5b763129e2bde5077c0cf3a5eb434021abaa6e0"}]}],"author":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2026-08-05 22:30:23.000000000","tz":-240},"committer":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2026-09-10 13:45:17.000000000","tz":480},"subject":"Fix data-to-zero corruption restoring an incremental backup chain","message":"Fix data-to-zero corruption restoring an incremental backup chain\n\nWhen restoring a chunked backup chain (Swift, S3, Google Cloud Storage,\nPOSIX/NFS) to a new or sparse volume, restore() applied the ``volume_is_new``\n\"skip all-zero chunks\" optimisation to every backup in the chain, not just\nthe base full backup. A region that changed from data to zero in an\nincremental was therefore skipped instead of being zeroed, leaving the base\nbackup\u0027s stale data at that offset in the restored volume.\n\nOnly the base backup, applied first to a pristine (zero) target, may skip\nall-zero chunks. Incrementals layered on top must be written in full so a\ndata-to-zero change overwrites the base\u0027s data. Gate the optimisation on the\nbase backup accordingly.\n\nThis is a pre-existing chunked-driver restore bug independent of Changed\nBlock Tracking; it is split out from the CBT framework change so it can be\nreviewed and backported on its own.\n\nCloses-Bug: #2162961\nCo-Authored-By: Claude Opus 4.8 \u003cnoreply@anthropic.com\u003e\nChange-Id: Id59d70da2326ce5a34e61b22c833903398ac58c4\nSigned-off-by: Simon Dodsley \u003csimon@purestorage.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/884d8ff7f7c2179bd4554e6d57bb4b69ed4a5b8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/884d8ff7f7c2179bd4554e6d57bb4b69ed4a5b8a"}]},"branch":"refs/heads/master"},"c900a0095e79bf04d3b6125cfaffbf63da663341":{"kind":"REWORK","_number":8,"created":"2026-09-11 02:20:15.000000000","uploader":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"ref":"refs/changes/11/999911/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/11/999911/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/999911/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/999911/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/999911/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/11/999911/8"}}},"commit":{"parents":[{"commit":"b5b763129e2bde5077c0cf3a5eb434021abaa6e0","subject":"Merge \"Centralize cooperative yield for eventlet removal\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b5b763129e2bde5077c0cf3a5eb434021abaa6e0"}]}],"author":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2026-08-05 22:30:23.000000000","tz":-240},"committer":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2026-09-11 02:19:24.000000000","tz":540},"subject":"Fix data-to-zero corruption restoring an incremental backup chain","message":"Fix data-to-zero corruption restoring an incremental backup chain\n\nWhen restoring a chunked backup chain (Swift, S3, Google Cloud Storage,\nPOSIX/NFS) to a new or sparse volume, restore() applied the ``volume_is_new``\n\"skip all-zero chunks\" optimisation to every backup in the chain, not just\nthe base full backup. A region that changed from data to zero in an\nincremental was therefore skipped instead of being zeroed, leaving the base\nbackup\u0027s stale data at that offset in the restored volume.\n\nOnly the base backup, applied first to a pristine (zero) target, may skip\nall-zero chunks. Incrementals layered on top must be written in full so a\ndata-to-zero change overwrites the base\u0027s data. Gate the optimisation on the\nbase backup accordingly.\n\nThis is a pre-existing chunked-driver restore bug independent of Changed\nBlock Tracking; it is split out from the CBT framework change so it can be\nreviewed and backported on its own.\n\nCloses-Bug: #2162961\nCo-Authored-By: Claude Opus 4.8 \u003cnoreply@anthropic.com\u003e\nChange-Id: Id59d70da2326ce5a34e61b22c833903398ac58c4\nSigned-off-by: Simon Dodsley \u003csimon@purestorage.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c900a0095e79bf04d3b6125cfaffbf63da663341"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c900a0095e79bf04d3b6125cfaffbf63da663341"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
