)]}'
{"id":"openstack%2Fcliff~511971","triplet_id":"openstack%2Fcliff~master~I1f2606cb43c8064e465e87d6801ed8d169daa26a","project":"openstack/cliff","branch":"master","topic":"errexit2","hashtags":[],"change_id":"I1f2606cb43c8064e465e87d6801ed8d169daa26a","subject":"Add an errexit attribute to InteractiveApp to exit on command errors","status":"MERGED","created":"2017-10-13 22:10:52.000000000","updated":"2019-10-02 19:35:54.000000000","submitted":"2019-10-02 19:34:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"511971-1570044873398-34c20870","meta_rev_id":"a3657373b1f59cfad975dac7c5c8937fd6e24381","_number":511971,"virtual_id_number":511971,"owner":{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"date":"2019-05-11 03:38:53.000000000","_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"date":"2019-08-05 22:05:56.000000000","_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},{"value":0,"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-02 19:34:33.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"recommended":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2019-09-13 01:46:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":2,"date":"2019-10-02 15:30:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2019-09-14 12:14:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":2,"date":"2019-05-06 20:33:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},{"value":1,"date":"2019-09-14 01:29:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-14 01:15:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":1,"date":"2019-10-02 15:30:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},{"value":0,"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-10-13 23:26:46.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2019-05-06 20:33:24.000000000","updated_by":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"reviewer":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"state":"REVIEWER"},{"updated":"2019-05-11 03:38:53.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2019-09-13 01:46:54.000000000","updated_by":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"reviewer":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"state":"REVIEWER"},{"updated":"2019-09-14 01:15:01.000000000","updated_by":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"reviewer":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"state":"REVIEWER"},{"updated":"2019-09-14 01:29:15.000000000","updated_by":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"reviewer":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"state":"REVIEWER"},{"updated":"2019-09-14 12:14:45.000000000","updated_by":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"reviewer":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"state":"REVIEWER"},{"updated":"2019-10-02 15:30:32.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-10-02 19:34:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"beff617d5fbba9385cf6123bdf87b41ffaaa9bdf","author":{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},"date":"2017-10-13 22:10:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ec08122b2bdd6943c94644608c5654ede31c29ab","author":{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},"date":"2017-10-13 22:16:04.000000000","message":"Patch Set 1:\n\n(2 comments)\n\nBoth of the interactive tests really need to send a few commands in via stdin to verify stuff is working, but it\u0027s late Friday afternoon and I wanted to get some feedback on the overall approach before diving in to that.\n\nThis in theory is backward-compatible, with the exception of not raising an exception that gets caught higher up in App.run_subcommand().","accounts_in_message":[],"_revision_number":1},{"id":"2cdc1f5e382b0b192b6308c3a7ee261edaf9f309","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2017-10-13 23:05:37.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"5f6c6da6063a2c53ed1a98850a41c90436d088f6","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2017-10-13 23:06:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"34bcf2815aa301888d8e573abfd8af68cbc88186","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-10-13 23:26:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-cliff-docs-ubuntu-xenial http://docs-draft.openstack.org/71/511971/1/check/gate-cliff-docs-ubuntu-xenial/7ee141b//doc/build/html/ : SUCCESS in 4m 32s\n- gate-cliff-pep8-ubuntu-xenial http://logs.openstack.org/71/511971/1/check/gate-cliff-pep8-ubuntu-xenial/ae2ec60/ : SUCCESS in 2m 36s\n- gate-cliff-python27-ubuntu-xenial http://logs.openstack.org/71/511971/1/check/gate-cliff-python27-ubuntu-xenial/ccd40b9/ : SUCCESS in 2m 47s\n- gate-cliff-python35 http://logs.openstack.org/71/511971/1/check/gate-cliff-python35/a6e9f23/ : SUCCESS in 2m 41s\n- gate-cliff-pypy http://logs.openstack.org/71/511971/1/check/gate-cliff-pypy/b220577/ : SUCCESS in 3m 32s (non-voting)\n- gate-tempest-dsvm-neutron-src-cliff-ubuntu-xenial http://logs.openstack.org/71/511971/1/check/gate-tempest-dsvm-neutron-src-cliff-ubuntu-xenial/4931a2e/ : SUCCESS in 1h 15m 42s\n- gate-cliff-tox-neutronclient-tip-ubuntu-xenial http://logs.openstack.org/71/511971/1/check/gate-cliff-tox-neutronclient-tip-ubuntu-xenial/caf0899/ : SUCCESS in 3m 25s\n- gate-cliff-tox-openstackclient-tip-ubuntu-xenial http://logs.openstack.org/71/511971/1/check/gate-cliff-tox-openstackclient-tip-ubuntu-xenial/863eb75/ : SUCCESS in 3m 48s","accounts_in_message":[],"_revision_number":1},{"id":"d13a7addb734a0a21a99390b7f57d3e59c865aaf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-14 04:24:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded.\n\n- openstack-tox-pep8 http://logs.openstack.org/71/511971/1/check/openstack-tox-pep8/30562a4/ : SUCCESS in 3m 38s\n- openstack-tox-py27 http://logs.openstack.org/71/511971/1/check/openstack-tox-py27/9924063/ : SUCCESS in 3m 40s\n- openstack-tox-py35 http://logs.openstack.org/71/511971/1/check/openstack-tox-py35/c95e11b/ : SUCCESS in 4m 13s\n- openstack-tox-pypy http://logs.openstack.org/71/511971/1/check/openstack-tox-pypy/a57f37a/ : SUCCESS in 4m 56s (non-voting)\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/71/511971/1/check/legacy-tempest-dsvm-neutron-src/b37d60d/ : SUCCESS in 1h 28m 56s\n- build-openstack-sphinx-docs http://logs.openstack.org/71/511971/1/check/build-openstack-sphinx-docs/f1be3bb/html/ : SUCCESS in 6m 08s\n- legacy-cliff-tox-neutronclient-tip http://logs.openstack.org/71/511971/1/check/legacy-cliff-tox-neutronclient-tip/f5339c1/ : SUCCESS in 4m 03s\n- legacy-cliff-tox-openstackclient-tip http://logs.openstack.org/71/511971/1/check/legacy-cliff-tox-openstackclient-tip/6e91d7c/ : SUCCESS in 4m 20s","accounts_in_message":[],"_revision_number":1},{"id":"be5371260a3c74b309bcf1959a67fb35cadc8311","author":{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},"date":"2017-10-14 14:25:28.000000000","message":"Patch Set 1: Workflow-1\n\n(2 comments)\n\nThanks Doug.  I do want to do the set command thing, but wanted to think it through for more general settings before rolling it out.  Things like being able to switch cloud configs between commands in an input file for an OSC-centric example.  I will re-work it to remove the options...","accounts_in_message":[],"_revision_number":1},{"id":"c7acf5f192d2074fe6c7375a3fcc6025384cf781","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2017-10-14 15:09:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3ad6f56b986387d238aea91e8001b287f923dd13","author":{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},"date":"2017-10-18 00:08:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"98162c5bda377ece81f693604d1ae46e5506a141","author":{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},"date":"2017-10-18 00:14:15.000000000","message":"Patch Set 2:\n\nDoug, https://review.openstack.org/512903 is what the OSC (osc-lib actually) side of this looks like.  I have to override interact() completely to change the constructor args for InteractiveApp, so you can change it but it\u0027s not completely usable.  Since I had to do that much I had it catch the TypeError in case an older cliff or another InteractiveApp class doesn\u0027t know errexit.\n\nI\u0027m all ears on a cleaner way to do this, it feels a bit hacky to me still.","accounts_in_message":[],"_revision_number":2},{"id":"b712a6698e1c2fbcf72009560237d40f596e62c3","author":{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},"date":"2017-10-18 02:51:39.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"36d532404678b0749f9c28e3d8a1ec44739b2ab9","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2017-10-18 12:42:45.000000000","message":"Patch Set 2:\n\n\u003e Doug, https://review.openstack.org/512903 is what the OSC (osc-lib\n \u003e actually) side of this looks like.  I have to override interact()\n \u003e completely to change the constructor args for InteractiveApp, so\n \u003e you can change it but it\u0027s not completely usable.  Since I had to\n \u003e do that much I had it catch the TypeError in case an older cliff or\n \u003e another InteractiveApp class doesn\u0027t know errexit.\n \u003e \n \u003e I\u0027m all ears on a cleaner way to do this, it feels a bit hacky to\n \u003e me still.\n\nI haven\u0027t looked deeply into this code in a while, but my first thought is to provide some way to override that interactive class. Maybe as a constructor argument to the main App class?\n\nOTOH, I don\u0027t really have an issue with building the short-cut stuff into the base implementation. It seems widely usable.","accounts_in_message":[],"_revision_number":2},{"id":"bbaaf76120dd8a0c818889a04deee5a21fd09ae6","author":{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},"date":"2019-05-04 20:32:27.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"757b572c954857ecb2525739fc1bc5e27e41d533","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 22:22:46.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/71/511971/3/check/tempest-full/d337487/ : SUCCESS in 1h 48m 22s\n- tempest-full-py3 http://logs.openstack.org/71/511971/3/check/tempest-full-py3/32f2b66/ : SUCCESS in 1h 49m 14s\n- openstack-tox-lower-constraints http://logs.openstack.org/71/511971/3/check/openstack-tox-lower-constraints/6f9a237/ : SUCCESS in 4m 21s\n- openstack-tox-pep8 http://logs.openstack.org/71/511971/3/check/openstack-tox-pep8/411bc54/ : SUCCESS in 4m 12s\n- openstack-tox-py27 http://logs.openstack.org/71/511971/3/check/openstack-tox-py27/d1ffe98/ : SUCCESS in 5m 09s\n- openstack-tox-py36 http://logs.openstack.org/71/511971/3/check/openstack-tox-py36/9c0cb18/ : SUCCESS in 3m 57s\n- openstack-tox-py37 http://logs.openstack.org/71/511971/3/check/openstack-tox-py37/e246569/ : SUCCESS in 3m 52s\n- openstack-tox-docs http://logs.openstack.org/71/511971/3/check/openstack-tox-docs/fc576ce/html/ : SUCCESS in 4m 13s\n- cliff-tox-py27-neutronclient-tip http://logs.openstack.org/71/511971/3/check/cliff-tox-py27-neutronclient-tip/c3b8ca4/ : SUCCESS in 6m 03s\n- osc-tox-unit-tips http://logs.openstack.org/71/511971/3/check/osc-tox-unit-tips/747e672/ : SUCCESS in 5m 14s","accounts_in_message":[],"_revision_number":3},{"id":"f6d9885441881b9ad1dbf8090c57daa208147bc8","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2019-05-06 20:33:24.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"1a64d442f6eab656ead3554f26af1fc87509d9dd","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-05-11 03:38:53.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"04514795edea943c90d3ceece1ad6486124a18ed","author":{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},"date":"2019-08-05 22:05:56.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"a1081edf31abdb7b180f31f8135c8217a06cd188","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2019-09-13 01:46:54.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"f98547b40bea0934b856993878c66ada18bba241","author":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"date":"2019-09-14 01:15:01.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"9d20f1361b3d5506d9c0aec248f3f2c7e4fd76d1","author":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"date":"2019-09-14 01:29:15.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"11f22b8fa741667225c6f8093039b661bb88bc95","author":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"date":"2019-09-14 12:14:45.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"7589400fea89b236074344781b3bb7a8ab652a39","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-10-02 15:30:32.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nThere\u0027s a TODO here but that can be addressed later","accounts_in_message":[],"_revision_number":3},{"id":"ec268fe0961bf7fc755969cbbbf6e1bfdf0c734c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 15:30:44.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"196c95275ad6e1e2d62eddbae2502e301efd478a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 19:34:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"63a9995a653bec5d4596565bb00ac16ada4cd68b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 19:34:33.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/d45ddd09c26e44d2b1d1cd6a48408f5f : SUCCESS in 1h 38m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4a7bf0cf35a9406e9ce3b58f30b5fe97 : SUCCESS in 1h 36m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f3d36fcddfd3478cb266acfa193ace00 : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89619f32d9064145be9f73a32a003ad7 : SUCCESS in 3m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f463995a567543ec8213a217400896bc : SUCCESS in 3m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a6cf4e0d160b4228aaf024ff9eb19662 : SUCCESS in 3m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/33f8a00f93cf4ef496bce032a7895a78 : SUCCESS in 3m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88c19b682759411c913ebaa88faba4e5 : SUCCESS in 3m 13s\n- cliff-tox-py27-neutronclient-tip https://zuul.opendev.org/t/openstack/build/6ae98dee7753496697b156e135189c60 : SUCCESS in 3m 41s\n- osc-tox-unit-tips https://zuul.opendev.org/t/openstack/build/543446748b1a486795cbb5a145e16f1a : SUCCESS in 4m 52s","accounts_in_message":[],"_revision_number":3},{"id":"5e41f3fcf8fcce99dcb688f01f91cc9782631bee","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 19:35:54.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/421e7d16801646ec948282d8b2d4c05e : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"9d77e2cb386dbb1dac089e5f7508b3c3f56fcd5d","revisions":{"edb96e9940c6c0421fde616d736287d7de4a5394":{"kind":"REWORK","_number":1,"created":"2017-10-13 22:10:52.000000000","uploader":{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},"ref":"refs/changes/71/511971/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/71/511971/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/71/511971/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/71/511971/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/71/511971/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/71/511971/1"}}},"commit":{"parents":[{"commit":"a9719bc23ea3ad90cbad9ab743e678f491e85f0a","subject":"handle more varied top_level.txt files in distributions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/a9719bc23ea3ad90cbad9ab743e678f491e85f0a"}]}],"author":{"name":"Dean Troyer","email":"dtroyer@gmail.com","date":"2017-10-13 21:52:15.000000000","tz":-300},"committer":{"name":"Dean Troyer","email":"dtroyer@gmail.com","date":"2017-10-13 22:08:46.000000000","tz":-300},"subject":"Add a --errexit option to exit the comamnd loop on command errors","message":"Add a --errexit option to exit the comamnd loop on command errors\n\nThis only affects interactive mode.\n\nThe desire is to enable a behaviour similar to the shell\u0027s set +errexit.\nTo do this required changing function returns in a number of places in\norder to propogate the exit code back up to the app.\n\nChange-Id: I1f2606cb43c8064e465e87d6801ed8d169daa26a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/edb96e9940c6c0421fde616d736287d7de4a5394"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/edb96e9940c6c0421fde616d736287d7de4a5394"}]},"branch":"refs/heads/master"},"76a539d1b73fd8d5335ed9ba5aaee4b362800b21":{"kind":"REWORK","_number":2,"created":"2017-10-18 00:08:47.000000000","uploader":{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},"ref":"refs/changes/71/511971/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/71/511971/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/71/511971/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/71/511971/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/71/511971/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/71/511971/2"}}},"commit":{"parents":[{"commit":"a9719bc23ea3ad90cbad9ab743e678f491e85f0a","subject":"handle more varied top_level.txt files in distributions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/a9719bc23ea3ad90cbad9ab743e678f491e85f0a"}]}],"author":{"name":"Dean Troyer","email":"dtroyer@gmail.com","date":"2017-10-13 21:52:15.000000000","tz":-300},"committer":{"name":"Dean Troyer","email":"dtroyer@gmail.com","date":"2017-10-16 22:38:44.000000000","tz":-300},"subject":"Add an errexit attribute to InteractiveApp to exit on command errors","message":"Add an errexit attribute to InteractiveApp to exit on command errors\n\nThis only affects interactive mode.\n\nThe desire is to enable a behaviour similar to the shell\u0027s set +errexit.\nTo do this required changing function returns in a number of places in\norder to propogate the exit code back up to the app.\n\nChange-Id: I1f2606cb43c8064e465e87d6801ed8d169daa26a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/76a539d1b73fd8d5335ed9ba5aaee4b362800b21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/76a539d1b73fd8d5335ed9ba5aaee4b362800b21"}]},"branch":"refs/heads/master"},"9d77e2cb386dbb1dac089e5f7508b3c3f56fcd5d":{"kind":"REWORK","_number":3,"created":"2019-05-04 20:32:27.000000000","uploader":{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},"ref":"refs/changes/71/511971/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/71/511971/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/71/511971/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/71/511971/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/71/511971/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/71/511971/3"}}},"commit":{"parents":[{"commit":"13801a59d3ce4256b4e83a7d907a48502d5e9bb8","subject":"Merge \"Updates for OpenDev transition\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/13801a59d3ce4256b4e83a7d907a48502d5e9bb8"}]}],"author":{"name":"Dean Troyer","email":"dtroyer@gmail.com","date":"2017-10-13 21:52:15.000000000","tz":-300},"committer":{"name":"Dean Troyer","email":"dtroyer@gmail.com","date":"2019-05-04 20:32:10.000000000","tz":-300},"subject":"Add an errexit attribute to InteractiveApp to exit on command errors","message":"Add an errexit attribute to InteractiveApp to exit on command errors\n\nThis only affects interactive mode.\n\nThe desire is to enable a behaviour similar to the shell\u0027s set +errexit.\nTo do this required changing function returns in a number of places in\norder to propogate the exit code back up to the app.\n\nChange-Id: I1f2606cb43c8064e465e87d6801ed8d169daa26a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/9d77e2cb386dbb1dac089e5f7508b3c3f56fcd5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/9d77e2cb386dbb1dac089e5f7508b3c3f56fcd5d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
