)]}'
{"id":"openstack%2Fcliff~552970","triplet_id":"openstack%2Fcliff~master~I713eab2bd9f949da01c03b65ff16a01de92e3e62","project":"openstack/cliff","branch":"master","hashtags":[],"change_id":"I713eab2bd9f949da01c03b65ff16a01de92e3e62","subject":"Allow finding command by partial name","status":"MERGED","created":"2018-03-14 16:35:56.000000000","updated":"2020-01-07 17:12:54.000000000","submitted":"2019-10-02 19:36:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"552970-1570045008037-aecd3d01","meta_rev_id":"e5072161053ae824e5eb05fce2dbf53d5b1ffd9f","_number":552970,"virtual_id_number":552970,"owner":{"_account_id":21813,"name":"Andrey Volkov","email":"m@amadev.ru","username":"avolkov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21813,"name":"Andrey Volkov","email":"m@amadev.ru","username":"avolkov"},{"value":0,"_account_id":27702,"name":"wanglei","email":"wanglei@awcloud.com","username":"lauanchpadid"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-02 19:36:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27997,"name":"Huifeng Qin","email":"huifeng_qin@163.com","username":"Darcy"},{"value":0,"_account_id":18446,"name":"vkulanov","email":"vitaliy@kulanov.org.ua","username":"vkulanov"},{"value":0,"date":"2018-03-26 04:45:37.000000000","_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"recommended":{"_account_id":27702,"name":"wanglei","email":"wanglei@awcloud.com","username":"lauanchpadid"},"all":[{"value":0,"_account_id":21813,"name":"Andrey Volkov","email":"m@amadev.ru","username":"avolkov"},{"value":1,"date":"2018-03-26 05:21:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27702,"name":"wanglei","email":"wanglei@awcloud.com","username":"lauanchpadid"},{"value":2,"date":"2019-10-02 15:31:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2018-03-20 14:36:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2018-03-23 13:49:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27997,"name":"Huifeng Qin","email":"huifeng_qin@163.com","username":"Darcy"},{"value":1,"date":"2018-04-06 04:08:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":18446,"name":"vkulanov","email":"vitaliy@kulanov.org.ua","username":"vkulanov"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":21813,"name":"Andrey Volkov","email":"m@amadev.ru","username":"avolkov"},{"value":0,"_account_id":27702,"name":"wanglei","email":"wanglei@awcloud.com","username":"lauanchpadid"},{"value":1,"date":"2019-10-02 15:31:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"date":"2019-05-02 21:20:01.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27997,"name":"Huifeng Qin","email":"huifeng_qin@163.com","username":"Darcy"},{"value":0,"_account_id":18446,"name":"vkulanov","email":"vitaliy@kulanov.org.ua","username":"vkulanov"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":18446,"name":"vkulanov","email":"vitaliy@kulanov.org.ua","username":"vkulanov"},{"_account_id":21813,"name":"Andrey Volkov","email":"m@amadev.ru","username":"avolkov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":27702,"name":"wanglei","email":"wanglei@awcloud.com","username":"lauanchpadid"},{"_account_id":27997,"name":"Huifeng Qin","email":"huifeng_qin@163.com","username":"Darcy"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-03-20 14:36:31.000000000","updated_by":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"reviewer":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"state":"REVIEWER"},{"updated":"2018-03-23 13:49:53.000000000","updated_by":{"_account_id":27997,"name":"Huifeng Qin","email":"huifeng_qin@163.com","username":"Darcy"},"reviewer":{"_account_id":27997,"name":"Huifeng Qin","email":"huifeng_qin@163.com","username":"Darcy"},"state":"REVIEWER"},{"updated":"2018-03-26 04:45:37.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2018-03-26 05:21:04.000000000","updated_by":{"_account_id":27702,"name":"wanglei","email":"wanglei@awcloud.com","username":"lauanchpadid"},"reviewer":{"_account_id":27702,"name":"wanglei","email":"wanglei@awcloud.com","username":"lauanchpadid"},"state":"REVIEWER"},{"updated":"2018-04-06 04:08:03.000000000","updated_by":{"_account_id":18446,"name":"vkulanov","email":"vitaliy@kulanov.org.ua","username":"vkulanov"},"reviewer":{"_account_id":18446,"name":"vkulanov","email":"vitaliy@kulanov.org.ua","username":"vkulanov"},"state":"REVIEWER"},{"updated":"2019-05-02 21:20:01.000000000","updated_by":{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},"reviewer":{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},"state":"REVIEWER"},{"updated":"2019-10-02 15:31:59.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-10-02 19:36:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fba20d73208b1f85d3ab6a0b99c0d0963dcabe1a","author":{"_account_id":21813,"name":"Andrey Volkov","email":"m@amadev.ru","username":"avolkov"},"date":"2018-03-14 16:35:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5f88167ffd2b063be11155c5d8e58028a24a9818","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2018-03-14 17:15:04.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"d19cafa9210939ae6aa4e0065a2457b7709dd5d3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-14 18:15:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/70/552970/1/check/openstack-tox-pep8/8d8bfa8/ : SUCCESS in 3m 18s\n- openstack-tox-py27 http://logs.openstack.org/70/552970/1/check/openstack-tox-py27/7a1773f/ : SUCCESS in 3m 15s\n- openstack-tox-py35 http://logs.openstack.org/70/552970/1/check/openstack-tox-py35/26ecc99/ : SUCCESS in 3m 47s\n- openstack-tox-pypy http://logs.openstack.org/70/552970/1/check/openstack-tox-pypy/6443e73/ : SUCCESS in 4m 21s (non-voting)\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/70/552970/1/check/legacy-tempest-dsvm-neutron-src/d2c1bdd/ : SUCCESS in 1h 20m 28s\n- build-openstack-sphinx-docs http://logs.openstack.org/70/552970/1/check/build-openstack-sphinx-docs/14c87ad/html/ : SUCCESS in 3m 34s\n- cliff-tox-py27-neutronclient-tip http://logs.openstack.org/70/552970/1/check/cliff-tox-py27-neutronclient-tip/78b47aa/ : SUCCESS in 4m 28s\n- osc-tox-unit-tips http://logs.openstack.org/70/552970/1/check/osc-tox-unit-tips/fcd699e/ : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":1},{"id":"16c1ad6397332e00d8a7c9803f5467c704ff2045","author":{"_account_id":21813,"name":"Andrey Volkov","email":"m@amadev.ru","username":"avolkov"},"date":"2018-03-15 08:35:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e65bf41cfff71fc071af866aa6c46b9e70e928b8","author":{"_account_id":21813,"name":"Andrey Volkov","email":"m@amadev.ru","username":"avolkov"},"date":"2018-03-15 08:40:43.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"6033be7f1ca172139f13d4ee10e3e1ec36df9fda","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-15 10:05:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/70/552970/2/check/openstack-tox-pep8/10c7d6a/ : SUCCESS in 4m 07s\n- openstack-tox-py27 http://logs.openstack.org/70/552970/2/check/openstack-tox-py27/1302997/ : SUCCESS in 4m 17s\n- openstack-tox-py35 http://logs.openstack.org/70/552970/2/check/openstack-tox-py35/d481b51/ : SUCCESS in 4m 25s\n- openstack-tox-pypy http://logs.openstack.org/70/552970/2/check/openstack-tox-pypy/ef8bce9/ : SUCCESS in 4m 28s (non-voting)\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/70/552970/2/check/legacy-tempest-dsvm-neutron-src/c73c548/ : SUCCESS in 1h 24m 47s\n- build-openstack-sphinx-docs http://logs.openstack.org/70/552970/2/check/build-openstack-sphinx-docs/2db3432/html/ : SUCCESS in 4m 23s\n- cliff-tox-py27-neutronclient-tip http://logs.openstack.org/70/552970/2/check/cliff-tox-py27-neutronclient-tip/ae35572/ : SUCCESS in 4m 31s\n- osc-tox-unit-tips http://logs.openstack.org/70/552970/2/check/osc-tox-unit-tips/8c0e89c/ : SUCCESS in 5m 41s","accounts_in_message":[],"_revision_number":2},{"id":"206d7bc2de5304d472c3bbafa7b866f230a8c30c","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2018-03-15 14:38:29.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"65e4b2c637baf4715d024022ebc63d0e61098b48","author":{"_account_id":21813,"name":"Andrey Volkov","email":"m@amadev.ru","username":"avolkov"},"date":"2018-03-15 14:49:21.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ec91c2a0c7f219c0e5b005d9085967ee318c40b2","author":{"_account_id":21813,"name":"Andrey Volkov","email":"m@amadev.ru","username":"avolkov"},"date":"2018-03-15 14:51:11.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"7d7b8e351f5110b26ee684c90bb9d2bc09a3e401","author":{"_account_id":21813,"name":"Andrey Volkov","email":"m@amadev.ru","username":"avolkov"},"date":"2018-03-15 14:52:24.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"275123615589eddb200205bba26c9dca9e6960cb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-15 16:22:35.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/70/552970/4/check/openstack-tox-pep8/640b046/ : SUCCESS in 3m 25s\n- openstack-tox-py27 http://logs.openstack.org/70/552970/4/check/openstack-tox-py27/8828e9a/ : SUCCESS in 3m 28s\n- openstack-tox-py35 http://logs.openstack.org/70/552970/4/check/openstack-tox-py35/0b65f79/ : SUCCESS in 3m 42s\n- openstack-tox-pypy http://logs.openstack.org/70/552970/4/check/openstack-tox-pypy/c7750bf/ : SUCCESS in 4m 09s (non-voting)\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/70/552970/4/check/legacy-tempest-dsvm-neutron-src/4660fcd/ : SUCCESS in 1h 18m 44s\n- build-openstack-sphinx-docs http://logs.openstack.org/70/552970/4/check/build-openstack-sphinx-docs/53fb8d8/html/ : SUCCESS in 3m 30s\n- cliff-tox-py27-neutronclient-tip http://logs.openstack.org/70/552970/4/check/cliff-tox-py27-neutronclient-tip/1107512/ : SUCCESS in 5m 03s\n- osc-tox-unit-tips http://logs.openstack.org/70/552970/4/check/osc-tox-unit-tips/dbbcaae/ : SUCCESS in 5m 04s","accounts_in_message":[],"_revision_number":4},{"id":"0449680b9540d32a52259c8f81dfdbaaf0e4e52e","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2018-03-20 14:36:31.000000000","message":"Patch Set 4: Code-Review+2\n\nThis could be done more efficiently, but this seems OK as a first pass.\n\nI would like to add some sort of tree structure to the command search logic to speed that up in general. We could make the prefix searching part of that tree structure.","accounts_in_message":[],"_revision_number":4},{"id":"4040b415b2592496e4a182a3973ea14d35815ab8","author":{"_account_id":27997,"name":"Huifeng Qin","email":"huifeng_qin@163.com","username":"Darcy"},"date":"2018-03-23 13:49:53.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"6b56583f5f72f49d289c06da4c0e6de95848b110","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2018-03-26 04:45:37.000000000","message":"Patch Set 4:\n\nI like this, but may cause some frustration if there are namespace conflicts with plugins.","accounts_in_message":[],"_revision_number":4},{"id":"3666391bb13318a8e154334c8c87a7718d836185","author":{"_account_id":27702,"name":"wanglei","email":"wanglei@awcloud.com","username":"lauanchpadid"},"date":"2018-03-26 05:21:04.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"3485168903df57397d86c3acfe6bc383b0dff846","author":{"_account_id":18446,"name":"vkulanov","email":"vitaliy@kulanov.org.ua","username":"vkulanov"},"date":"2018-04-06 04:08:03.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"516fe6e4e573eff47121c423032f052c76b5b480","author":{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},"date":"2019-05-02 21:20:01.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"fffa78d2ce45f88bbe241af36e03e60d095c42fa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-02 23:17:04.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/70/552970/4/check/tempest-full/46f8642/ : SUCCESS in 1h 51m 25s\n- tempest-full-py3 http://logs.openstack.org/70/552970/4/check/tempest-full-py3/42b3a1b/ : SUCCESS in 1h 41m 02s\n- openstack-tox-lower-constraints http://logs.openstack.org/70/552970/4/check/openstack-tox-lower-constraints/b1677c5/ : SUCCESS in 4m 44s\n- openstack-tox-pep8 http://logs.openstack.org/70/552970/4/check/openstack-tox-pep8/fbe79bd/ : SUCCESS in 3m 49s\n- openstack-tox-py27 http://logs.openstack.org/70/552970/4/check/openstack-tox-py27/9980ba4/ : SUCCESS in 3m 46s\n- openstack-tox-py35 http://logs.openstack.org/70/552970/4/check/openstack-tox-py35/676fee6/ : SUCCESS in 4m 59s\n- openstack-tox-py36 http://logs.openstack.org/70/552970/4/check/openstack-tox-py36/532a932/ : SUCCESS in 6m 00s\n- openstack-tox-py37 http://logs.openstack.org/70/552970/4/check/openstack-tox-py37/3aa4d21/ : SUCCESS in 4m 16s\n- openstack-tox-docs http://logs.openstack.org/70/552970/4/check/openstack-tox-docs/113ac56/html/ : SUCCESS in 4m 15s\n- cliff-tox-py27-neutronclient-tip http://logs.openstack.org/70/552970/4/check/cliff-tox-py27-neutronclient-tip/f7f1a2d/ : SUCCESS in 4m 17s\n- osc-tox-unit-tips http://logs.openstack.org/70/552970/4/check/osc-tox-unit-tips/c4d703f/ : SUCCESS in 7m 40s","accounts_in_message":[],"_revision_number":4},{"id":"cc56c455da953e5f34abb8ea22421d90ed2ba94a","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-10-02 15:31:59.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"8e7b1276761e22a3c79d56a851120bae62dd9b50","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 15:32:11.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"9c09208196b122cc81a8093af9441b4a056a782c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 19:36:47.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/9991fb0a535640e1890f432fe9c3a024 : SUCCESS in 1h 36m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f6ea500450674d3fb25414e75d2f4a9d : SUCCESS in 1h 37m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f57386af292d435485fbf1c206e54dcc : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/193e92da4f9b44839f97574e1dc66edf : SUCCESS in 2m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/632c2c34f1384dfdab93574fb1cd434f : SUCCESS in 3m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/20451b33fdd14563b07b0d8904e8ec63 : SUCCESS in 3m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fc5e562f1cb14cc9a0a230d0f937d87b : SUCCESS in 3m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c68eb7131d64699882720f72d7342ee : SUCCESS in 4m 48s\n- cliff-tox-py27-neutronclient-tip https://zuul.opendev.org/t/openstack/build/842d05626e284fc6a8676b5c8bbd96f5 : SUCCESS in 3m 48s\n- osc-tox-unit-tips https://zuul.opendev.org/t/openstack/build/2dbb1e1aec0d4488adec17901962db32 : SUCCESS in 6m 24s","accounts_in_message":[],"_revision_number":4},{"id":"e785af3834fc80ad5c8e7656c8ccf33aadb8cf25","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 19:36:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"32d93b36d855d2e9605e01c8230beecaa3c2d59f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 19:37:56.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98b863e245584e7fabbabe9fdf007aae : SUCCESS in 48s","accounts_in_message":[],"_revision_number":4},{"id":"5b1f062047306e2b62287d2b8e0f501e6cc4b126","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-01-03 17:15:28.000000000","message":"Patch Set 4:\n\nThis has somehow broken python-openstackclient doc builds. Version 2.17.0 of cliff will need to be held back until the issue has been figured out. Not apparent to me how this is causing the problem.","accounts_in_message":[],"_revision_number":4},{"id":"e5072161053ae824e5eb05fce2dbf53d5b1ffd9f","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-01-07 17:12:54.000000000","message":"Patch Set 4:\n\n\u003e This has somehow broken python-openstackclient doc builds. Version\n \u003e 2.17.0 of cliff will need to be held back until the issue has been\n \u003e figured out. Not apparent to me how this is causing the problem.\n\nAs noted at [1], it\u0027s not this specifically that\u0027s broken things but rather the fact that it\u0027s this commit, rather than a commit on master after it had been merged, that\u0027s been tagged as 2.17.0. This means 2.17.0 is version 2.11.0 + some commits instead of 2.16.0 plus some commits.","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"8bcd068e876ddd48ae61c1803449d666f5e28ba0","revisions":{"c4d8f3f0d2912b86c3f353d7accc8deb22416002":{"kind":"REWORK","_number":1,"created":"2018-03-14 16:35:56.000000000","uploader":{"_account_id":21813,"name":"Andrey Volkov","email":"m@amadev.ru","username":"avolkov"},"ref":"refs/changes/70/552970/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/70/552970/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/70/552970/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/70/552970/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/70/552970/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/70/552970/1"}}},"commit":{"parents":[{"commit":"fa5303f058cdd1bc90a86d2dc2fff70a9cbd5de1","subject":"Merge \"Fix cmd2 doc URL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/fa5303f058cdd1bc90a86d2dc2fff70a9cbd5de1"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2018-03-14 16:30:06.000000000","tz":180},"committer":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2018-03-14 16:35:54.000000000","tz":180},"subject":"Allow finding command by partial name","message":"Allow finding command by partial name\n\nThis small improvement helps to decrease the amount of typing.\n\n$ openstack resource provider list\n...\n^ too long?\n\n$ alias os\u003dopenstack\n$ export CLIFF_PARTIAL_NAME\u003d1\n$ os r p l\n...\n^ much better!\n\nChange-Id: I713eab2bd9f949da01c03b65ff16a01de92e3e62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/c4d8f3f0d2912b86c3f353d7accc8deb22416002"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/c4d8f3f0d2912b86c3f353d7accc8deb22416002"}]},"branch":"refs/heads/master"},"977c1b0627ec2cd9c65b3443ed18d20090ece011":{"kind":"REWORK","_number":2,"created":"2018-03-15 08:35:35.000000000","uploader":{"_account_id":21813,"name":"Andrey Volkov","email":"m@amadev.ru","username":"avolkov"},"ref":"refs/changes/70/552970/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/70/552970/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/70/552970/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/70/552970/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/70/552970/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/70/552970/2"}}},"commit":{"parents":[{"commit":"fa5303f058cdd1bc90a86d2dc2fff70a9cbd5de1","subject":"Merge \"Fix cmd2 doc URL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/fa5303f058cdd1bc90a86d2dc2fff70a9cbd5de1"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2018-03-14 16:30:06.000000000","tz":180},"committer":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2018-03-15 08:33:20.000000000","tz":180},"subject":"Allow finding command by partial name","message":"Allow finding command by partial name\n\nThis small improvement helps to decrease the amount of typing.\n\n$ openstack resource provider list\n...\n^ too long?\n\n$ alias os\u003dopenstack\n$ os r p l\n...\n^ much better!\n\nChange-Id: I713eab2bd9f949da01c03b65ff16a01de92e3e62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/977c1b0627ec2cd9c65b3443ed18d20090ece011"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/977c1b0627ec2cd9c65b3443ed18d20090ece011"}]},"branch":"refs/heads/master"},"fc112a54be8d0dd3791f3baa04888caf2ccd1a3a":{"kind":"REWORK","_number":3,"created":"2018-03-15 14:49:21.000000000","uploader":{"_account_id":21813,"name":"Andrey Volkov","email":"m@amadev.ru","username":"avolkov"},"ref":"refs/changes/70/552970/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/70/552970/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/70/552970/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/70/552970/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/70/552970/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/70/552970/3"}}},"commit":{"parents":[{"commit":"fa5303f058cdd1bc90a86d2dc2fff70a9cbd5de1","subject":"Merge \"Fix cmd2 doc URL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/fa5303f058cdd1bc90a86d2dc2fff70a9cbd5de1"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2018-03-14 16:30:06.000000000","tz":180},"committer":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2018-03-15 14:48:12.000000000","tz":180},"subject":"Allow finding command by partial name","message":"Allow finding command by partial name\n\nThis small improvement helps to decrease the amount of typing.\n\n$ openstack resource provider list\n...\n^ too long?\n\n$ alias os\u003dopenstack\n$ os r p l\n...\n^ much better!\n\nChange-Id: I713eab2bd9f949da01c03b65ff16a01de92e3e62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/fc112a54be8d0dd3791f3baa04888caf2ccd1a3a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/fc112a54be8d0dd3791f3baa04888caf2ccd1a3a"}]},"branch":"refs/heads/master"},"8bcd068e876ddd48ae61c1803449d666f5e28ba0":{"kind":"REWORK","_number":4,"created":"2018-03-15 14:51:11.000000000","uploader":{"_account_id":21813,"name":"Andrey Volkov","email":"m@amadev.ru","username":"avolkov"},"ref":"refs/changes/70/552970/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/70/552970/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/70/552970/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/70/552970/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/70/552970/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/70/552970/4"}}},"commit":{"parents":[{"commit":"fa5303f058cdd1bc90a86d2dc2fff70a9cbd5de1","subject":"Merge \"Fix cmd2 doc URL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/fa5303f058cdd1bc90a86d2dc2fff70a9cbd5de1"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2018-03-14 16:30:06.000000000","tz":180},"committer":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2018-03-15 14:50:11.000000000","tz":180},"subject":"Allow finding command by partial name","message":"Allow finding command by partial name\n\nThis small improvement helps to decrease the amount of typing.\n\n$ openstack resource provider list\n...\n^ too long?\n\n$ alias os\u003dopenstack\n$ os r p l\n...\n^ much better!\n\nChange-Id: I713eab2bd9f949da01c03b65ff16a01de92e3e62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/8bcd068e876ddd48ae61c1803449d666f5e28ba0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/8bcd068e876ddd48ae61c1803449d666f5e28ba0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
