)]}'
{"id":"openstack%2Fcliff~581564","triplet_id":"openstack%2Fcliff~master~I09aa6a330e988070ad5b3a20b56b1c301386d657","project":"openstack/cliff","branch":"master","topic":"stestr","hashtags":[],"change_id":"I09aa6a330e988070ad5b3a20b56b1c301386d657","subject":"Switch to stestr","status":"ABANDONED","created":"2018-07-11 04:28:58.000000000","updated":"2020-06-09 15:40:22.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fef39a991251133c3cd1d32dd5b158e8b5e8a0eb","_number":581564,"virtual_id_number":581564,"owner":{"_account_id":25903,"name":"Vu Cong Tuan","email":"tuanvc@vn.fujitsu.com","username":"tuanvu"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-10-02 17:38:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28956,"name":"Nguyen Hai Truong","email":"truongnh@vn.fujitsu.com","username":"truongnh"},{"_account_id":25903,"name":"Vu Cong Tuan","email":"tuanvc@vn.fujitsu.com","username":"tuanvu"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"date":"2019-10-02 15:43:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28956,"name":"Nguyen Hai Truong","email":"truongnh@vn.fujitsu.com","username":"truongnh"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25903,"name":"Vu Cong Tuan","email":"tuanvc@vn.fujitsu.com","username":"tuanvu"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28956,"name":"Nguyen Hai Truong","email":"truongnh@vn.fujitsu.com","username":"truongnh"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":25903,"name":"Vu Cong Tuan","email":"tuanvc@vn.fujitsu.com","username":"tuanvu"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25903,"name":"Vu Cong Tuan","email":"tuanvc@vn.fujitsu.com","username":"tuanvu"},{"_account_id":28956,"name":"Nguyen Hai Truong","email":"truongnh@vn.fujitsu.com","username":"truongnh"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-11-02 09:44:00.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2018-11-05 06:15:54.000000000","updated_by":{"_account_id":28956,"name":"Nguyen Hai Truong","email":"truongnh@vn.fujitsu.com","username":"truongnh"},"reviewer":{"_account_id":28956,"name":"Nguyen Hai Truong","email":"truongnh@vn.fujitsu.com","username":"truongnh"},"state":"REVIEWER"},{"updated":"2018-11-05 13:45:34.000000000","updated_by":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"reviewer":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"state":"REVIEWER"},{"updated":"2019-10-02 15:43:38.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-10-02 17:38:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"431a5ad9c23c6da589e7c35b18f780ed2653e34f","author":{"_account_id":25903,"name":"Vu Cong Tuan","email":"tuanvc@vn.fujitsu.com","username":"tuanvu"},"date":"2018-07-11 04:28:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"684f7207fc4e461430cae18aeaae9e02df094c0e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-11 06:17:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/64/581564/1/check/openstack-tox-pep8/6dbaf24/ : SUCCESS in 5m 00s\n- openstack-tox-py27 http://logs.openstack.org/64/581564/1/check/openstack-tox-py27/1143492/ : FAILURE in 3m 59s\n- openstack-tox-py35 http://logs.openstack.org/64/581564/1/check/openstack-tox-py35/33df14f/ : FAILURE in 4m 23s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/64/581564/1/check/legacy-tempest-dsvm-neutron-src/e46ffa9/ : SUCCESS in 1h 42m 31s\n- build-openstack-sphinx-docs http://logs.openstack.org/64/581564/1/check/build-openstack-sphinx-docs/7149f2e/html/ : SUCCESS in 3m 49s\n- requirements-check http://logs.openstack.org/64/581564/1/check/requirements-check/596227a/ : FAILURE in 2m 50s\n- cliff-tox-py27-neutronclient-tip http://logs.openstack.org/64/581564/1/check/cliff-tox-py27-neutronclient-tip/037641c/ : SUCCESS in 5m 14s\n- osc-tox-unit-tips http://logs.openstack.org/64/581564/1/check/osc-tox-unit-tips/ed1aa1c/ : SUCCESS in 6m 50s\n- openstack-tox-lower-constraints http://logs.openstack.org/64/581564/1/check/openstack-tox-lower-constraints/1dd67b7/ : FAILURE in 4m 52s","accounts_in_message":[],"_revision_number":1},{"id":"fd63f0364eb8670182f11795cc6a048fe1245486","author":{"_account_id":25903,"name":"Vu Cong Tuan","email":"tuanvc@vn.fujitsu.com","username":"tuanvu"},"date":"2018-07-11 06:31:44.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e9405578f0ae2f74b5455490449af5e110e4e503","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-11 08:02:32.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/64/581564/2/check/openstack-tox-pep8/340e3b3/ : SUCCESS in 4m 39s\n- openstack-tox-py27 http://logs.openstack.org/64/581564/2/check/openstack-tox-py27/e278e86/ : SUCCESS in 3m 46s\n- openstack-tox-py35 http://logs.openstack.org/64/581564/2/check/openstack-tox-py35/ddabac9/ : SUCCESS in 4m 01s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/64/581564/2/check/legacy-tempest-dsvm-neutron-src/0b8fa8b/ : SUCCESS in 1h 24m 11s\n- build-openstack-sphinx-docs http://logs.openstack.org/64/581564/2/check/build-openstack-sphinx-docs/ac8bafb/html/ : SUCCESS in 4m 34s\n- requirements-check http://logs.openstack.org/64/581564/2/check/requirements-check/46866bc/ : FAILURE in 3m 09s\n- cliff-tox-py27-neutronclient-tip http://logs.openstack.org/64/581564/2/check/cliff-tox-py27-neutronclient-tip/4a4659e/ : SUCCESS in 5m 05s\n- osc-tox-unit-tips http://logs.openstack.org/64/581564/2/check/osc-tox-unit-tips/087be0c/ : SUCCESS in 6m 47s\n- openstack-tox-lower-constraints http://logs.openstack.org/64/581564/2/check/openstack-tox-lower-constraints/4d0447d/ : SUCCESS in 5m 05s","accounts_in_message":[],"_revision_number":2},{"id":"9fd09313bc0993b198f1211d4f8f4d03ba8359d1","author":{"_account_id":25903,"name":"Vu Cong Tuan","email":"tuanvc@vn.fujitsu.com","username":"tuanvu"},"date":"2018-07-11 08:06:23.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"af704f3df88ec6c25326e0b90a68224f74589f69","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-11 09:55:15.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/64/581564/3/check/openstack-tox-pep8/c1eebbb/ : SUCCESS in 4m 08s\n- openstack-tox-py27 http://logs.openstack.org/64/581564/3/check/openstack-tox-py27/379f4e2/ : SUCCESS in 3m 43s\n- openstack-tox-py35 http://logs.openstack.org/64/581564/3/check/openstack-tox-py35/22c83a1/ : SUCCESS in 3m 57s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/64/581564/3/check/legacy-tempest-dsvm-neutron-src/b7c33c1/ : SUCCESS in 1h 36m 16s\n- build-openstack-sphinx-docs http://logs.openstack.org/64/581564/3/check/build-openstack-sphinx-docs/df5eb9d/html/ : SUCCESS in 4m 08s\n- requirements-check http://logs.openstack.org/64/581564/3/check/requirements-check/8047495/ : SUCCESS in 2m 59s\n- cliff-tox-py27-neutronclient-tip http://logs.openstack.org/64/581564/3/check/cliff-tox-py27-neutronclient-tip/3715e90/ : SUCCESS in 4m 36s\n- osc-tox-unit-tips http://logs.openstack.org/64/581564/3/check/osc-tox-unit-tips/878653d/ : SUCCESS in 5m 44s\n- openstack-tox-lower-constraints http://logs.openstack.org/64/581564/3/check/openstack-tox-lower-constraints/ce95137/ : SUCCESS in 5m 34s","accounts_in_message":[],"_revision_number":3},{"id":"fd32fa46e238dcbb71010c2af9cd62f275dea2cf","author":{"_account_id":25903,"name":"Vu Cong Tuan","email":"tuanvc@vn.fujitsu.com","username":"tuanvu"},"date":"2018-07-11 10:04:02.000000000","message":"Patch Set 3:\n\nReady for review","accounts_in_message":[],"_revision_number":3},{"id":"62f0b2172334a5c970c8f5ac6b774066c9d938a4","author":{"_account_id":25903,"name":"Vu Cong Tuan","email":"tuanvc@vn.fujitsu.com","username":"tuanvu"},"date":"2018-11-02 09:44:07.000000000","message":"Patch Set 3:\n\nWaiting for review","accounts_in_message":[],"_revision_number":3},{"id":"7ee10966aed2f2c70321fe1479039299a449a3cb","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2018-11-02 14:26:32.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"cc0d3a090e109b0d34fe7040cfa8a4eec1de8674","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-02 14:26:47.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f5e6e9a37f839476ab84a9a65f5ee5c922eca662","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-02 16:13:11.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/64/581564/3/gate/openstack-tox-pep8/40a4468/ : SUCCESS in 4m 45s\n- openstack-tox-py27 http://logs.openstack.org/64/581564/3/gate/openstack-tox-py27/a0270d5/ : SUCCESS in 6m 10s\n- openstack-tox-py35 http://logs.openstack.org/64/581564/3/gate/openstack-tox-py35/0f13507/ : SUCCESS in 5m 08s\n- openstack-tox-py36 http://logs.openstack.org/64/581564/3/gate/openstack-tox-py36/7907d85/ : SUCCESS in 7m 18s\n- tempest-full http://logs.openstack.org/64/581564/3/gate/tempest-full/09c7d40/ : SUCCESS in 1h 43m 25s\n- tempest-full-py3 http://logs.openstack.org/64/581564/3/gate/tempest-full-py3/f541424/ : SUCCESS in 1h 12m 19s\n- openstack-tox-docs http://logs.openstack.org/64/581564/3/gate/openstack-tox-docs/a6f556c/html/ : SUCCESS in 5m 40s\n- requirements-check http://logs.openstack.org/64/581564/3/gate/requirements-check/d4f3b1c/ : SUCCESS in 4m 38s\n- cliff-tox-py27-neutronclient-tip http://logs.openstack.org/64/581564/3/gate/cliff-tox-py27-neutronclient-tip/d703d0c/ : FAILURE in 4m 41s\n- osc-tox-unit-tips http://logs.openstack.org/64/581564/3/gate/osc-tox-unit-tips/43ac299/ : SUCCESS in 6m 01s\n- openstack-tox-lower-constraints http://logs.openstack.org/64/581564/3/gate/openstack-tox-lower-constraints/109f0a0/ : SUCCESS in 7m 18s","accounts_in_message":[],"_revision_number":3},{"id":"d990eaf8e3d80aa5f0c86da388cae9ca155dd47b","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2018-11-02 18:33:59.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"79243a62f6a7f6ba604917f84dfa5b216f1d04ba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-02 20:18:58.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/64/581564/3/check/openstack-tox-pep8/5c4bd74/ : SUCCESS in 6m 04s\n- openstack-tox-py27 http://logs.openstack.org/64/581564/3/check/openstack-tox-py27/66b7877/ : SUCCESS in 6m 18s\n- openstack-tox-py35 http://logs.openstack.org/64/581564/3/check/openstack-tox-py35/8df9dad/ : SUCCESS in 6m 25s\n- openstack-tox-py36 http://logs.openstack.org/64/581564/3/check/openstack-tox-py36/2e55fef/ : SUCCESS in 7m 46s\n- tempest-full http://logs.openstack.org/64/581564/3/check/tempest-full/149e8f6/ : SUCCESS in 1h 26m 48s\n- tempest-full-py3 http://logs.openstack.org/64/581564/3/check/tempest-full-py3/74f659a/ : SUCCESS in 1h 35m 26s\n- openstack-tox-docs http://logs.openstack.org/64/581564/3/check/openstack-tox-docs/fa9dd0d/html/ : SUCCESS in 5m 43s\n- requirements-check http://logs.openstack.org/64/581564/3/check/requirements-check/4dae53f/ : SUCCESS in 6m 52s\n- cliff-tox-py27-neutronclient-tip http://logs.openstack.org/64/581564/3/check/cliff-tox-py27-neutronclient-tip/afe7934/ : FAILURE in 6m 48s\n- osc-tox-unit-tips http://logs.openstack.org/64/581564/3/check/osc-tox-unit-tips/3215203/ : SUCCESS in 7m 27s\n- openstack-tox-lower-constraints http://logs.openstack.org/64/581564/3/check/openstack-tox-lower-constraints/7af0bc5/ : SUCCESS in 7m 06s","accounts_in_message":[],"_revision_number":3},{"id":"ea6d3e3ae4b68c486a1983ab98ac7c388110b42f","author":{"_account_id":25903,"name":"Vu Cong Tuan","email":"tuanvc@vn.fujitsu.com","username":"tuanvu"},"date":"2018-11-05 03:03:19.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"d9a1c434dc0b681a008f8df7481eb3aaa73600c6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-05 04:47:36.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/64/581564/3/check/openstack-tox-pep8/b759d88/ : SUCCESS in 4m 17s\n- openstack-tox-py27 http://logs.openstack.org/64/581564/3/check/openstack-tox-py27/a9bbdad/ : SUCCESS in 3m 40s\n- openstack-tox-py35 http://logs.openstack.org/64/581564/3/check/openstack-tox-py35/1af9748/ : SUCCESS in 5m 17s\n- openstack-tox-py36 http://logs.openstack.org/64/581564/3/check/openstack-tox-py36/a9bc4f5/ : SUCCESS in 3m 40s\n- tempest-full http://logs.openstack.org/64/581564/3/check/tempest-full/a1d3d63/ : SUCCESS in 1h 39m 13s\n- tempest-full-py3 http://logs.openstack.org/64/581564/3/check/tempest-full-py3/504fe25/ : SUCCESS in 1h 30m 38s\n- openstack-tox-docs http://logs.openstack.org/64/581564/3/check/openstack-tox-docs/387e2b7/html/ : SUCCESS in 4m 07s\n- requirements-check http://logs.openstack.org/64/581564/3/check/requirements-check/58523d5/ : SUCCESS in 3m 37s\n- cliff-tox-py27-neutronclient-tip http://logs.openstack.org/64/581564/3/check/cliff-tox-py27-neutronclient-tip/e013889/ : FAILURE in 4m 41s\n- osc-tox-unit-tips http://logs.openstack.org/64/581564/3/check/osc-tox-unit-tips/c438c25/ : SUCCESS in 5m 09s\n- openstack-tox-lower-constraints http://logs.openstack.org/64/581564/3/check/openstack-tox-lower-constraints/6151819/ : SUCCESS in 6m 11s","accounts_in_message":[],"_revision_number":3},{"id":"9ac7da12842ad9cc7b8cc671b4cc90cf5322a2d5","author":{"_account_id":28956,"name":"Nguyen Hai Truong","email":"truongnh@vn.fujitsu.com","username":"truongnh"},"date":"2018-11-05 06:15:54.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"72115ecbbeb4eef2e5fa9018283da2f9c27e664d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-05 08:07:59.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/64/581564/3/check/openstack-tox-pep8/e4a8e03/ : SUCCESS in 4m 40s\n- openstack-tox-py27 http://logs.openstack.org/64/581564/3/check/openstack-tox-py27/4c1ff82/ : SUCCESS in 4m 54s\n- openstack-tox-py35 http://logs.openstack.org/64/581564/3/check/openstack-tox-py35/26de15d/ : SUCCESS in 5m 27s\n- openstack-tox-py36 http://logs.openstack.org/64/581564/3/check/openstack-tox-py36/8012c9a/ : SUCCESS in 5m 14s\n- tempest-full http://logs.openstack.org/64/581564/3/check/tempest-full/98dfe7c/ : SUCCESS in 1h 45m 41s\n- tempest-full-py3 http://logs.openstack.org/64/581564/3/check/tempest-full-py3/a289b1f/ : SUCCESS in 1h 13m 59s\n- openstack-tox-docs http://logs.openstack.org/64/581564/3/check/openstack-tox-docs/0816aa9/html/ : SUCCESS in 5m 03s\n- requirements-check http://logs.openstack.org/64/581564/3/check/requirements-check/f56ea58/ : SUCCESS in 4m 10s\n- cliff-tox-py27-neutronclient-tip http://logs.openstack.org/64/581564/3/check/cliff-tox-py27-neutronclient-tip/8054a32/ : FAILURE in 5m 00s\n- osc-tox-unit-tips http://logs.openstack.org/64/581564/3/check/osc-tox-unit-tips/815dc53/ : SUCCESS in 5m 35s\n- openstack-tox-lower-constraints http://logs.openstack.org/64/581564/3/check/openstack-tox-lower-constraints/f08d9d0/ : SUCCESS in 6m 01s","accounts_in_message":[],"_revision_number":3},{"id":"e635bef41a668398c27e955f147ed81235c64196","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2018-11-05 13:45:34.000000000","message":"Patch Set 3: -Code-Review -Workflow\n\nSince we have the same test failing repeatedly, I think there is something legitimately wrong. I don\u0027t know if it is with the test, the use of stestr, the code here, or with code in the neutron client.","accounts_in_message":[],"_revision_number":3},{"id":"a093cebaf94fc755f33b5ebc80ea290f2cf5ea4c","author":{"_account_id":25903,"name":"Vu Cong Tuan","email":"tuanvc@vn.fujitsu.com","username":"tuanvu"},"date":"2018-11-06 01:54:44.000000000","message":"Patch Set 3:\n\nHi Doug Hellmann,\nThanks for your comment.\nI\u0027ll check for more details.\nWish you a great day ahead :)","accounts_in_message":[],"_revision_number":3},{"id":"13adb3b4910da0490696f10559575ffcdd5cdd9b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-04 15:44:02.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"d8fe19d84ff6a47b3c38796f8fad113952ce38f5","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-04 15:56:35.000000000","message":"Patch Set 4: Code-Review+2\n\nHopefully the issues dhellmann was seeing are resolved now, but we shall see","accounts_in_message":[],"_revision_number":4},{"id":"12a89819e4db4ca338909b5f52fd4ad4cfa78c22","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 17:44:37.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/64/581564/4/check/requirements-check/545bf2a/ : FAILURE in 6m 03s\n- tempest-full http://logs.openstack.org/64/581564/4/check/tempest-full/8213a2a/ : SUCCESS in 1h 57m 11s\n- tempest-full-py3 http://logs.openstack.org/64/581564/4/check/tempest-full-py3/0cef178/ : SUCCESS in 1h 30m 04s\n- openstack-tox-lower-constraints http://logs.openstack.org/64/581564/4/check/openstack-tox-lower-constraints/702596e/ : SUCCESS in 4m 49s\n- openstack-tox-pep8 http://logs.openstack.org/64/581564/4/check/openstack-tox-pep8/82f42b7/ : SUCCESS in 4m 30s\n- openstack-tox-py27 http://logs.openstack.org/64/581564/4/check/openstack-tox-py27/3279fe1/ : SUCCESS in 4m 04s\n- openstack-tox-py36 http://logs.openstack.org/64/581564/4/check/openstack-tox-py36/9438128/ : SUCCESS in 4m 03s\n- openstack-tox-py37 http://logs.openstack.org/64/581564/4/check/openstack-tox-py37/562ae38/ : SUCCESS in 4m 25s\n- openstack-tox-docs http://logs.openstack.org/64/581564/4/check/openstack-tox-docs/7fd56a4/html/ : SUCCESS in 3m 59s\n- cliff-tox-py27-neutronclient-tip http://logs.openstack.org/64/581564/4/check/cliff-tox-py27-neutronclient-tip/42c1f88/ : SUCCESS in 4m 31s\n- osc-tox-unit-tips http://logs.openstack.org/64/581564/4/check/osc-tox-unit-tips/33d497b/ : SUCCESS in 5m 24s","accounts_in_message":[],"_revision_number":4},{"id":"30723703841d4c772a11d545a6f7b33ba6338ffe","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-10-02 15:41:46.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"370266e0a7c9ef47a89de287add0285289d3ca54","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-10-02 15:42:25.000000000","message":"Patch Set 5: Code-Review+2\n\nFixed the trivial merge conflict","accounts_in_message":[],"_revision_number":5},{"id":"40e1c2fadd1ec5b1ce3602ea051fe5c12a243058","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-10-02 15:43:38.000000000","message":"Patch Set 5: -Code-Review\n\nActually, shouldn\u0027t there be an .stestr.conf file here?","accounts_in_message":[],"_revision_number":5},{"id":"6e47abc8848dc21016424343c166083839c6fe1b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 17:38:49.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/478cc526e68c495184b311738181d00f : FAILURE in 2m 24s\n- tempest-full https://zuul.opendev.org/t/openstack/build/0d34a1d2950d45838548c4ad9da33172 : SUCCESS in 1h 40m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4940bc6494c145b28774cd61a7f2a019 : SUCCESS in 1h 41m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/981d89aba4864d0f8100e5a8c37e5400 : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d98bf5cc81e84dae9a45b52ca7f8cb9a : SUCCESS in 3m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/20da9dc6b8b04696950d0022ce87f305 : SUCCESS in 3m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7ea36ea5188b467c93c678c2757d191a : SUCCESS in 3m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8b8e0d5eaaa5482ab74da4cf0f0f808d : SUCCESS in 3m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/994a2937bef442fb9e223856d12086c3 : SUCCESS in 3m 53s\n- cliff-tox-py27-neutronclient-tip https://zuul.opendev.org/t/openstack/build/2eaf77fc85f54c5f9624502af79dbabb : SUCCESS in 4m 44s\n- osc-tox-unit-tips https://zuul.opendev.org/t/openstack/build/2284a5b2dc9642b495a721aee79c8382 : SUCCESS in 4m 13s","accounts_in_message":[],"_revision_number":5},{"id":"bc27d4e2409acd8d8508026fcdf71942db7abfdf","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-06-09 15:40:22.000000000","message":"Abandoned\n\nFixed in a different patch.","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"36d7d1cfc5372a1cd557493964cd71b08b478f75","revisions":{"fad4b8ee55498d8327f3ea4b69c6fabe78bad038":{"kind":"REWORK","_number":1,"created":"2018-07-11 04:28:58.000000000","uploader":{"_account_id":25903,"name":"Vu Cong Tuan","email":"tuanvc@vn.fujitsu.com","username":"tuanvu"},"ref":"refs/changes/64/581564/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/64/581564/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/64/581564/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/64/581564/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/64/581564/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/64/581564/1"}}},"commit":{"parents":[{"commit":"dd6abfffba239ff1be945dbd2b53a402d2757b46","subject":"Merge \"Assure executable name is kept when app is called as module\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/dd6abfffba239ff1be945dbd2b53a402d2757b46"}]}],"author":{"name":"Vu Cong Tuan","email":"tuanvc@vn.fujitsu.com","date":"2018-07-11 04:31:05.000000000","tz":420},"committer":{"name":"Vu Cong Tuan","email":"tuanvc@vn.fujitsu.com","date":"2018-07-11 04:31:05.000000000","tz":420},"subject":"Switch to stestr","message":"Switch to stestr\n\nAccording to Openstack summit session [1],\nstestr is maintained project to which all Openstack projects should migrate.\nLet\u0027s switch to stestr as other projects have already moved to it.\n\n[1] https://etherpad.openstack.org/p/YVR-python-pti\n\nChange-Id: I09aa6a330e988070ad5b3a20b56b1c301386d657\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/fad4b8ee55498d8327f3ea4b69c6fabe78bad038"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/fad4b8ee55498d8327f3ea4b69c6fabe78bad038"}]},"branch":"refs/heads/master"},"ddd4ab9de0500613c3461cec7bb50c652a47533e":{"kind":"REWORK","_number":2,"created":"2018-07-11 06:31:44.000000000","uploader":{"_account_id":25903,"name":"Vu Cong Tuan","email":"tuanvc@vn.fujitsu.com","username":"tuanvu"},"ref":"refs/changes/64/581564/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/64/581564/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/64/581564/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/64/581564/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/64/581564/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/64/581564/2"}}},"commit":{"parents":[{"commit":"dd6abfffba239ff1be945dbd2b53a402d2757b46","subject":"Merge \"Assure executable name is kept when app is called as module\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/dd6abfffba239ff1be945dbd2b53a402d2757b46"}]}],"author":{"name":"Vu Cong Tuan","email":"tuanvc@vn.fujitsu.com","date":"2018-07-11 04:31:05.000000000","tz":420},"committer":{"name":"Vu Cong Tuan","email":"tuanvc@vn.fujitsu.com","date":"2018-07-11 06:34:23.000000000","tz":420},"subject":"Switch to stestr","message":"Switch to stestr\n\nAccording to Openstack summit session [1],\nstestr is maintained project to which all Openstack projects should migrate.\nLet\u0027s switch to stestr as other projects have already moved to it.\n\n[1] https://etherpad.openstack.org/p/YVR-python-pti\n\nChange-Id: I09aa6a330e988070ad5b3a20b56b1c301386d657\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/ddd4ab9de0500613c3461cec7bb50c652a47533e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/ddd4ab9de0500613c3461cec7bb50c652a47533e"}]},"branch":"refs/heads/master"},"b964896fc0d9489a804152058c40ded6960efab4":{"kind":"REWORK","_number":3,"created":"2018-07-11 08:06:23.000000000","uploader":{"_account_id":25903,"name":"Vu Cong Tuan","email":"tuanvc@vn.fujitsu.com","username":"tuanvu"},"ref":"refs/changes/64/581564/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/64/581564/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/64/581564/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/64/581564/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/64/581564/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/64/581564/3"}}},"commit":{"parents":[{"commit":"dd6abfffba239ff1be945dbd2b53a402d2757b46","subject":"Merge \"Assure executable name is kept when app is called as module\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/dd6abfffba239ff1be945dbd2b53a402d2757b46"}]}],"author":{"name":"Vu Cong Tuan","email":"tuanvc@vn.fujitsu.com","date":"2018-07-11 04:31:05.000000000","tz":420},"committer":{"name":"Vu Cong Tuan","email":"tuanvc@vn.fujitsu.com","date":"2018-07-11 08:08:48.000000000","tz":420},"subject":"Switch to stestr","message":"Switch to stestr\n\nAccording to Openstack summit session [1],\nstestr is maintained project to which all Openstack projects should migrate.\nLet\u0027s switch to stestr as other projects have already moved to it.\n\n[1] https://etherpad.openstack.org/p/YVR-python-pti\n\nChange-Id: I09aa6a330e988070ad5b3a20b56b1c301386d657\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/b964896fc0d9489a804152058c40ded6960efab4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/b964896fc0d9489a804152058c40ded6960efab4"}]},"branch":"refs/heads/master"},"8034e45aabdcf149bf94f2a84303ea014c9abe3e":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-05-04 15:44:02.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/64/581564/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/64/581564/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/64/581564/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/64/581564/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/64/581564/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/64/581564/4"}}},"commit":{"parents":[{"commit":"6e52ba5bcca0b032df89f215d3ebf22a97b66c61","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/6e52ba5bcca0b032df89f215d3ebf22a97b66c61"}]}],"author":{"name":"Vu Cong Tuan","email":"tuanvc@vn.fujitsu.com","date":"2018-07-11 04:31:05.000000000","tz":420},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2019-05-04 15:44:02.000000000","tz":0},"subject":"Switch to stestr","message":"Switch to stestr\n\nAccording to Openstack summit session [1],\nstestr is maintained project to which all Openstack projects should migrate.\nLet\u0027s switch to stestr as other projects have already moved to it.\n\n[1] https://etherpad.openstack.org/p/YVR-python-pti\n\nChange-Id: I09aa6a330e988070ad5b3a20b56b1c301386d657\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/8034e45aabdcf149bf94f2a84303ea014c9abe3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/8034e45aabdcf149bf94f2a84303ea014c9abe3e"}]},"branch":"refs/heads/master"},"36d7d1cfc5372a1cd557493964cd71b08b478f75":{"kind":"REWORK","_number":5,"created":"2019-10-02 15:41:46.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/64/581564/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/64/581564/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/64/581564/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/64/581564/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/64/581564/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/64/581564/5"}}},"commit":{"parents":[{"commit":"a033aeb84befbb85b3c6490a0e30e977abc95256","subject":"Merge \"Stop wildcard importing argparse\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/a033aeb84befbb85b3c6490a0e30e977abc95256"}]}],"author":{"name":"Vu Cong Tuan","email":"tuanvc@vn.fujitsu.com","date":"2019-10-02 15:40:33.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-10-02 15:41:05.000000000","tz":60},"subject":"Switch to stestr","message":"Switch to stestr\n\nAccording to Openstack summit session [1],\nstestr is maintained project to which all Openstack projects should migrate.\nLet\u0027s switch to stestr as other projects have already moved to it.\n\n[1] https://etherpad.openstack.org/p/YVR-python-pti\n\nChange-Id: I09aa6a330e988070ad5b3a20b56b1c301386d657\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/36d7d1cfc5372a1cd557493964cd71b08b478f75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/36d7d1cfc5372a1cd557493964cd71b08b478f75"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
