)]}'
{"id":"openstack%2Fcliff~592060","triplet_id":"openstack%2Fcliff~master~I3bde60d16f92d6805ff5f6c1eee88bb3e474a07a","project":"openstack/cliff","branch":"master","hashtags":[],"change_id":"I3bde60d16f92d6805ff5f6c1eee88bb3e474a07a","subject":"cap cmd2 also for python3","status":"ABANDONED","created":"2018-08-15 14:18:23.000000000","updated":"2019-05-04 15:44:49.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"70862cac92d24bd54cfc20fca8badae545ca0ffa","_number":592060,"virtual_id_number":592060,"owner":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"actions":{},"labels":{"Verified":{"all":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-08-15 14:29:52.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2018-08-16 09:38:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9195b9cec7d9f6d9eaf05d31795ce04fae0a7065","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2018-08-15 14:18:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e531c65bff779c082a238e716b6ddee527e9542f","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2018-08-15 14:29:52.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"908cebe0f5499e1c810a38816ceba129a7e87dbd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-15 16:19:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/60/592060/1/check/openstack-tox-pep8/bd78090/ : SUCCESS in 4m 14s\n- openstack-tox-py27 http://logs.openstack.org/60/592060/1/check/openstack-tox-py27/5c71ead/ : SUCCESS in 3m 56s\n- openstack-tox-py35 http://logs.openstack.org/60/592060/1/check/openstack-tox-py35/74d3ec5/ : SUCCESS in 4m 53s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/60/592060/1/check/legacy-tempest-dsvm-neutron-src/5d1cbd2/ : SUCCESS in 1h 54m 37s\n- build-openstack-sphinx-docs http://logs.openstack.org/60/592060/1/check/build-openstack-sphinx-docs/08605dd/html/ : SUCCESS in 4m 09s\n- requirements-check http://logs.openstack.org/60/592060/1/check/requirements-check/d6b1a68/ : FAILURE in 4m 12s\n- cliff-tox-py27-neutronclient-tip http://logs.openstack.org/60/592060/1/check/cliff-tox-py27-neutronclient-tip/66dbdb0/ : FAILURE in 5m 07s\n- osc-tox-unit-tips http://logs.openstack.org/60/592060/1/check/osc-tox-unit-tips/73dcbeb/ : SUCCESS in 6m 17s\n- openstack-tox-lower-constraints http://logs.openstack.org/60/592060/1/check/openstack-tox-lower-constraints/60d8860/ : SUCCESS in 6m 02s","accounts_in_message":[],"_revision_number":1},{"id":"b203650d3032b0be2d9b62653f2d2a5070b0850b","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2018-08-16 07:40:09.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"aa21213ef27e775c3b56899b31759865c2e6f104","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-16 07:56:53.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/60/592060/2/check/openstack-tox-pep8/81a3f8b/ : FAILURE in 2m 59s\n- openstack-tox-py27 http://logs.openstack.org/60/592060/2/check/openstack-tox-py27/bc052aa/ : FAILURE in 4m 21s\n- openstack-tox-py35 http://logs.openstack.org/60/592060/2/check/openstack-tox-py35/fd1ac42/ : FAILURE in 4m 25s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/60/592060/2/check/legacy-tempest-dsvm-neutron-src/cf4d973/ : FAILURE in 12m 37s\n- build-openstack-sphinx-docs http://logs.openstack.org/60/592060/2/check/build-openstack-sphinx-docs/475a932/ : POST_FAILURE in 3m 42s\n- requirements-check http://logs.openstack.org/60/592060/2/check/requirements-check/c938baf/ : FAILURE in 3m 27s\n- cliff-tox-py27-neutronclient-tip http://logs.openstack.org/60/592060/2/check/cliff-tox-py27-neutronclient-tip/51aa919/ : FAILURE in 4m 22s\n- osc-tox-unit-tips http://logs.openstack.org/60/592060/2/check/osc-tox-unit-tips/3b3e954/ : FAILURE in 5m 30s\n- openstack-tox-lower-constraints http://logs.openstack.org/60/592060/2/check/openstack-tox-lower-constraints/9282e82/ : FAILURE in 4m 20s","accounts_in_message":[],"_revision_number":2},{"id":"0f8f1159a9c7a0146a6790b4366bc46b61861ecb","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2018-08-16 07:59:39.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e9f87ee337247cf5df94a382b5fd6d7a4a4ea595","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-16 09:38:08.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/60/592060/3/check/openstack-tox-pep8/878be6f/ : SUCCESS in 4m 08s\n- openstack-tox-py27 http://logs.openstack.org/60/592060/3/check/openstack-tox-py27/322cfdd/ : SUCCESS in 4m 26s\n- openstack-tox-py35 http://logs.openstack.org/60/592060/3/check/openstack-tox-py35/5a60afa/ : SUCCESS in 4m 42s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/60/592060/3/check/legacy-tempest-dsvm-neutron-src/959a07c/ : SUCCESS in 1h 36m 19s\n- build-openstack-sphinx-docs http://logs.openstack.org/60/592060/3/check/build-openstack-sphinx-docs/6e1946e/html/ : SUCCESS in 6m 34s\n- requirements-check http://logs.openstack.org/60/592060/3/check/requirements-check/b12d5bf/ : FAILURE in 3m 11s\n- cliff-tox-py27-neutronclient-tip http://logs.openstack.org/60/592060/3/check/cliff-tox-py27-neutronclient-tip/c62c295/ : FAILURE in 4m 56s\n- osc-tox-unit-tips http://logs.openstack.org/60/592060/3/check/osc-tox-unit-tips/511f7ac/ : SUCCESS in 6m 15s\n- openstack-tox-lower-constraints http://logs.openstack.org/60/592060/3/check/openstack-tox-lower-constraints/de0f5c0/ : SUCCESS in 5m 46s","accounts_in_message":[],"_revision_number":3},{"id":"f64d091eccc4d48e205840378d13c998be0c2e3f","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-04 15:43:39.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"7c488182eb9bd47cf799caf6e2c628f9c30d25c4","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-04 15:44:39.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":4},{"id":"70862cac92d24bd54cfc20fca8badae545ca0ffa","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-04 15:44:49.000000000","message":"Patch Set 4:\n\nAbandoned in favour of https://review.opendev.org/#/c/629283/","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"6e12a1d6aeb6ac34f2b4bc6425306591494599f8","revisions":{"29d06750b1f34515664969668c153fee7ec7441f":{"kind":"REWORK","_number":1,"created":"2018-08-15 14:18:23.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/60/592060/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/60/592060/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/60/592060/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/60/592060/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/60/592060/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/60/592060/1"}}},"commit":{"parents":[{"commit":"dd6abfffba239ff1be945dbd2b53a402d2757b46","subject":"Merge \"Assure executable name is kept when app is called as module\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/dd6abfffba239ff1be945dbd2b53a402d2757b46"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2018-08-15 14:18:07.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2018-08-15 14:18:07.000000000","tz":120},"subject":"cap cmd2 also for python3","message":"cap cmd2 also for python3\n\nWith python 3.6.6 osc is not able to work in an interactive mode \n(unrecognized argument). This can be reproduced by installing osc into \nthe virtualenv created with python 3.6.6\n\nChange-Id: I3bde60d16f92d6805ff5f6c1eee88bb3e474a07a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/29d06750b1f34515664969668c153fee7ec7441f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/29d06750b1f34515664969668c153fee7ec7441f"}]},"branch":"refs/heads/master"},"13aa70881bcf86256766643933ceb5f4407a7862":{"kind":"REWORK","_number":2,"created":"2018-08-16 07:40:09.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/60/592060/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/60/592060/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/60/592060/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/60/592060/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/60/592060/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/60/592060/2"}}},"commit":{"parents":[{"commit":"dd6abfffba239ff1be945dbd2b53a402d2757b46","subject":"Merge \"Assure executable name is kept when app is called as module\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/dd6abfffba239ff1be945dbd2b53a402d2757b46"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2018-08-16 07:39:49.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2018-08-16 07:39:49.000000000","tz":120},"subject":"cap cmd2 also for python3","message":"cap cmd2 also for python3\n\nWith python 3.6.6 osc is not able to work in an interactive mode \n(unrecognized argument). This can be reproduced by installing osc into \nthe virtualenv created with python 3.6.6, 3.5.5, 3.4.8. Manual \ndowngrading of cmd2 to 0.8.8 helps\n\nChange-Id: I3bde60d16f92d6805ff5f6c1eee88bb3e474a07a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/13aa70881bcf86256766643933ceb5f4407a7862"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/13aa70881bcf86256766643933ceb5f4407a7862"}]},"branch":"refs/heads/master"},"db1187026e93f0967ea799c6fe2b0e824bf7789c":{"kind":"REWORK","_number":3,"created":"2018-08-16 07:59:39.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/60/592060/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/60/592060/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/60/592060/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/60/592060/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/60/592060/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/60/592060/3"}}},"commit":{"parents":[{"commit":"dd6abfffba239ff1be945dbd2b53a402d2757b46","subject":"Merge \"Assure executable name is kept when app is called as module\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/dd6abfffba239ff1be945dbd2b53a402d2757b46"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2018-08-16 07:59:29.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2018-08-16 07:59:29.000000000","tz":120},"subject":"cap cmd2 also for python3","message":"cap cmd2 also for python3\n\nWith python 3.6.6 osc is not able to work in an interactive mode \n(unrecognized argument). This can be reproduced by installing osc into \nthe virtualenv created with python 3.6.6, 3.5.5, 3.4.8. Manual \ndowngrading of cmd2 to 0.8.8 helps\n\nChange-Id: I3bde60d16f92d6805ff5f6c1eee88bb3e474a07a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/db1187026e93f0967ea799c6fe2b0e824bf7789c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/db1187026e93f0967ea799c6fe2b0e824bf7789c"}]},"branch":"refs/heads/master"},"6e12a1d6aeb6ac34f2b4bc6425306591494599f8":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-05-04 15:43:39.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/60/592060/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/60/592060/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/60/592060/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/60/592060/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/60/592060/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/60/592060/4"}}},"commit":{"parents":[{"commit":"6e52ba5bcca0b032df89f215d3ebf22a97b66c61","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/6e52ba5bcca0b032df89f215d3ebf22a97b66c61"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2018-08-16 07:59:29.000000000","tz":120},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2019-05-04 15:43:39.000000000","tz":0},"subject":"cap cmd2 also for python3","message":"cap cmd2 also for python3\n\nWith python 3.6.6 osc is not able to work in an interactive mode \n(unrecognized argument). This can be reproduced by installing osc into \nthe virtualenv created with python 3.6.6, 3.5.5, 3.4.8. Manual \ndowngrading of cmd2 to 0.8.8 helps\n\nChange-Id: I3bde60d16f92d6805ff5f6c1eee88bb3e474a07a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/6e12a1d6aeb6ac34f2b4bc6425306591494599f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/6e12a1d6aeb6ac34f2b4bc6425306591494599f8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
