)]}'
{"id":"openstack%2Fcliff~611889","triplet_id":"openstack%2Fcliff~master~I3e8264f689f128d48042ba88c231351e5ab02b42","project":"openstack/cliff","branch":"master","topic":"envvar-integration","hashtags":[],"change_id":"I3e8264f689f128d48042ba88c231351e5ab02b42","subject":"Stop wildcard importing argparse","status":"MERGED","created":"2018-10-19 14:34:34.000000000","updated":"2019-09-23 15:39:12.000000000","submitted":"2019-09-23 15:38:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"611889-1569253081032-805a2d6b","meta_rev_id":"178f93787b12fc9710f3ad4a48502368df51601b","_number":611889,"virtual_id_number":611889,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},{"value":0,"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-23 15:38:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":2,"date":"2019-05-09 15:29:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":1,"date":"2019-09-13 01:49:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":2,"date":"2019-09-23 13:28:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2019-09-14 12:15:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},{"value":1,"date":"2019-09-14 01:27:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-14 01:13:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":1,"date":"2019-09-23 13:28:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},{"value":0,"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-03-19 14:03:38.000000000","updated_by":{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},"reviewer":{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},"state":"REVIEWER"},{"updated":"2019-05-09 15:29:04.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2019-09-13 01:49:04.000000000","updated_by":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"reviewer":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"state":"REVIEWER"},{"updated":"2019-09-14 01:13:25.000000000","updated_by":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"reviewer":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"state":"REVIEWER"},{"updated":"2019-09-14 01:27:34.000000000","updated_by":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"reviewer":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"state":"REVIEWER"},{"updated":"2019-09-14 12:15:12.000000000","updated_by":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"reviewer":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"state":"REVIEWER"},{"updated":"2019-09-23 15:38:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f0e43d20a9a21af669e6b5fb2ae1716df9ca4ef6","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2018-10-19 14:34:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3774e3bbdb661f640f577334ac5b70094cca2f13","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-19 17:20:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/89/611889/1/check/openstack-tox-pep8/f3e1394/ : SUCCESS in 4m 24s\n- openstack-tox-py27 http://logs.openstack.org/89/611889/1/check/openstack-tox-py27/a9c8c8d/ : SUCCESS in 3m 59s\n- openstack-tox-py35 http://logs.openstack.org/89/611889/1/check/openstack-tox-py35/5cc0d7a/ : SUCCESS in 3m 55s\n- openstack-tox-py36 http://logs.openstack.org/89/611889/1/check/openstack-tox-py36/f6453ac/ : SUCCESS in 6m 34s\n- tempest-full http://logs.openstack.org/89/611889/1/check/tempest-full/baa45b2/ : SUCCESS in 1h 40m 03s\n- tempest-full-py3 http://logs.openstack.org/89/611889/1/check/tempest-full-py3/219af96/ : SUCCESS in 1h 48m 11s\n- openstack-tox-docs http://logs.openstack.org/89/611889/1/check/openstack-tox-docs/c9c10af/html/ : SUCCESS in 3m 41s\n- cliff-tox-py27-neutronclient-tip http://logs.openstack.org/89/611889/1/check/cliff-tox-py27-neutronclient-tip/15c2d0c/ : SUCCESS in 5m 00s\n- osc-tox-unit-tips http://logs.openstack.org/89/611889/1/check/osc-tox-unit-tips/441fc0f/ : SUCCESS in 6m 06s\n- openstack-tox-lower-constraints http://logs.openstack.org/89/611889/1/check/openstack-tox-lower-constraints/3e5b4eb/ : SUCCESS in 6m 31s","accounts_in_message":[],"_revision_number":1},{"id":"7e8ff385f02ee744a3291f3ff62b03f750b95553","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-02-28 15:14:58.000000000","message":"Patch Set 1: Code-Review+1\n\nThis looks reasonable to me and it\u0027s still passing the unit tests added for this behavior. I\u0027ve added Julien to the review since he wrote this originally. Let\u0027s see if he has any input.","accounts_in_message":[],"_revision_number":1},{"id":"8148df016092b44c33a5ef98b951bb237ef4944a","author":{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},"date":"2019-03-19 14:03:38.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a0a204fc731333bf3077f6556f6326fefadf0974","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-03-19 19:17:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nstable/stein has been cut, so it\u0027s probably as good a time as any to merge this.","accounts_in_message":[],"_revision_number":1},{"id":"919327955eeeae2036d78515854d3a4418f5043e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-05 19:27:50.000000000","message":"Patch Set 1: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"96f139547554ce73124a776579d5bffa6901f4a7","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-09 11:09:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7aa92897615871aa3b659ba9bcf5fcf7433a82fe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 13:22:27.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/89/611889/2/check/tempest-full/8b1bb7a/ : TIMED_OUT in 2h 07m 42s\n- tempest-full-py3 http://logs.openstack.org/89/611889/2/check/tempest-full-py3/25348d3/ : SUCCESS in 1h 32m 24s\n- openstack-tox-lower-constraints http://logs.openstack.org/89/611889/2/check/openstack-tox-lower-constraints/9174e02/ : SUCCESS in 5m 09s\n- openstack-tox-pep8 http://logs.openstack.org/89/611889/2/check/openstack-tox-pep8/77304ad/ : FAILURE in 3m 42s\n- openstack-tox-py27 http://logs.openstack.org/89/611889/2/check/openstack-tox-py27/36836e2/ : SUCCESS in 4m 06s\n- openstack-tox-py36 http://logs.openstack.org/89/611889/2/check/openstack-tox-py36/50708d1/ : SUCCESS in 4m 05s\n- openstack-tox-py37 http://logs.openstack.org/89/611889/2/check/openstack-tox-py37/ec1c40a/ : SUCCESS in 4m 20s\n- openstack-tox-docs http://logs.openstack.org/89/611889/2/check/openstack-tox-docs/70dffca/html/ : SUCCESS in 7m 35s\n- cliff-tox-py27-neutronclient-tip http://logs.openstack.org/89/611889/2/check/cliff-tox-py27-neutronclient-tip/a4c43c7/ : SUCCESS in 4m 50s\n- osc-tox-unit-tips http://logs.openstack.org/89/611889/2/check/osc-tox-unit-tips/a5b3e21/ : SUCCESS in 6m 01s","accounts_in_message":[],"_revision_number":2},{"id":"ea91014ea168d8da8e2fdbcfe118e6dd968afa75","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-05-09 15:29:04.000000000","message":"Patch Set 2: Code-Review+2\n\nLooks like this is failing on the bandit 1.6.0 issue. Probably easiest to just wait until 1.6.1 and recheck.","accounts_in_message":[],"_revision_number":2},{"id":"2dfdfbdba49984b44e2c8ac2ccad26a53393168b","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2019-09-13 01:49:04.000000000","message":"Patch Set 2: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"dd0679381e34b7288113c852093f26183249fdf3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 03:33:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/6b49da970ad64319afef9cf335b8ecfe : SUCCESS in 1h 41m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8ee704b5d6354b33915c63c4c578ab46 : SUCCESS in 1h 24m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/39e8b84ff2e04ec4853204c03d61d5a3 : SUCCESS in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/606c47e585df485e920e46e9e70977c3 : SUCCESS in 2m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7ab380ae64244234bfe814ac0d580416 : SUCCESS in 2m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/93d077ea8cc74c8a9d71ffd1df84f19b : SUCCESS in 2m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dca1b4e91584496581d5221b2024676f : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f2bc82074f64814b4c6f701888020bf : SUCCESS in 3m 21s\n- cliff-tox-py27-neutronclient-tip https://zuul.opendev.org/t/openstack/build/cdb77bdcc1f04ec0946621fc61544e39 : SUCCESS in 3m 59s\n- osc-tox-unit-tips https://zuul.opendev.org/t/openstack/build/668ff7b06a514a869a3524055afacb07 : SUCCESS in 5m 31s","accounts_in_message":[],"_revision_number":2},{"id":"c5390ee138658fa16a0c94cdf5d4fb6dc12f0edd","author":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"date":"2019-09-14 01:13:25.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"3c5c304de4177995ce8e58a8488079dccac25c93","author":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"date":"2019-09-14 01:27:34.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"08cdd4868ec66618bcc02753e6a99799c9a11366","author":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"date":"2019-09-14 12:15:12.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"5720499baffa21fa68788b4038404aedf9b7ae3d","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-09-23 13:28:28.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nThis has been sitting here for a long time with a lot of reviews. cliff is pretty low activity so I\u0027m going to do a bad thing and proxy bnemec\u0027s +2 as a +W and approve my own patch. Don\u0027t hate me.","accounts_in_message":[],"_revision_number":2},{"id":"609bb4adfed537c7583975936982c6f5a8b10070","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 13:28:39.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"07d7e9309405628627ac21b79cd7ce4d4feada0e","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-09-23 14:24:49.000000000","message":"Patch Set 2:\n\nYeah, I was just waiting for stable/train to be cut. Since Dean had already +2\u0027d before the rebase I\u0027d say this is fine.","accounts_in_message":[],"_revision_number":2},{"id":"b10e7fdc1273ecd9aaf17aa864f2b7263b1add30","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 15:38:00.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/aa17d178a72543d082f3f35cf99844b3 : SUCCESS in 2h 02m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e9c852ea271849fd914d59a95cbb9a68 : SUCCESS in 1h 35m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/37879dd4af1e482a89a2f64c493a7761 : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03dba7dc382644c38bc2efe6099b3d9f : SUCCESS in 3m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7f19471b685041f689da5b72b9a83b82 : SUCCESS in 3m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e0e4abd056cc47ab89c41a3987dc4a71 : SUCCESS in 3m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/063c084dfa3a494b92217a2c449e0f80 : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a1f7169276141efbcdc9c06487945fd : SUCCESS in 4m 19s\n- cliff-tox-py27-neutronclient-tip https://zuul.opendev.org/t/openstack/build/34a2022c33194a6ea5ad6e4f54a64357 : SUCCESS in 4m 36s\n- osc-tox-unit-tips https://zuul.opendev.org/t/openstack/build/0a17d934930b4327b3667cbc08b5ee73 : SUCCESS in 6m 43s","accounts_in_message":[],"_revision_number":2},{"id":"bf361a7f7375f4a4a0e5b2c12b5ba349e8f7e16d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 15:38:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"7260ddfaec8489263c7bab10ebc7d1760a53e9c5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 15:39:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c53bd6d4608e4c8093a98831d6b3d038 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"584352dcd008d58c433136539b22a6ae9d6c45cc","revisions":{"ff87a07ef96e44813b5f22d475602d11298ef2a1":{"kind":"REWORK","_number":1,"created":"2018-10-19 14:34:34.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/89/611889/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/89/611889/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/89/611889/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/89/611889/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/89/611889/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/89/611889/1"}}},"commit":{"parents":[{"commit":"a4ff48c63b21d9fc466d7fec394896b03c84b500","subject":"Remove dead files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/a4ff48c63b21d9fc466d7fec394896b03c84b500"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2018-10-19 13:55:33.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2018-10-19 14:17:21.000000000","tz":60},"subject":"Stop wildcard importing argparse","message":"Stop wildcard importing argparse\n\nThere\u0027s no point in doing this when we only need two things from the\nmodule. Better to just centralize what we need to do with argparse in\nhere.\n\nChange-Id: I3e8264f689f128d48042ba88c231351e5ab02b42\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/ff87a07ef96e44813b5f22d475602d11298ef2a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/ff87a07ef96e44813b5f22d475602d11298ef2a1"}]},"branch":"refs/heads/master"},"584352dcd008d58c433136539b22a6ae9d6c45cc":{"kind":"REWORK","_number":2,"created":"2019-05-09 11:09:34.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/89/611889/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/89/611889/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/89/611889/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/89/611889/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/89/611889/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/89/611889/2"}}},"commit":{"parents":[{"commit":"b39b3276ec1d8ba1e0df36791051e24ba2af2fb5","subject":"Merge \"Remove dead files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/b39b3276ec1d8ba1e0df36791051e24ba2af2fb5"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2018-10-19 13:55:33.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-05-09 11:06:47.000000000","tz":60},"subject":"Stop wildcard importing argparse","message":"Stop wildcard importing argparse\n\nThere\u0027s no point in doing this when we only need two things from the\nmodule. Better to just centralize what we need to do with argparse in\nhere.\n\nChange-Id: I3e8264f689f128d48042ba88c231351e5ab02b42\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/584352dcd008d58c433136539b22a6ae9d6c45cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/584352dcd008d58c433136539b22a6ae9d6c45cc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
