)]}'
{"id":"openstack%2Fcliff~611890","triplet_id":"openstack%2Fcliff~master~I0850235825fc15d0bd8c64fff7b4856e34f5b3e1","project":"openstack/cliff","branch":"master","topic":"envvar-integration","hashtags":[],"change_id":"I0850235825fc15d0bd8c64fff7b4856e34f5b3e1","subject":"WIP: Add support for environment variables","status":"ABANDONED","created":"2018-10-19 14:34:34.000000000","updated":"2019-10-02 15:20:48.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"522078c957ade0bbdc746c0efb70bb7f1468f737","_number":611890,"virtual_id_number":611890,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":-1,"date":"2019-05-09 13:20:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-10-19 14:35:13.000000000","updated_by":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"reviewer":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"state":"REVIEWER"},{"updated":"2018-10-19 14:35:33.000000000","updated_by":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"reviewer":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"state":"REVIEWER"},{"updated":"2019-05-09 13:20:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5f0a0b1d9bd6b83d06fde7b7619b75f7fc1a645e","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2018-10-19 14:34:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ebb5a5ec598badf85539861be0ed9f87895d674b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2018-10-19 14:38:16.000000000","message":"Patch Set 1:\n\nmtreinish: I\u0027ve started drafting this mostly for use in stestr (though I\u0027m sure it can be used elsewhere). Instead of writing:\n\n  [testenv]\n  commands \u003d\n    stestr run {posargs}\n\n  [testenv:functional]\n  commands \u003d\n    stestr --test-path\u003d./placement/tests/functional run {posargs}\n\nI\u0027d like to be able to do\n\n  [testenv:functional]\n  setenv \u003d\n    TEST_PATH\u003d./placement/tests/functional\n\nWhich makes more sense in my head and fits in quite nicely with what pytest lets you do. Thoughts?","accounts_in_message":[],"_revision_number":1},{"id":"b669aca0cf8835e3fddee6649561e272c3169897","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2018-10-19 14:38:35.000000000","message":"Patch Set 1: Workflow-1\n\nWIP because I need to handle nargs!\u003d1, add docs and add tests","accounts_in_message":[],"_revision_number":1},{"id":"bf0a63d4f4f662066296e10cae97fa7ad01b8f4c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-19 17:07:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/90/611890/1/check/openstack-tox-pep8/1663984/ : FAILURE in 4m 12s\n- openstack-tox-py27 http://logs.openstack.org/90/611890/1/check/openstack-tox-py27/2b4adc2/ : SUCCESS in 3m 57s\n- openstack-tox-py35 http://logs.openstack.org/90/611890/1/check/openstack-tox-py35/9025faf/ : SUCCESS in 4m 49s\n- openstack-tox-py36 http://logs.openstack.org/90/611890/1/check/openstack-tox-py36/9060e24/ : SUCCESS in 4m 04s\n- tempest-full http://logs.openstack.org/90/611890/1/check/tempest-full/afbe79f/ : SUCCESS in 1h 44m 02s\n- tempest-full-py3 http://logs.openstack.org/90/611890/1/check/tempest-full-py3/981bd5b/ : SUCCESS in 1h 29m 26s\n- openstack-tox-docs http://logs.openstack.org/90/611890/1/check/openstack-tox-docs/650b7c7/html/ : SUCCESS in 4m 34s\n- cliff-tox-py27-neutronclient-tip http://logs.openstack.org/90/611890/1/check/cliff-tox-py27-neutronclient-tip/95f7477/ : SUCCESS in 6m 26s\n- osc-tox-unit-tips http://logs.openstack.org/90/611890/1/check/osc-tox-unit-tips/4468b26/ : SUCCESS in 5m 55s\n- openstack-tox-lower-constraints http://logs.openstack.org/90/611890/1/check/openstack-tox-lower-constraints/6523d51/ : SUCCESS in 6m 22s","accounts_in_message":[],"_revision_number":1},{"id":"3ec125116d6ca315237bea55cb09d4fabfae6b5b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-09 11:09:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c4382390b49fd7514283461d22b1d23ca661456f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 13:20:24.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/90/611890/2/check/tempest-full/6597711/ : TIMED_OUT in 2h 08m 15s\n- tempest-full-py3 http://logs.openstack.org/90/611890/2/check/tempest-full-py3/5923af8/ : SUCCESS in 1h 42m 23s\n- openstack-tox-lower-constraints http://logs.openstack.org/90/611890/2/check/openstack-tox-lower-constraints/8da5671/ : SUCCESS in 5m 34s\n- openstack-tox-pep8 http://logs.openstack.org/90/611890/2/check/openstack-tox-pep8/1884723/ : FAILURE in 4m 23s\n- openstack-tox-py27 http://logs.openstack.org/90/611890/2/check/openstack-tox-py27/55c3509/ : SUCCESS in 7m 46s\n- openstack-tox-py36 http://logs.openstack.org/90/611890/2/check/openstack-tox-py36/78ee467/ : SUCCESS in 4m 07s\n- openstack-tox-py37 http://logs.openstack.org/90/611890/2/check/openstack-tox-py37/e0466cf/ : SUCCESS in 8m 03s\n- openstack-tox-docs http://logs.openstack.org/90/611890/2/check/openstack-tox-docs/334e452/html/ : SUCCESS in 6m 49s\n- cliff-tox-py27-neutronclient-tip http://logs.openstack.org/90/611890/2/check/cliff-tox-py27-neutronclient-tip/14c528d/ : SUCCESS in 4m 08s\n- osc-tox-unit-tips http://logs.openstack.org/90/611890/2/check/osc-tox-unit-tips/f9c58ec/ : SUCCESS in 6m 47s","accounts_in_message":[],"_revision_number":2},{"id":"61f318382aee20b4d66d9ecd10f34653ae32bf79","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-10-02 15:20:48.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4bee98ad80b63f22f1955c4fd6fc9bba02566239","revisions":{"1be6af3eb787e778cdbce80ac9eb6b8f080706e3":{"kind":"REWORK","_number":1,"created":"2018-10-19 14:34:34.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/90/611890/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/90/611890/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/90/611890/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/90/611890/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/90/611890/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/90/611890/1"}}},"commit":{"parents":[{"commit":"ff87a07ef96e44813b5f22d475602d11298ef2a1","subject":"Stop wildcard importing argparse","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/ff87a07ef96e44813b5f22d475602d11298ef2a1"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2018-10-19 14:11:57.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2018-10-19 14:33:22.000000000","tz":60},"subject":"WIP: Add support for environment variables","message":"WIP: Add support for environment variables\n\nIt can be helpful to use environment variables as default for some\nargument. click supports this. It\u0027s time cliff did too.\n\nChange-Id: I0850235825fc15d0bd8c64fff7b4856e34f5b3e1\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/1be6af3eb787e778cdbce80ac9eb6b8f080706e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/1be6af3eb787e778cdbce80ac9eb6b8f080706e3"}]},"branch":"refs/heads/master"},"4bee98ad80b63f22f1955c4fd6fc9bba02566239":{"kind":"REWORK","_number":2,"created":"2019-05-09 11:09:34.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/90/611890/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/90/611890/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/90/611890/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/90/611890/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/90/611890/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/90/611890/2"}}},"commit":{"parents":[{"commit":"584352dcd008d58c433136539b22a6ae9d6c45cc","subject":"Stop wildcard importing argparse","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/584352dcd008d58c433136539b22a6ae9d6c45cc"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2018-10-19 14:11:57.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-05-09 11:09:08.000000000","tz":60},"subject":"WIP: Add support for environment variables","message":"WIP: Add support for environment variables\n\nIt can be helpful to use environment variables as default for some\nargument. click supports this. It\u0027s time cliff did too.\n\nChange-Id: I0850235825fc15d0bd8c64fff7b4856e34f5b3e1\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/4bee98ad80b63f22f1955c4fd6fc9bba02566239"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/4bee98ad80b63f22f1955c4fd6fc9bba02566239"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
