)]}'
{"id":"openstack%2Fcliff~670954","triplet_id":"openstack%2Fcliff~master~I310de5eb30689ab0f03d333a32291754a5f9d708","project":"openstack/cliff","branch":"master","hashtags":[],"change_id":"I310de5eb30689ab0f03d333a32291754a5f9d708","subject":"Add Python 3 Train unit tests","status":"ABANDONED","created":"2019-07-16 06:08:47.000000000","updated":"2019-10-02 15:37:37.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"421c8109b92a9542361ef211eaa5112ad47baed3","_number":670954,"virtual_id_number":670954,"owner":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2019-07-16 10:20:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-07-16 08:30:28.000000000","_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-16 10:20:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2cb0ec4772ba3c3c2812b0a064f5fd8f383bd469","author":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"date":"2019-07-16 06:08:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2648a972520939c3874ce9c422bda9ce0ad158f1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 08:23:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/54/670954/1/check/test-release-openstack/d511b28/ : SUCCESS in 2m 44s\n- tempest-full http://logs.openstack.org/54/670954/1/check/tempest-full/20ddefb/ : TIMED_OUT in 2h 12m 09s\n- tempest-full-py3 http://logs.openstack.org/54/670954/1/check/tempest-full-py3/16625ac/ : SUCCESS in 1h 38m 27s\n- openstack-tox-lower-constraints http://logs.openstack.org/54/670954/1/check/openstack-tox-lower-constraints/b10e877/ : SUCCESS in 4m 37s\n- openstack-tox-pep8 http://logs.openstack.org/54/670954/1/check/openstack-tox-pep8/bbf55ff/ : SUCCESS in 2m 39s\n- openstack-tox-py27 http://logs.openstack.org/54/670954/1/check/openstack-tox-py27/b2bae4a/ : SUCCESS in 3m 29s\n- openstack-tox-py36 http://logs.openstack.org/54/670954/1/check/openstack-tox-py36/a23e0f2/ : SUCCESS in 3m 59s\n- openstack-tox-py37 http://logs.openstack.org/54/670954/1/check/openstack-tox-py37/81461c1/ : SUCCESS in 3m 21s\n- openstack-tox-docs http://logs.openstack.org/54/670954/1/check/openstack-tox-docs/9374481/html/ : SUCCESS in 3m 33s\n- cliff-tox-py27-neutronclient-tip http://logs.openstack.org/54/670954/1/check/cliff-tox-py27-neutronclient-tip/67f00e7/ : SUCCESS in 3m 11s\n- osc-tox-unit-tips http://logs.openstack.org/54/670954/1/check/osc-tox-unit-tips/71bd7a4/ : SUCCESS in 5m 49s","accounts_in_message":[],"_revision_number":1},{"id":"07f8acf65aa50110830ddf02d0f7b0853e3de7c7","author":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"date":"2019-07-16 08:30:28.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2ec6d7b0adc4e5c7cd1cadbc84c93c7eee76b5e5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 10:20:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/54/670954/1/check/test-release-openstack/625ef65/ : SUCCESS in 2m 13s\n- tempest-full http://logs.openstack.org/54/670954/1/check/tempest-full/4af7582/ : SUCCESS in 1h 44m 26s\n- tempest-full-py3 http://logs.openstack.org/54/670954/1/check/tempest-full-py3/3ec24b1/ : SUCCESS in 1h 31m 42s\n- openstack-tox-lower-constraints http://logs.openstack.org/54/670954/1/check/openstack-tox-lower-constraints/75226d2/ : SUCCESS in 3m 32s\n- openstack-tox-pep8 http://logs.openstack.org/54/670954/1/check/openstack-tox-pep8/88ec513/ : SUCCESS in 3m 19s\n- openstack-tox-py27 http://logs.openstack.org/54/670954/1/check/openstack-tox-py27/a4baa6f/ : SUCCESS in 3m 40s\n- openstack-tox-py36 http://logs.openstack.org/54/670954/1/check/openstack-tox-py36/9e642c9/ : SUCCESS in 4m 10s\n- openstack-tox-py37 http://logs.openstack.org/54/670954/1/check/openstack-tox-py37/624cc39/ : SUCCESS in 3m 18s\n- openstack-tox-docs http://logs.openstack.org/54/670954/1/check/openstack-tox-docs/13cd97a/html/ : SUCCESS in 2m 41s\n- cliff-tox-py27-neutronclient-tip http://logs.openstack.org/54/670954/1/check/cliff-tox-py27-neutronclient-tip/6f51a8e/ : SUCCESS in 3m 37s\n- osc-tox-unit-tips http://logs.openstack.org/54/670954/1/check/osc-tox-unit-tips/4bb5bea/ : SUCCESS in 5m 06s","accounts_in_message":[],"_revision_number":1},{"id":"421c8109b92a9542361ef211eaa5112ad47baed3","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-10-02 15:37:37.000000000","message":"Abandoned\n\nLooks like this was done by I5b08ef6ac98eb4650a2f3df5cb559144172fb29c","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1b6a68507c48cea1d0dc2949022ea564260ad45b","revisions":{"1b6a68507c48cea1d0dc2949022ea564260ad45b":{"kind":"REWORK","_number":1,"created":"2019-07-16 06:08:47.000000000","uploader":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"ref":"refs/changes/54/670954/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/54/670954/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/54/670954/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/54/670954/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/54/670954/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/54/670954/1"}}},"commit":{"parents":[{"commit":"b39b3276ec1d8ba1e0df36791051e24ba2af2fb5","subject":"Merge \"Remove dead files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/b39b3276ec1d8ba1e0df36791051e24ba2af2fb5"}]}],"author":{"name":"qiufossen","email":"qiujunting@inspur.com","date":"2019-07-16 06:06:40.000000000","tz":-240},"committer":{"name":"qiufossen","email":"qiujunting@inspur.com","date":"2019-07-16 06:08:33.000000000","tz":-240},"subject":"Add Python 3 Train unit tests","message":"Add Python 3 Train unit tests\n\nThis is a mechanically generated patch to ensure unit testing is in place\nfor all of the Tested Runtimes for Train.\n\nSee the Train python3-updates goal document for details:\nhttps://governance.openstack.org/tc/goals/train/python3-updates.html\n\nChange-Id: I310de5eb30689ab0f03d333a32291754a5f9d708\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/1b6a68507c48cea1d0dc2949022ea564260ad45b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/1b6a68507c48cea1d0dc2949022ea564260ad45b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
