)]}'
{"id":"openstack%2Fcliff~687451","triplet_id":"openstack%2Fcliff~master~Ia991403e2393f2a64b361d0986326d045986df58","project":"openstack/cliff","branch":"master","hashtags":[],"change_id":"Ia991403e2393f2a64b361d0986326d045986df58","subject":"Add docutils to requirements.txt","status":"ABANDONED","created":"2019-10-09 04:14:05.000000000","updated":"2020-02-19 13:58:19.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3e60e6e04cf192a26b41d0e3f19ab85ae5d056b1","_number":687451,"virtual_id_number":687451,"owner":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-10-09 05:35:17.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-09 05:35:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"837f337318337f1ed72e7111df1ce0c5f938551a","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2019-10-09 04:14:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"905246f7a69d2772f26d0d47860dfe9310f12de4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 05:35:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/6f9ece12919f4a5cadf3c90674d268c3 : FAILURE in 1m 46s\n- tempest-full https://zuul.opendev.org/t/openstack/build/1f06ad06086e4b85b802a092c9eebb9b : SUCCESS in 1h 20m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d06f9a0eb5b04bc899ac7cf2a811c5b9 : SUCCESS in 1h 07m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1a3c517ece0a4f2da7905e0f8dd6bc55 : SUCCESS in 3m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a921275c821249b4a3d50169f3fa35bc : SUCCESS in 2m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/31770f24c3cd4e86b176e1ace11321d4 : SUCCESS in 2m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6b1f5f432bb248a3aca2e4b07afc72e8 : SUCCESS in 2m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c785ca7204114a1299f93e90288edcb0 : SUCCESS in 2m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d41ac250e6b44ccbc27d22b4a2ebb19 : SUCCESS in 2m 23s\n- cliff-tox-py27-neutronclient-tip https://zuul.opendev.org/t/openstack/build/c2736760ef8d47df8bd907fc86211855 : SUCCESS in 2m 43s\n- osc-tox-unit-tips https://zuul.opendev.org/t/openstack/build/eea23a0f626b4df78f7e9ff9e3431cca : SUCCESS in 3m 47s","accounts_in_message":[],"_revision_number":1},{"id":"3e60e6e04cf192a26b41d0e3f19ab85ae5d056b1","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-02-19 13:58:19.000000000","message":"Abandoned\n\nWe should be getting this for free from the sphinx dependency, no? In any case, this has been in -Verified for some time now so I\u0027m going to assume it\u0027s stalled and close it out. Please reopen (or ask someone in oslo-core to do so) if that\u0027s not the case :)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"40f6674d2a5fcfad2ddc9bf9e2e0ea964bcc3a0c","revisions":{"40f6674d2a5fcfad2ddc9bf9e2e0ea964bcc3a0c":{"kind":"REWORK","_number":1,"created":"2019-10-09 04:14:05.000000000","uploader":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"ref":"refs/changes/51/687451/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/51/687451/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/51/687451/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/51/687451/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/51/687451/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/51/687451/1"}}},"commit":{"parents":[{"commit":"9c561ef6c3e8f1de6b1ff8ce11fd660573f631c9","subject":"Merge \"Allow finding command by partial name\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/9c561ef6c3e8f1de6b1ff8ce11fd660573f631c9"}]}],"author":{"name":"Thomas Bechtold","email":"tbechtold@suse.com","date":"2019-10-09 04:13:17.000000000","tz":120},"committer":{"name":"Thomas Bechtold","email":"tbechtold@suse.com","date":"2019-10-09 04:13:17.000000000","tz":120},"subject":"Add docutils to requirements.txt","message":"Add docutils to requirements.txt\n\ndocutils is used in cliff/sphinxext.py so it must be listed in\nrequirements.txt .\n\nChange-Id: Ia991403e2393f2a64b361d0986326d045986df58\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/40f6674d2a5fcfad2ddc9bf9e2e0ea964bcc3a0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/40f6674d2a5fcfad2ddc9bf9e2e0ea964bcc3a0c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
