)]}'
{"id":"openstack%2Fcliff~705612","triplet_id":"openstack%2Fcliff~master~Icb08c704cb5d0d8004d0a5408822c9d53ae9c728","project":"openstack/cliff","branch":"master","topic":"drop-py27-support","hashtags":[],"change_id":"Icb08c704cb5d0d8004d0a5408822c9d53ae9c728","subject":"[ussuri][goal] Drop python 2.7 support and testing","status":"MERGED","created":"2020-02-04 07:44:41.000000000","updated":"2020-02-04 16:55:00.000000000","submitted":"2020-02-04 16:49:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"705612-1580834942245-a46f390a","meta_rev_id":"3d7e90ee4dc7bcb7fdfd8ed969640a03bff05cc7","_number":705612,"virtual_id_number":705612,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-02-04 14:05:49.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-04 16:49:02.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2020-02-04 14:41:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2020-02-04 14:06:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2020-02-04 14:41:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-04 14:06:37.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-02-04 14:41:26.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-02-04 16:49:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"abc8024877b4c7d34e2580d0af8a8b8df165ca6e","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-02-04 07:44:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"da30a5679f86c2475916bcdf0a4409814e973bee","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-02-04 07:46:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a9d549b87e49bda9691f68bafd377ff6b659b133","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-04 09:18:07.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/874326b5cbfa454582e5ca2e994d9683 : SUCCESS in 3m 00s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8f06ddfb56b046398179ab658aeb0b45 : SUCCESS in 3m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4ea77d1d92b04948a4d62f0bef40dc6b : SUCCESS in 1h 25m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cd076da09ff14c0b9613854769ce571c : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5adf357801ff40f3989bf50cec852aa3 : FAILURE in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/03132c60128b4681ab26f90a639daa6d : SUCCESS in 5m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/11bcc46af46a42998827a8e880379592 : SUCCESS in 5m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/63922d9273c54fcb9c87fb6ca4518ec9 : SUCCESS in 3m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1b70206068d4cbc9a3fa3c425601923 : FAILURE in 5m 03s\n- cliff-tox-py37-neutronclient-tip https://zuul.opendev.org/t/openstack/build/36dfd2cf9f5240e796a52d2ebd06c1f1 : SUCCESS in 4m 30s\n- osc-tox-unit-tips https://zuul.opendev.org/t/openstack/build/433f369d441e488dbb31d5319e5be829 : FAILURE in 5m 05s","accounts_in_message":[],"_revision_number":2},{"id":"e55547f7ae2f8f43ac79c6de4275287b43175f32","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-02-04 10:31:59.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"30bbae7a5a608e76f23be8aac8a98022359c5d8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-04 11:59:48.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b40aa6e0e65e4f4883524bbb60b746b5 : SUCCESS in 3m 14s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6a9c34ea8a474362b33f0cf9908ba2f2 : SUCCESS in 2m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f66076af71ad427785e127b3f853d7d5 : SUCCESS in 1h 25m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/939fc50ce8af4162908cb60d94be78ea : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e1d88ac45f44c308a9aef3ba574232c : SUCCESS in 2m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/39edbe7abbbd428185287d08bd12c07d : SUCCESS in 3m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/89f927d746f042c8ae6b89b2c3c841d7 : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7259fdd23e4747719a6efdc221778b90 : SUCCESS in 4m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e9a8acc13814bfa92410694dbaf9d10 : SUCCESS in 3m 17s\n- cliff-tox-py37-neutronclient-tip https://zuul.opendev.org/t/openstack/build/9ddfade4c4cd49a9bcf4d194fb07c97f : SUCCESS in 5m 10s\n- osc-tox-unit-tips https://zuul.opendev.org/t/openstack/build/75bc706af786458e9628518740a5b64d : FAILURE in 5m 40s","accounts_in_message":[],"_revision_number":3},{"id":"7685b8d97c8b988b74f9d7a0cb7b4c3f87a1a3b8","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-02-04 12:27:54.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"4620b28170858288a411c70e723f762c2faf5504","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-04 14:00:02.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0cef834bf3ed43718a87580b047c0bd4 : SUCCESS in 2m 19s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8c2443e168944e67a9a0de992b8f05b4 : SUCCESS in 2m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/96ed6a646f6e4520be4cf17a6fb06819 : SUCCESS in 1h 26m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6e42195f54fd43688e93d8e99155f8cc : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1daf8b327164c2a9d723cf98e00bc55 : SUCCESS in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fbd6d163bff94a1eaa52ec3722044e52 : SUCCESS in 3m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b0b01deb8db4486ebaaff35ea9bb2de8 : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/99b3e28882d345fab5af5ba87139631d : SUCCESS in 4m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/306d2b01e2af4f0ba290588b7f4202d6 : SUCCESS in 3m 33s\n- cliff-tox-py37-neutronclient-tip https://zuul.opendev.org/t/openstack/build/bd4037296fe8480dbadea0b895236737 : SUCCESS in 4m 41s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/40ec09bc8c92428f9b8b89a323c09657 : SUCCESS in 4m 53s","accounts_in_message":[],"_revision_number":4},{"id":"0fae978e6ef12c9c8b1d48c81d696ff477071cdb","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-02-04 14:02:43.000000000","message":"Patch Set 4: Code-Review-1\n\nShould be removed too \u003d\u003e https://github.com/openstack/cliff/blob/master/setup.py#L19,L25","accounts_in_message":[],"_revision_number":4},{"id":"69199b5932dda23afe89b312018d6c69046826a3","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-02-04 14:05:38.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"cc18792590d70f78fbd08056a35ff7f4727ecca8","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-02-04 14:05:49.000000000","message":"Patch Set 5:\n\nremoved setup.py lines","accounts_in_message":[],"_revision_number":5},{"id":"a0d4a36213a7b72dc43bdf720a0ab31c2781c53a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-02-04 14:06:37.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"f90d3f711bbeec5fc7ad89ee7166191759db1d27","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-02-04 14:41:26.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"06817d527879e4623c37a30c718128006ca19bea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-04 15:26:56.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9da39bb4a6c24cd6a9d9cff54e611ddf : SUCCESS in 3m 01s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d0323d051eba4a03bc6c69217493906f : SUCCESS in 4m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5c90201b67cc4c549fc15f66bab08bd2 : SUCCESS in 1h 19m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4a9c999383a6482ea6876f6e8efe2cbf : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/750d81b79158494f9dbfc2be0b35757e : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/75834b59d76e4fb28da41442d13e5571 : SUCCESS in 6m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7dba551d30ed4bdd9f5eee2234f97045 : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bf9394f0cb774e999a4abfca8d722569 : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb1ea0d61c904fe5b779cbdf610c9b8f : SUCCESS in 4m 05s\n- cliff-tox-py37-neutronclient-tip https://zuul.opendev.org/t/openstack/build/f068f361764645259d9cfb694247bca6 : SUCCESS in 6m 41s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/90525a28bf2547dca8e78c1d8350fde8 : SUCCESS in 6m 44s","accounts_in_message":[],"_revision_number":5},{"id":"74ea229d1fe75c1164c1ae8bc6d1c9dae69ac2b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-04 15:27:13.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"96b0bd3e8db6f121c1064f1522a11e56c503ea9a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-04 16:49:02.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/03046cd7da484a02830a45b27b992991 : SUCCESS in 3m 10s\n- requirements-check https://zuul.opendev.org/t/openstack/build/de7fbc2b8bc94c988e7859e43678ae91 : SUCCESS in 3m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d7af74662bf9469ab70970d7fe77fc30 : SUCCESS in 1h 20m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2a576970fb1b4f21b3f711ddaf090a6b : SUCCESS in 6m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce7e06c5105f467abf4a400d9342684c : SUCCESS in 6m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ad836e81115a4921be30af15642ddee8 : SUCCESS in 4m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a206e26d51a64d299d188a2795a3f9a5 : SUCCESS in 3m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f7609d70d5242e2a208fa92b2e0643e : SUCCESS in 5m 18s\n- cliff-tox-py37-neutronclient-tip https://zuul.opendev.org/t/openstack/build/9d0ee76f483e434da5cac293090b18ba : SUCCESS in 7m 34s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/a274f79a0e3241f3a8f80c9b95c55231 : SUCCESS in 6m 02s","accounts_in_message":[],"_revision_number":5},{"id":"f26f0a2f756eeb97f9df899a76e801bb70025bc1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-04 16:49:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"8bdc208ff59b0f382d8e884b79bbd57d0a926941","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-04 16:55:00.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f488e24678664e41817c6b9d35db6cd3 : SUCCESS in 2m 57s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"972410871009d44916a8857d6132dad8fbc96866","revisions":{"afa789ff46f693a8d7a6ba161efcdb07b655439a":{"kind":"REWORK","_number":1,"created":"2020-02-04 07:44:41.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/12/705612/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/12/705612/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/12/705612/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/12/705612/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/12/705612/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/12/705612/1"}}},"commit":{"parents":[{"commit":"009b13b2df8cff141ec74c208d7cd46df0992fb8","subject":"Merge \"Add autoprogram_cliff_app_dist_name config opt\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/009b13b2df8cff141ec74c208d7cd46df0992fb8"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-02-04 07:34:09.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-02-04 07:44:05.000000000","tz":60},"subject":"[ussuri][goal] Drop python 2.7 support and testing","message":"[ussuri][goal] Drop python 2.7 support and testing\n\nOpenStack is dropping the py2.7 support in ussuri cycle.\n\nComplete discussion \u0026 schedule can be found in\n- http://lists.openstack.org/pipermail/openstack-discuss/2019-October/010142.html\n- https://etherpad.openstack.org/p/drop-python2-support\n\nUssuri Communtiy-wide goal:\nhttps://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html\n\nCleanup setup.cfg, conf.py, and tox.ini from cruft.\n\nChange-Id: Icb08c704cb5d0d8004d0a5408822c9d53ae9c728\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/afa789ff46f693a8d7a6ba161efcdb07b655439a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/afa789ff46f693a8d7a6ba161efcdb07b655439a"}]},"branch":"refs/heads/master"},"c2478771d39bf2826fd121a3dfe7bf9ce6e3ccb6":{"kind":"REWORK","_number":2,"created":"2020-02-04 07:46:02.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/12/705612/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/12/705612/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/12/705612/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/12/705612/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/12/705612/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/12/705612/2"}}},"commit":{"parents":[{"commit":"009b13b2df8cff141ec74c208d7cd46df0992fb8","subject":"Merge \"Add autoprogram_cliff_app_dist_name config opt\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/009b13b2df8cff141ec74c208d7cd46df0992fb8"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-02-04 07:34:09.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-02-04 07:45:53.000000000","tz":60},"subject":"[ussuri][goal] Drop python 2.7 support and testing","message":"[ussuri][goal] Drop python 2.7 support and testing\n\nOpenStack is dropping the py2.7 support in ussuri cycle.\n\nComplete discussion \u0026 schedule can be found in\n- http://lists.openstack.org/pipermail/openstack-discuss/2019-October/010142.html\n- https://etherpad.openstack.org/p/drop-python2-support\n\nUssuri Communtiy-wide goal:\nhttps://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html\n\nCleanup setup.cfg, conf.py, and tox.ini from cruft.\n\nChange-Id: Icb08c704cb5d0d8004d0a5408822c9d53ae9c728\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/c2478771d39bf2826fd121a3dfe7bf9ce6e3ccb6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/c2478771d39bf2826fd121a3dfe7bf9ce6e3ccb6"}]},"branch":"refs/heads/master"},"31a366e4875c549233174c4a7a5cbde0e8501616":{"kind":"REWORK","_number":3,"created":"2020-02-04 10:31:59.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/12/705612/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/12/705612/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/12/705612/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/12/705612/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/12/705612/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/12/705612/3"}}},"commit":{"parents":[{"commit":"009b13b2df8cff141ec74c208d7cd46df0992fb8","subject":"Merge \"Add autoprogram_cliff_app_dist_name config opt\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/009b13b2df8cff141ec74c208d7cd46df0992fb8"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-02-04 07:34:09.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-02-04 10:31:56.000000000","tz":60},"subject":"[ussuri][goal] Drop python 2.7 support and testing","message":"[ussuri][goal] Drop python 2.7 support and testing\n\nOpenStack is dropping the py2.7 support in ussuri cycle.\n\nComplete discussion \u0026 schedule can be found in\n- http://lists.openstack.org/pipermail/openstack-discuss/2019-October/010142.html\n- https://etherpad.openstack.org/p/drop-python2-support\n\nUssuri Communtiy-wide goal:\nhttps://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html\n\nCleanup setup.cfg, conf.py, and tox.ini from cruft.\n\nChange-Id: Icb08c704cb5d0d8004d0a5408822c9d53ae9c728\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/31a366e4875c549233174c4a7a5cbde0e8501616"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/31a366e4875c549233174c4a7a5cbde0e8501616"}]},"branch":"refs/heads/master"},"d88d61e2e7c9b8ed6c5082bbdbee0e13b40d6f95":{"kind":"REWORK","_number":4,"created":"2020-02-04 12:27:54.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/12/705612/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/12/705612/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/12/705612/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/12/705612/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/12/705612/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/12/705612/4"}}},"commit":{"parents":[{"commit":"009b13b2df8cff141ec74c208d7cd46df0992fb8","subject":"Merge \"Add autoprogram_cliff_app_dist_name config opt\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/009b13b2df8cff141ec74c208d7cd46df0992fb8"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-02-04 07:34:09.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-02-04 12:27:39.000000000","tz":60},"subject":"[ussuri][goal] Drop python 2.7 support and testing","message":"[ussuri][goal] Drop python 2.7 support and testing\n\nOpenStack is dropping the py2.7 support in ussuri cycle.\n\nComplete discussion \u0026 schedule can be found in\n- http://lists.openstack.org/pipermail/openstack-discuss/2019-October/010142.html\n- https://etherpad.openstack.org/p/drop-python2-support\n\nUssuri Communtiy-wide goal:\nhttps://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html\n\nCleanup setup.cfg, conf.py, and tox.ini from cruft.\n\nChange-Id: Icb08c704cb5d0d8004d0a5408822c9d53ae9c728\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/d88d61e2e7c9b8ed6c5082bbdbee0e13b40d6f95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/d88d61e2e7c9b8ed6c5082bbdbee0e13b40d6f95"}]},"branch":"refs/heads/master"},"972410871009d44916a8857d6132dad8fbc96866":{"kind":"REWORK","_number":5,"created":"2020-02-04 14:05:38.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/12/705612/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/12/705612/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/12/705612/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/12/705612/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/12/705612/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/12/705612/5"}}},"commit":{"parents":[{"commit":"009b13b2df8cff141ec74c208d7cd46df0992fb8","subject":"Merge \"Add autoprogram_cliff_app_dist_name config opt\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/009b13b2df8cff141ec74c208d7cd46df0992fb8"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-02-04 07:34:09.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-02-04 14:05:33.000000000","tz":60},"subject":"[ussuri][goal] Drop python 2.7 support and testing","message":"[ussuri][goal] Drop python 2.7 support and testing\n\nOpenStack is dropping the py2.7 support in ussuri cycle.\n\nComplete discussion \u0026 schedule can be found in\n- http://lists.openstack.org/pipermail/openstack-discuss/2019-October/010142.html\n- https://etherpad.openstack.org/p/drop-python2-support\n\nUssuri Communtiy-wide goal:\nhttps://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html\n\nCleanup setup.cfg, conf.py, and tox.ini from cruft.\n\nChange-Id: Icb08c704cb5d0d8004d0a5408822c9d53ae9c728\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/972410871009d44916a8857d6132dad8fbc96866"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/972410871009d44916a8857d6132dad8fbc96866"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
