)]}'
{"id":"openstack%2Fcliff~708882","triplet_id":"openstack%2Fcliff~master~I517c61f24ba6194ff6181e115a3a23adbce3ea53","project":"openstack/cliff","branch":"master","hashtags":[],"change_id":"I517c61f24ba6194ff6181e115a3a23adbce3ea53","subject":"Fix nested argument groups with ignore conflict handler","status":"MERGED","created":"2020-02-20 13:54:22.000000000","updated":"2020-03-24 17:56:27.000000000","submitted":"2020-03-24 17:54:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"708882-1585072450135-81340386","meta_rev_id":"aa81a3850b7575f78e3a2fe1548a6ff16e161dd9","_number":708882,"virtual_id_number":708882,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-24 17:54:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2020-03-23 16:51:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2020-03-24 16:13:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2020-03-24 16:15:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-23 16:51:34.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-03-24 16:15:07.000000000","updated_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"},{"updated":"2020-03-24 17:54:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c552a60faa49acb0edffbf343dba7afdd546dcea","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-20 13:54:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"90b5ed1999dd60fe8fd4e50ce41c13c442375ab3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-20 15:37:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ce3cf94cf0eb466eb818affa8cdb6537 : SUCCESS in 1h 37m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0031bfc728d7456281ed578b2d682de8 : SUCCESS in 7m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80a206c95dd541cf92c45b16cf43ff9a : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/85d2b097dbc940f1b746d75cbed6af1f : SUCCESS in 6m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5cc916dee81c46e3976820f52c0ecb42 : SUCCESS in 6m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/48f2744f28664734bff1e7f139200cd2 : SUCCESS in 6m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/214709197bde41e59a332842bd0557ab : SUCCESS in 5m 24s\n- cliff-tox-py37-neutronclient-tip https://zuul.opendev.org/t/openstack/build/b0f37cdd4edf4ec8bd3eeecbe4aee987 : SUCCESS in 7m 55s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/50cbd656de364984b691e241a0516d52 : SUCCESS in 7m 26s","accounts_in_message":[],"_revision_number":1},{"id":"2fe03be0af8a7076cbc5d886986d6afc5f8fcbf0","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-03-23 16:51:34.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"33d13f5a4a86a139e0791b297e502caf537a2336","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2020-03-24 16:13:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"af12daf3542eead0a492e549145a69c6f87815a5","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2020-03-24 16:15:07.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"938b2039abca55a511dbea02dd114bc600410947","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 16:17:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0d5edd559302a4886e4a1c60a5a9442ed535657e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 17:54:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/05abcc9b5f3347e78aa47617d5227a7f : SUCCESS in 1h 32m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/81ba76c5d20b4507b84afed0437b4276 : SUCCESS in 9m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9363d3dae514089ba096619063598fb : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/35985b0dcd664739aa2c4a7d1cfb089d : SUCCESS in 4m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ef179e3323504bbbb582287c865f2fb8 : SUCCESS in 5m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/045f19d6c8c34e28ba10e3c5eada8e3e : SUCCESS in 4m 53s\n- cliff-tox-py37-neutronclient-tip https://zuul.opendev.org/t/openstack/build/def9ae662d984521aeeca94443587a93 : SUCCESS in 6m 43s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/c5327d8e236d4bcb967622dbe67c4259 : SUCCESS in 6m 19s","accounts_in_message":[],"_revision_number":1},{"id":"796ec8de659241defcc780caf2efa61926fda21d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 17:54:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"aa81a3850b7575f78e3a2fe1548a6ff16e161dd9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 17:56:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80a7dd6aca774213ae524ae4b91f1fd9 : SUCCESS in 1m 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1ae0201aacbb83235c6c78dd569c9dd8fa69ceb0","revisions":{"1ae0201aacbb83235c6c78dd569c9dd8fa69ceb0":{"kind":"REWORK","_number":1,"created":"2020-02-20 13:54:22.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/82/708882/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/82/708882/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/82/708882/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/82/708882/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/82/708882/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/82/708882/1"}}},"commit":{"parents":[{"commit":"1e04cb683eae98917b628df810becc13216f0605","subject":"adding missing releasenote for the drop of py27 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/1e04cb683eae98917b628df810becc13216f0605"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-20 13:52:37.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-20 13:54:20.000000000","tz":0},"subject":"Fix nested argument groups with ignore conflict handler","message":"Fix nested argument groups with ignore conflict handler\n\nCliff provides an additional conflict handler for argparse - \u0027ignore\u0027.\nTo make this work it subclasses the argparse argument container classes\nand adds a handle_conflict_ignore method. It also overrides the group\ncreation methods to return a subclassed argument group or mutually\nexclusive argument group objects. This works, until you add a nested\nargument group, at which point a standard argument group object is\nreturned which does not support the ignore conflict handler.\n\nThis change fixes the issue by returning subclassed group objects for\nnested groups.\n\nChange-Id: I517c61f24ba6194ff6181e115a3a23adbce3ea53\nStory: 2005891\nTask: 33749\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/1ae0201aacbb83235c6c78dd569c9dd8fa69ceb0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/1ae0201aacbb83235c6c78dd569c9dd8fa69ceb0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
