)]}'
{"id":"openstack%2Fcliff~715305","triplet_id":"openstack%2Fcliff~master~I14fca07853677d041215649cb55caf1e0dc49753","project":"openstack/cliff","branch":"master","hashtags":[],"change_id":"I14fca07853677d041215649cb55caf1e0dc49753","subject":"Re-add support for python 3.5","status":"MERGED","created":"2020-03-26 20:34:29.000000000","updated":"2020-03-26 23:45:56.000000000","submitted":"2020-03-26 23:43:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"715305-1585266229201-b8390b35","meta_rev_id":"6da7cba341b8cc38dfbfde847228b801762c81ff","_number":715305,"virtual_id_number":715305,"owner":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":16068,"name":"Tobias Henkel","email":"tobias.henkel@bmw.de","username":"tobias.henkel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-26 23:43:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-03-26 20:52:36.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-03-26 20:38:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":1,"date":"2020-03-26 20:38:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2020-03-26 20:38:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16068,"name":"Tobias Henkel","email":"tobias.henkel@bmw.de","username":"tobias.henkel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":1,"date":"2020-03-26 20:39:04.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":16068,"name":"Tobias Henkel","email":"tobias.henkel@bmw.de","username":"tobias.henkel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":16068,"name":"Tobias Henkel","email":"tobias.henkel@bmw.de","username":"tobias.henkel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-26 20:38:35.000000000","updated_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"reviewer":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"state":"REVIEWER"},{"updated":"2020-03-26 20:38:40.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-03-26 20:38:41.000000000","updated_by":{"_account_id":16068,"name":"Tobias Henkel","email":"tobias.henkel@bmw.de","username":"tobias.henkel"},"reviewer":{"_account_id":16068,"name":"Tobias Henkel","email":"tobias.henkel@bmw.de","username":"tobias.henkel"},"state":"REVIEWER"},{"updated":"2020-03-26 20:52:36.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-03-26 23:43:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cf4fd75cca3f30a89d8f0c83dd0423833e2941dd","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-03-26 20:34:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f8f9dc80d9e034d701858c19de00c1cb21d94566","author":{"_account_id":16068,"name":"Tobias Henkel","email":"tobias.henkel@bmw.de","username":"tobias.henkel"},"date":"2020-03-26 20:35:00.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f467b81c66264f8fa67766c47aa7d0161c23d8ec","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2020-03-26 20:36:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5cab8c2a6fe98f22bf34ce5cff751ddc52ce20c5","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-03-26 20:37:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3156bb8102af50328c79d5d5382ef44c2ed786ed","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2020-03-26 20:37:59.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ada046b55c2a65475590a61e2440ea375f971f60","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-26 20:38:09.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"052b7254e8d35ba859945d4f1dd99dbd5f68120a","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-03-26 20:38:19.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5bd14d3de8c858725ea786372d3d9ac475474ac4","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2020-03-26 20:38:35.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"5fb55da49c69f74ad114a22f31e069211043c38b","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-26 20:38:40.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"82033685289ffb9685c03f4d5ec9e54679257b32","author":{"_account_id":16068,"name":"Tobias Henkel","email":"tobias.henkel@bmw.de","username":"tobias.henkel"},"date":"2020-03-26 20:38:41.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"5db6703f5e73559f7bdd1adb970df774cada21f6","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-03-26 20:39:04.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"2230c61b48d558dfe3ca62fcdd6291fd6600468c","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-26 20:52:36.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"50c02459e725f39ce125d9b7f6522d06c283276b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-26 22:14:11.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e5bba58e864e493a82434509dc47467e : SUCCESS in 3m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fec81899575d44e6a7722cd4f6edeaf4 : SUCCESS in 1h 33m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b4f232a34e5d4d73b3d3d6af948a0a32 : SUCCESS in 4m 56s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/72c67072bb5f4ac48eab9287b151e4dc : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1684fa6bbe124966bfcfc5ab09bf78e7 : SUCCESS in 3m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/04286e1a2a7e423182af9697abbc9669 : SUCCESS in 3m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f16c79ff3ad34b1dac735206d3010bf6 : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/785b49491a1c48418c690bba333972a7 : SUCCESS in 3m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e759a715df7b468da2d1c31806bb5da0 : SUCCESS in 3m 31s\n- cliff-tox-py37-neutronclient-tip https://zuul.opendev.org/t/openstack/build/108ec856a1fa43d59db3047e304737ee : SUCCESS in 4m 10s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/a4ef681d42904c4eadde7f3c41cd77ea : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":3},{"id":"d8ee0431af4e9aae8e07c247bfd7eb25b1261161","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-26 22:14:26.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ef3d62bd6198151b6c972f6ac341f2aac942ca6c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-26 23:43:49.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/39e3f976d243453a9fd334f5af1f6430 : SUCCESS in 2m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5938ece59b4348c0885df5e2d914041a : SUCCESS in 1h 24m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3c3a7a301dd44982a52512948afe1c0f : SUCCESS in 4m 57s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a65c0eed97c54ae28d699ab233342869 : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f1e55b8520046bcb5081ccd08a50e97 : SUCCESS in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f529892038e14aff8e38c359cfa8fb33 : SUCCESS in 4m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a8e8dbe78a1c4988913a0f65303c5947 : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96d54dc8098045a79f0dbf83e257e67c : SUCCESS in 3m 45s\n- cliff-tox-py37-neutronclient-tip https://zuul.opendev.org/t/openstack/build/46499eb485474c0e909f166d14cfa50b : SUCCESS in 5m 15s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/c939c9414065428581494c4e9cb44bdf : SUCCESS in 6m 33s","accounts_in_message":[],"_revision_number":3},{"id":"624c8f13a6a27b0fca34ef1b0b8b30b8da4ad5c3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-26 23:43:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"6da7cba341b8cc38dfbfde847228b801762c81ff","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-26 23:45:56.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac3ae498e3914e9185e5ceee210e8b59 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"5405c3dfd76a48a36f8c847dcd8daa38cfb7668c","revisions":{"726ea0066dd1c1b7c7e073283d31dc1c2de70da2":{"kind":"REWORK","_number":1,"created":"2020-03-26 20:34:29.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/05/715305/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/05/715305/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/05/715305/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/05/715305/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/05/715305/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/05/715305/1"}}},"commit":{"parents":[{"commit":"1ae0201aacbb83235c6c78dd569c9dd8fa69ceb0","subject":"Fix nested argument groups with ignore conflict handler","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/1ae0201aacbb83235c6c78dd569c9dd8fa69ceb0"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2020-03-26 20:33:45.000000000","tz":-300},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2020-03-26 20:34:28.000000000","tz":-300},"subject":"Re-add support for python 3.5","message":"Re-add support for python 3.5\n\ncliff is used by stestr which is used by more things than just\nopenstack projects - one of them being zuul which still supports\npython 3.5. Let\u0027s keep supporting 3.5 here for now.\n\nChange-Id: I14fca07853677d041215649cb55caf1e0dc49753\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/726ea0066dd1c1b7c7e073283d31dc1c2de70da2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/726ea0066dd1c1b7c7e073283d31dc1c2de70da2"}]},"branch":"refs/heads/master"},"1f9b67daa4ec5f4b0c38bb93e67cc3b1ef9ed4f6":{"kind":"REWORK","_number":2,"created":"2020-03-26 20:37:25.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/05/715305/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/05/715305/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/05/715305/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/05/715305/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/05/715305/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/05/715305/2"}}},"commit":{"parents":[{"commit":"1ae0201aacbb83235c6c78dd569c9dd8fa69ceb0","subject":"Fix nested argument groups with ignore conflict handler","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/1ae0201aacbb83235c6c78dd569c9dd8fa69ceb0"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2020-03-26 20:33:45.000000000","tz":-300},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2020-03-26 20:37:13.000000000","tz":-300},"subject":"Re-add support for python 3.5","message":"Re-add support for python 3.5\n\ncliff is used by stestr which is used by more things than just\nopenstack projects - one of them being zuul which still supports\npython 3.5. Let\u0027s keep supporting 3.5 here for now.\n\nChange-Id: I14fca07853677d041215649cb55caf1e0dc49753\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/1f9b67daa4ec5f4b0c38bb93e67cc3b1ef9ed4f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/1f9b67daa4ec5f4b0c38bb93e67cc3b1ef9ed4f6"}]},"branch":"refs/heads/master"},"5405c3dfd76a48a36f8c847dcd8daa38cfb7668c":{"kind":"REWORK","_number":3,"created":"2020-03-26 20:38:19.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/05/715305/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/05/715305/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/05/715305/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/05/715305/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/05/715305/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/05/715305/3"}}},"commit":{"parents":[{"commit":"1ae0201aacbb83235c6c78dd569c9dd8fa69ceb0","subject":"Fix nested argument groups with ignore conflict handler","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/1ae0201aacbb83235c6c78dd569c9dd8fa69ceb0"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2020-03-26 20:33:45.000000000","tz":-300},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2020-03-26 20:38:14.000000000","tz":-300},"subject":"Re-add support for python 3.5","message":"Re-add support for python 3.5\n\ncliff is used by stestr which is used by more things than just\nopenstack projects - one of them being zuul which still supports\npython 3.5. Let\u0027s keep supporting 3.5 here for now.\n\nChange-Id: I14fca07853677d041215649cb55caf1e0dc49753\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/5405c3dfd76a48a36f8c847dcd8daa38cfb7668c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/5405c3dfd76a48a36f8c847dcd8daa38cfb7668c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
