)]}'
{"id":"openstack%2Fcliff~735364","triplet_id":"openstack%2Fcliff~master~Ie15e3f5058c4bd104978d9f31f0590d6c795024b","project":"openstack/cliff","branch":"master","topic":"remove_cmd2_cap","hashtags":[],"change_id":"Ie15e3f5058c4bd104978d9f31f0590d6c795024b","subject":"Remove cap on cmd2","status":"MERGED","created":"2020-06-12 15:49:02.000000000","updated":"2020-06-13 01:26:06.000000000","submitted":"2020-06-13 01:24:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"735364-1592011498870-4f1731f4","meta_rev_id":"1ca52477e8e7196030adb82bbcd0bef64f0cc36a","_number":735364,"virtual_id_number":735364,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-13 01:24:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-06-12 21:53:34.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2020-06-12 21:43:16.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2020-06-12 21:43:16.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-12 16:14:56.000000000","updated_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"},{"updated":"2020-06-12 21:43:16.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2020-06-12 21:53:34.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-06-13 01:24:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2f812e88dd5e1b16fc2d5cdb60d75b9a8203433d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-06-12 15:49:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"739925b2cb8e62c2d9e52fbcb5ab7ef2d25e99b3","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-12 15:51:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"912fdc2412a2125e450f830a2d68247a8cd7d7af","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-06-12 16:05:23.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"43f1835919d563278ad3799619806ff7f842b90b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-06-12 16:07:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"541510bdc2b3eb775f8bdf6d130ffb2ebee11a5d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-06-12 16:08:28.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"46a22e25cbf85020bb44c6a38bad0740992e153c","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-06-12 16:14:33.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"291187f62f341a7f20865b8b827944b007f859f5","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2020-06-12 16:14:56.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"4f2fa4d7f568eb227c9e7ea26954cf1f9a5d69a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-12 17:23:15.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/ee202e04adb241eead5db14d6afb7726 : SUCCESS in 2m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5ec9215a5bd24aa6986588361055d8c4 : FAILURE in 1h 13m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dc731219746448eeb4aa932a06dee08e : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8579c4cf06024d5abe461ce6cbaec25f : SUCCESS in 3m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/751b5527c17948039b45e0e6b59c7f3f : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a8fb03effb574794af1dca14f275c23f : SUCCESS in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/336a492baba147389ee3997fab4b94ae : FAILURE in 3m 16s\n- cliff-tox-py37-neutronclient-tip https://zuul.opendev.org/t/openstack/build/ead6611ef1104f089c0cc674253aefde : SUCCESS in 4m 16s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/bfb94f9ab28444978bb87a8b767f635e : SUCCESS in 5m 43s","accounts_in_message":[],"_revision_number":2},{"id":"e32c8e2ca944a65e059c922de205ff974fb89e1e","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-06-12 17:43:43.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"742a787978628116ff6783c855c3f3b98b277c4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-12 19:07:09.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/81ba157aefcb4398b939fcfbd224f063 : SUCCESS in 3m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ab76cba460e9440f8c542ab3f208a61a : SUCCESS in 1h 21m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c542374af9a943e38acd80dcd5bd6c5c : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d71efd7ac6b0489483c2a11097a11a51 : SUCCESS in 3m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c4b86932343f4eaab818c629dfb8be64 : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4039c99ed653460c9cf89eca7d0db2c4 : FAILURE in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab29462733664a029991f2b194957d45 : FAILURE in 3m 40s\n- cliff-tox-py37-neutronclient-tip https://zuul.opendev.org/t/openstack/build/b69153b9b6574d71909030a346c4d2d3 : SUCCESS in 4m 01s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/68d8371142284a3aab7b017e778a3bc2 : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":2},{"id":"043e43c5d34490c22753e26b0f23d27c39fc1f52","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-06-12 21:43:03.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1116bbea56b6d430c67a21faf032b3b9b5dac0e6","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-06-12 21:43:16.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"1f91cc49d810f42ed2b2d9afc61a52967b36ef5c","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-12 21:53:34.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"d5b522d5e6af16aab290d579e364f14c377f912b","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-06-12 22:46:11.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"de74acd8120e226025c5edb2940f2c8c2ec3fdaa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-13 00:07:26.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/b0b8e1bde4a24fab8a31322fc54b2c21 : SUCCESS in 1m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f1db10537417443dbd67d09781fa1509 : SUCCESS in 1h 20m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4e1a616a7aeb40aeaf52d2fc39e9b560 : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3c15898ec41471abda17223448d5f5f : SUCCESS in 2m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3f0604885c2a4f5990b8d83d83514866 : SUCCESS in 2m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0ad6bf8d29984cc79ec7ea32568b5159 : SUCCESS in 3m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32194f0e4bed475fbe3b87ef550fc141 : SUCCESS in 3m 35s\n- cliff-tox-py37-neutronclient-tip https://zuul.opendev.org/t/openstack/build/504872f3215b498399969628e59e1fe7 : SUCCESS in 4m 00s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/245b2f673a8b40f68ecf7127d17321bd : SUCCESS in 4m 37s","accounts_in_message":[],"_revision_number":3},{"id":"05ab7006eec0c2fcd0fb5eb405925919c306133d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-13 00:07:37.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f29f19a4c97c71b8f6f83807eb50c692174a0486","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-13 01:24:58.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/307c270e5b084807a0ded388ce2e6087 : SUCCESS in 2m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ce768a474097417ab142a43db6816fca : SUCCESS in 1h 12m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/751df49bea7c4d1f911d1c9a526073c7 : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8dbf343f94d4908b4295e616dab04c0 : SUCCESS in 2m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3021d81761be45598e42490a56d85dc3 : SUCCESS in 2m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/52aa9da666e34865949f03bbc103466b : SUCCESS in 3m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0584dac9514547c38f5fdd86f8631fdf : SUCCESS in 3m 15s\n- cliff-tox-py37-neutronclient-tip https://zuul.opendev.org/t/openstack/build/889e17981c384dbe9bdbf9e2b3373e2b : SUCCESS in 11m 26s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/dfae50da21d24351a6d1363e9ad05f82 : SUCCESS in 4m 39s","accounts_in_message":[],"_revision_number":3},{"id":"7264be62eafc852372bea69afe18463f43d43813","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-13 01:24:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"b854f16f5aa5d36ae298bec14860eb24820d176d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-13 01:26:06.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47d6046f7f764a8c85c543d6beada968 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8099c98249dfc9caa9cdbac4cbe7afba4279c734","revisions":{"399793ebf28e12a72bef5fd043ccda45cd277fd8":{"kind":"REWORK","_number":1,"created":"2020-06-12 15:49:02.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/64/735364/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/64/735364/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/64/735364/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/64/735364/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/64/735364/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/64/735364/1"}}},"commit":{"parents":[{"commit":"6989b47cc234444b7357b47b61c9a9308d0ea06e","subject":"Merge \"Fix compatibility with new cmd2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/6989b47cc234444b7357b47b61c9a9308d0ea06e"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-06-12 15:45:48.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-06-12 15:48:36.000000000","tz":0},"subject":"Remove cap on cmd2","message":"Remove cap on cmd2\n\nAfter [1] and [2], there is no need to cap the version of cmd2.\n\nThis will also fix the current \"rally_openstack\" import problems\nexperienced in the CI: http://paste.openstack.org/show/794701/\n\nChange-Id: Ie15e3f5058c4bd104978d9f31f0590d6c795024b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/399793ebf28e12a72bef5fd043ccda45cd277fd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/399793ebf28e12a72bef5fd043ccda45cd277fd8"}]},"branch":"refs/heads/master"},"03d0c460abd64aa707dc90ca0494975594a13d4c":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-06-12 16:08:28.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/64/735364/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/64/735364/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/64/735364/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/64/735364/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/64/735364/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/64/735364/2"}}},"commit":{"parents":[{"commit":"6989b47cc234444b7357b47b61c9a9308d0ea06e","subject":"Merge \"Fix compatibility with new cmd2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/6989b47cc234444b7357b47b61c9a9308d0ea06e"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-06-12 15:45:48.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-06-12 16:08:25.000000000","tz":0},"subject":"Remove cap on cmd2","message":"Remove cap on cmd2\n\nAfter [1] and [2], there is no need to cap the version of cmd2.\n\nThis will also fix the current \"rally_openstack\" import problems\nexperienced in the CI: http://paste.openstack.org/show/794701/\n\n[1]https://review.opendev.org/#/c/712591/\n[2]https://review.opendev.org/#/c/734629/\n\nChange-Id: Ie15e3f5058c4bd104978d9f31f0590d6c795024b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/03d0c460abd64aa707dc90ca0494975594a13d4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/03d0c460abd64aa707dc90ca0494975594a13d4c"}]},"branch":"refs/heads/master"},"8099c98249dfc9caa9cdbac4cbe7afba4279c734":{"kind":"REWORK","_number":3,"created":"2020-06-12 21:43:03.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/64/735364/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/64/735364/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/64/735364/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/64/735364/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/64/735364/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/64/735364/3"}}},"commit":{"parents":[{"commit":"6989b47cc234444b7357b47b61c9a9308d0ea06e","subject":"Merge \"Fix compatibility with new cmd2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/6989b47cc234444b7357b47b61c9a9308d0ea06e"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-06-12 15:45:48.000000000","tz":0},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2020-06-12 21:42:56.000000000","tz":-300},"subject":"Remove cap on cmd2","message":"Remove cap on cmd2\n\nAfter [1] and [2], there is no need to cap the version of cmd2.\n\nThis will also fix the current \"rally_openstack\" import problems\nexperienced in the CI: http://paste.openstack.org/show/794701/\n\n[1]https://review.opendev.org/#/c/712591/\n[2]https://review.opendev.org/#/c/734629/\n\nChange-Id: Ie15e3f5058c4bd104978d9f31f0590d6c795024b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/8099c98249dfc9caa9cdbac4cbe7afba4279c734"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/8099c98249dfc9caa9cdbac4cbe7afba4279c734"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
