)]}'
{"id":"openstack%2Fcliff~735465","triplet_id":"openstack%2Fcliff~master~I0d5d5c56a447e990b2fe72ff7fa220ac19c1dae9","project":"openstack/cliff","branch":"master","topic":"osc-plugin-optimize","hashtags":[],"change_id":"I0d5d5c56a447e990b2fe72ff7fa220ac19c1dae9","subject":"Add method to load builtins by string","status":"NEW","created":"2020-06-14 13:59:10.000000000","updated":"2022-11-08 08:19:53.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f881cb94117e8cae5dc53be7ddd176b6c503ff13","_number":735465,"virtual_id_number":735465,"owner":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-11-08 08:19:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-15 17:33:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9c554ab0e85fe16b088227217b276df12b5ec948","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-06-14 13:59:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"94e88d9fdc24e2ea4e12e416bc2aeadca72448d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-14 15:25:57.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8d622d5676844b16b54e7d22c77ad13e : SUCCESS in 1h 25m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f7bc7edf86d14b8ea88c099f538dcba0 : FAILURE in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5be8879c5b0447c8eae50fffdfbb16f : FAILURE in 3m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0f2997bda05744428e82e461b0523e29 : FAILURE in 3m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0d72c3f3843a4e89a67add94d54c8d7d : FAILURE in 3m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6b3077be8124a99b85ea236372e1a6d : SUCCESS in 3m 40s\n- cliff-tox-py37-neutronclient-tip https://zuul.opendev.org/t/openstack/build/f4d7c5b237104978ba30972e0e7c97bc : SUCCESS in 4m 01s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/d2dfd14b33b842b5a5fe1e9e872dbe8f : SUCCESS in 5m 05s","accounts_in_message":[],"_revision_number":1},{"id":"367b575c90984ee60758364d637291118985b288","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-06-15 15:33:37.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"42ae52beeb09e6252a3fe86876edb77ba8a8fa61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 17:33:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ebc92e17a6d74150803beb27e494efd4 : SUCCESS in 1h 34m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2b8ef6e0617e4194aba78630c91881eb : SUCCESS in 6m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fcbe75c0d8046f089590388e4bdf421 : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/44de1b45b60d454ab99b4c5925d46ced : SUCCESS in 6m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1c69cdb728324da9ad347c853ebb53b2 : SUCCESS in 8m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c186cccbe69410baa7a813387363a32 : SUCCESS in 7m 24s\n- cliff-tox-py37-neutronclient-tip https://zuul.opendev.org/t/openstack/build/dc7b23e57142452f919ba80b6a6d53fb : SUCCESS in 8m 32s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/aefe559e9a8648c7b6b42058bb0b341a : SUCCESS in 7m 46s","accounts_in_message":[],"_revision_number":2},{"id":"b896324636f22e9d83a8ce4120d6eec3567d14ed","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-10-10 15:56:00.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2},{"id":"c1153fe6e7b8e508bd4410d761228051bd0dbb1c","tag":"autogenerated:gerrit:restore","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2022-11-07 16:24:29.000000000","message":"Restored","accounts_in_message":[],"_revision_number":2},{"id":"27534dbfabb6fa529f5c0b960c89a6c4e090bf64","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2022-11-07 16:24:40.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8e161c6cc4a3325e5aaccc92cc2b26c0869ca6cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2022-11-07 16:26:42.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"969fc6768963fc90f551ad99de67a867a32d77b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 18:08:56.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29ad4cae8866483e90da5ad2f7a39703\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/74b8654fc3484188a2828e8626c9b5fc : SUCCESS in 1h 35m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f88c138bed64fddb17cb3e14912cc24 : SUCCESS in 2m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/337e888c70514a1c9a4c979467d7658c : SUCCESS in 3m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a33d3d3222764ce7ad371274ecd1adaf : SUCCESS in 3m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb15afe16a3d4a60ba2bd3d521e0c9c5 : SUCCESS in 3m 11s\n- cliff-tox-py38-neutronclient-tip https://zuul.opendev.org/t/openstack/build/3ac59ce8c03f4b28984a1a4b0290990d : SUCCESS in 3m 48s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/64492e8638f345668b65ba28a50d44c0 : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":4},{"id":"8f25b947a14c905032dcb294b1effd9f00ca2aaa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-11-08 07:07:33.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"f881cb94117e8cae5dc53be7ddd176b6c503ff13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-08 08:19:53.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65f3f9ae9ca649b48ad624ec4e164c80\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b0d0a07a091f4e1592cb304e051a9627 : SUCCESS in 1h 09m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8907bebbe63c4dd7850fd4828d1b7f84 : SUCCESS in 4m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4ce192140ebf434e8fc8fb7ae0fd6ef4 : SUCCESS in 5m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e40e7d2c79a349468c8b4a6d5ec397b4 : SUCCESS in 3m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4803a0d4f7345f58f9db7a24b76cb24 : SUCCESS in 2m 32s\n- cliff-tox-py38-neutronclient-tip https://zuul.opendev.org/t/openstack/build/c402da636f534b778949188ba5643ac8 : SUCCESS in 6m 25s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/6b9d28c58d7e4a44be2dfc17c127c8f6 : SUCCESS in 4m 13s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"800a6c287e059fa72dc5a912fcefca8f46869931","revisions":{"39ec00b48b2f2491b8564edb33aacb0711f868bb":{"kind":"REWORK","_number":1,"created":"2020-06-14 13:59:10.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/65/735465/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/65/735465/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/65/735465/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/65/735465/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/65/735465/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/65/735465/1"}}},"commit":{"parents":[{"commit":"71cedd6a208f7c4ad6f8ac2385a364409ca582ae","subject":"Add load_plugins option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/71cedd6a208f7c4ad6f8ac2385a364409ca582ae"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2020-06-13 15:01:14.000000000","tz":-300},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2020-06-14 13:46:24.000000000","tz":-300},"subject":"Add method to load builtins by string","message":"Add method to load builtins by string\n\nTo make it easy to keep a list of builtins and then to load them,\nallow passing builtins by name similar to how they work in setup.cfg.\n\nChange-Id: I0d5d5c56a447e990b2fe72ff7fa220ac19c1dae9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/39ec00b48b2f2491b8564edb33aacb0711f868bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/39ec00b48b2f2491b8564edb33aacb0711f868bb"}]},"branch":"refs/heads/master"},"6bf788c9620c9e7481941945d6e3bd09d028d1a2":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-06-15 15:33:37.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/65/735465/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/65/735465/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/65/735465/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/65/735465/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/65/735465/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/65/735465/2"}}},"commit":{"parents":[{"commit":"2395ac5fd9536c403669e82cffdf2ff26d258bf5","subject":"Add load_plugins option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/2395ac5fd9536c403669e82cffdf2ff26d258bf5"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2020-06-13 15:01:14.000000000","tz":-300},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2020-06-15 15:33:10.000000000","tz":-300},"subject":"Add method to load builtins by string","message":"Add method to load builtins by string\n\nTo make it easy to keep a list of builtins and then to load them,\nallow passing builtins by name similar to how they work in setup.cfg.\n\nChange-Id: I0d5d5c56a447e990b2fe72ff7fa220ac19c1dae9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/6bf788c9620c9e7481941945d6e3bd09d028d1a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/6bf788c9620c9e7481941945d6e3bd09d028d1a2"}]},"branch":"refs/heads/master"},"676c963107ad3bb29f1b78fd3016a9d96fa12129":{"kind":"REWORK","_number":3,"created":"2022-11-07 16:24:40.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/65/735465/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/65/735465/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/65/735465/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/65/735465/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/65/735465/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/65/735465/3"}}},"commit":{"parents":[{"commit":"ca11fe185c40e877eddb1d59fac6f609fa3c60cb","subject":"Add load_plugins option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/ca11fe185c40e877eddb1d59fac6f609fa3c60cb"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2020-06-13 15:01:14.000000000","tz":-300},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2022-11-07 16:23:38.000000000","tz":60},"subject":"Add method to load builtins by string","message":"Add method to load builtins by string\n\nTo make it easy to keep a list of builtins and then to load them,\nallow passing builtins by name similar to how they work in setup.cfg.\n\nChange-Id: I0d5d5c56a447e990b2fe72ff7fa220ac19c1dae9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/676c963107ad3bb29f1b78fd3016a9d96fa12129"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/676c963107ad3bb29f1b78fd3016a9d96fa12129"}]},"branch":"refs/heads/master"},"aa813ed335d995a43c5de53206623398f1fb3f61":{"kind":"REWORK","_number":4,"created":"2022-11-07 16:26:42.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/65/735465/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/65/735465/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/65/735465/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/65/735465/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/65/735465/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/65/735465/4"}}},"commit":{"parents":[{"commit":"ca11fe185c40e877eddb1d59fac6f609fa3c60cb","subject":"Add load_plugins option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/ca11fe185c40e877eddb1d59fac6f609fa3c60cb"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2020-06-13 15:01:14.000000000","tz":-300},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2022-11-07 16:26:34.000000000","tz":60},"subject":"Add method to load builtins by string","message":"Add method to load builtins by string\n\nTo make it easy to keep a list of builtins and then to load them,\nallow passing builtins by name similar to how they work in setup.cfg.\n\nChange-Id: I0d5d5c56a447e990b2fe72ff7fa220ac19c1dae9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/aa813ed335d995a43c5de53206623398f1fb3f61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/aa813ed335d995a43c5de53206623398f1fb3f61"}]},"branch":"refs/heads/master"},"800a6c287e059fa72dc5a912fcefca8f46869931":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2022-11-08 07:07:33.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/735465/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/65/735465/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/65/735465/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/65/735465/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/65/735465/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/65/735465/5"}}},"commit":{"parents":[{"commit":"5079a9239fd6c709aed4976a7519c6b8607637b9","subject":"Add load_plugins option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/5079a9239fd6c709aed4976a7519c6b8607637b9"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2020-06-13 15:01:14.000000000","tz":-300},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2022-11-08 07:06:43.000000000","tz":60},"subject":"Add method to load builtins by string","message":"Add method to load builtins by string\n\nTo make it easy to keep a list of builtins and then to load them,\nallow passing builtins by name similar to how they work in setup.cfg.\n\nChange-Id: I0d5d5c56a447e990b2fe72ff7fa220ac19c1dae9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/800a6c287e059fa72dc5a912fcefca8f46869931"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/800a6c287e059fa72dc5a912fcefca8f46869931"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
