)]}'
{"id":"openstack%2Fcliff~739329","triplet_id":"openstack%2Fcliff~master~Ib7bf53091470b55ab87082d315ca283d3600a636","project":"openstack/cliff","branch":"master","topic":"osc-performance","hashtags":[],"change_id":"Ib7bf53091470b55ab87082d315ca283d3600a636","subject":"switch to stevedore for loading entry points","status":"MERGED","created":"2020-07-04 18:44:49.000000000","updated":"2020-07-06 19:11:59.000000000","submitted":"2020-07-06 19:09:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"739329-1594062587217-f3086ce0","meta_rev_id":"d6f2302cd25e16aa8aea527df6b695998ee12a51","_number":739329,"virtual_id_number":739329,"owner":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-06 19:09:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":2,"date":"2020-07-06 17:32:10.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":2,"date":"2020-07-06 09:26:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":1,"date":"2020-07-06 17:32:10.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-06 09:26:23.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-07-06 17:32:10.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2020-07-06 19:09:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1b641ac4713f0b615d98cd937b11584bd53bd337","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2020-07-04 18:44:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5a77439b6c840c41eb648f424ec288b297ec37de","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2020-07-04 18:45:08.000000000","message":"Patch Set 1: Code-Review-2","accounts_in_message":[],"_revision_number":1},{"id":"4eaac915a4026ffbf4766f7f6c5dc8312d6d76d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-04 19:01:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/a230ba42d2444bbdb3e7735eae7d8e7e : FAILURE in 2m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5c092a8d932d4289ac3738876cc229c9 : FAILURE in 16m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bb050f9b4ee14f129df1db5d238cf118 : FAILURE in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d7bb3a9cdbb41d4bd42d1d830c3c636 : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e9b309013ba143278d2f570f7187b45d : FAILURE in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4bef4b41ef0f4e1e82aa8ac8efd2e1b4 : FAILURE in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2551fa435c84434888c0c431d3f5953 : FAILURE in 3m 30s\n- cliff-tox-py37-neutronclient-tip https://zuul.opendev.org/t/openstack/build/7bc83462cf5749c5b1c3bcd3ab8b4507 : FAILURE in 4m 40s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/8058349cde474004be054a57e8506229 : FAILURE in 4m 24s","accounts_in_message":[],"_revision_number":1},{"id":"bbb33417f4c04ee27787c978546d3297d7383794","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2020-07-04 19:15:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"586b1586a891342dba1812ab0f2d14960a541492","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-04 19:37:10.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/84727c7b65d84360b6490825de7957d9 : FAILURE in 2m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/01351aa4e8c8473e955f79225019eb3b : FAILURE in 20m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f0979bc6dae543f79e8906ab01e6a3b6 : FAILURE in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3388849afcae449198ae091cdcc51046 : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d33b89a3cb674b35b1f1b36bb53d3c81 : FAILURE in 3m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/31b2b7c628a64561bfb91d1d4d8b9374 : FAILURE in 3m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9aca2d0e0dc149199239ab18caf4f16a : FAILURE in 3m 26s\n- cliff-tox-py37-neutronclient-tip https://zuul.opendev.org/t/openstack/build/843148a547d44587a0b59ba247a173f2 : FAILURE in 3m 25s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/351c9890f7024067ac3c45edb9652a56 : FAILURE in 5m 17s","accounts_in_message":[],"_revision_number":2},{"id":"c63e494011ff964f43cfa83bbe8c9c09107fa5d3","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2020-07-04 21:26:19.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"37bc62f1d8c663df1a35f8d338a7fbe79811dc46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-04 21:53:25.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/f95b1dbc7459490fb6aad10347e6b20d : SUCCESS in 1m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a8bea8d8820a49088203753f81b658e5 : FAILURE in 21m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dae51c16458b41398424f031a7c2da9c : FAILURE in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/369dd2c2c6234da690920c454eb394a4 : SUCCESS in 2m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3c97a7d09a7d4089b76944bf82ca259f : FAILURE in 2m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cd63061247da42d0b73fdf81e1160d5d : FAILURE in 3m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fbae79c625f4fb1a0385fca6014f64b : FAILURE in 2m 51s\n- cliff-tox-py37-neutronclient-tip https://zuul.opendev.org/t/openstack/build/ed48e099aa0d4614ada3626228b1b9c0 : FAILURE in 3m 26s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/8c7cbc2594244cb79f1feb2dd7e10c78 : FAILURE in 4m 18s","accounts_in_message":[],"_revision_number":3},{"id":"e71ece53d233f79e1b8411399db98db9958c0a24","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2020-07-05 19:38:05.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"6fb15904007db45c3fc4856abe59f78041928fa8","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2020-07-05 19:57:54.000000000","message":"Patch Set 4: -Code-Review","accounts_in_message":[],"_revision_number":4},{"id":"9076423592254a57a097a3e2b44da125f746c3a7","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2020-07-05 19:59:33.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"ac1efce537f4abf86170e10c5cad88fe9f57ffac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-05 21:20:53.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/a8be534ede4c4cc58cce9e24b0eddabc : SUCCESS in 2m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dd322b0a80924e8f8658b2da984fa75c : SUCCESS in 1h 19m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/260f173624444765a5866b8662e94cd5 : SUCCESS in 3m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5466fc2c20974db09b875dce206248c0 : SUCCESS in 3m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7ba378a816fc4d6f855272fbb8c60c0c : SUCCESS in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8db36104e6e94d96966e4056c8fb9f25 : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19e80d09e2a1411ca07dacb3c278a2e7 : SUCCESS in 3m 59s\n- cliff-tox-py37-neutronclient-tip https://zuul.opendev.org/t/openstack/build/5fb88e8ef1314cac9b071fcedf61c481 : SUCCESS in 4m 31s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/efc16fa9d6064dfc84034d7a4f5cd29a : SUCCESS in 4m 51s","accounts_in_message":[],"_revision_number":5},{"id":"a63c5b5c63eac8213bd389001457558353ea0215","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-06 09:26:23.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"839c66a5d0763eee1f3bd5fea65bdfd4a3cace14","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-07-06 17:32:10.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"1fcc29b97c7360e755d552f66575fa3d42fbdd33","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 17:32:22.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"25037854d5fefc5a7bfc36dbf039529bd2a3688b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 19:09:46.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/e6b958a411c14ec4a4489d33a49a0420 : SUCCESS in 5m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d6a9729a95294575b8b199f4a1261a7f : SUCCESS in 1h 31m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/569d221b93fa4a0f8d2031ae3127dde9 : SUCCESS in 7m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26b351e286ea43f89428c46ed51b4039 : SUCCESS in 6m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f1b77bd3f23f43caa2d51c38884f737e : SUCCESS in 7m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e19f69b6bfff4f1390d8f9abe8e26e97 : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffe27015200d4a3f960e3b4f9aea1d70 : SUCCESS in 4m 44s\n- cliff-tox-py37-neutronclient-tip https://zuul.opendev.org/t/openstack/build/1901d5e664664f8b9f6d1ddaf0b8584a : SUCCESS in 7m 53s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/0720794f6dc441b59b06da787416c967 : SUCCESS in 6m 34s","accounts_in_message":[],"_revision_number":5},{"id":"a5414c0433633cf4e31cb34391a70a558532e45e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 19:09:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"7d0bc11826ee414dfffa1703f51c4b28c204a1c3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 19:11:59.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17cca15a7e9642e790a6a0f222fceb40 : SUCCESS in 1m 51s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"6ac510d0cd25c8c3c76d56dd8e100ca2a6f8b962","revisions":{"9706579a52d902c41c32b3f22223322c78769c64":{"kind":"REWORK","_number":1,"created":"2020-07-04 18:44:49.000000000","uploader":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"ref":"refs/changes/29/739329/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/29/739329/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/29/739329/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/29/739329/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/29/739329/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/29/739329/1"}}},"commit":{"parents":[{"commit":"8099c98249dfc9caa9cdbac4cbe7afba4279c734","subject":"Remove cap on cmd2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/8099c98249dfc9caa9cdbac4cbe7afba4279c734"}]}],"author":{"name":"Doug Hellmann","email":"doug@doughellmann.com","date":"2020-07-04 18:43:49.000000000","tz":-240},"committer":{"name":"Doug Hellmann","email":"doug@doughellmann.com","date":"2020-07-04 18:43:49.000000000","tz":-240},"subject":"switch to stevedore for loading entry points","message":"switch to stevedore for loading entry points\n\nUsing the entrypoints cache API in stevedore should make command line\nprograms run more quickly, since they won\u0027t need to scan every\ndistribution every time they start.\n\nChange-Id: Ib7bf53091470b55ab87082d315ca283d3600a636\nDepends-On: https://review.opendev.org/739326\nSigned-off-by: Doug Hellmann \u003cdoug@doughellmann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/9706579a52d902c41c32b3f22223322c78769c64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/9706579a52d902c41c32b3f22223322c78769c64"}]},"branch":"refs/heads/master"},"83b5c2ba412b275d467299e456e25e508c9ad62e":{"kind":"REWORK","_number":2,"created":"2020-07-04 19:15:42.000000000","uploader":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"ref":"refs/changes/29/739329/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/29/739329/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/29/739329/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/29/739329/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/29/739329/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/29/739329/2"}}},"commit":{"parents":[{"commit":"8099c98249dfc9caa9cdbac4cbe7afba4279c734","subject":"Remove cap on cmd2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/8099c98249dfc9caa9cdbac4cbe7afba4279c734"}]}],"author":{"name":"Doug Hellmann","email":"doug@doughellmann.com","date":"2020-07-04 18:43:49.000000000","tz":-240},"committer":{"name":"Doug Hellmann","email":"doug@doughellmann.com","date":"2020-07-04 19:15:35.000000000","tz":-240},"subject":"switch to stevedore for loading entry points","message":"switch to stevedore for loading entry points\n\nUsing the entrypoints cache API in stevedore should make command line\nprograms run more quickly, since they won\u0027t need to scan every\ndistribution every time they start.\n\nChange-Id: Ib7bf53091470b55ab87082d315ca283d3600a636\nDepends-On: https://review.opendev.org/739326\nSigned-off-by: Doug Hellmann \u003cdoug@doughellmann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/83b5c2ba412b275d467299e456e25e508c9ad62e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/83b5c2ba412b275d467299e456e25e508c9ad62e"}]},"branch":"refs/heads/master"},"3b25bf1357b0345b3588d3d802b763cb7eb59449":{"kind":"REWORK","_number":3,"created":"2020-07-04 21:26:19.000000000","uploader":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"ref":"refs/changes/29/739329/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/29/739329/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/29/739329/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/29/739329/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/29/739329/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/29/739329/3"}}},"commit":{"parents":[{"commit":"8099c98249dfc9caa9cdbac4cbe7afba4279c734","subject":"Remove cap on cmd2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/8099c98249dfc9caa9cdbac4cbe7afba4279c734"}]}],"author":{"name":"Doug Hellmann","email":"doug@doughellmann.com","date":"2020-07-04 18:43:49.000000000","tz":-240},"committer":{"name":"Doug Hellmann","email":"doug@doughellmann.com","date":"2020-07-04 21:26:10.000000000","tz":-240},"subject":"switch to stevedore for loading entry points","message":"switch to stevedore for loading entry points\n\nUsing the entrypoints cache API in stevedore should make command line\nprograms run more quickly, since they won\u0027t need to scan every\ndistribution every time they start.\n\nChange-Id: Ib7bf53091470b55ab87082d315ca283d3600a636\nDepends-On: https://review.opendev.org/739326\nSigned-off-by: Doug Hellmann \u003cdoug@doughellmann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/3b25bf1357b0345b3588d3d802b763cb7eb59449"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/3b25bf1357b0345b3588d3d802b763cb7eb59449"}]},"branch":"refs/heads/master"},"81b5b8eb17607349be1a82b00c054f599c5169bc":{"kind":"REWORK","_number":4,"created":"2020-07-05 19:38:05.000000000","uploader":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"ref":"refs/changes/29/739329/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/29/739329/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/29/739329/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/29/739329/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/29/739329/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/29/739329/4"}}},"commit":{"parents":[{"commit":"8099c98249dfc9caa9cdbac4cbe7afba4279c734","subject":"Remove cap on cmd2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/8099c98249dfc9caa9cdbac4cbe7afba4279c734"}]}],"author":{"name":"Doug Hellmann","email":"doug@doughellmann.com","date":"2020-07-04 18:43:49.000000000","tz":-240},"committer":{"name":"Doug Hellmann","email":"doug@doughellmann.com","date":"2020-07-05 19:37:25.000000000","tz":-240},"subject":"switch to stevedore for loading entry points","message":"switch to stevedore for loading entry points\n\nUsing the entrypoints cache API in stevedore should make command line\nprograms run more quickly, since they won\u0027t need to scan every\ndistribution every time they start.\n\nChange-Id: Ib7bf53091470b55ab87082d315ca283d3600a636\nDepends-On: I8b08f289d446f4775eac1e1a91997fa96f25f641\nDepends-On: Ic6db7af34c87a636bfe55bacae03c42154f4b9c7\nSigned-off-by: Doug Hellmann \u003cdoug@doughellmann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/81b5b8eb17607349be1a82b00c054f599c5169bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/81b5b8eb17607349be1a82b00c054f599c5169bc"}]},"branch":"refs/heads/master"},"6ac510d0cd25c8c3c76d56dd8e100ca2a6f8b962":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2020-07-05 19:59:33.000000000","uploader":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"ref":"refs/changes/29/739329/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/29/739329/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/29/739329/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/29/739329/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/29/739329/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/29/739329/5"}}},"commit":{"parents":[{"commit":"8099c98249dfc9caa9cdbac4cbe7afba4279c734","subject":"Remove cap on cmd2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/8099c98249dfc9caa9cdbac4cbe7afba4279c734"}]}],"author":{"name":"Doug Hellmann","email":"doug@doughellmann.com","date":"2020-07-04 18:43:49.000000000","tz":-240},"committer":{"name":"Doug Hellmann","email":"doug@doughellmann.com","date":"2020-07-05 19:58:46.000000000","tz":-240},"subject":"switch to stevedore for loading entry points","message":"switch to stevedore for loading entry points\n\nSwitch to using stevedore now so that when the cache implementation is\nreleased all cliff applications can take advantage of the performance\nbenefits.\n\nChange-Id: Ib7bf53091470b55ab87082d315ca283d3600a636\nSigned-off-by: Doug Hellmann \u003cdoug@doughellmann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/6ac510d0cd25c8c3c76d56dd8e100ca2a6f8b962"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/6ac510d0cd25c8c3c76d56dd8e100ca2a6f8b962"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
