)]}'
{"id":"openstack%2Fcliff~750588","triplet_id":"openstack%2Fcliff~master~I478dd2e6b539d9e7bd0470c9cd4c764dc90c4448","project":"openstack/cliff","branch":"master","attention_set":{},"removed_from_attention_set":{"27900":{"account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"last_update":"2024-08-07 15:32:24.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I478dd2e6b539d9e7bd0470c9cd4c764dc90c4448","subject":"Enable osc-functional-devstack-tips job","status":"ABANDONED","created":"2020-09-09 07:47:50.000000000","updated":"2024-08-07 15:32:24.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5a13e36c6a8c87d4f0eeea64ce8adfa275194435","_number":750588,"virtual_id_number":750588,"owner":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"date":"2020-11-06 08:24:30.000000000","_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-03-23 15:10:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2022-03-23 12:04:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2020-11-17 18:12:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"date":"2022-03-23 13:43:30.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-09 15:46:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-17 18:12:59.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2022-03-04 19:42:20.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2022-03-23 12:04:26.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"d46f58dd08121bce4723e23b8681ab9125605a4a","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2020-09-09 07:47:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ec00b22393b8e4d2722d9fd1c18edeab89ab363d","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2020-09-09 07:49:02.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"f73742020a53845367dd7c27a48d3061d35a5c2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 09:09:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/45711314b83c4946b33c6baa36383035 : SUCCESS in 1h 13m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b0964031eb1647e6b86135f7ff8ba705 : SUCCESS in 5m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c882f5262b44198bf6b9d34d8ebbde4 : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/540ae419b4ca494f8237c0de26eda849 : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/de82a95294db41b1942c95742df452af : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b6d54ac56654613b535c10c6637a3a5 : SUCCESS in 5m 25s\n- cliff-tox-py37-neutronclient-tip https://zuul.opendev.org/t/openstack/build/f7568de94d804a638e875b0bdb4a6e6b : SUCCESS in 4m 57s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/996b5ee7edf54ee88d5a7c2989e3eba3 : SUCCESS in 6m 10s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/dcbaeb65067d416e9e9116cd2a93abed : FAILURE in 47m 04s","accounts_in_message":[],"_revision_number":2},{"id":"fbf7ca4bfbbc18d236be158c57f0be4184c7e945","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2020-09-09 09:34:01.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"ae2e539d27880f94725313906e3758f36d794eaa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 09:34:57.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"1e6d357a8fe006f3b1b88f4b6713227ad0e768a7","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2020-09-09 12:26:12.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"2ee00cf5da76a03385bffe0de27854d5137d450b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 14:07:27.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a7f504330a034a66bfa9a67388db70c1 : SUCCESS in 1h 32m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/48ab4fff38f34b92a9d4a8a0cf8c7259 : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee83e9c1988447e5a57948b31403b77c : SUCCESS in 3m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5bd265009b024412aef60aa8ee3c8921 : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c21eb7636859490882394368d9825a30 : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f338c580bd2b4abdb98d0ba2d7d695c9 : SUCCESS in 3m 59s\n- cliff-tox-py37-neutronclient-tip https://zuul.opendev.org/t/openstack/build/d9778ae65ab34ea38ecd2e2c677be731 : SUCCESS in 4m 29s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/823a00d72b1d425789b4f9684cdd296e : SUCCESS in 7m 23s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/442f5076a2054efd86e7c76076852fb8 : FAILURE in 50m 27s","accounts_in_message":[],"_revision_number":3},{"id":"9bc134e1650128fc6421628b1b0fbc99bd0fa5bb","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2020-09-09 14:27:49.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"6a912ec0290b54759882469d88996650023ed68b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 15:46:27.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fad11bbd0b504a3ca2c5d19c4fee6584 : SUCCESS in 1h 15m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/60ef945bca954f86bcc3cb87a6cf1718 : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c54c3dc84ee94b5588a4f46ae83fbb0f : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b1c6561a31554891bd3272fda141d638 : SUCCESS in 3m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/51b18ecbe3954e25b20fdbe9a57683a4 : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98f40715076d43a6ac52a815fd67c1aa : SUCCESS in 5m 27s\n- cliff-tox-py37-neutronclient-tip https://zuul.opendev.org/t/openstack/build/458e09b06379468a8b2d5b4c1b51c552 : SUCCESS in 5m 23s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/e6b124b2ff324c4e80218de615d60867 : SUCCESS in 6m 02s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/1c8c469991c74c859c812ffa31fbfd29 : SUCCESS in 57m 21s","accounts_in_message":[],"_revision_number":4},{"id":"ba584ba9d01dfcbbc4652461d3ffd21b5b0a0806","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-10-29 11:37:09.000000000","message":"Patch Set 4: Code-Review-1\n\ncliff changes very slowly nowadays, if at all, but the functional jobs are slow. Is there a recent example of it breaking OSC? If not, I\u0027d be tempted to say this is premature optimization of sorts and we\u0027d be better off waiting until we have a demonstrable example of this need","accounts_in_message":[],"_revision_number":4},{"id":"7decd67d807f43f244d83dbd71f150960b447926","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2020-11-06 08:24:30.000000000","message":"Patch Set 4:\n\n\u003e cliff changes very slowly nowadays, if at all, but the functional\n \u003e jobs are slow. Is there a recent example of it breaking OSC? If\n \u003e not, I\u0027d be tempted to say this is premature optimization of sorts\n \u003e and we\u0027d be better off waiting until we have a demonstrable example\n \u003e of this need\n\nYes, we had this recently (the depends-on issue was a fix)","accounts_in_message":[],"_revision_number":4},{"id":"4faf3131c76bd1db56de64b324534ff2fdfccb07","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-11-17 18:12:59.000000000","message":"Patch Set 4: Code-Review+2\n\n\u003e \u003e cliff changes very slowly nowadays, if at all, but the functional\n \u003e \u003e jobs are slow. Is there a recent example of it breaking OSC? If\n \u003e \u003e not, I\u0027d be tempted to say this is premature optimization of\n \u003e sorts\n \u003e \u003e and we\u0027d be better off waiting until we have a demonstrable\n \u003e example\n \u003e \u003e of this need\n \u003e \n \u003e Yes, we had this recently (the depends-on issue was a fix)\n\nOkay then, let\u0027s do it","accounts_in_message":[],"_revision_number":4},{"id":"74b01744d16d83378547cde85335a37dd1bd1f8f","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-03-04 19:42:20.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2fa1126af262af782f5d34e1854ead59030b2e84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-04 21:37:23.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4c3b698dbfa1427fb6f969585700ed41 : SUCCESS in 1h 53m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2303c1b659fa4051974eafcf84bb0d78 : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d3e93de079c943f8a80bbcadf0ad2c7e : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d86312ff425f4f72b788201c433cbac5 : SUCCESS in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0aaae0e165b94e11a6c2a5d2a8c8a5f0 : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c674c22d03164d869df05be988a22978 : SUCCESS in 5m 43s\n- cliff-tox-py37-neutronclient-tip https://zuul.opendev.org/t/openstack/build/ef07d7da8ecc476cbec7a16ac4fe3dcc : SUCCESS in 5m 47s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/1e0f01a665b5496895e51b85a85659ee : SUCCESS in 6m 43s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/0e09443416f345c0bd18d78bb359ec0b : SUCCESS in 59m 11s","accounts_in_message":[],"_revision_number":4},{"id":"1f60ad1d8044810cc44d1b97b1c7c188ec3b1ba3","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-03-23 12:04:26.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"6b5fa68ac38c5f07e8499ae7260f84b155a8b1a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-23 12:05:25.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"7dacba5b4832a1b2089c0a8e6f61f31cfc2b405a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-23 13:38:27.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1ac664c1420b4878b8c5bf5fec4505ba : SUCCESS in 1h 26m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/057db47cba4647ca871cec2cb5301ea8 : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0ecb81f304394cbbbd771fcb22d85fc6 : SUCCESS in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9ab112ba35e249ce898892a1964bc274 : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/545b5f2502ba40f69311b489ad1a9d36 : SUCCESS in 5m 18s\n- cliff-tox-py37-neutronclient-tip https://zuul.opendev.org/t/openstack/build/739f6331ea4b41aa954e8c389bb4b974 : SUCCESS in 6m 19s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/0eedf4657bda496f98b8a38aba554b53 : SUCCESS in 7m 02s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/521533d92dca4a2f870d0d5e194af3c5 : FAILURE in 1h 00m 13s","accounts_in_message":[],"_revision_number":4},{"id":"6ecd2fa04e8d5c38ce142427728e3b6378ca31d5","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-03-23 13:43:30.000000000","message":"Patch Set 4: -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"30cdcb918354127e1d3dcb9bcc7445215218c448","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-23 15:10:50.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8c6bd6cf7da24337b8a389e62962ccc0 : SUCCESS in 1h 24m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa75c1759d14414c926f30f2bea6ae4d : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2b2aefce55d343f0a1de13da0a9c8acb : SUCCESS in 5m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c919f7895f884a2f858d79f3a78404ca : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b556aa946def4941b30c6a441cda96ad : SUCCESS in 5m 20s\n- cliff-tox-py37-neutronclient-tip https://zuul.opendev.org/t/openstack/build/1d82ce898c884ce293ad8c7c43c05efd : SUCCESS in 6m 58s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/200c9ccd152f48c6bb1226b732491b1b : SUCCESS in 7m 49s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/70860e50f671416f8f51234d60ab908e : SUCCESS in 56m 37s","accounts_in_message":[],"_revision_number":4},{"id":"5a13e36c6a8c87d4f0eeea64ce8adfa275194435","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-08-07 15:32:24.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"649cb3c2a63c284d8a4cf7f847b45d7232997c8b","revisions":{"de830dfc8bca84751a23bbfd2ff361335a774c0b":{"kind":"REWORK","_number":1,"created":"2020-09-09 07:47:50.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/88/750588/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/88/750588/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/88/750588/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/88/750588/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/88/750588/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/88/750588/1"}}},"commit":{"parents":[{"commit":"28c172f70133208f3f7aacc1f45b0125ae4ec240","subject":"Capturing argparse errors due to problem with cmd2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/28c172f70133208f3f7aacc1f45b0125ae4ec240"}]}],"author":{"name":"Artem Goncharov","email":"Artem.goncharov@gmail.com","date":"2020-09-09 07:42:23.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"Artem.goncharov@gmail.com","date":"2020-09-09 07:44:14.000000000","tz":120},"subject":"Enable osc-functional-devstack-tips tests","message":"Enable osc-functional-devstack-tips tests\n\nCliff might break osc. Do not let this and test it in tips.\nThis is not optimal, but reality tells us it is worth it.\n\nChange-Id: I478dd2e6b539d9e7bd0470c9cd4c764dc90c4448\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/de830dfc8bca84751a23bbfd2ff361335a774c0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/de830dfc8bca84751a23bbfd2ff361335a774c0b"}]},"branch":"refs/heads/master"},"aba974cca8502f2494e2beba3f939118ed5ae7bf":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-09-09 07:49:02.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/88/750588/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/88/750588/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/88/750588/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/88/750588/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/88/750588/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/88/750588/2"}}},"commit":{"parents":[{"commit":"28c172f70133208f3f7aacc1f45b0125ae4ec240","subject":"Capturing argparse errors due to problem with cmd2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/28c172f70133208f3f7aacc1f45b0125ae4ec240"}]}],"author":{"name":"Artem Goncharov","email":"Artem.goncharov@gmail.com","date":"2020-09-09 07:42:23.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2020-09-09 07:48:49.000000000","tz":0},"subject":"Enable osc-functional-devstack-tips job","message":"Enable osc-functional-devstack-tips job\n\nCliff might break osc. Do not let this and test it in tips.\nThis is not optimal, but reality tells us it is worth it.\n\nChange-Id: I478dd2e6b539d9e7bd0470c9cd4c764dc90c4448\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/aba974cca8502f2494e2beba3f939118ed5ae7bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/aba974cca8502f2494e2beba3f939118ed5ae7bf"}]},"branch":"refs/heads/master"},"86aadfe003fd676e3a206cd951279381497523b8":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-09-09 09:34:01.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/88/750588/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/88/750588/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/88/750588/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/88/750588/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/88/750588/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/88/750588/3"}}},"commit":{"parents":[{"commit":"28c172f70133208f3f7aacc1f45b0125ae4ec240","subject":"Capturing argparse errors due to problem with cmd2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/28c172f70133208f3f7aacc1f45b0125ae4ec240"}]}],"author":{"name":"Artem Goncharov","email":"Artem.goncharov@gmail.com","date":"2020-09-09 07:42:23.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2020-09-09 09:33:55.000000000","tz":0},"subject":"Enable osc-functional-devstack-tips job","message":"Enable osc-functional-devstack-tips job\n\nCliff might break osc. Do not let this and test it in tips.\nThis is not optimal, but reality tells us it is worth it.\n\nDepends-On: https://review.opendev.org/#/c/750585/\n\nChange-Id: I478dd2e6b539d9e7bd0470c9cd4c764dc90c4448\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/86aadfe003fd676e3a206cd951279381497523b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/86aadfe003fd676e3a206cd951279381497523b8"}]},"branch":"refs/heads/master"},"649cb3c2a63c284d8a4cf7f847b45d7232997c8b":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-09-09 14:27:49.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/88/750588/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/88/750588/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/88/750588/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/88/750588/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/88/750588/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/88/750588/4"}}},"commit":{"parents":[{"commit":"28c172f70133208f3f7aacc1f45b0125ae4ec240","subject":"Capturing argparse errors due to problem with cmd2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/28c172f70133208f3f7aacc1f45b0125ae4ec240"}]}],"author":{"name":"Artem Goncharov","email":"Artem.goncharov@gmail.com","date":"2020-09-09 07:42:23.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2020-09-09 14:27:14.000000000","tz":0},"subject":"Enable osc-functional-devstack-tips job","message":"Enable osc-functional-devstack-tips job\n\nCliff might break osc. Do not let this and test it in tips.\nThis is not optimal, but reality tells us it is worth it.\n\nDepends-On: https://review.opendev.org/#/c/750648/\n\nChange-Id: I478dd2e6b539d9e7bd0470c9cd4c764dc90c4448\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/649cb3c2a63c284d8a4cf7f847b45d7232997c8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/649cb3c2a63c284d8a4cf7f847b45d7232997c8b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
