)]}'
{"id":"openstack%2Fcliff~756188","triplet_id":"openstack%2Fcliff~master~I9bd4886ebe05f920c4a23d0f4ffe5878aeacb92c","project":"openstack/cliff","branch":"master","hashtags":[],"change_id":"I9bd4886ebe05f920c4a23d0f4ffe5878aeacb92c","subject":"Remove six","status":"ABANDONED","created":"2020-10-06 03:41:37.000000000","updated":"2020-10-29 11:42:37.000000000","total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"77004fe9e50ebfcc9ba5ec9650c6fe501a878459","_number":756188,"virtual_id_number":756188,"owner":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"date":"2020-10-06 14:16:34.000000000","_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-10-06 09:09:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"all":[{"value":-1,"date":"2020-10-29 11:13:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2020-10-29 11:38:55.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-06 03:43:06.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2020-10-06 09:09:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-06 14:16:34.000000000","updated_by":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"reviewer":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"state":"REVIEWER"},{"updated":"2020-10-29 11:13:43.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-10-29 11:38:55.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"}],"messages":[{"id":"035b2149e354cceae10e32147dc36b4c23f27bc8","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2020-10-06 03:41:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5c722740bce043d554ac82a0a150f48d0ec548bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 05:07:24.000000000","message":"Patch Set 1: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/5ee0728c5bad4720bb9069156a276f62 : SUCCESS in 1m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fd395eab81d74ed69b9ebcce796f8d74 : SUCCESS in 1h 20m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/30090113bc944ce48f9d6a16b8ca9510 : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7dbdb0defc1b4a6cb07dd6e1e29f1686 : FAILURE in 3m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5e6143c6dcbe418db3d2582cc8f2412d : SUCCESS in 2m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c4446fc5fd14df2b79e767983280bb3 : SUCCESS in 3m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc2d4d90de0940f99a5355f49e11a7e0 : SUCCESS in 3m 04s\n- cliff-tox-py37-neutronclient-tip https://zuul.opendev.org/t/openstack/build/332c49ab48cf452e909c841318a2e8a3 : SUCCESS in 3m 34s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/336b268ee8c24818900c903400a11b6b : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":1},{"id":"151e6b0f38ce95cba9c9635c9846d368d7593c43","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2020-10-06 07:32:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d55937bc23cd63223164740c1e0baa759b1ad3ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 09:09:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/79ae0f4f15fe4731a127d3d50659c6fb : SUCCESS in 2m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/313a70efd3534bff91ef720a694814cc : SUCCESS in 1h 23m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4704746902a74fda84990bf4678b2a16 : SUCCESS in 6m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97e1712fe08b4f148ee9557bd9f5c9b2 : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c2280495f6474da9a465c8b7deb0486c : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fc0f86ecfd5d4c93ad62c6b3963e3427 : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fcfbfd58eac44738ac0f21fb0ec5e3df : SUCCESS in 4m 53s\n- cliff-tox-py37-neutronclient-tip https://zuul.opendev.org/t/openstack/build/f875ad5b6a8b40d989aecb70a39172f2 : SUCCESS in 4m 57s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/26742d4fba9e4fbabe8ffe3f5207cbbb : SUCCESS in 6m 44s","accounts_in_message":[],"_revision_number":2},{"id":"845b169c3ef43e805e8ffde424f718603352e914","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2020-10-06 14:16:34.000000000","message":"Patch Set 2:\n\nThis is a duplicate of https://review.opendev.org/#/c/754683/","accounts_in_message":[],"_revision_number":2},{"id":"f8751d9b9df9db49277fa3b31efae92d48068425","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-10-29 11:13:43.000000000","message":"Patch Set 2: Code-Review-1\n\n(5 comments)\n\nthis is a duplicate of https://review.opendev.org/#/c/754683/, btw","accounts_in_message":[],"_revision_number":2},{"id":"f56f5bdb2a3e7814ded5a384c4eef21ec2bc36f9","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-10-29 11:38:55.000000000","message":"Patch Set 2: Code-Review-1","accounts_in_message":[],"_revision_number":2},{"id":"77004fe9e50ebfcc9ba5ec9650c6fe501a878459","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2020-10-29 11:42:37.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8dbe15ee265acb984eee09af4866742973f92ad3","revisions":{"c244b6c62c9ad35ad5781946b1b4e5140c8b1bf4":{"kind":"REWORK","_number":1,"created":"2020-10-06 03:41:37.000000000","uploader":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"ref":"refs/changes/88/756188/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/88/756188/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/88/756188/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/88/756188/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/88/756188/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/88/756188/1"}}},"commit":{"parents":[{"commit":"ff4ee455773cb3a66999495ee20ba015dee722c4","subject":"Merge \"Add Python3 wallaby unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/ff4ee455773cb3a66999495ee20ba015dee722c4"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-10-06 03:40:59.000000000","tz":480},"committer":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-10-06 03:41:30.000000000","tz":480},"subject":"Remove six","message":"Remove six\n\nChange-Id: I9bd4886ebe05f920c4a23d0f4ffe5878aeacb92c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/c244b6c62c9ad35ad5781946b1b4e5140c8b1bf4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/c244b6c62c9ad35ad5781946b1b4e5140c8b1bf4"}]},"branch":"refs/heads/master"},"8dbe15ee265acb984eee09af4866742973f92ad3":{"kind":"REWORK","_number":2,"created":"2020-10-06 07:32:02.000000000","uploader":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"ref":"refs/changes/88/756188/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/88/756188/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/88/756188/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/88/756188/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/88/756188/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/88/756188/2"}}},"commit":{"parents":[{"commit":"ff4ee455773cb3a66999495ee20ba015dee722c4","subject":"Merge \"Add Python3 wallaby unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/ff4ee455773cb3a66999495ee20ba015dee722c4"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-10-06 03:40:59.000000000","tz":480},"committer":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-10-06 07:31:08.000000000","tz":480},"subject":"Remove six","message":"Remove six\n\nWith python3.x, classes can use \u0027metaclass\u003d\u0027 instead of\n\u0027six.add_metaclass\u0027, \u0027six.iteritems\u0027 and \u0027six.iterkeys\u0027 can\nbe replaced by \u0027items\u0027 and \u0027keys\u0027, \u0027six.moves.urllib.parse\u0027\ncan be replaced by \u0027urllib.parse\u0027, \u0027six.StringIO\u0027 and\n\u0027six.moves.cStringIO\u0027 can be replaced by \u0027io.StringIO\u0027,\n\u0027six.text_type\u0027 and \u0027six.string_type\u0027 are just \u0027str\u0027.\n\nChange-Id: I9bd4886ebe05f920c4a23d0f4ffe5878aeacb92c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/8dbe15ee265acb984eee09af4866742973f92ad3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/8dbe15ee265acb984eee09af4866742973f92ad3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
