)]}'
{"id":"openstack%2Fcliff~772866","triplet_id":"openstack%2Fcliff~master~I08e1f1bc75fa6452f19dfb9d221c1daec194d58d","project":"openstack/cliff","branch":"master","topic":"sort-dir","hashtags":[],"change_id":"I08e1f1bc75fa6452f19dfb9d221c1daec194d58d","subject":"Make \u0027FormattableColumn\u0027 comparable","status":"MERGED","created":"2021-01-28 11:44:10.000000000","updated":"2021-02-10 12:13:49.000000000","submitted":"2021-02-10 12:12:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"772866-sort-dir","meta_rev_id":"5dd64f4e847f0a173dc31bead055a1bf59a8f030","_number":772866,"virtual_id_number":772866,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-10 12:12:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"all":[{"value":2,"date":"2021-02-10 12:13:49.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-02-09 16:43:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"all":[{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-02-09 16:43:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-28 13:21:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-28 15:27:03.000000000","updated_by":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"reviewer":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"state":"REVIEWER"},{"updated":"2021-02-09 16:43:58.000000000","updated_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"}],"messages":[{"id":"32b1b01daebc3c0d463fa1cf17107a8632144f19","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-01-28 11:44:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"85df568065ea21f47bfb50b4843f227aedfc5ab7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 13:21:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/93cee272819e40ffa29cd08781374e92 : SUCCESS in 1h 26m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67eb4ff64c7a426aba1042a76cafb674 : SUCCESS in 3m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9af3d88dfb854cd49da6868d2978c4b3 : SUCCESS in 3m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/30689ebb11ff4c33ba5d3a8004323f4a : SUCCESS in 3m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/16a1fcd7fcb3471c8b3e6880a1953f5b : SUCCESS in 3m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8912a98ce804e268593f4afa6a872a0 : SUCCESS in 3m 35s\n- cliff-tox-py37-neutronclient-tip https://zuul.opendev.org/t/openstack/build/987fcbf1f62c4b88ac004237b376abe4 : SUCCESS in 4m 34s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/144ac9d02ba1434796882fc05ca0e4db : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":1},{"id":"72b9e19ce5db0dc19c139cfb40ec792fe45a6c28","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2021-01-28 15:27:03.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9f490683148d628d03f71e2e950df92623eb6ab2","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-01-28 15:49:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e08ea63ac26f236a458a5282e580bc923544129e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-01-28 16:14:06.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"7bf3859744a35ad27953a2f98002e901b3cd5131","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 17:47:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/56153862bec94d68b86bbd1a1213a309 : SUCCESS in 1h 23m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/699bbc79ae3546589e3ca2963add68d9 : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dccd325e27104b2f99eaee592f24df6d : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a5262b61a43b4c6c984493433f898682 : SUCCESS in 3m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6244bce4e5044f338a03823e1c0642b9 : SUCCESS in 4m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bc0b071c88c4f188d44ec14da63869f : SUCCESS in 5m 00s\n- cliff-tox-py37-neutronclient-tip https://zuul.opendev.org/t/openstack/build/2f6f5ffa36b745bba2e74b68efc319f6 : SUCCESS in 4m 35s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/31c1247ffba54ac5b8629bb2a1da3cca : SUCCESS in 5m 06s","accounts_in_message":[],"_revision_number":2},{"id":"881fd1f6598af130baed82c39d7b446239a8a6e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-01-29 15:40:54.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"52121f9ea2bb819a7680453b95ab9b4df5473980","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-29 17:18:31.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cae900dec6034a98a934953c6d802948 : SUCCESS in 1h 31m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8999b86305114e4e98a6ac6d250c6eb3 : SUCCESS in 3m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0c85a39d87bc4df09314e8476725515f : SUCCESS in 3m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e618c1c1913a4ee8b7f0594ca83b328a : SUCCESS in 3m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7eb959e1c70d4a0eb4137a155c27a8a4 : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/784b5af4c048495d8e418cbf81f05a93 : SUCCESS in 4m 31s\n- cliff-tox-py37-neutronclient-tip https://zuul.opendev.org/t/openstack/build/1faeb0859f5349bb86ad12bc069033c1 : SUCCESS in 4m 22s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/c6d542ffebaa483aa3c5412d37bb1acf : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":3},{"id":"a9a6b20bb2bfcf91b4ebd3d7b2f01498c4b75855","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2021-02-09 16:43:58.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"dacb0ede8dca2debe5e5f34e77e4d2a2d73c0148","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 16:46:48.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"08bbb0597d46e5cd81ca11fab180aa1e84cad97b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 18:35:59.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d5633dc1f2e344df9f63215f2af99bc4 : FAILURE in 1h 39m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6b940564fc74c13a0cb10778a22b944 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/841ca482559548bbb6c75e8a202dad60 : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/362b5b92c64c4c9aa90c207875fa8817 : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb69d34ad48c42d6966ed70b57209e58 : SUCCESS in 3m 21s\n- cliff-tox-py37-neutronclient-tip https://zuul.opendev.org/t/openstack/build/cd07b7c90daa49deb3974724aa3bfb46 : SUCCESS in 4m 37s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/08f96b6b479c47a9bf5bc93d1991f681 : SUCCESS in 7m 19s","accounts_in_message":[],"_revision_number":3},{"id":"ec296fd58124e9e2e80aaeb5884ffa38b988ef21","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-02-10 09:17:02.000000000","message":"Patch Set 3:\n\nrecheck failed to delete volume","accounts_in_message":[],"_revision_number":3},{"id":"483bb30bdf82f676434da3488eaf86b6d8a57f9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 10:26:46.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c00455d8910e4cabbc01ccd10c7fe917 : SUCCESS in 1h 07m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f2ca94f68104e3bb9b94c3eb04e2c88 : SUCCESS in 2m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/89d505ecd3d04b87b96907aaa2c6e6cb : SUCCESS in 3m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a2451e5efa624718b7696176cb06e9cc : SUCCESS in 3m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/79011e32fb9b4479b26402be882aa575 : SUCCESS in 2m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc757b4eb6e74c528c6e3d89fd41e421 : SUCCESS in 3m 09s\n- cliff-tox-py37-neutronclient-tip https://zuul.opendev.org/t/openstack/build/4465883c89194e17bed67bea63acaa56 : SUCCESS in 3m 21s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/f85e39f8e2b64c45a9d7b918b0df7143 : SUCCESS in 5m 14s","accounts_in_message":[],"_revision_number":3},{"id":"8f227706e2ef377f4bf552b66d6576c77dc5b999","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 10:27:08.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"81235d8ef253a1b1a39c2b5cfc9f49bf0b7c1abb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 12:12:30.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2a018a96341b4979be14670d347fe5ef : SUCCESS in 1h 42m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/098196137b8348bbbaa13fe2b94ca458 : SUCCESS in 2m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/00f092ccbd7143c8b88d0cd5ea78bd6e : SUCCESS in 3m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0164defe220c48ccb1d31c62a5e2dafd : SUCCESS in 3m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f82773aa57b64c5e9ef318391e823965 : SUCCESS in 3m 00s\n- cliff-tox-py37-neutronclient-tip https://zuul.opendev.org/t/openstack/build/fe9ed6856ee44cfd9af0522f03c37259 : SUCCESS in 4m 28s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/ba50b5f6c988488a869c52673c288017 : SUCCESS in 5m 23s","accounts_in_message":[],"_revision_number":3},{"id":"51c91e038fc24446f56aa74955de262db0fffb74","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 12:12:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"c2509ef2a36d6c9e8383de7d6a77b28ec9147792","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 12:13:49.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36009eeada3e42a095a098167fd2ed12 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c1c991045cece85dd55494f9d4670429e370e131","revisions":{"88a4774eeada0ab5432100b584985c00f0aa711a":{"kind":"REWORK","_number":1,"created":"2021-01-28 11:44:10.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/66/772866/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/66/772866/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/66/772866/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/66/772866/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/66/772866/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/66/772866/1"}}},"commit":{"parents":[{"commit":"609b2fc987664b77a51ab9f099461440a04202f9","subject":"Handle null values when sorting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/609b2fc987664b77a51ab9f099461440a04202f9"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2021-01-28 11:34:44.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2021-01-28 11:43:34.000000000","tz":0},"subject":"Make \u0027FormattableColumn\u0027 comparable","message":"Make \u0027FormattableColumn\u0027 comparable\n\nImplement the \u0027__lt__\u0027 magic method, thus providing the minimal set of\nrich comparison methods necessary to support sorting. This will allows\nusers using these formatters for the more basic types (i.e. not dicts)\nto sort their output using the standard \u0027--sort-column\u0027 option.\n\nChange-Id: I08e1f1bc75fa6452f19dfb9d221c1daec194d58d\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/88a4774eeada0ab5432100b584985c00f0aa711a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/88a4774eeada0ab5432100b584985c00f0aa711a"}]},"branch":"refs/heads/master"},"a18b3fa60a2c4f1a15567f4fa5406fb5032b01fd":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-01-28 16:14:06.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/66/772866/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/66/772866/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/66/772866/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/66/772866/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/66/772866/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/66/772866/2"}}},"commit":{"parents":[{"commit":"e86f653b91c210ddcc9136d8fba61ae1c44b8e92","subject":"Handle null values when sorting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/e86f653b91c210ddcc9136d8fba61ae1c44b8e92"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2021-01-28 11:34:44.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2021-01-28 16:13:50.000000000","tz":0},"subject":"Make \u0027FormattableColumn\u0027 comparable","message":"Make \u0027FormattableColumn\u0027 comparable\n\nImplement the \u0027__lt__\u0027 magic method, thus providing the minimal set of\nrich comparison methods necessary to support sorting. This will allows\nusers using these formatters for the more basic types (i.e. not dicts)\nto sort their output using the standard \u0027--sort-column\u0027 option.\n\nChange-Id: I08e1f1bc75fa6452f19dfb9d221c1daec194d58d\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/a18b3fa60a2c4f1a15567f4fa5406fb5032b01fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/a18b3fa60a2c4f1a15567f4fa5406fb5032b01fd"}]},"branch":"refs/heads/master"},"c1c991045cece85dd55494f9d4670429e370e131":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-01-29 15:40:54.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/66/772866/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/66/772866/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/66/772866/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/66/772866/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/66/772866/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/66/772866/3"}}},"commit":{"parents":[{"commit":"4f45f9a30e657265c5b5ac119af3aaa0e5ec7184","subject":"Handle null values when sorting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/4f45f9a30e657265c5b5ac119af3aaa0e5ec7184"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2021-01-28 11:34:44.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2021-01-29 15:40:42.000000000","tz":0},"subject":"Make \u0027FormattableColumn\u0027 comparable","message":"Make \u0027FormattableColumn\u0027 comparable\n\nImplement the \u0027__lt__\u0027 magic method, thus providing the minimal set of\nrich comparison methods necessary to support sorting. This will allows\nusers using these formatters for the more basic types (i.e. not dicts)\nto sort their output using the standard \u0027--sort-column\u0027 option.\n\nChange-Id: I08e1f1bc75fa6452f19dfb9d221c1daec194d58d\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/c1c991045cece85dd55494f9d4670429e370e131"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/c1c991045cece85dd55494f9d4670429e370e131"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
