)]}'
{"id":"openstack%2Fcliff~787931","triplet_id":"openstack%2Fcliff~master~I328d2e7843e3ba62a6de4ef221c4f8564a64f234","project":"openstack/cliff","branch":"master","attention_set":{},"removed_from_attention_set":{"30092":{"account":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"last_update":"2022-05-25 18:07:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I328d2e7843e3ba62a6de4ef221c4f8564a64f234","subject":"setup.cfg: Replace dashes with underscores","status":"MERGED","created":"2021-04-26 07:50:32.000000000","updated":"2022-05-25 18:10:26.000000000","submitted":"2022-05-25 18:07:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"787931","meta_rev_id":"dca423c592bd9e5b7a5e9128b83cad72aa8bd866","_number":787931,"virtual_id_number":787931,"owner":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-25 18:07:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2021-04-26 09:22:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2022-05-25 16:27:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-26 09:22:59.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2021-04-26 09:27:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"23e10f9d6d2ab73e6e0db52a450a3ceee5179f09","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2021-04-26 07:50:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"882bcc42816dbabdcf4be94c5c1a13f302f5ab98","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-04-26 09:22:59.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a9343805ee78c6840722e2cd8e2b45e7a5c50430","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 09:27:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7e5b0d3572fd49a9909e4e97c8202cd0 : SUCCESS in 3m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/04e016030c9a4ed38197ae4cea69c318 : SUCCESS in 1h 31m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b8aca426887422e890469b71de5453b : SUCCESS in 3m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ada1a1cf639b47b0a9cc3e7e36171e09 : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5cbb9d6f0f854f74ade4a9f212c2cbb3 : SUCCESS in 3m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/801ae7c7c4d04961891f117f02dda6fd : SUCCESS in 4m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84987ba212144e16baf218aa01775d7c : SUCCESS in 3m 27s\n- cliff-tox-py37-neutronclient-tip https://zuul.opendev.org/t/openstack/build/a4ac53fb0c7a4827adde67873a7a8d2c : SUCCESS in 4m 16s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/f944d16127864e25ae0f70c8505fb585 : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":1},{"id":"a5ed3c3d85dcdd7172f63ab4c048299fcd5562c0","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-05-25 16:27:33.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e09159976d1f6b1a73875d54e4968d6a1e836872","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-25 16:28:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2adb59d43cd7db046b2dca728fdb9633362b35a9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-25 18:07:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6d675b797ef74e5ba8ab59832b977111 : SUCCESS in 2m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4e49a9d4b2164a2886171225464d1094 : SUCCESS in 1h 33m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/edfc3e6955a2425fa9859dc594f00103 : SUCCESS in 3m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e62cc50c7ea34569a8caa73a6d7fdfeb : SUCCESS in 3m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ae68b0a482514b3cb83295459ed159d5 : SUCCESS in 3m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cf667ef541f432ea69397ffe79af9b8 : SUCCESS in 3m 30s\n- cliff-tox-py38-neutronclient-tip https://zuul.opendev.org/t/openstack/build/019de68a264e48fcac6c9a110cfe10f5 : SUCCESS in 4m 25s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/c08f6b367ae74ca99af1966cb800a951 : SUCCESS in 5m 36s","accounts_in_message":[],"_revision_number":1},{"id":"94cbb9e1f027ebe5a4abec7559bf95cb5066fa40","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-25 18:07:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"dca423c592bd9e5b7a5e9128b83cad72aa8bd866","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-25 18:10:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8f1eea611e94f44a227d66f5616c6b0 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fbff6de978fefa8272f7ef0c0fd39599062d4867","revisions":{"fbff6de978fefa8272f7ef0c0fd39599062d4867":{"kind":"REWORK","_number":1,"created":"2021-04-26 07:50:32.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/31/787931/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/31/787931/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/31/787931/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/31/787931/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/31/787931/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/31/787931/1"}}},"commit":{"parents":[{"commit":"b1f0f8a001e73fccc8dbf1a2e13be06f60cf1e86","subject":"Merge \"trivial: Remove references to Python 2.7\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/b1f0f8a001e73fccc8dbf1a2e13be06f60cf1e86"}]}],"author":{"name":"Yandong Xuan","email":"xuanyandong@inspur.com","date":"2021-04-26 07:50:22.000000000","tz":480},"committer":{"name":"Yandong Xuan","email":"xuanyandong@inspur.com","date":"2021-04-26 07:50:22.000000000","tz":480},"subject":"setup.cfg: Replace dashes with underscores","message":"setup.cfg: Replace dashes with underscores\n\nResolves warnings like the following:\n\n  UserWarning: Usage of dash-separated \u0027description-file\u0027 will not be\n  supported in future versions. Please use the underscore name\n  \u0027description_file\u0027 instead\n\nChange-Id: I328d2e7843e3ba62a6de4ef221c4f8564a64f234\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/fbff6de978fefa8272f7ef0c0fd39599062d4867"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/fbff6de978fefa8272f7ef0c0fd39599062d4867"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
