)]}'
{"id":"openstack%2Fcliff~790543","triplet_id":"openstack%2Fcliff~master~I7a1692d9979e9ffaf781de1f39f5bfa59a01cf3c","project":"openstack/cliff","branch":"master","topic":"deprecation-warning","hashtags":[],"change_id":"I7a1692d9979e9ffaf781de1f39f5bfa59a01cf3c","subject":"Replace getargspec with getfullargspec","status":"MERGED","created":"2021-05-11 01:59:23.000000000","updated":"2021-05-13 14:03:25.000000000","submitted":"2021-05-13 14:01:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"790543-deprecation-warning","meta_rev_id":"1838fef3a120605eda80bf8908c158d7d3e8ae41","_number":790543,"virtual_id_number":790543,"owner":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-13 14:01:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2021-05-13 08:46:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2021-05-13 12:42:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2021-05-13 12:42:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-11 03:22:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-11 12:52:12.000000000","updated_by":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"reviewer":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"state":"CC"},{"updated":"2021-05-12 08:54:01.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2021-05-12 10:48:52.000000000","updated_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"},{"updated":"2021-05-13 12:42:04.000000000","updated_by":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"reviewer":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"state":"REVIEWER"}],"messages":[{"id":"186d3341e5e6fec1314b2198c05317d5a7616869","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2021-05-11 01:59:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6a5eb2c5fa16f503127cd90c55ae8f1b8a99737b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2021-05-11 02:02:47.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"1d793006c0649818548afe7be0e89dd3eb343a79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 03:22:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f1889625af58488fa94d2483b86dd279 : SUCCESS in 1h 17m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0995fc68512f41cdada41aaef33c50ff : SUCCESS in 3m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e2fb8fdabdd9476393db59820c873993 : SUCCESS in 3m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8c624acde18144c998cfe0f01858abf6 : SUCCESS in 2m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a8024de72d7a4091968fb1a2e9129857 : SUCCESS in 3m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c61cb0728c4845cebaf9201e3268b62c : SUCCESS in 3m 07s\n- cliff-tox-py37-neutronclient-tip https://zuul.opendev.org/t/openstack/build/8e2661580cd94fe9bbff066fb27cbdfd : SUCCESS in 3m 51s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/3bf8c519df2b4556852b4719e4835a65 : SUCCESS in 4m 57s","accounts_in_message":[],"_revision_number":2},{"id":"5e8f661454f236aee7f525a213039c599395d653","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2021-05-11 12:52:12.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"7fe5ac60365ea8ca55a2aaed8ba341ac82dc502f","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-05-12 08:54:01.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5650b9105b93209c43f0ee4c45e420b603f6e835","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2021-05-12 09:20:39.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"dc95881ff472cf90f457754e98aa4815fe880791","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2021-05-12 09:20:53.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9b3b4b69c6eb48e7d998dfb00282a0fe087199e0","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-05-12 09:40:18.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a85d026569092f385c75ede42c2a9283c390942f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2021-05-12 09:44:41.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"d5d61fb8a338ebdb1298309721fe0ca6a7465945","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-05-12 09:45:16.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"b391814e797cfef35d14b17348d814c5347026b6","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2021-05-12 09:45:51.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"62112fc7691020429c51a3b14cd8b3cd07e7d0d8","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2021-05-12 10:48:52.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"2f4a895f1f0ad54fa698cf0620694ce341551362","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 11:18:05.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/474603b7d8484f9cb904bd4cfe7abc4e : SUCCESS in 1h 32m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13b1f2e306934e609b97e0ba2b350e24 : FAILURE in 3m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/91fb3c44063a4eada0cf42a09ef247b6 : SUCCESS in 3m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/38d8d11b3932474fa625900ee8f99339 : SUCCESS in 3m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/931cd34820e543e5a7e3e0f18117086c : SUCCESS in 3m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2a2ec39026d44d3bcc6e8659bbe9d82 : SUCCESS in 3m 39s\n- cliff-tox-py37-neutronclient-tip https://zuul.opendev.org/t/openstack/build/747d74f6808f45959629f3c427d9b3ad : SUCCESS in 4m 46s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/c66b8fc7ff2e4b29a1c0c639d49c4673 : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":4},{"id":"32fbebbc63dd51faf3bf1d00405f51b627654ca6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2021-05-13 01:12:54.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"47094e572afeeba2f05705e8eea357bbdb2072fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 02:40:34.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6846ce1c641a4c3e876f02159249470b : SUCCESS in 1h 26m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/918621f82ebf4849ac663488024d6e51 : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4b59d7b3cb3c4f6493387fc78db012c0 : SUCCESS in 3m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/723413f6efa84aa98ab3d266694502f0 : SUCCESS in 3m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b4a3dafdd210467f9d0ebc56519fb57a : SUCCESS in 3m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2db4cacb9aaa414fbc2ba391063b28a4 : SUCCESS in 4m 01s\n- cliff-tox-py37-neutronclient-tip https://zuul.opendev.org/t/openstack/build/6225067574bf4ee994691663d10360f9 : SUCCESS in 4m 41s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/83fbafad3740402b948c3dfff71f3c7c : SUCCESS in 4m 52s","accounts_in_message":[],"_revision_number":5},{"id":"4789783f40dd0e80bbcc75754a643add4caf5a59","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-05-13 08:46:34.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"d577bc98b0efd3c2922c220856aff95a4b7eb116","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2021-05-13 12:42:04.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\nThanks!","accounts_in_message":[],"_revision_number":5},{"id":"6caab478443b49eb26cb5b1ffabe70c79ae69e6a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 12:42:46.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"4e7940ed952e2998d182820e1b0b23394273bce0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 14:01:12.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bfaa68f12142467aa87ba2fc7ca83570 : SUCCESS in 1h 17m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9924f901f7ea46a6ab84a3d8bf4e3b1c : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/89287c92cc9244f3890a8fac7cd58fac : SUCCESS in 6m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/47032068abd14698be6d0052f87b65fe : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d1cc297fecc4cdd86eefc55c47a4a84 : SUCCESS in 4m 31s\n- cliff-tox-py37-neutronclient-tip https://zuul.opendev.org/t/openstack/build/8e7ff2242a3d4dd0ad60486698e62c6c : SUCCESS in 5m 47s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/78d9d53f15464121be22f2f2eca1af90 : SUCCESS in 4m 52s","accounts_in_message":[],"_revision_number":5},{"id":"f4e84d873bcd683205ebf44d236684e87e1a6def","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 14:01:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"1838fef3a120605eda80bf8908c158d7d3e8ae41","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 14:03:25.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a83af866a88b4e85a16675819d991a37 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"075345196d0747168a27d71aae4e3d97c5746907","revisions":{"18403a5d247b7846d47d0ca3bbc0c3837ffc6bb0":{"kind":"REWORK","_number":1,"created":"2021-05-11 01:59:23.000000000","uploader":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"ref":"refs/changes/43/790543/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/43/790543/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/43/790543/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/43/790543/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/43/790543/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/43/790543/1"}}},"commit":{"parents":[{"commit":"fa0535a88831cb7b91c0bf9568679a21c77f9e7c","subject":"Use py3 as the default runtime for tox","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/fa0535a88831cb7b91c0bf9568679a21c77f9e7c"}]}],"author":{"name":"likui","email":"likui@yovole.com","date":"2021-05-11 01:58:58.000000000","tz":480},"committer":{"name":"likui","email":"likui@yovole.com","date":"2021-05-11 01:58:58.000000000","tz":480},"subject":"Replace getargspec with getfullargspec","message":"Replace getargspec with getfullargspec\n\ninspect.getargspec() is deprecated since py3\n\n[1] https://docs.python.org/3/library/inspect.html#inspect.getargspec\n\nChange-Id: I7a1692d9979e9ffaf781de1f39f5bfa59a01cf3c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/18403a5d247b7846d47d0ca3bbc0c3837ffc6bb0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/18403a5d247b7846d47d0ca3bbc0c3837ffc6bb0"}]},"branch":"refs/heads/master"},"265850307b09ebac7a7b98fcad7b521fc766540a":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-05-11 02:02:47.000000000","uploader":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"ref":"refs/changes/43/790543/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/43/790543/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/43/790543/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/43/790543/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/43/790543/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/43/790543/2"}}},"commit":{"parents":[{"commit":"fa0535a88831cb7b91c0bf9568679a21c77f9e7c","subject":"Use py3 as the default runtime for tox","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/fa0535a88831cb7b91c0bf9568679a21c77f9e7c"}]}],"author":{"name":"likui","email":"likui@yovole.com","date":"2021-05-11 01:58:58.000000000","tz":480},"committer":{"name":"likui","email":"likui@yovole.com","date":"2021-05-11 02:02:00.000000000","tz":480},"subject":"Replace getargspec with getfullargspec","message":"Replace getargspec with getfullargspec\n\nThe Python 2.x Support has been dropped since Ussuri.\nSo remove hacking rules for compatibility between python 2 and 3\ninspect.getargspec() is deprecated since py3\n\n[1] https://docs.python.org/3/library/inspect.html#inspect.getargspec\n\nChange-Id: I7a1692d9979e9ffaf781de1f39f5bfa59a01cf3c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/265850307b09ebac7a7b98fcad7b521fc766540a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/265850307b09ebac7a7b98fcad7b521fc766540a"}]},"branch":"refs/heads/master"},"4fb37d05dd8037c7fbc79d79b30bd8a38a74fdd3":{"kind":"REWORK","_number":3,"created":"2021-05-12 09:20:39.000000000","uploader":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"ref":"refs/changes/43/790543/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/43/790543/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/43/790543/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/43/790543/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/43/790543/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/43/790543/3"}}},"commit":{"parents":[{"commit":"fa0535a88831cb7b91c0bf9568679a21c77f9e7c","subject":"Use py3 as the default runtime for tox","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/fa0535a88831cb7b91c0bf9568679a21c77f9e7c"}]}],"author":{"name":"likui","email":"likui@yovole.com","date":"2021-05-11 01:58:58.000000000","tz":480},"committer":{"name":"likui","email":"likui@yovole.com","date":"2021-05-12 09:20:16.000000000","tz":480},"subject":"Replace getargspec with getfullargspec","message":"Replace getargspec with getfullargspec\n\nThe Python 2.x Support has been dropped since Ussuri.\nSo remove hacking rules for compatibility between python 2 and 3\ninspect.getargspec() is deprecated since py3\n\n[1] https://docs.python.org/3/library/inspect.html#inspect.getargspec\n\nChange-Id: I7a1692d9979e9ffaf781de1f39f5bfa59a01cf3c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/4fb37d05dd8037c7fbc79d79b30bd8a38a74fdd3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/4fb37d05dd8037c7fbc79d79b30bd8a38a74fdd3"}]},"branch":"refs/heads/master"},"813d04686a3732e7c0bf09f75f81013aca0e6e14":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2021-05-12 09:44:41.000000000","uploader":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"ref":"refs/changes/43/790543/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/43/790543/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/43/790543/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/43/790543/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/43/790543/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/43/790543/4"}}},"commit":{"parents":[{"commit":"fa0535a88831cb7b91c0bf9568679a21c77f9e7c","subject":"Use py3 as the default runtime for tox","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/fa0535a88831cb7b91c0bf9568679a21c77f9e7c"}]}],"author":{"name":"likui","email":"likui@yovole.com","date":"2021-05-11 01:58:58.000000000","tz":480},"committer":{"name":"likui","email":"likui@yovole.com","date":"2021-05-12 09:44:03.000000000","tz":480},"subject":"Replace getargspec with getfullargspec","message":"Replace getargspec with getfullargspec\n\ninspect.getargspec() is deprecated since py3\n\n[1] https://docs.python.org/3/library/inspect.html#inspect.getargspec\n\nChange-Id: I7a1692d9979e9ffaf781de1f39f5bfa59a01cf3c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/813d04686a3732e7c0bf09f75f81013aca0e6e14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/813d04686a3732e7c0bf09f75f81013aca0e6e14"}]},"branch":"refs/heads/master"},"075345196d0747168a27d71aae4e3d97c5746907":{"kind":"REWORK","_number":5,"created":"2021-05-13 01:12:54.000000000","uploader":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"ref":"refs/changes/43/790543/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/43/790543/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/43/790543/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/43/790543/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/43/790543/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/43/790543/5"}}},"commit":{"parents":[{"commit":"fa0535a88831cb7b91c0bf9568679a21c77f9e7c","subject":"Use py3 as the default runtime for tox","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/fa0535a88831cb7b91c0bf9568679a21c77f9e7c"}]}],"author":{"name":"likui","email":"likui@yovole.com","date":"2021-05-11 01:58:58.000000000","tz":480},"committer":{"name":"likui","email":"likui@yovole.com","date":"2021-05-13 01:12:39.000000000","tz":480},"subject":"Replace getargspec with getfullargspec","message":"Replace getargspec with getfullargspec\n\ninspect.getargspec() is deprecated since py3\n\n[1] https://docs.python.org/3/library/inspect.html#inspect.getargspec\n\nChange-Id: I7a1692d9979e9ffaf781de1f39f5bfa59a01cf3c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/075345196d0747168a27d71aae4e3d97c5746907"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/075345196d0747168a27d71aae4e3d97c5746907"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
