)]}'
{"id":"openstack%2Fcliff~791075","triplet_id":"openstack%2Fcliff~master~Icccc9cc2b3a0d236746c4b58a8815d25d8b0a443","project":"openstack/cliff","branch":"master","topic":"setup-cfg","hashtags":[],"change_id":"Icccc9cc2b3a0d236746c4b58a8815d25d8b0a443","subject":"setup.cfg: Replace dashes with underscores","status":"MERGED","created":"2021-05-13 01:30:14.000000000","updated":"2021-05-14 04:08:47.000000000","submitted":"2021-05-14 04:07:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"791075-setup-cfg","meta_rev_id":"976cb920f2a68365023177c17e1fd8997ee3b431","_number":791075,"virtual_id_number":791075,"owner":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-14 04:07:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2021-05-13 08:46:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2021-05-13 12:54:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-05-13 11:25:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2021-05-13 12:54:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-13 02:57:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-13 08:46:53.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2021-05-13 11:25:16.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2021-05-13 12:54:45.000000000","updated_by":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"reviewer":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"state":"REVIEWER"}],"messages":[{"id":"3a2a5db341dc7fdbc1955e18bd56163dbbdfaf3d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2021-05-13 01:30:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"10d39f16073c0c44ef029e502de48ec620f63ce6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 02:57:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/915fe74f75b54f28932dba405f078157 : SUCCESS in 2m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/57fb1ef8ea7d44eebece43fbbbf73f25 : SUCCESS in 1h 25m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b52adbc5d03479daee5c1fbe5602c85 : SUCCESS in 3m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/975823d269bc4de7bfe359f829d84796 : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2dd8133569b54abcaf905f1896c8bebc : SUCCESS in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5d865ff7170245fd9bc0699dea7efd0e : SUCCESS in 3m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/542855322e454ca696d99629d9b7fe45 : SUCCESS in 3m 14s\n- cliff-tox-py37-neutronclient-tip https://zuul.opendev.org/t/openstack/build/2477d532f67d463d990beeab280f3e35 : SUCCESS in 4m 24s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/b76bb500af114976bc99ecd46f59f3b1 : SUCCESS in 5m 28s","accounts_in_message":[],"_revision_number":1},{"id":"b6afcb337aed371933b621c9c1029a61df44b66a","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-05-13 08:46:53.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"47967cbac130e7cb732bf46dcb6512b61ac805ac","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2021-05-13 11:25:16.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"1fa9c1642252cf80ce8ac37cfa343a3de885e9b6","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2021-05-13 12:54:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a929f128adc16cdb8787c3998b46d6dacccabff4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 12:55:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9adfe9eaa85ecb2d4f4588c21ba13414a6930d88","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 15:03:58.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/cde7888bacf5487d8fee0bdb4eb8abe9 : SUCCESS in 3m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9891fb2e693e476396fe62e40b459989 : TIMED_OUT in 2h 07m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1ddd27d88d94afa8e4b0d0d31afd514 : SUCCESS in 3m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/71f7a57fb16b4529b0fa2babb04d89bb : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d4855698895246ee9dc5f53fbe01e911 : SUCCESS in 3m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13e4725c54d242a9a46b0f03e1299965 : SUCCESS in 3m 40s\n- cliff-tox-py37-neutronclient-tip https://zuul.opendev.org/t/openstack/build/34bb6d12c077455ca7b1ad1a7c7c4d58 : SUCCESS in 4m 03s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/21b9b506885849178feb57902e53ccb4 : SUCCESS in 7m 45s","accounts_in_message":[],"_revision_number":1},{"id":"82f6015cd557dba4e9f29f35a4813eb329d51603","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2021-05-14 01:12:55.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"723967962b0f4ef9a6f0e0aebbe23b51e3e9005d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 02:42:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ff3e22a2f3cc4f25abc31fa023496f01 : SUCCESS in 4m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/36643de79ffb40d5abf83564c2cd0462 : SUCCESS in 1h 28m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8a7d0d046194ca493e76b3387891fab : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/57a8790585934de09e1f026ae57f349e : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d0e6246919594fa49f9c73cf144aa98a : SUCCESS in 3m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c569d4e9e5c34665b181c97a275d33e2 : SUCCESS in 4m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a62b43ce946643e3aedfd4501aacf393 : SUCCESS in 3m 32s\n- cliff-tox-py37-neutronclient-tip https://zuul.opendev.org/t/openstack/build/90d7736554f046ba8feff09910f54a8d : SUCCESS in 4m 31s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/aeca546bf1fa48b3b897411891b42d18 : SUCCESS in 7m 25s","accounts_in_message":[],"_revision_number":1},{"id":"1185d99065ce3dd60c45dfc20a6dd3b5234659ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 02:43:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c3474e092257166982351ccd751993fd9631b499","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 04:07:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/84c84b280576431690fa3cb3f4ba2141 : SUCCESS in 3m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c3c4c5ad90204392992207f21a9b4936 : SUCCESS in 1h 23m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9a4d4c9eddd4855a332929b98e635d6 : SUCCESS in 3m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/02022cf60f4747459bc771eeb822b671 : SUCCESS in 4m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8a3d529fbb0c4d1f8b52f8bb505c754d : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72d98653fa814e21b44b9d3a7f489d5a : SUCCESS in 4m 21s\n- cliff-tox-py37-neutronclient-tip https://zuul.opendev.org/t/openstack/build/2081ae15a3f2475e98452d1d1cb4c0c6 : SUCCESS in 4m 49s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/6d2444911d63421abd2a858fd763e263 : SUCCESS in 5m 49s","accounts_in_message":[],"_revision_number":1},{"id":"bb963768e7f2b6910a08fd0d1e157924c08a8f45","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 04:07:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"976cb920f2a68365023177c17e1fd8997ee3b431","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 04:08:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67fd103e0b5a47e085ca9c0ae5d5dd58 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2fdf358b02105f596c382477542e7cc2bf9decf9","revisions":{"2fdf358b02105f596c382477542e7cc2bf9decf9":{"kind":"REWORK","_number":1,"created":"2021-05-13 01:30:14.000000000","uploader":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"ref":"refs/changes/75/791075/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/75/791075/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/75/791075/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/75/791075/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/75/791075/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/75/791075/1"}}},"commit":{"parents":[{"commit":"fa0535a88831cb7b91c0bf9568679a21c77f9e7c","subject":"Use py3 as the default runtime for tox","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/fa0535a88831cb7b91c0bf9568679a21c77f9e7c"}]}],"author":{"name":"likui","email":"likui@yovole.com","date":"2021-05-13 01:29:53.000000000","tz":480},"committer":{"name":"likui","email":"likui@yovole.com","date":"2021-05-13 01:29:53.000000000","tz":480},"subject":"setup.cfg: Replace dashes with underscores","message":"setup.cfg: Replace dashes with underscores\n\nSetuptools v54.1.0 introduces a warning that the use of\ndash-separated options in \u0027setup.cfg\u0027 will not be supported\nin a future version [1].\nGet ahead of the issue by replacing the dashes with underscores.\nWithout this, we see \u0027UserWarning\u0027 messages\nlike the following on new enough\nversions of setuptools:\n\n  UserWarning: Usage of dash-separated \u0027description-file\u0027 will not be\n  supported in future versions. Please use the underscore name\n  \u0027description_file\u0027 instead\n\n[1] https://github.com/pypa/setuptools/commit/a2e9ae4cb\n\nChange-Id: Icccc9cc2b3a0d236746c4b58a8815d25d8b0a443\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/2fdf358b02105f596c382477542e7cc2bf9decf9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/2fdf358b02105f596c382477542e7cc2bf9decf9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
