)]}'
{"id":"openstack%2Fcliff~799344","triplet_id":"openstack%2Fcliff~master~If9e1aa5166da32c58cc0fa617f4f81eaa9b2c470","project":"openstack/cliff","branch":"master","topic":"autopage","hashtags":[],"change_id":"If9e1aa5166da32c58cc0fa617f4f81eaa9b2c470","subject":"Colourise and automatically page help output","status":"MERGED","created":"2021-07-03 04:39:07.000000000","updated":"2021-08-06 18:31:11.000000000","submitted":"2021-08-06 18:29:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"799344-autopage","meta_rev_id":"895381b3e81a655bed78004df6a1470a259bf348","_number":799344,"virtual_id_number":799344,"owner":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-06 18:29:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":2,"date":"2021-08-06 17:02:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":2,"date":"2021-07-13 15:22:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":1,"date":"2021-08-06 17:02:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-03 06:10:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-08 17:58:26.000000000","updated_by":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"reviewer":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"state":"CC"},{"updated":"2021-07-13 15:22:45.000000000","updated_by":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"reviewer":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"state":"REVIEWER"},{"updated":"2021-08-06 17:02:01.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"}],"messages":[{"id":"2f321eb3fd6ec73e82e52134f2c3fd29e0bc9962","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2021-07-03 04:39:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"40d9ff2fd97af63885c78baa551651fe8333f6d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-03 06:10:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/83639fafbcd24d5b85950c827aa7189e : SUCCESS in 2m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0d4c6f96dada4f2684deb0ea5b11f1e3 : SUCCESS in 1h 25m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebfc382ddd3a48be89a431370def4568 : SUCCESS in 3m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ee13cb2614e24913953e77e16121e9c4 : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ab1a537569cf44e5b639610ac98b7ee3 : SUCCESS in 3m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/57a0d685e5b74383a09a2d155f115f69 : SUCCESS in 3m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5421501a719941cb8ac0dfbf9d8de489 : SUCCESS in 3m 49s\n- cliff-tox-py37-neutronclient-tip https://zuul.opendev.org/t/openstack/build/3c0ac0592c78437fa4e0f0db1c146f20 : SUCCESS in 4m 28s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/1104f6d359834652b1a35d4c4c20cd11 : SUCCESS in 5m 10s","accounts_in_message":[],"_revision_number":1},{"id":"6b6b3fdc2155180243815bb6375e812fa1dba1b3","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2021-07-08 17:58:26.000000000","message":"Patch Set 1:\n\n(2 comments)\n\nThis looks like it would be a nice addition. I have a few questions about details, but nothing blocking.","accounts_in_message":[],"_revision_number":1},{"id":"9597e54ebb85d4d1a2336e589491bea672340719","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2021-07-10 04:10:16.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"386fab474ec31b3225761c3eff4612b32735a56e","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2021-07-12 11:05:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5b615b485c109de170e5f1a80433825adf334330","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2021-07-13 02:23:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c1bb6de307e1e70539bb39a506fc038ddb13e573","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2021-07-13 02:24:54.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e2b8461b2fc0fea8c05b24fd92d3c2d9ed65a401","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-13 03:57:34.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/7f057c23a050454cb2cd17325ab4f98d : FAILURE in 3m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/abcaa7c45de64ee1a365cb96ec55eb0e : SUCCESS in 1h 28m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9526d8e8d9ae46fb8caa8c3b85c3496d : SUCCESS in 3m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cc9cd6e709e64e9d90f330f7f35086c4 : SUCCESS in 3m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9c08b65b41224429a59e055319ed253a : SUCCESS in 3m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e554db3b3e364f828d4e435cc69d1300 : SUCCESS in 3m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53518f02ee654a31b6e179411858e778 : SUCCESS in 3m 54s\n- cliff-tox-py37-neutronclient-tip https://zuul.opendev.org/t/openstack/build/4a099547ed144b83b2ee2d7b91b24748 : SUCCESS in 4m 39s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/744de636e425487f925741b66715291b : SUCCESS in 5m 34s","accounts_in_message":[],"_revision_number":2},{"id":"5e65fad3451d4c5796768647a4edce0f798171a8","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2021-07-13 12:42:24.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"54f3a261029380a51ce76cec7ac2dfbc60d269c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-13 13:58:38.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/df4146d211cf464d975ab4aa3cd33516 : FAILURE in 3m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6041f64aeb0c44ad94f5307567da0898 : SUCCESS in 1h 13m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/604d08b4cd2c461fa47afd78f7e3fd68 : SUCCESS in 3m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/595857cc46bb42928fdc2baa7ca0129e : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/32d46a194f7441038936be3a4f58bfcb : SUCCESS in 4m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8c1045f29a4c4cfb92323c72e1c0eaa2 : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/541df13984714c779c3bcc9fce6c2341 : SUCCESS in 4m 02s\n- cliff-tox-py37-neutronclient-tip https://zuul.opendev.org/t/openstack/build/597a6f1587e948deb0ad10107bf31c8b : SUCCESS in 4m 40s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/0d307194bb09418a8318dcf330271234 : SUCCESS in 5m 00s","accounts_in_message":[],"_revision_number":2},{"id":"cb7e4719cc955f50a36a679884cfc209d78739f4","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2021-07-13 15:22:45.000000000","message":"Patch Set 2: Code-Review+2\n\n(2 comments)\n\nI have one nit that isn\u0027t worth holding up the change. Otherwise this looks good.\n\nThanks!","accounts_in_message":[],"_revision_number":2},{"id":"690dbeb708657bc63d9cea51d0a85b7cb0b94746","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2021-07-13 16:15:52.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nI\u0027m not sure what\u0027s happening with the requirements job here. Zuul is not pulling in the Depends-On patch:\n\n    2021-07-13 12:45:37.349687 | ubuntu-focal | opendev.org/openstack/requirements checked out to:\n    2021-07-13 12:45:37.349818 | ubuntu-focal | 791550915b1217b5e2c93cb467e9c41601ee9084 Merge \"Updated from generate-constraints\"\n\nWhereas in the previous patchset it did:\n\n    2021-07-03 04:44:31.547667 | ubuntu-focal | opendev.org/openstack/requirements checked out to:\n    2021-07-03 04:44:31.547739 | ubuntu-focal | 90c570a0a6bbd40c7d2c465cb736906a813495b3 Add autopage to global-requirements\n\nand the commit message hasn\u0027t changed. Both tests were run after the Zuul 2.6.0 upgrade.","accounts_in_message":[],"_revision_number":2},{"id":"ea134ccc3e81366915f6e1ab1e0cb1a2d0eb1c72","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2021-07-26 14:18:40.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"8e0a4808daf11910975175941a89237eb81c4cf4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-26 15:27:21.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/5f6a4f0d659c40e898adb70e89aaf624 : SUCCESS in 4m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b8e5e0ace94e4858a730a49edbbdf81c : FAILURE in 1h 01m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2540f4fa6fe46bc9f0a4a1d141c36d3 : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1ff91d7ebb3c47349a1983ed15b552b2 : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/35fd9e7af4464cb1ade9e50186d0593d : SUCCESS in 3m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/95758c1d81fa4942b004c1a86e9c70c4 : SUCCESS in 4m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/454c39a5c70d4dddb7673764cbfddb38 : SUCCESS in 4m 26s\n- cliff-tox-py37-neutronclient-tip https://zuul.opendev.org/t/openstack/build/5d0d321a516c43578777b567564df818 : FAILURE in 4m 17s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/d3b66f00b4c94cb8b0db8f12233a945d : SUCCESS in 5m 12s","accounts_in_message":[],"_revision_number":2},{"id":"5460dbc4585279846b1158937b9ecb98628d3685","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2021-07-26 20:06:54.000000000","message":"Patch Set 2:\n\nrequirements change merged, but waiting on https://review.opendev.org/c/openstack/python-neutronclient/+/802306 now due to new testtools release.","accounts_in_message":[],"_revision_number":2},{"id":"66f37483e1f75e2852ed7b74516e7ace0d286bb6","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2021-07-27 17:00:16.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"71186b55e592ec6c775708da97c8739d279ca309","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 18:21:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/9b9471ab8ce6489d9a2d05d5f8649bba : SUCCESS in 3m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0d24c3a953db4215b82ae80c8bf0e841 : SUCCESS in 1h 16m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5bf9188e4eb040b7adc9aa0e3fea0838 : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ed374d9a147a49008733a5b4f3c423d2 : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/86a54545f7a346b9a2787946c545200d : SUCCESS in 4m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d68eacece1494e1da4b6404cac4914f1 : SUCCESS in 3m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6ca4c073a144b558cbb8bed4cf8c1b9 : SUCCESS in 3m 55s\n- cliff-tox-py37-neutronclient-tip https://zuul.opendev.org/t/openstack/build/829ef333118c4b72b0585f4d17534291 : SUCCESS in 4m 30s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/e2200741cce14e628883a9c20d6387fa : SUCCESS in 7m 15s","accounts_in_message":[],"_revision_number":2},{"id":"6382b4678fd14ec460a98af3beac0a3ffb169862","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2021-08-06 17:02:01.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nCool!","accounts_in_message":[],"_revision_number":2},{"id":"68cfb9a370d09698186bf35f711d1c69f5ff5521","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-06 17:02:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b797f25b083ce646f781ba405014bd85f2a15526","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-06 18:29:46.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/c83a89fb48564eb6a66355648a68a5a7 : SUCCESS in 2m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/600517355c28418db86542d1bbe6935c : SUCCESS in 1h 22m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/362015bfc23141c4ae7667152e255978 : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7e68a917816740e2a784486868aaf4ef : SUCCESS in 3m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a4fdbd6bb87c457cb72b7684b2e3df3e : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88f5fc33156a485d926b1f17dae22d2a : SUCCESS in 4m 02s\n- cliff-tox-py37-neutronclient-tip https://zuul.opendev.org/t/openstack/build/fa59938e856b42bcbe25f1f91c12c50e : SUCCESS in 5m 19s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/c07b9646b19c4a66aaa54d52ec83f29e : SUCCESS in 5m 25s","accounts_in_message":[],"_revision_number":2},{"id":"e6b2748d95a0b9ab24d6ed7345be987b7e8ebddd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-06 18:29:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"895381b3e81a655bed78004df6a1470a259bf348","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-06 18:31:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a00c5a9775f405f86026a63edd33965 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8fa916e9169f507ce7b861f3d9df9a14734672d2","revisions":{"4085c328d5a94fa371d6130e4977c42faa5297ed":{"kind":"REWORK","_number":1,"created":"2021-07-03 04:39:07.000000000","uploader":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"ref":"refs/changes/44/799344/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/44/799344/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/44/799344/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/44/799344/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/44/799344/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/44/799344/1"}}},"commit":{"parents":[{"commit":"392f3b2e7cb2dad8036ebbaefbb75dd758914421","subject":"Handle SIGPIPE exit gracefully","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/392f3b2e7cb2dad8036ebbaefbb75dd758914421"}]}],"author":{"name":"Zane Bitter","email":"zbitter@redhat.com","date":"2021-06-05 02:44:13.000000000","tz":-240},"committer":{"name":"Zane Bitter","email":"zbitter@redhat.com","date":"2021-07-03 04:37:26.000000000","tz":-240},"subject":"Colourise and automatically page help output","message":"Colourise and automatically page help output\n\nUsing the autopage library we can automatically send the help output to\na pager (less, by default), git-style. The pager is configured to not\nreset the terminal on exit, avoiding the problem when piping to less\nmanually that the help text you want to refer to disappears off the\nscreen when you go to use it. The pager is only invoked when the output\nis to the terminal.\n\nSince we invoke the pager, we can ensure that it is correctly set up to\ninterpret ANSI escape codes, so it is safe to use colour to make the\noutput easier to read. The autopage library provides light styling of\nthe default argparse help output, and some additional colour\nhighlighting is added here for the command list (which is generated by\ncliff, not using argparse\u0027s formatting code).\n\nChange-Id: If9e1aa5166da32c58cc0fa617f4f81eaa9b2c470\nDepends-On: https://review.opendev.org/c/openstack/requirements/+/799343\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/4085c328d5a94fa371d6130e4977c42faa5297ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/4085c328d5a94fa371d6130e4977c42faa5297ed"}]},"branch":"refs/heads/master"},"8fa916e9169f507ce7b861f3d9df9a14734672d2":{"kind":"REWORK","_number":2,"created":"2021-07-13 02:23:13.000000000","uploader":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"ref":"refs/changes/44/799344/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/44/799344/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/44/799344/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/44/799344/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/44/799344/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/44/799344/2"}}},"commit":{"parents":[{"commit":"392f3b2e7cb2dad8036ebbaefbb75dd758914421","subject":"Handle SIGPIPE exit gracefully","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/392f3b2e7cb2dad8036ebbaefbb75dd758914421"}]}],"author":{"name":"Zane Bitter","email":"zbitter@redhat.com","date":"2021-06-05 02:44:13.000000000","tz":-240},"committer":{"name":"Zane Bitter","email":"zbitter@redhat.com","date":"2021-07-13 01:03:32.000000000","tz":-240},"subject":"Colourise and automatically page help output","message":"Colourise and automatically page help output\n\nUsing the autopage library we can automatically send the help output to\na pager (less, by default), git-style. The pager is configured to not\nreset the terminal on exit, avoiding the problem when piping to less\nmanually that the help text you want to refer to disappears off the\nscreen when you go to use it. The pager is only invoked when the output\nis to the terminal.\n\nSince we invoke the pager, we can ensure that it is correctly set up to\ninterpret ANSI escape codes, so it is safe to use colour to make the\noutput easier to read. The autopage library provides light styling of\nthe default argparse help output, and some additional colour\nhighlighting is added here for the command list (which is generated by\ncliff, not using argparse\u0027s formatting code).\n\nChange-Id: If9e1aa5166da32c58cc0fa617f4f81eaa9b2c470\nDepends-On: https://review.opendev.org/c/openstack/requirements/+/799343\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/8fa916e9169f507ce7b861f3d9df9a14734672d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/8fa916e9169f507ce7b861f3d9df9a14734672d2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
