)]}'
{"id":"openstack%2Fcliff~822454","triplet_id":"openstack%2Fcliff~master~I0b77a709c7fa3766c0cdb674185017bd1823c7e3","project":"openstack/cliff","branch":"master","attention_set":{},"removed_from_attention_set":{"33881":{"account":{"_account_id":33881,"name":"LiZekun","email":"2954674728@qq.com","username":"joeylee666666"},"last_update":"2022-05-25 16:27:59.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I0b77a709c7fa3766c0cdb674185017bd1823c7e3","subject":"Update python testing classifier","status":"ABANDONED","created":"2021-12-21 09:50:43.000000000","updated":"2022-05-25 16:27:59.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ab1dafeb77641eb38304c9bd2dc4d23d05e419ae","_number":822454,"virtual_id_number":822454,"owner":{"_account_id":33881,"name":"LiZekun","email":"2954674728@qq.com","username":"joeylee666666"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-12-21 11:23:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-21 11:23:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-25 16:27:56.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"CC"}],"messages":[{"id":"ee618903f467f05e1828d48ebe94d2500730a5d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33881,"name":"LiZekun","email":"2954674728@qq.com","username":"joeylee666666"},"date":"2021-12-21 09:50:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cfb156b6a358de91339f9e2348c30fca92b9810f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-21 11:23:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e4899df05deb4e558256214f2d3683c0 : SUCCESS in 2m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/095e1197db9247b6b3442008b851f4ac : SUCCESS in 1h 26m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2de92da8b88540afbe8ed02597dd224e : SUCCESS in 3m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/066a1fe55e014215b21e507b255a45d8 : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f0368e4ba1514223a5a7dbb61a9e4aa2 : SUCCESS in 3m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/44531b154e1048fa8ced18ceccae1822 : SUCCESS in 3m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b986048d563f44df80599e6b58a79e52 : SUCCESS in 3m 51s\n- cliff-tox-py37-neutronclient-tip https://zuul.opendev.org/t/openstack/build/00bf7c3965f141d0aae7aebcf04e6a9d : SUCCESS in 4m 13s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/1a3ec265f2824238b2563f4779703285 : SUCCESS in 5m 22s","accounts_in_message":[],"_revision_number":1},{"id":"e5ad48adca5d2c959d43a885de046ebe4cfab56a","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-05-25 16:27:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ab1dafeb77641eb38304c9bd2dc4d23d05e419ae","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-05-25 16:27:59.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3b3cfbcfb6114e64b5e7c4b964699899b7fa24c7","revisions":{"3b3cfbcfb6114e64b5e7c4b964699899b7fa24c7":{"kind":"REWORK","_number":1,"created":"2021-12-21 09:50:43.000000000","uploader":{"_account_id":33881,"name":"LiZekun","email":"2954674728@qq.com","username":"joeylee666666"},"ref":"refs/changes/54/822454/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/54/822454/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/54/822454/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/54/822454/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/54/822454/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/54/822454/1"}}},"commit":{"parents":[{"commit":"a04a48f4f7dc72b1bcc95a5c6a550c7650e35ab3","subject":"Removing modindex link from docs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/a04a48f4f7dc72b1bcc95a5c6a550c7650e35ab3"}]}],"author":{"name":"LiZekun","email":"2954674728@qq.com","date":"2021-12-21 09:49:51.000000000","tz":480},"committer":{"name":"LiZekun","email":"2954674728@qq.com","date":"2021-12-21 09:49:51.000000000","tz":480},"subject":"Update python testing classifier","message":"Update python testing classifier\n\nChange-Id: I0b77a709c7fa3766c0cdb674185017bd1823c7e3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/3b3cfbcfb6114e64b5e7c4b964699899b7fa24c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/3b3cfbcfb6114e64b5e7c4b964699899b7fa24c7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
