)]}'
{"id":"openstack%2Fcliff~824305","triplet_id":"openstack%2Fcliff~stable%2Fussuri~I327ece99a04bc8b2ebfa554dea643b1f2a456336","project":"openstack/cliff","branch":"stable/ussuri","topic":"conflict/handler-stable/wallaby-stable/victoria-stable/ussuri","attention_set":{},"removed_from_attention_set":{"2":{"account":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"last_update":"2022-10-10 15:55:15.000000000","reason":"Change was abandoned"},"16515":{"account":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"last_update":"2022-10-10 15:55:15.000000000","reason":"Change was abandoned"},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2022-10-10 15:55:15.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I327ece99a04bc8b2ebfa554dea643b1f2a456336","subject":"Add conflict_handler parameter as attribut in Command class","status":"ABANDONED","created":"2022-01-12 08:36:22.000000000","updated":"2022-10-10 15:55:15.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":824304,"cherry_pick_of_patch_set":1,"meta_rev_id":"5bd8cc46c75e3dbf2edb3e9ce3c07ae27967c534","_number":824305,"virtual_id_number":824305,"owner":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-01-12 10:19:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-12 09:09:33.000000000","updated_by":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2022-01-12 09:09:33.000000000","updated_by":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2022-01-12 10:19:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4b5b8661ff4a2ae34ad5e2fa87282e4457b7ede0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2022-01-12 08:36:22.000000000","message":"Patch Set 1: Cherry Picked from branch stable/victoria.","accounts_in_message":[],"_revision_number":1},{"id":"49705ad0e297fde5b47f9fa05a9f7567e691ee74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 10:19:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f90e2c1eed2a4cd0b828d89994e53fa2 : SUCCESS in 1h 31m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ae00bcbb5e2944cc95552da5747d3deb : SUCCESS in 5m 01s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/151abd3104d343318dc488b286481c6a : FAILURE in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11fb76955de14552bac0226c04ca55b3 : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9e291cb63246482a865fff60c1dc01f6 : SUCCESS in 4m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8dafd9fd8b2a4b5082826462865f1806 : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c4d0d15641d457bb7f727c5ec16a8a9 : SUCCESS in 4m 50s","accounts_in_message":[],"_revision_number":1},{"id":"5bd8cc46c75e3dbf2edb3e9ce3c07ae27967c534","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-10-10 15:55:15.000000000","message":"Abandoned\n\nLooks like stable/ussuri is broken. I don\u0027t think this is important enough to justify the effort in fixing, though if someone wants to then by all means, feel free","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d82b9181b5aa9d7b0c32f9b3bf95827eefddebc2","revisions":{"d82b9181b5aa9d7b0c32f9b3bf95827eefddebc2":{"kind":"REWORK","_number":1,"created":"2022-01-12 08:36:22.000000000","uploader":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"ref":"refs/changes/05/824305/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/05/824305/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/05/824305/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/05/824305/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/05/824305/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/05/824305/1"}}},"commit":{"parents":[{"commit":"9ae55732efd88379271e879d09bceaac6627b863","subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/9ae55732efd88379271e879d09bceaac6627b863"}]}],"author":{"name":"matbu","email":"mat.bultel@gmail.com","date":"2021-06-04 12:21:44.000000000","tz":120},"committer":{"name":"mbu","email":"mat.bultel@gmail.com","date":"2022-01-12 08:36:22.000000000","tz":0},"subject":"Add conflict_handler parameter as attribut in Command class","message":"Add conflict_handler parameter as attribut in Command class\n\nAdding conflict_handler as attribut in the Command class in order to be\nable to take control of this parameter and change to different behavior\nthat argparse is handling: error / resolve / ignore.\n\nCallers will be able to override it and get a proper Parser object.\n\nChange-Id: I327ece99a04bc8b2ebfa554dea643b1f2a456336\n(cherry picked from commit 452fff3aadb01c41355a9254c6f21b76b5ed1609)\n(cherry picked from commit e36a71b365b6058022272b27483fcba275c18aa5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/d82b9181b5aa9d7b0c32f9b3bf95827eefddebc2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/d82b9181b5aa9d7b0c32f9b3bf95827eefddebc2"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
