)]}'
{"id":"openstack%2Fcliff~842348","triplet_id":"openstack%2Fcliff~master~I2356dc9803b4d0eef3528c6d057207509932e6b2","project":"openstack/cliff","branch":"master","topic":"slow-imports","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2022-07-18 15:01:19.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"hashtags":[],"change_id":"I2356dc9803b4d0eef3528c6d057207509932e6b2","subject":"Defer loading cmd2","status":"MERGED","created":"2022-05-18 12:03:35.000000000","updated":"2022-07-18 18:43:32.000000000","submitted":"2022-07-18 18:40:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"842348-slow-imports","meta_rev_id":"5160ac43ad94026f9c1b68bf8442a940b0641288","_number":842348,"virtual_id_number":842348,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-18 18:40:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-05-27 12:05:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2022-07-14 14:18:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-18 13:56:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-27 12:05:24.000000000","updated_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"}],"messages":[{"id":"236a73b21e5772cd167ccc16f2bb1cf44c8b3b9e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-05-18 12:03:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6397c6b2f086ea7e7c08bd7d261ca7ca1bf60390","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-18 13:56:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b899332bb5564f2f9e51554630e0fab8 : SUCCESS in 1h 44m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b538408efd14251a0ca60dc8b6fec89 : SUCCESS in 3m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f7a579f57d594b31bfc1b162f80dc3e6 : SUCCESS in 3m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3626ed9cb7c24dc99176c5c8119e0ff6 : SUCCESS in 3m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa0b9653c20048b7a406b944d0542e57 : SUCCESS in 3m 37s\n- cliff-tox-py37-neutronclient-tip https://zuul.opendev.org/t/openstack/build/3cfcdf33d63745debbd9b02ef598af1f : FAILURE in 3m 12s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/13560d35c5d349518f9f69ffaf3db120 : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":1},{"id":"603560d3a9ec5b8dc34536c9eeb6f767a384516d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-05-25 16:18:26.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"91b7314653c1479fd22c53b360088e3a5bdfc0a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-25 17:38:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2e0d626931824fad8483aeede1dec10b : SUCCESS in 1h 15m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66f5fe1c26df4e4fb05d1e5b1c35ebbe : SUCCESS in 3m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5b6642f856ff4cc59dc5c4d22d5e91b6 : SUCCESS in 4m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/42e0c65e20984cbfb6dab8683d38eb17 : SUCCESS in 4m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5c7fe05defc24840a06bd9b333d0de85 : SUCCESS in 4m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de51a95db57242558c63d714f1ea64f8 : SUCCESS in 4m 30s\n- cliff-tox-py38-neutronclient-tip https://zuul.opendev.org/t/openstack/build/32b69b110aea4a8384ed03c0fd95f64e : SUCCESS in 4m 49s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/e162c416b9304797a8d355732241327f : SUCCESS in 6m 29s","accounts_in_message":[],"_revision_number":2},{"id":"84efb9a8f96dcdf80c021d7faef351ce2ac10e1f","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2022-05-27 12:05:24.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e9700c4e5770bdc34b440c0e60fd09a4ef9ee383","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-07-14 14:18:26.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d45ebdc147734f9ce11674bd7d27e4bc7c881cf7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-14 14:19:06.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"dcd9e9b11f69a601302f240f335d8db7a2087c83","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-14 16:04:17.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"2010ee17369eaa623b9d42898c52047ffc067baa","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-07-15 14:25:43.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"695be4f46e5707683c8dcb3971351af4b2cd537a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 16:02:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f4f67efee7d475088c4dffc9821457c\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/50d031bdfe184e2aaab7b877508832b3 : SUCCESS in 1h 34m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/208b5461567c46f6922532e20fce4d08 : SUCCESS in 3m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5ec18144e2bb419cbf9c1f102427e6cc : SUCCESS in 4m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f3de1042808244a3a198c52cb2edaadd : SUCCESS in 4m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3f0c43392dac4c739d38c56698c23a9f : SUCCESS in 5m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78bc7321b0a8411caec62f1a1bdb9fef : SUCCESS in 4m 38s\n- cliff-tox-py38-neutronclient-tip https://zuul.opendev.org/t/openstack/build/b6418e8b217d4e2cae2764ce1d67fcc5 : SUCCESS in 4m 45s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/3fd5b44036dd4cbf9993bf811abd10e0 : SUCCESS in 6m 25s","accounts_in_message":[],"_revision_number":2},{"id":"92b95f93ee6f6db9b8f303bfb391e52339b22c54","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 16:15:07.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7820fdb87371d7932210f8482ef6e7ba59fb62da","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 17:34:09.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"77c3c083d05f6381e23b9468e267256df191aa14","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-07-18 15:01:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3ebc77e3192d9b28bebf4cee5a3017f9fcaac5ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-18 16:42:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c72aaf6c116d4154aead8169485aaab7\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7dc9161c189a4463b43cb27c579be36d : SUCCESS in 1h 39m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e43a342e2a1c4b0ca94684899afc2262 : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a855f604eeb84d78b0fc79ece978e3f4 : SUCCESS in 6m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/102e3a6c608b46258f2c404be12b2e75 : SUCCESS in 6m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0b7fc3c8bdbf48b7be785c96eec8b905 : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21d1b6240bdf4f1f9774ad0e22f5c982 : SUCCESS in 6m 22s\n- cliff-tox-py38-neutronclient-tip https://zuul.opendev.org/t/openstack/build/3baab72ceb684dac9a5be5da0da796c3 : SUCCESS in 7m 08s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/8b3f4c682501485282c747d41d89a8fe : SUCCESS in 6m 21s","accounts_in_message":[],"_revision_number":2},{"id":"3642705d306ee3025ab47ae6358ae713ed872e90","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-18 16:48:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8cda128240761bf9bbcba82b6280136e131bc656","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-18 18:40:53.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/67a176dff2b74327bfb14af5cca619fd\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a7fa9677d8d74914b129f63730a5b8cb : SUCCESS in 1h 37m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a0391381a8f418e96e78dfcb82a1201 : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/904324fbbd07472fb26ec035a5fe4da1 : SUCCESS in 4m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c8fe7fe2a41548fcbab692894f368a6a : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f6c1c3fb8d8471590781d6aac39f5cf : SUCCESS in 4m 10s\n- cliff-tox-py38-neutronclient-tip https://zuul.opendev.org/t/openstack/build/8570adfacaad4e1b96be5528f4a055d3 : SUCCESS in 4m 42s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/4b741c24f2774871bf7dfdfcb9cc3fc7 : SUCCESS in 6m 26s","accounts_in_message":[],"_revision_number":2},{"id":"90f8c21b9082f27f06a2c8d4978c6d5ff2a37457","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-18 18:40:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"5160ac43ad94026f9c1b68bf8442a940b0641288","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-18 18:43:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eff076c08d794dd89a5fa5e617e7298e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38a5a569f238482c80fc5d680a235d92 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6811218817ccc034161dfe1b14d51eb9edc05579","revisions":{"c9858df5a25a0a23a47276d445130dd1f45ba44f":{"kind":"REWORK","_number":1,"created":"2022-05-18 12:03:35.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/48/842348/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/48/842348/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/48/842348/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/48/842348/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/48/842348/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/48/842348/1"}}},"commit":{"parents":[{"commit":"44a6dc4a89ab30b8d1d6a3487bf8b2bfaf149fcc","subject":"requirements: Remove explicit python-subunit dependency","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/44a6dc4a89ab30b8d1d6a3487bf8b2bfaf149fcc"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2022-05-18 11:02:14.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2022-05-18 11:56:10.000000000","tz":60},"subject":"Defer loading cmd2","message":"Defer loading cmd2\n\nWe were importing cmd2 purely so we could do some exception\ntransformation. However, this is only needed if we\u0027re in interactive\nmode. Avoid both the import of cmd2 and the transformation of the\nexceptions unless this is the case. This speeds up import time by ~30%\nfor the demoapp on my machine (~160mS after compared to ~210mS before)\n\nChange-Id: I2356dc9803b4d0eef3528c6d057207509932e6b2\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/c9858df5a25a0a23a47276d445130dd1f45ba44f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/c9858df5a25a0a23a47276d445130dd1f45ba44f"}]},"branch":"refs/heads/master"},"6811218817ccc034161dfe1b14d51eb9edc05579":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-05-25 16:18:26.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/48/842348/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/48/842348/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/48/842348/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/48/842348/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/48/842348/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/48/842348/2"}}},"commit":{"parents":[{"commit":"98095d406ac5b363807385404afce207f29e620b","subject":"requirements: Remove explicit python-subunit dependency","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/98095d406ac5b363807385404afce207f29e620b"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2022-05-18 11:02:14.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2022-05-25 16:09:04.000000000","tz":60},"subject":"Defer loading cmd2","message":"Defer loading cmd2\n\nWe were importing cmd2 purely so we could do some exception\ntransformation. However, this is only needed if we\u0027re in interactive\nmode. Avoid both the import of cmd2 and the transformation of the\nexceptions unless this is the case. This speeds up import time by ~30%\nfor the demoapp on my machine (~160mS after compared to ~210mS before)\n\nChange-Id: I2356dc9803b4d0eef3528c6d057207509932e6b2\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/6811218817ccc034161dfe1b14d51eb9edc05579"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/6811218817ccc034161dfe1b14d51eb9edc05579"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}]}],"submit_requirements":[]}
