)]}'
{"id":"openstack%2Fcliff~855726","triplet_id":"openstack%2Fcliff~master~I6f3cb7e6f31f882486b8352be10abd4b45fc00b6","project":"openstack/cliff","branch":"master","topic":"release-health-check","hashtags":[],"change_id":"I6f3cb7e6f31f882486b8352be10abd4b45fc00b6","subject":"DNM: dummy change to test gate health","status":"ABANDONED","created":"2022-09-03 20:27:12.000000000","updated":"2022-09-07 09:19:52.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c085f2cd84378934c6d74a102476fe8bef4971b0","_number":855726,"virtual_id_number":855726,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-09-03 21:59:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-03 21:59:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"417032161fb4f1ead69d87cc7b484d61c0b6cfde","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-09-03 20:27:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"64c7bfc3458d1b1057fb5781d372793f1775ecf0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-03 21:59:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d661058c1144db3a55adf0901b30e46\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9ceaa964423941f5b158a32fefaabf6f : SUCCESS in 1h 31m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/beba2ea80b1441eb9cf7c884f871b3f9 : SUCCESS in 3m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c0975c09ac3a4333a03087d0b7a8ddbf : SUCCESS in 3m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5f5df3f4a2cd4de0bbe6cc6a25b65a0c : SUCCESS in 4m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8e0673565f5142ea89f66ca6a5d1f3ae : SUCCESS in 3m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bdbfebb81be43d7a71dc8c1f3ac3893 : SUCCESS in 3m 48s\n- cliff-tox-py38-neutronclient-tip https://zuul.opendev.org/t/openstack/build/3b95cdcebeb545049faf450020d9b1f8 : SUCCESS in 3m 52s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/13203c9acc084c02bec1978db89bc399 : SUCCESS in 6m 01s","accounts_in_message":[],"_revision_number":1},{"id":"c085f2cd84378934c6d74a102476fe8bef4971b0","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-09-07 09:19:52.000000000","message":"Abandoned\n\nGate check passed, no further need for this patch.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c3ecb293f8b0811985cc78f61c58b4d96757687e","revisions":{"c3ecb293f8b0811985cc78f61c58b4d96757687e":{"kind":"REWORK","_number":1,"created":"2022-09-03 20:27:12.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/26/855726/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/26/855726/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/26/855726/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/26/855726/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/26/855726/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/26/855726/1"}}},"commit":{"parents":[{"commit":"58c853d1015c8bde2afa777a9bb1246a65645f3f","subject":"Merge \"Removing brackets around tested conditional\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/58c853d1015c8bde2afa777a9bb1246a65645f3f"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-09-03 20:27:00.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-09-03 20:27:11.000000000","tz":120},"subject":"DNM: dummy change to test gate health","message":"DNM: dummy change to test gate health\n\nChange-Id: I6f3cb7e6f31f882486b8352be10abd4b45fc00b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/c3ecb293f8b0811985cc78f61c58b4d96757687e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/c3ecb293f8b0811985cc78f61c58b4d96757687e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
