)]}'
{"id":"openstack%2Fcliff~859579","triplet_id":"openstack%2Fcliff~master~I624fc0059057de05bb2bacd9c1238cba71ebad2a","project":"openstack/cliff","branch":"master","topic":"importlib-metadata","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-12-19 13:03:21.000000000","reason":"Change was submitted"},"2472":{"account":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"last_update":"2023-12-19 13:03:21.000000000","reason":"Change was submitted"},"35352":{"account":{"_account_id":35352,"name":"Benjamin Alan Beasley","email":"code@musicinmybrain.net","username":"music"},"last_update":"2023-12-19 13:03:21.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-11-29 13:13:04.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I624fc0059057de05bb2bacd9c1238cba71ebad2a","subject":"Use importlib.metadata on Python 3.10+","status":"MERGED","created":"2022-09-28 12:55:34.000000000","updated":"2023-12-19 13:06:01.000000000","submitted":"2023-12-19 13:03:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"859579-importlib-metadata","meta_rev_id":"26fe1ea2264a5054be584a08c04be29e8b408e0d","_number":859579,"virtual_id_number":859579,"owner":{"_account_id":35352,"name":"Benjamin Alan Beasley","email":"code@musicinmybrain.net","username":"music"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-19 13:03:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2023-12-01 01:09:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2023-12-19 11:36:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"}],"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-28 14:33:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-01 14:09:34.000000000","updated_by":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"reviewer":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"state":"CC"},{"updated":"2022-10-04 16:46:23.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2023-11-07 16:43:31.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2023-11-28 10:25:33.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"036fb07c4833194872d44f7bce25ccea5d383471","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35352,"name":"Benjamin Alan Beasley","email":"code@musicinmybrain.net","username":"music"},"date":"2022-09-28 12:55:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"df68f64b5534acd872151edb6e9dd5f2ec963122","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-28 14:33:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/875f348e6e7548f190f2966b97b5ce24\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/fea9e6c59d564794adc2560592c5bfd0 : FAILURE in 1m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/663029293a554f9f933dc3ede6695ae0 : SUCCESS in 1h 33m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf60348b4bdc40b3831b9433efde94d5 : SUCCESS in 2m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/044ef51eb2c84c6694c23b014b297cdd : SUCCESS in 3m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/21b601b3b01f4c8b8a716311e531e792 : SUCCESS in 2m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32a46a8c70644a41981ac444d5e9b6f8 : SUCCESS in 3m 04s\n- cliff-tox-py38-neutronclient-tip https://zuul.opendev.org/t/openstack/build/7c0c05d26fec412dbc7d5573c0dcb862 : SUCCESS in 2m 58s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/ef67aaa5fd674d51ba6afb86525f9a04 : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":1},{"id":"aca3a7f089bb756ed4a7d9950fd83fba8b9a6b40","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2022-10-01 14:09:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3074c145ee5932570931c79602eff9307babc245","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-10-04 16:46:23.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ee100f2e04162bfa41f3530a2614eaa8cb2cf34d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35352,"name":"Benjamin Alan Beasley","email":"code@musicinmybrain.net","username":"music"},"date":"2022-10-05 15:12:21.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"6b3d6e2fcdff438563b7ccce48996499652ae8e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-05 16:52:16.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/29761636f54c43989add54cee8e1635c\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/5db589c346a148749c2269596ba6adb6 : FAILURE in 2m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7bdf6d6e84514fa1a04c065f0fc5ff29 : SUCCESS in 1h 34m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53a29180c66e48c99279d7f23dfc67a3 : SUCCESS in 3m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/660c0672db504ffe89e7b2f22324f32a : SUCCESS in 3m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a7323241d6e543ff95352dbf0f7814d9 : SUCCESS in 2m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/843a0423d18c462fbaf7ad8f469df3ed : SUCCESS in 3m 46s\n- cliff-tox-py38-neutronclient-tip https://zuul.opendev.org/t/openstack/build/15f07d5527cc4597ae5fc3cc27073521 : SUCCESS in 3m 33s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/12cdd0d88e8242418d5f4456b944f215 : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":2},{"id":"33c4880c206de3d3e2b6ee80e6a55a76de20729a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35352,"name":"Benjamin Alan Beasley","email":"code@musicinmybrain.net","username":"music"},"date":"2023-01-04 19:43:43.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"395e78aded68438e75e130f37e044fd1fd9c2543","author":{"_account_id":35352,"name":"Benjamin Alan Beasley","email":"code@musicinmybrain.net","username":"music"},"date":"2023-01-04 19:50:30.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"a1d6cd7652327a51d788a49987eef947caf91e0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-04 21:15:02.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/929abfd859ae4aaab439a7989bdad78e\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/12af2df67c224eada5b577d52b82cd9b : FAILURE in 1m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c3bfc09e95384109b9e87310b74e3e3b : SUCCESS in 1h 28m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d26a72a4845041419ed1055e421c1008 : FAILURE in 2m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6661ce77d0e14961b28c0d26221dd4ea : FAILURE in 2m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a86fc50462364329ad391f894310a5d4 : FAILURE in 3m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/080ffa2f0f994dbcae598cfb8aa65ba9 : FAILURE in 2m 33s\n- cliff-tox-py38-neutronclient-tip https://zuul.opendev.org/t/openstack/build/e45cc9161033498ea024c639bcb63ccd : FAILURE in 3m 01s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/0aa93581eb974facb8fe733fea0d175e : SUCCESS in 4m 24s","accounts_in_message":[],"_revision_number":3},{"id":"b5ef38161bd2f9af52ee3600e6324921571fae6b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-02-28 15:08:32.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b1d9b2532d2cff53eadafd7a8b45195995f89b98","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-02-28 15:10:14.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"36d055f7efce9b321bc39c883bbe5a9c1cca54f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 16:42:47.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/787d440b5a1e41698aab5639157b5ec2\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/f2e2713a7bb54ac1900a1cc14b7d5622 : FAILURE in 2m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fb70043c03d84cc288918920e42b0d38 : SUCCESS in 1h 27m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79cbc27180b349e5a346d5c22c2d7221 : SUCCESS in 2m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1882965ad3d447119c06f24dd4c64ea4 : SUCCESS in 2m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/49b71e1a244a429687a289f0a0f24270 : SUCCESS in 3m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1eced951a584157bf87981fb7c48d12 : SUCCESS in 3m 08s\n- cliff-tox-py38-neutronclient-tip https://zuul.opendev.org/t/openstack/build/2aad0a4575e2476897c28c7556acca6f : SUCCESS in 2m 59s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/69328f9a8d1a4886b4fa0ccf9a536ea3 : SUCCESS in 4m 34s","accounts_in_message":[],"_revision_number":3},{"id":"dbecdc6b7b46da9d9499ee26cd8979e4c11eb4bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35352,"name":"Benjamin Alan Beasley","email":"code@musicinmybrain.net","username":"music"},"date":"2023-02-28 19:11:09.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"581d5f2171045a450b064bf9cb68a3b788b2b7fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 20:32:36.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d73a8574267f42e88cf7fa83574e6960\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/0c3cf1f5703446aea4c4db47040d81f7 : FAILURE in 3m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1b9060d90289491d83ed722f45ce7699 : SUCCESS in 1h 18m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d966c2ac380486ba97a115501be2447 : SUCCESS in 2m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/954939ca891b49708d312ca4a92081ae : FAILURE in 2m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0642461870e34469818ef502d0ddc8e7 : SUCCESS in 2m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa60f16328f94740bc5e289750dce714 : SUCCESS in 3m 06s\n- cliff-tox-py38-neutronclient-tip https://zuul.opendev.org/t/openstack/build/e9b82982221447448d35fdf3a612fa9e : FAILURE in 3m 07s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/e4919007488740bbacbacb64644e64b2 : FAILURE in 3m 57s","accounts_in_message":[],"_revision_number":4},{"id":"7710019e053fc53e232ec74fd643296f76f4f451","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-11 07:41:47.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"b07a3021fbefd04670811ca13237dd88418988c7","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-11 07:42:06.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"a888d6f70c5f37eea3a9213c26f0151875990a64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 09:02:10.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/66a1c79e4a134f7c809ab41dff58b65e\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/0fd37fd8e45d412ba5a108b66fe787ba : FAILURE in 1m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/280c1e4cb13844249c199604433e3750 : SUCCESS in 1h 19m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a40d26a6e2c4846880c19ba1da82439 : SUCCESS in 2m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/903e263163ea4d00a96f41793c699302 : FAILURE in 2m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4757e16e83c64898a87ff97a5a9e3045 : SUCCESS in 3m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f5b9074aa0c4e4994d0067b0b29bf51 : SUCCESS in 3m 12s\n- cliff-tox-py38-neutronclient-tip https://zuul.opendev.org/t/openstack/build/11451ccdd79c4a729e30c3d826b1f02b : FAILURE in 2m 47s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/d258ac0247604ee1ad98d3f85477d50e : FAILURE in 4m 53s","accounts_in_message":[],"_revision_number":5},{"id":"96c905f42e1b1112de42c36b8d03f926c0738c3f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-07 16:43:31.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"24eb2973fb868a16dae1afdb14865f9324529dff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-07 18:23:38.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ee0b2fbf23834814b4a28d3f8dd507b9\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/e50a71131a644c618546fbfd75e8b4dd : FAILURE in 1m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/846c222945df42adacbec7f140b1518c : SUCCESS in 1h 34m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7234d37454c4a2f9b3759a4e4bc1b8e : SUCCESS in 3m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4008b6084a1e47a9b324a3702fcf8181 : FAILURE in 2m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/085dbeabb6da4d1182566c15f9376eec : SUCCESS in 3m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54cbeff253cf4290acdb0aad0bc9591f : SUCCESS in 3m 00s\n- cliff-tox-py38-neutronclient-tip https://zuul.opendev.org/t/openstack/build/15fa21a14e4b44f89f943700d532c780 : FAILURE in 3m 15s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/025d8d2430204ad4a65a748b2be87171 : FAILURE in 3m 47s","accounts_in_message":[],"_revision_number":5},{"id":"74be06f124585d0aab3d7fa059dca5017188bc4b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-28 10:25:33.000000000","message":"Patch Set 5: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"e2132abcdcde75ec27daf121ea0badb86c19a407","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"real_author":{"_account_id":35352,"name":"Benjamin Alan Beasley","email":"code@musicinmybrain.net","username":"music"},"date":"2023-11-28 16:54:59.000000000","message":"Patch Set 6: Patch Set 5 was rebased on behalf of \u003cGERRIT_ACCOUNT_15334\u003e\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}],"_revision_number":6},{"id":"03fac0e001d472bf0863931df49481dfe6483b23","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35352,"name":"Benjamin Alan Beasley","email":"code@musicinmybrain.net","username":"music"},"date":"2023-11-28 17:00:26.000000000","message":"Patch Set 7: Published edit on patch set 6.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":7},{"id":"4036d2185aeb2a494c393298c3a21a06cfe0ea18","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35352,"name":"Benjamin Alan Beasley","email":"code@musicinmybrain.net","username":"music"},"date":"2023-11-28 17:08:07.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"e0a195839e1d9d737b3b7152d86bd56f40443234","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 17:25:06.000000000","message":"Patch Set 8: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/df5ce763defb4c529c3e171acd9254cf\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/a7e9e7f1c4324dbc9ecef6b3fa14664d : FAILURE in 2m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c1f29b8e29dc44e0b231cdeaf123c28b : FAILURE in 14m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f20fcc940b0440d9ece74c57586423e : FAILURE in 3m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/491df791dbb74daa90234a334d3eeb80 : FAILURE in 2m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f2a74a574a2d4f6cbf277fd2c112b13b : FAILURE in 2m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/baba270c8a734acc9a2e9fac278cd7a5 : FAILURE in 2m 37s\n- cliff-tox-py38-neutronclient-tip https://zuul.opendev.org/t/openstack/build/833aad6a61654ae2b6b807a6b1abced3 : FAILURE in 4m 05s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/f230d7d91ac643dc9e858e511a7369d2 : FAILURE in 4m 09s","accounts_in_message":[],"_revision_number":8},{"id":"d4ae7854a29f1fd55b3ad339f14e7ef4fa5ebe70","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-29 13:12:12.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"b686d2b25b3dab4e77b03f750f6919923178db18","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-29 13:12:55.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"2c211b92aed52f2a08bed7849542534fa78f30ad","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-29 13:13:04.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"e1a4839480b363379a1dc3fb612d65dcbb914e15","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-29 13:13:49.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"ffdfd6d2961aa75ab797208b5c0bc661f678d72e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-29 14:55:11.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b6260f1f3ef94a2a85e4a1b4e93fc81f\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/cfa4d41f28e949aa8888318f3263f385 : SUCCESS in 1m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6c9321b21871405a99b5e074a4ed4688 : SUCCESS in 1h 39m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d378ca5c50604dd7922e58d1addc813c : SUCCESS in 2m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c817e9679ea54c41a4a42783d9d55e36 : SUCCESS in 2m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ca0a95036ccd42d48542e37d690d196a : SUCCESS in 2m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/470d1b42bc584a4282058bd23d414a33 : SUCCESS in 3m 11s\n- cliff-tox-py38-neutronclient-tip https://zuul.opendev.org/t/openstack/build/bbebccf8889f43a789a7349ca24f029b : SUCCESS in 4m 00s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/c608f837b0224733aa962393d2cf0319 : SUCCESS in 5m 48s","accounts_in_message":[],"_revision_number":11},{"id":"5e809aa2e441f0eac9b599632d6cd802bfb26fa2","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-12-01 01:09:55.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"c82d7f2caf4223955b7d815ac4fd54d0d9558609","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-12-19 11:36:08.000000000","message":"Patch Set 11: Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"217ddac6270f7289797a163a53d8cf890a044803","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-19 11:36:42.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"03ac63b502cc4ec626b56aaa608e274ebf3af8e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-19 13:03:21.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43db02952ef34913a8e95f69a28fcc04\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/64b7b72481364d4ba75ac60bbc38d64d : SUCCESS in 2m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bb8ac93b0714444ab467957275600311 : SUCCESS in 1h 26m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9486bdfd02a745a2a2017c9490e4abc2 : SUCCESS in 2m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/74c67441a9f548d5964ee7aaba66dfc0 : SUCCESS in 2m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a393faa169774d2eafb82ba27dbf314e : SUCCESS in 2m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/144fac5b764d4e7db8ec2e4020908ccd : SUCCESS in 2m 48s\n- cliff-tox-py38-neutronclient-tip https://zuul.opendev.org/t/openstack/build/469de4a857dc4ac19bfc46e854071317 : SUCCESS in 3m 19s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/4f48096864eb469884450eeeffcffe81 : SUCCESS in 4m 20s","accounts_in_message":[],"_revision_number":11},{"id":"5b9c59715b099a3bfa7407ec1bc5b86a01fe3ef5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-19 13:03:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":11},{"id":"26fe1ea2264a5054be584a08c04be29e8b408e0d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-19 13:06:01.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/67ec53780e86457d90e13824ebb2251c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d295fbac02814291a7df0ec3a229e365 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"156e1eb0b9b9ad5feaa4b3e772f2cbc6a7694cf5","revisions":{"6af10b8905061856d4654dbe380f1de6c5a0eb76":{"kind":"REWORK","_number":1,"created":"2022-09-28 12:55:34.000000000","uploader":{"_account_id":35352,"name":"Benjamin Alan Beasley","email":"code@musicinmybrain.net","username":"music"},"ref":"refs/changes/79/859579/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/79/859579/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/79/859579/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/79/859579/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/79/859579/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/79/859579/1"}}},"commit":{"parents":[{"commit":"9614d30450c642dd176ba00e6b6c2a893d742171","subject":"Add Python3 antelope unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/9614d30450c642dd176ba00e6b6c2a893d742171"}]}],"author":{"name":"Benjamin A. Beasley","email":"code@musicinmybrain.net","date":"2022-09-28 12:28:18.000000000","tz":-240},"committer":{"name":"Benjamin A. Beasley","email":"code@musicinmybrain.net","date":"2022-09-28 12:54:44.000000000","tz":-240},"subject":"Use importlib.metadata on Python 3.10+","message":"Use importlib.metadata on Python 3.10+\n\nFavor importlib.metadata.packages_distributions() over\nimportlib_metadata.packages_distributions() where it is available, i.e.,\non Python 3.10 and later.\n\nDepend on PyPI importlib_metadata only on Python 3.9 and older:\nconditionalize the requirements.txt entry on Python version.\n\nChange-Id: I624fc0059057de05bb2bacd9c1238cba71ebad2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/6af10b8905061856d4654dbe380f1de6c5a0eb76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/6af10b8905061856d4654dbe380f1de6c5a0eb76"}]},"branch":"refs/heads/master"},"fe58e0ac781bc581acd1a0451a6439db8916fd42":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-10-05 15:12:21.000000000","uploader":{"_account_id":35352,"name":"Benjamin Alan Beasley","email":"code@musicinmybrain.net","username":"music"},"ref":"refs/changes/79/859579/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/79/859579/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/79/859579/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/79/859579/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/79/859579/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/79/859579/2"}}},"commit":{"parents":[{"commit":"91c62985fb6c529cecfed5354ae687d62c25a821","subject":"Merge \"columns: Useful __str__, __repr__ implementation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/91c62985fb6c529cecfed5354ae687d62c25a821"}]}],"author":{"name":"Benjamin A. Beasley","email":"code@musicinmybrain.net","date":"2022-09-28 12:28:18.000000000","tz":-240},"committer":{"name":"Benjamin Alan Beasley","email":"code@musicinmybrain.net","date":"2022-10-05 15:12:21.000000000","tz":0},"subject":"Use importlib.metadata on Python 3.10+","message":"Use importlib.metadata on Python 3.10+\n\nFavor importlib.metadata.packages_distributions() over\nimportlib_metadata.packages_distributions() where it is available, i.e.,\non Python 3.10 and later.\n\nDepend on PyPI importlib_metadata only on Python 3.9 and older:\nconditionalize the requirements.txt entry on Python version.\n\nChange-Id: I624fc0059057de05bb2bacd9c1238cba71ebad2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/fe58e0ac781bc581acd1a0451a6439db8916fd42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/fe58e0ac781bc581acd1a0451a6439db8916fd42"}]},"branch":"refs/heads/master","description":"Rebase"},"29e4f030a68f18f77e504d51203b4899cfce113e":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-01-04 19:43:43.000000000","uploader":{"_account_id":35352,"name":"Benjamin Alan Beasley","email":"code@musicinmybrain.net","username":"music"},"ref":"refs/changes/79/859579/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/79/859579/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/79/859579/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/79/859579/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/79/859579/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/79/859579/3"}}},"commit":{"parents":[{"commit":"fb9a3a9b2daa21317d0adceca86215c393e975e6","subject":"Strip trailing periods when getting description","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/fb9a3a9b2daa21317d0adceca86215c393e975e6"}]}],"author":{"name":"Benjamin A. Beasley","email":"code@musicinmybrain.net","date":"2022-09-28 12:28:18.000000000","tz":-240},"committer":{"name":"Benjamin Alan Beasley","email":"code@musicinmybrain.net","date":"2023-01-04 19:43:43.000000000","tz":0},"subject":"Use importlib.metadata on Python 3.10+","message":"Use importlib.metadata on Python 3.10+\n\nFavor importlib.metadata.packages_distributions() over\nimportlib_metadata.packages_distributions() where it is available, i.e.,\non Python 3.10 and later.\n\nDepend on PyPI importlib_metadata only on Python 3.9 and older:\nconditionalize the requirements.txt entry on Python version.\n\nChange-Id: I624fc0059057de05bb2bacd9c1238cba71ebad2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/29e4f030a68f18f77e504d51203b4899cfce113e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/29e4f030a68f18f77e504d51203b4899cfce113e"}]},"branch":"refs/heads/master","description":"Rebase"},"35f96cbf439a77c4dd6b4ee7376ffdbd626ad0a2":{"kind":"REWORK","_number":4,"created":"2023-02-28 19:11:09.000000000","uploader":{"_account_id":35352,"name":"Benjamin Alan Beasley","email":"code@musicinmybrain.net","username":"music"},"ref":"refs/changes/79/859579/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/79/859579/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/79/859579/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/79/859579/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/79/859579/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/79/859579/4"}}},"commit":{"parents":[{"commit":"fb9a3a9b2daa21317d0adceca86215c393e975e6","subject":"Strip trailing periods when getting description","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/fb9a3a9b2daa21317d0adceca86215c393e975e6"}]}],"author":{"name":"Benjamin A. Beasley","email":"code@musicinmybrain.net","date":"2022-09-28 12:28:18.000000000","tz":-240},"committer":{"name":"Benjamin Alan Beasley","email":"code@musicinmybrain.net","date":"2023-02-28 19:10:35.000000000","tz":0},"subject":"Use importlib.metadata on Python 3.10+","message":"Use importlib.metadata on Python 3.10+\n\nFavor importlib.metadata.packages_distributions() over\nimportlib_metadata.packages_distributions() where it is available, i.e.,\non Python 3.10 and later.\n\nDepend on PyPI importlib_metadata only on Python 3.9 and older:\nconditionalize the requirements.txt entry on Python version.\n\nChange-Id: I624fc0059057de05bb2bacd9c1238cba71ebad2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/35f96cbf439a77c4dd6b4ee7376ffdbd626ad0a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/35f96cbf439a77c4dd6b4ee7376ffdbd626ad0a2"}]},"branch":"refs/heads/master"},"de0b23ef1aeee8a62aadf0779e142d47bf9ee89e":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-09-11 07:41:47.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/79/859579/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/79/859579/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/79/859579/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/79/859579/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/79/859579/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/79/859579/5"}}},"commit":{"parents":[{"commit":"6096869f7060bf74586b4f491affd3dea6630a8b","subject":"Fix flake8 violation E721","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/6096869f7060bf74586b4f491affd3dea6630a8b"}]}],"author":{"name":"Benjamin A. Beasley","email":"code@musicinmybrain.net","date":"2022-09-28 12:28:18.000000000","tz":-240},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-09-11 07:41:47.000000000","tz":0},"subject":"Use importlib.metadata on Python 3.10+","message":"Use importlib.metadata on Python 3.10+\n\nFavor importlib.metadata.packages_distributions() over\nimportlib_metadata.packages_distributions() where it is available, i.e.,\non Python 3.10 and later.\n\nDepend on PyPI importlib_metadata only on Python 3.9 and older:\nconditionalize the requirements.txt entry on Python version.\n\nChange-Id: I624fc0059057de05bb2bacd9c1238cba71ebad2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/de0b23ef1aeee8a62aadf0779e142d47bf9ee89e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/de0b23ef1aeee8a62aadf0779e142d47bf9ee89e"}]},"branch":"refs/heads/master","description":"Rebase"},"44f084a8ec0a60618459f73305a504ae6dbbe135":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2023-11-28 16:54:59.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"real_uploader":{"_account_id":35352,"name":"Benjamin Alan Beasley","email":"code@musicinmybrain.net","username":"music"},"ref":"refs/changes/79/859579/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/79/859579/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/79/859579/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/79/859579/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/79/859579/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/79/859579/6"}}},"commit":{"parents":[{"commit":"3ee0725a094b0f80fa72f6bfc8bc97fe20b8fde8","subject":"Merge \"Handle complex objects in yaml formatter better\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/3ee0725a094b0f80fa72f6bfc8bc97fe20b8fde8"}]}],"author":{"name":"Benjamin A. Beasley","email":"code@musicinmybrain.net","date":"2022-09-28 12:28:18.000000000","tz":-240},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-11-28 16:54:59.000000000","tz":0},"subject":"Use importlib.metadata on Python 3.10+","message":"Use importlib.metadata on Python 3.10+\n\nFavor importlib.metadata.packages_distributions() over\nimportlib_metadata.packages_distributions() where it is available, i.e.,\non Python 3.10 and later.\n\nDepend on PyPI importlib_metadata only on Python 3.9 and older:\nconditionalize the requirements.txt entry on Python version.\n\nChange-Id: I624fc0059057de05bb2bacd9c1238cba71ebad2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/44f084a8ec0a60618459f73305a504ae6dbbe135"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/44f084a8ec0a60618459f73305a504ae6dbbe135"}]},"branch":"refs/heads/master","description":"Rebase"},"1e8097b2ad34571f22ca2cd2c18fb32c6e83dd17":{"kind":"REWORK","_number":7,"created":"2023-11-28 17:00:26.000000000","uploader":{"_account_id":35352,"name":"Benjamin Alan Beasley","email":"code@musicinmybrain.net","username":"music"},"ref":"refs/changes/79/859579/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/79/859579/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/79/859579/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/79/859579/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/79/859579/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/79/859579/7"}}},"commit":{"parents":[{"commit":"3ee0725a094b0f80fa72f6bfc8bc97fe20b8fde8","subject":"Merge \"Handle complex objects in yaml formatter better\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/3ee0725a094b0f80fa72f6bfc8bc97fe20b8fde8"}]}],"author":{"name":"Benjamin A. Beasley","email":"code@musicinmybrain.net","date":"2022-09-28 12:28:18.000000000","tz":-240},"committer":{"name":"Benjamin Alan Beasley","email":"code@musicinmybrain.net","date":"2023-11-28 16:59:42.000000000","tz":0},"subject":"Use importlib.metadata on Python 3.10+","message":"Use importlib.metadata on Python 3.10+\n\nFavor importlib.metadata.packages_distributions() over\nimportlib_metadata.packages_distributions() where it is available, i.e.,\non Python 3.10 and later.\n\nDepend on PyPI importlib_metadata only on Python 3.9 and older:\nconditionalize the requirements.txt entry on Python version.\n\nChange-Id: I624fc0059057de05bb2bacd9c1238cba71ebad2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/1e8097b2ad34571f22ca2cd2c18fb32c6e83dd17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/1e8097b2ad34571f22ca2cd2c18fb32c6e83dd17"}]},"branch":"refs/heads/master"},"2bf5aaf9f38bdbe51c42805b5602b51e08fbdccc":{"kind":"REWORK","_number":8,"created":"2023-11-28 17:08:07.000000000","uploader":{"_account_id":35352,"name":"Benjamin Alan Beasley","email":"code@musicinmybrain.net","username":"music"},"ref":"refs/changes/79/859579/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/79/859579/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/79/859579/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/79/859579/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/79/859579/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/79/859579/8"}}},"commit":{"parents":[{"commit":"3ee0725a094b0f80fa72f6bfc8bc97fe20b8fde8","subject":"Merge \"Handle complex objects in yaml formatter better\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/3ee0725a094b0f80fa72f6bfc8bc97fe20b8fde8"}]}],"author":{"name":"Benjamin A. Beasley","email":"code@musicinmybrain.net","date":"2022-09-28 12:28:18.000000000","tz":-240},"committer":{"name":"Benjamin Alan Beasley","email":"code@musicinmybrain.net","date":"2023-11-28 17:08:07.000000000","tz":0},"subject":"Use importlib.metadata on Python 3.10+","message":"Use importlib.metadata on Python 3.10+\n\nFavor importlib.metadata.packages_distributions() over\nimportlib_metadata.packages_distributions() where it is available, i.e.,\non Python 3.10 and later.\n\nDepend on PyPI importlib_metadata only on Python 3.9 and older:\nconditionalize the requirements.txt entry on Python version.\n\nChange-Id: I624fc0059057de05bb2bacd9c1238cba71ebad2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/2bf5aaf9f38bdbe51c42805b5602b51e08fbdccc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/2bf5aaf9f38bdbe51c42805b5602b51e08fbdccc"}]},"branch":"refs/heads/master"},"de0634f390b7d26d08627c51fd8caf977e96c833":{"kind":"REWORK","_number":9,"created":"2023-11-29 13:12:12.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/79/859579/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/79/859579/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/79/859579/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/79/859579/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/79/859579/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/79/859579/9"}}},"commit":{"parents":[{"commit":"3ee0725a094b0f80fa72f6bfc8bc97fe20b8fde8","subject":"Merge \"Handle complex objects in yaml formatter better\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/3ee0725a094b0f80fa72f6bfc8bc97fe20b8fde8"}]}],"author":{"name":"Benjamin A. Beasley","email":"code@musicinmybrain.net","date":"2022-09-28 12:28:18.000000000","tz":-240},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-29 13:12:09.000000000","tz":540},"subject":"Use importlib.metadata on Python 3.10+","message":"Use importlib.metadata on Python 3.10+\n\nFavor importlib.metadata.packages_distributions() over\nimportlib_metadata.packages_distributions() where it is available, i.e.,\non Python 3.10 and later.\n\nDepend on PyPI importlib_metadata only on Python 3.9 and older:\nconditionalize the requirements.txt entry on Python version.\n\nChange-Id: I624fc0059057de05bb2bacd9c1238cba71ebad2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/de0634f390b7d26d08627c51fd8caf977e96c833"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/de0634f390b7d26d08627c51fd8caf977e96c833"}]},"branch":"refs/heads/master"},"732ac31dae15f5fac31c17026c88bb573a12b3dd":{"kind":"REWORK","_number":10,"created":"2023-11-29 13:12:55.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/79/859579/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/79/859579/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/79/859579/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/79/859579/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/79/859579/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/79/859579/10"}}},"commit":{"parents":[{"commit":"3ee0725a094b0f80fa72f6bfc8bc97fe20b8fde8","subject":"Merge \"Handle complex objects in yaml formatter better\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/3ee0725a094b0f80fa72f6bfc8bc97fe20b8fde8"}]}],"author":{"name":"Benjamin A. Beasley","email":"code@musicinmybrain.net","date":"2022-09-28 12:28:18.000000000","tz":-240},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-29 13:12:47.000000000","tz":540},"subject":"Use importlib.metadata on Python 3.10+","message":"Use importlib.metadata on Python 3.10+\n\nFavor importlib.metadata.packages_distributions() over\nimportlib_metadata.packages_distributions() where it is available, i.e.,\non Python 3.10 and later.\n\nDepend on PyPI importlib_metadata only on Python 3.9 and older:\nconditionalize the requirements.txt entry on Python version.\n\nChange-Id: I624fc0059057de05bb2bacd9c1238cba71ebad2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/732ac31dae15f5fac31c17026c88bb573a12b3dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/732ac31dae15f5fac31c17026c88bb573a12b3dd"}]},"branch":"refs/heads/master"},"156e1eb0b9b9ad5feaa4b3e772f2cbc6a7694cf5":{"kind":"REWORK","_number":11,"created":"2023-11-29 13:13:49.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/79/859579/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/79/859579/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/79/859579/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/79/859579/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/79/859579/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/79/859579/11"}}},"commit":{"parents":[{"commit":"3ee0725a094b0f80fa72f6bfc8bc97fe20b8fde8","subject":"Merge \"Handle complex objects in yaml formatter better\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/3ee0725a094b0f80fa72f6bfc8bc97fe20b8fde8"}]}],"author":{"name":"Benjamin A. Beasley","email":"code@musicinmybrain.net","date":"2022-09-28 12:28:18.000000000","tz":-240},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-29 13:13:41.000000000","tz":540},"subject":"Use importlib.metadata on Python 3.10+","message":"Use importlib.metadata on Python 3.10+\n\nFavor importlib.metadata.packages_distributions() over\nimportlib_metadata.packages_distributions() where it is available, i.e.,\non Python 3.10 and later.\n\nDepend on PyPI importlib_metadata only on Python 3.9 and older:\nconditionalize the requirements.txt entry on Python version.\n\nChange-Id: I624fc0059057de05bb2bacd9c1238cba71ebad2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/156e1eb0b9b9ad5feaa4b3e772f2cbc6a7694cf5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/156e1eb0b9b9ad5feaa4b3e772f2cbc6a7694cf5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
