)]}'
{"id":"openstack%2Fcliff~862690","triplet_id":"openstack%2Fcliff~master~I6cc2499c371b86aaf6636d7a46416bc947c11bb2","project":"openstack/cliff","branch":"master","topic":"damerau-levenshtein/clarification","attention_set":{},"removed_from_attention_set":{"27900":{"account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"last_update":"2023-09-11 07:45:50.000000000","reason":"\u003cGERRIT_ACCOUNT_27900\u003e replied on the change","reason_account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}},"32926":{"account":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"last_update":"2023-09-11 08:49:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6cc2499c371b86aaf6636d7a46416bc947c11bb2","subject":"Clarification of the algorithm used","status":"MERGED","created":"2022-10-26 13:15:05.000000000","updated":"2023-09-11 08:50:57.000000000","submitted":"2023-09-11 08:49:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"862690-damerau-levenshtein/clarification","meta_rev_id":"2aad4fb70925729c43719daec460002e9b5aa654","_number":862690,"virtual_id_number":862690,"owner":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-11 08:49:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2023-09-11 07:41:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-09-11 07:45:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2023-09-11 07:41:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-26 14:29:26.000000000","updated_by":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"},{"updated":"2022-10-26 14:39:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-11 07:41:12.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"20356cf195e73f9b9e119aaadc12e032fc72bc7d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2022-10-26 13:15:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bfd10893a23400263db1fff9ede651277f7a80e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-26 14:39:37.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ec7d0c1355ad41efb8bf8155276142ae\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a728d925da294ae3bf36bcd258abc38d : SUCCESS in 1h 22m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d57b07c0d3a4979afa44afadd1f0d99 : FAILURE in 3m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/69d4e75034cf4e69adb44c5b436b22b2 : SUCCESS in 2m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7bee4193c76043b89a241b708b51c55c : SUCCESS in 2m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0042f227e21a4e58a9d6c5dbdf62fbdc : SUCCESS in 3m 00s\n- cliff-tox-py38-neutronclient-tip https://zuul.opendev.org/t/openstack/build/9e9c0a155fe0488b93e6d162e5dc99e9 : SUCCESS in 3m 49s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/cff8e358dfb4485a989422c5edabb122 : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":1},{"id":"ef772439e54f2ab1f2650b5141ac93eb97d00f78","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2022-10-27 06:58:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"df9641a0fb352747d6892abdd796af7d2a14c177","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-27 08:23:47.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d62d3241f6b84bd3a001f946b14cad50\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2974e0a590d2417588322a726e4b2e59 : SUCCESS in 1h 24m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c55402d297624df0a590dafd4e062ffb : FAILURE in 2m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a9268b80f07f4195b5fe063e4e5ffd2e : SUCCESS in 2m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7fb14e493a374618836426a76b7d33d1 : SUCCESS in 2m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fad34e1285d84947b682f44422f8d7a9 : SUCCESS in 3m 22s\n- cliff-tox-py38-neutronclient-tip https://zuul.opendev.org/t/openstack/build/a9350d98e78b49fc9eae61ac19b238ba : SUCCESS in 2m 47s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/f78e6f492c634dafae591bd9d33e709e : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":2},{"id":"35d4933153f3d47a84f402adaa184ea6aea7e515","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2022-10-27 08:47:16.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"df7f1dbb01bc0d18008135dbd884139267ac31c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-27 10:14:56.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5350afb6a9ec429787761f1d7c166684\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4e4d839f97bf48f0a545fd495c950455 : SUCCESS in 1h 26m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c89f22957eea4aa799ae5debdefdfcfc : SUCCESS in 2m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f4d47891d1e047aebba0e359c62cbb6c : SUCCESS in 3m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/567ed2e803c246f48a714ce6dda63689 : SUCCESS in 2m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8c1e17f1b05494a9ab34abadc0b8ac6 : SUCCESS in 3m 39s\n- cliff-tox-py38-neutronclient-tip https://zuul.opendev.org/t/openstack/build/c2ae701b5492409f81c51e26f6747393 : SUCCESS in 3m 06s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/bba00679a3304cf4a5ef971d445ac50f : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":3},{"id":"308a70a0b2db3592ea1c9ab9ba30ffab45ee0037","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-11 07:41:12.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"02eb0bf5704c7183e278608b870008018e921541","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 07:41:45.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"16b49e0c858c2027117fa96862d70d588916cbb1","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2023-09-11 07:45:50.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"5c7c075996a7b92b4bc64aad17538b93ca3f3e90","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 08:49:56.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/88da75b98de94ef59d8f2cddf34620ba\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5c364c0115ba460ba4265e1a265d6457 : SUCCESS in 1h 06m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7771a59eba164ecb92257e9eafd92bed : SUCCESS in 2m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5d4a19d6bdba4eff8066cdbc6598a52d : SUCCESS in 3m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0c562083a797410ea1b8dd2e0f206cb2 : SUCCESS in 3m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2bb875ae55e34d05bcd2b97e1a6a6a7d : SUCCESS in 3m 06s\n- cliff-tox-py38-neutronclient-tip https://zuul.opendev.org/t/openstack/build/47283b4543384507a09083fdd6257725 : SUCCESS in 3m 09s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/6459e76ecd05460d946d1c8e2c638b20 : SUCCESS in 5m 24s","accounts_in_message":[],"_revision_number":3},{"id":"ad2fccbb48b6cae51f8efaac457391f56e31ed52","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 08:49:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"2aad4fb70925729c43719daec460002e9b5aa654","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 08:50:57.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/650cb9e045b145fbbb9da1015467cacd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebca1cbd738f41c1bd6151a87de34921 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f2215ba0e7d641d550bcd860d6414319fa37090e","revisions":{"9e8936b74e00374c03e113a7541449da099a6828":{"kind":"REWORK","_number":1,"created":"2022-10-26 13:15:05.000000000","uploader":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"ref":"refs/changes/90/862690/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/90/862690/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/90/862690/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/90/862690/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/90/862690/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/90/862690/1"}}},"commit":{"parents":[{"commit":"56b700afcab95080182687f945c62cc96f650e45","subject":"Removing brackets around tested conditional","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/56b700afcab95080182687f945c62cc96f650e45"}]}],"author":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2022-10-26 11:27:33.000000000","tz":120},"committer":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2022-10-26 11:27:33.000000000","tz":120},"subject":"Clarification of the algorithm used","message":"Clarification of the algorithm used\n\nLink was pointing to a wrong wikipedia page, we are using\nDamerau-Levenshtein metric, which allows for additional \"swap\"\noperation. A subtle but important distinction.\n\nSigned-off-by: Jiri Podivin \u003cjpodivin@redhat.com\u003e\nChange-Id: I6cc2499c371b86aaf6636d7a46416bc947c11bb2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/9e8936b74e00374c03e113a7541449da099a6828"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/9e8936b74e00374c03e113a7541449da099a6828"}]},"branch":"refs/heads/master"},"835013d09c7d35487a4c15e9dc31b6d8baa424db":{"kind":"REWORK","_number":2,"created":"2022-10-27 06:58:16.000000000","uploader":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"ref":"refs/changes/90/862690/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/90/862690/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/90/862690/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/90/862690/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/90/862690/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/90/862690/2"}}},"commit":{"parents":[{"commit":"56b700afcab95080182687f945c62cc96f650e45","subject":"Removing brackets around tested conditional","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/56b700afcab95080182687f945c62cc96f650e45"}]}],"author":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2022-10-26 11:27:33.000000000","tz":120},"committer":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2022-10-27 06:58:03.000000000","tz":120},"subject":"Clarification of the algorithm used","message":"Clarification of the algorithm used\n\nLink was pointing to a wrong wikipedia page, we are using\nDamerau-Levenshtein metric, which allows for additional \"swap\"\noperation. A subtle but important distinction.\n\nSigned-off-by: Jiri Podivin \u003cjpodivin@redhat.com\u003e\nChange-Id: I6cc2499c371b86aaf6636d7a46416bc947c11bb2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/835013d09c7d35487a4c15e9dc31b6d8baa424db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/835013d09c7d35487a4c15e9dc31b6d8baa424db"}]},"branch":"refs/heads/master"},"f2215ba0e7d641d550bcd860d6414319fa37090e":{"kind":"REWORK","_number":3,"created":"2022-10-27 08:47:16.000000000","uploader":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"ref":"refs/changes/90/862690/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/90/862690/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/90/862690/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/90/862690/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/90/862690/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/90/862690/3"}}},"commit":{"parents":[{"commit":"56b700afcab95080182687f945c62cc96f650e45","subject":"Removing brackets around tested conditional","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/56b700afcab95080182687f945c62cc96f650e45"}]}],"author":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2022-10-26 11:27:33.000000000","tz":120},"committer":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2022-10-27 08:47:05.000000000","tz":120},"subject":"Clarification of the algorithm used","message":"Clarification of the algorithm used\n\nLink was pointing to a wrong wikipedia page, we are using\nDamerau-Levenshtein metric, which allows for additional \"swap\"\noperation. A subtle but important distinction.\n\nSigned-off-by: Jiri Podivin \u003cjpodivin@redhat.com\u003e\nChange-Id: I6cc2499c371b86aaf6636d7a46416bc947c11bb2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/f2215ba0e7d641d550bcd860d6414319fa37090e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/f2215ba0e7d641d550bcd860d6414319fa37090e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
