)]}'
{"id":"openstack%2Fcliff~918102","triplet_id":"openstack%2Fcliff~master~I43e601f9752b4de775c1ee4b6210f358f2e904c2","project":"openstack/cliff","branch":"master","topic":"pre-commit","hashtags":[],"change_id":"I43e601f9752b4de775c1ee4b6210f358f2e904c2","subject":"requirements: Bump minimums","status":"MERGED","created":"2024-05-03 16:43:43.000000000","updated":"2024-05-03 19:41:02.000000000","submitted":"2024-05-03 19:39:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"918102-pre-commit","meta_rev_id":"600d9e8630d6e544ce5486770513161c37c710be","_number":918102,"virtual_id_number":918102,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-03 19:39:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2024-05-03 16:44:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2024-05-03 16:44:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-03 18:12:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"de2413272ea6e7c18d46dd2a7de85ae8399a152a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-05-03 16:43:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7e7ba7bd4e3d18f4d9f1f6e960dbee3f776231de","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-05-03 16:44:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"63ab5626a13e57e1637d9e656d12d9c3d199a6dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-03 18:12:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c742399324894862adc571c726274830\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/75fbf363caef46f5bd4713e60a0dd747 : SUCCESS in 1m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e4fdc3e9eeb246edaf4f747e2c3a9afe : SUCCESS in 1h 27m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a54968e9a4314e6c8f9e18976ce9c27a : SUCCESS in 2m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f00ea410e7e0424db7bba3f4b822f665 : SUCCESS in 3m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e0bc4eab6ba042ed85a5537fe9b8de11 : SUCCESS in 3m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/268c392abde54a1ea7353219111b63d8 : FAILURE in 6m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bcab2c91c9d4e8697729ec2783670fa : SUCCESS in 3m 08s\n- cliff-tox-py38-neutronclient-tip https://zuul.opendev.org/t/openstack/build/5a4062dfc5684bd0b2380189c367bf11 : SUCCESS in 3m 21s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/3d87794efae54860a00a0156628ff42f : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":1},{"id":"2b47d57738477958eb68b415c7a6f56e110be631","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-03 18:13:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"194442756ca5b70aa777e03459094fe4443ed444","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-03 19:39:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed4b8205fbd34f78a6c67eabf25d20bd\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/8d861efca2bd4d1b9d08a7cf9c25f075 : SUCCESS in 2m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e9717d337c2144a09143ae9587d86f39 : SUCCESS in 1h 21m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea6c7bc871dd40fc87be02e36ad610b3 : SUCCESS in 4m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cac0d8fbcebf40a5b54035761df4d0bc : SUCCESS in 3m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bbfb265734d94feebe022eb1a633f810 : SUCCESS in 3m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91e67c1e2d3940f1a87287ba3fa34527 : SUCCESS in 3m 29s\n- cliff-tox-py38-neutronclient-tip https://zuul.opendev.org/t/openstack/build/5f2ea72d86c04d088b9c1144e3e4592c : SUCCESS in 4m 13s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/d52a8e778b53423396bf07457481aa6f : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":1},{"id":"46e8c9ffc221b25af1636be364f06ba23a5b3a4e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-03 19:39:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"600d9e8630d6e544ce5486770513161c37c710be","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-03 19:41:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea78fa4118fe4f2dad44fa8eccbe5543\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b720e439b214a4191a6b5c5317a4d50 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bc56deae9141adcb57f68a9c2604c90aed02de50","revisions":{"bc56deae9141adcb57f68a9c2604c90aed02de50":{"kind":"REWORK","_number":1,"created":"2024-05-03 16:43:43.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/02/918102/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/02/918102/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/02/918102/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/02/918102/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/02/918102/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/02/918102/1"}}},"commit":{"parents":[{"commit":"b6119d504deca17e8c092b12fae209d944c6fa73","subject":"Test python 3.10 and 3.11","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/b6119d504deca17e8c092b12fae209d944c6fa73"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2024-05-03 16:42:45.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2024-05-03 16:43:10.000000000","tz":60},"subject":"requirements: Bump minimums","message":"requirements: Bump minimums\n\nMake these a little more realistic\n\nChange-Id: I43e601f9752b4de775c1ee4b6210f358f2e904c2\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/bc56deae9141adcb57f68a9c2604c90aed02de50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/bc56deae9141adcb57f68a9c2604c90aed02de50"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
