)]}'
{"id":"openstack%2Fcliff~918841","triplet_id":"openstack%2Fcliff~master~I30dcc4dbe8f6ec79ed7c110fd5c84eade8300f4e","project":"openstack/cliff","branch":"master","topic":"bump_sphinx","hashtags":[],"change_id":"I30dcc4dbe8f6ec79ed7c110fd5c84eade8300f4e","subject":"[DNM] Testing docs bump with new Sphinx","status":"ABANDONED","created":"2024-05-09 19:23:29.000000000","updated":"2024-05-09 23:08:01.000000000","total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":false,"meta_rev_id":"8b2dd7f4b5ce152dc54f70fa6b5f8a5a247fbadb","_number":918841,"virtual_id_number":918841,"owner":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-05-09 21:01:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2024-05-09 21:01:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ffb1a3da45172e158ea8ec957e4615050fc495f4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-05-09 19:23:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6dbaca52388493716a0328d9dc2e99403186b1c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 21:01:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/27897a505db44f688ae1bfafe69688d7\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a1f65eac7e40478e9198a66eae61d16f : SUCCESS in 2m 15s\n- requirements-check https://zuul.opendev.org/t/openstack/build/9ad1bd9d017d48259ef166ae9bf94f36 : SUCCESS in 1m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7202ba639d3f459e9330978abcb916fe : SUCCESS in 1h 31m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3374c5799f174ea38dff06c1308a2e39 : FAILURE in 3m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3c961163dd364a0f9d7c69a036b67578 : SUCCESS in 3m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cc33d627a6c2432bb75aba6842ac9486 : SUCCESS in 3m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/dcf3c85364204b4191ab5dd172f633a3 : FAILURE in 5m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8613ab22b0ff4d1c813a8304898ec224 : SUCCESS in 3m 49s\n- cliff-tox-py38-neutronclient-tip https://zuul.opendev.org/t/openstack/build/c4ae41f1ed484074b0951a50c5cefae7 : SUCCESS in 3m 06s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/9c2e4ebfad904dd389534fea457a4436 : SUCCESS in 4m 38s","accounts_in_message":[],"_revision_number":1},{"id":"8b2dd7f4b5ce152dc54f70fa6b5f8a5a247fbadb","tag":"autogenerated:gerrit:abandon","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-05-09 23:08:01.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cf37d8d19777cd00a2edd58caca00b0e651276a0","revisions":{"cf37d8d19777cd00a2edd58caca00b0e651276a0":{"kind":"REWORK","_number":1,"created":"2024-05-09 19:23:29.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/41/918841/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/41/918841/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/41/918841/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/41/918841/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/41/918841/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/41/918841/1"}}},"commit":{"parents":[{"commit":"b75afa0ecad51c7eb150b2b0c93aec1c2a159052","subject":"Merge \"Add fixtures explicitly in test requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/b75afa0ecad51c7eb150b2b0c93aec1c2a159052"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-05-09 19:23:27.000000000","tz":-300},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-05-09 19:23:27.000000000","tz":-300},"subject":"[DNM] Testing docs bump with new Sphinx","message":"[DNM] Testing docs bump with new Sphinx\n\nSphinx updates can be scary.  This change leverages a constraints bump to see what breaks.\n\nDepends-On: https://review.opendev.org/c/openstack/requirements/+/916565\nChange-Id: I30dcc4dbe8f6ec79ed7c110fd5c84eade8300f4e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/cf37d8d19777cd00a2edd58caca00b0e651276a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/cf37d8d19777cd00a2edd58caca00b0e651276a0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
