)]}'
{"id":"openstack%2Fcliff~926294","triplet_id":"openstack%2Fcliff~master~I56ce1eb843c5af74cec420e0b2f4916f58152146","project":"openstack/cliff","branch":"master","topic":"release-health-check-cwi-dalmatian","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2024-09-12 13:45:21.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I56ce1eb843c5af74cec420e0b2f4916f58152146","subject":"DNM: gate health test","status":"ABANDONED","created":"2024-08-14 15:57:06.000000000","updated":"2024-09-12 13:45:21.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"89b38f0695c9d427822ff3956d0db9779e60379f","_number":926294,"virtual_id_number":926294,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-08-14 17:31:52.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-14 17:31:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-15 05:55:46.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2024-08-19 11:28:55.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"}],"messages":[{"id":"3f02ac34713f77796508629194a8704b3c63f7b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-08-14 15:57:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3c5380b3e3308c5752025001c8106cd57fa561f2","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-08-14 15:57:31.000000000","message":"Topic release-health-check-cwi-caracal removed","accounts_in_message":[],"_revision_number":1},{"id":"bacb190e5c4688977df50f644bf95d123d4b7ebc","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-08-14 15:57:40.000000000","message":"Topic set to release-health-check-cwi-dalmatian","accounts_in_message":[],"_revision_number":1},{"id":"0edc74d04099ec6a2bcec4ccc2abf3a9b50e3d30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-14 17:31:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/91735cd0faa5492898a67c4ed40f7ab0\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/da19580def2140649e11ee819f0a1692 : SUCCESS in 1h 34m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab2bfabf899f41e2a60057106022e399 : FAILURE in 2m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aa16e338b8dc40bfa41bdffeccab2adb : SUCCESS in 2m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6df7248c519d4139b0dde9a528fb6138 : SUCCESS in 3m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/68f9646f8f6d4aa49c25f6dad31748f6 : FAILURE in 6m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e01546e7307d4981a6a0f901ca96ee5c : SUCCESS in 3m 12s\n- cliff-tox-py38-neutronclient-tip https://zuul.opendev.org/t/openstack/build/f21ae6024f4d443c854bd9cb2ea93f53 : SUCCESS in 3m 49s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/c7d5c5bda90145cbbba7ae1a23908ce5 : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":1},{"id":"7868eb3953b3d9134f5ea6915b9ea68cb37ea398","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-08-15 05:55:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a10a55cab51aee78f4f089377da2fe5cba0a229b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-08-15 09:11:48.000000000","message":"Abandoned\n\nGate looks healthy.","accounts_in_message":[],"_revision_number":1},{"id":"5067effdec4951b68087e58b1f053abbed5eb3d4","tag":"autogenerated:gerrit:restore","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-08-19 11:28:27.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"bacbcb23d453a64bf16b07d70ef788abfadb243c","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-08-19 11:28:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8de2f5b9fa676f563c6989fb506f39896f801e58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-19 13:11:17.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3740064c95174f6391b2f5ef9bc3b75a\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/235ed068f3704f8994392c187266f0fe : SUCCESS in 1h 37m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14f39e3fcd34484fbc26b2be8e5b9f84 : FAILURE in 2m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d3cda714e5f7400eb383640db5342089 : SUCCESS in 2m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/97823019f206428eb91251402965477f : SUCCESS in 2m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/11d1571d316e4409bb27803fd1b9c1d5 : SUCCESS in 3m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9182f35ff6e4a01b9fe4dd2cb56c398 : SUCCESS in 2m 54s\n- cliff-tox-py38-neutronclient-tip https://zuul.opendev.org/t/openstack/build/433cd9925b0e4268bcf7b6e23fc1646c : SUCCESS in 4m 34s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/21a014d9ecfa45918fc3de24c1fb81df : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":1},{"id":"89b38f0695c9d427822ff3956d0db9779e60379f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-09-12 13:45:21.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ec7c8d9397c5695ce6cb38dcd702792b237ca934","revisions":{"ec7c8d9397c5695ce6cb38dcd702792b237ca934":{"kind":"REWORK","_number":1,"created":"2024-08-14 15:57:06.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/94/926294/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/94/926294/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/94/926294/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/94/926294/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/94/926294/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/94/926294/1"}}},"commit":{"parents":[{"commit":"b75afa0ecad51c7eb150b2b0c93aec1c2a159052","subject":"Merge \"Add fixtures explicitly in test requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/b75afa0ecad51c7eb150b2b0c93aec1c2a159052"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2024-08-14 15:56:53.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2024-08-14 15:57:04.000000000","tz":120},"subject":"DNM: gate health test","message":"DNM: gate health test\n\nChange-Id: I56ce1eb843c5af74cec420e0b2f4916f58152146\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/ec7c8d9397c5695ce6cb38dcd702792b237ca934"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/ec7c8d9397c5695ce6cb38dcd702792b237ca934"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
