)]}'
{"id":"openstack%2Fcliff~929157","triplet_id":"openstack%2Fcliff~master~I74d19aaef5cab04d84d95d10d29ada4d454263d0","project":"openstack/cliff","branch":"master","topic":"ruff","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2024-09-17 17:20:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I74d19aaef5cab04d84d95d10d29ada4d454263d0","subject":"Migrate to ruff, ruff-format","status":"MERGED","created":"2024-09-12 16:40:59.000000000","updated":"2024-09-17 17:21:48.000000000","submitted":"2024-09-17 17:20:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"929157-ruff","meta_rev_id":"a9e9ec59f412e7a45d559fbf983ab5428370c2e9","_number":929157,"virtual_id_number":929157,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-17 17:20:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2024-09-17 15:33:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2024-09-17 17:12:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2024-09-17 15:33:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-12 18:13:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-17 17:12:23.000000000","updated_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"}],"messages":[{"id":"e1b1066b4bc8e5dfce930618685f5e1f164d236d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-09-12 16:40:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c1f066a78ec0a12d9deff8ed909ff882a593408f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-12 18:13:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/edd3faaeef634999bbac26cdfb5d829a\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a8b1c646d46d4d95ae9c6974c74c7bfe : FAILURE in 1h 30m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b67fa0da42b4cf384d58b07c46068f6 : SUCCESS in 2m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/05d2505a479a431eacf288430e0f8e29 : SUCCESS in 3m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/14700c51955a4a12a709ced2f791e815 : SUCCESS in 2m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/72a22d74d6a347d0afc6fe4ce26609e0 : SUCCESS in 3m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c475e5315308448bab122802242da84a : SUCCESS in 3m 56s\n- cliff-tox-py38-neutronclient-tip https://zuul.opendev.org/t/openstack/build/c3183086f4ab4f4caabe192cd25442ce : SUCCESS in 2m 33s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/8c0ad1a4081a4cae8cf5991daa1b2bbb : FAILURE in 2m 12s","accounts_in_message":[],"_revision_number":1},{"id":"a203683b359a5846d38f30ad3728691efcef9c5f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-09-13 08:49:28.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"9cfcccf12e0a11d635600a244efbdb264f4f935f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-13 10:08:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e0c8cdcadab44267883619c0ff4d50b4\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1cc0a8d962cc425caf30b86d3430961b : SUCCESS in 1h 17m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc52b2ce02ac4cc1bc9ba1cda373f747 : SUCCESS in 2m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8e564585d6d34f5eba4faf4a8f4a761f : SUCCESS in 3m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/52a3bf3c273140698d527835ac85c2f5 : SUCCESS in 3m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cd2d9f8c4c1b4677bfce287d3391f4d3 : SUCCESS in 3m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e483163a632e4a3ab690f40e817920e6 : SUCCESS in 2m 57s\n- cliff-tox-py312-neutronclient-tip https://zuul.opendev.org/t/openstack/build/4377feac1e3746188a33544d9eabffd2 : SUCCESS in 3m 42s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/ef85e5b9ec1a4c6798a9e26c7d29b0dc : SUCCESS in 4m 39s","accounts_in_message":[],"_revision_number":2},{"id":"3079263a8b7e3d33689276711c704100ab1f9c5c","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-09-17 15:33:09.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"951c46d67d0dbcbac27fd4e57518a30931ea557c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-17 15:33:44.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"87ee402d68acc7b69bd0c773407d7505bb7abb66","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-09-17 17:12:23.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"91b6320fd8f8cd4da29118f80a8c55c43df66661","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-17 17:20:48.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27d0aca85c5d4c739beb779150dbb057\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6e6217f27eab46f0bda9af9264037487 : SUCCESS in 1h 41m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af85717dfd8a4e28a556b1ffc77b5913 : SUCCESS in 3m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a4eb5861644942e6919a9d909bcb56a0 : SUCCESS in 2m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/db59d913a53e4a8496f8ae76c920a213 : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de4d2794601648569a8d7a02ce3040eb : SUCCESS in 2m 36s\n- cliff-tox-py312-neutronclient-tip https://zuul.opendev.org/t/openstack/build/0b7f8acea4994a8cb917a8ae638acdf5 : SUCCESS in 3m 44s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/7ffbe2b3b14e4ceca9d9f533c1957e95 : SUCCESS in 7m 20s","accounts_in_message":[],"_revision_number":2},{"id":"d897e54f0fb34b9953fdfd5d37ec9c156902044c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-17 17:20:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"a9e9ec59f412e7a45d559fbf983ab5428370c2e9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-17 17:21:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b619836b7aab4a11bfaa1fd6a0a57244\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa02df9bd3534a5d8f35f2e0d85fc3de : SUCCESS in 40s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b35e895a6233de2e584580fcd7b9fb9ecc5c2de4","revisions":{"66c4d34828a53aa01136127c0ecd5d6bb723b772":{"kind":"REWORK","_number":1,"created":"2024-09-12 16:40:59.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/57/929157/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/57/929157/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/57/929157/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/57/929157/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/57/929157/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/57/929157/1"}}},"commit":{"parents":[{"commit":"3757bb323f546ad93324226cc62c25bd84590385","subject":"pre-commit: Bump versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/3757bb323f546ad93324226cc62c25bd84590385"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2024-09-12 16:30:18.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2024-09-12 16:37:53.000000000","tz":60},"subject":"Migrate to ruff, ruff-format","message":"Migrate to ruff, ruff-format\n\nDrop black, flake8 and bandit.\n\nChange-Id: I74d19aaef5cab04d84d95d10d29ada4d454263d0\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/66c4d34828a53aa01136127c0ecd5d6bb723b772"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/66c4d34828a53aa01136127c0ecd5d6bb723b772"}]},"branch":"refs/heads/master"},"b35e895a6233de2e584580fcd7b9fb9ecc5c2de4":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-09-13 08:49:28.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/57/929157/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/57/929157/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/57/929157/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/57/929157/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/57/929157/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/57/929157/2"}}},"commit":{"parents":[{"commit":"ea4f13df7b92883098913b5ff37a297ae6c80626","subject":"Drop support for Python 3.8, add Python 3.12","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/ea4f13df7b92883098913b5ff37a297ae6c80626"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2024-09-12 16:30:18.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2024-09-13 08:49:28.000000000","tz":0},"subject":"Migrate to ruff, ruff-format","message":"Migrate to ruff, ruff-format\n\nDrop black, flake8 and bandit.\n\nChange-Id: I74d19aaef5cab04d84d95d10d29ada4d454263d0\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/b35e895a6233de2e584580fcd7b9fb9ecc5c2de4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/b35e895a6233de2e584580fcd7b9fb9ecc5c2de4"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
