)]}'
{"id":"openstack%2Fcliff~929158","triplet_id":"openstack%2Fcliff~master~I6cd2d5e1a3fe79eb41199059cb98c5c535c6154b","project":"openstack/cliff","branch":"master","topic":"ruff","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2024-09-17 17:20:50.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6cd2d5e1a3fe79eb41199059cb98c5c535c6154b","subject":"ruff: Enable pyupgrade rules","status":"MERGED","created":"2024-09-12 16:40:59.000000000","updated":"2024-09-17 17:22:33.000000000","submitted":"2024-09-17 17:20:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"929158-ruff","meta_rev_id":"a4c5f0a1b8a89d0f25d605f74d2553fac05aba38","_number":929158,"virtual_id_number":929158,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-17 17:20:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2024-09-17 15:33:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2024-09-17 17:12:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2024-09-17 15:33:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-12 18:21:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-17 17:12:49.000000000","updated_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"}],"messages":[{"id":"7758245e0bb8e9e2aa124ac0a41a1f0e75d20de8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-09-12 16:40:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dcc0e8f3c150b33b477bcb48de24a05f44249d6e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-12 18:21:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4f94c4905dce4df9aefc0f16ab6d6b02\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0d6e2d8fb33f425a8fabee751745d0e4 : SUCCESS in 1h 38m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cdc622715ff84909918a5f043919c963 : SUCCESS in 2m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0b74fdacf81d4b33b6563360ac64e1a4 : SUCCESS in 3m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/faf0743211544e6e908297cc3eaf43de : SUCCESS in 4m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/75049929570e484b9c24747eb90f5f08 : SUCCESS in 2m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1248f3fa5d0b4556a78ca83e6ef2a0a2 : SUCCESS in 3m 42s\n- cliff-tox-py38-neutronclient-tip https://zuul.opendev.org/t/openstack/build/6b1fb6b6dc4f4b568aa73555394565be : SUCCESS in 3m 22s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/711c2f6f5e224836a5cb9cd8f2492a38 : FAILURE in 2m 44s","accounts_in_message":[],"_revision_number":1},{"id":"ed0ae5b50f4e666d9afe9db41479f2057856046d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-09-13 08:49:33.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"65d9a1b8c379fd9790e88762d65191dac7d42651","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-13 09:57:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f2e34dff38443619a9cf1a058b5cdfd\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/45aea994caa646a587d43454914e7206 : SUCCESS in 1h 04m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/562a6b4e6c614ab1b3efc3cb4594596a : SUCCESS in 2m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/141c1307781946deb15a59a496afd63e : SUCCESS in 2m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/58d166602543403fb0b5b21f9c42f229 : SUCCESS in 2m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/147ebb856f3b493ca6b15999fc6d81ca : SUCCESS in 3m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/901f15cf3968442caf020c1ce3d749dc : SUCCESS in 2m 53s\n- cliff-tox-py312-neutronclient-tip https://zuul.opendev.org/t/openstack/build/8bffe4107aa249128e7887b9f7c90186 : SUCCESS in 3m 55s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/f5a88dd3e437446d83407aec371b87da : SUCCESS in 4m 36s","accounts_in_message":[],"_revision_number":2},{"id":"ae2603bdef00b8cdb511fc8ea0cd190d2abd0aa5","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-09-17 15:33:14.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"bb3fd63797adc9be25b35d47cbc1441a151e6737","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-17 15:33:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"82514d77b891debd64056d6ef07eeb10525bbf2a","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-09-17 17:12:49.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"3bf6ef8f6d29f7089c61db67c70b4e60a1725bda","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-17 17:20:50.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/365a47be4d1b4b92a89e3420f108be60\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/965affaeeae24d99b5ba5c8cc525f425 : SUCCESS in 1h 36m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b5b8908b8824a22971a40d4f33b6669 : SUCCESS in 2m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4886515e8463433fb3bec4edc14e8d53 : SUCCESS in 4m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5f4071b8e671469ca6c5af1c25d50d98 : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/070b773d309043e7bd553eaba1df28ae : SUCCESS in 2m 41s\n- cliff-tox-py312-neutronclient-tip https://zuul.opendev.org/t/openstack/build/1c55f4b81e2e432280a3a670fb5675aa : SUCCESS in 4m 11s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/5240e3aa80a14f19b4242882f3990290 : SUCCESS in 3m 52s","accounts_in_message":[],"_revision_number":2},{"id":"aeb8e1df15d5e340029f50bab27c62e28a7bff2a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-17 17:20:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"a4c5f0a1b8a89d0f25d605f74d2553fac05aba38","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-17 17:22:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3fae1b77242543d49cc9233cd0694e83\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74166023d6204a8d8f0d2cf4149427aa : SUCCESS in 37s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f812e262c3d0b7e7d5fdbc2b2850f1a8a4c59507","revisions":{"35db6531f8345ae2d8443fa7b45aef8a834a8e56":{"kind":"REWORK","_number":1,"created":"2024-09-12 16:40:59.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/58/929158/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/58/929158/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/58/929158/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/58/929158/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/58/929158/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/58/929158/1"}}},"commit":{"parents":[{"commit":"66c4d34828a53aa01136127c0ecd5d6bb723b772","subject":"Migrate to ruff, ruff-format","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/66c4d34828a53aa01136127c0ecd5d6bb723b772"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2024-09-12 16:38:39.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2024-09-12 16:39:43.000000000","tz":60},"subject":"ruff: Enable pyupgrade rules","message":"ruff: Enable pyupgrade rules\n\nChange-Id: I6cd2d5e1a3fe79eb41199059cb98c5c535c6154b\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/35db6531f8345ae2d8443fa7b45aef8a834a8e56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/35db6531f8345ae2d8443fa7b45aef8a834a8e56"}]},"branch":"refs/heads/master"},"f812e262c3d0b7e7d5fdbc2b2850f1a8a4c59507":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-09-13 08:49:33.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/58/929158/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/58/929158/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/58/929158/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/58/929158/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/58/929158/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/58/929158/2"}}},"commit":{"parents":[{"commit":"b35e895a6233de2e584580fcd7b9fb9ecc5c2de4","subject":"Migrate to ruff, ruff-format","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/b35e895a6233de2e584580fcd7b9fb9ecc5c2de4"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2024-09-12 16:38:39.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2024-09-13 08:49:33.000000000","tz":0},"subject":"ruff: Enable pyupgrade rules","message":"ruff: Enable pyupgrade rules\n\nChange-Id: I6cd2d5e1a3fe79eb41199059cb98c5c535c6154b\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/f812e262c3d0b7e7d5fdbc2b2850f1a8a4c59507"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/f812e262c3d0b7e7d5fdbc2b2850f1a8a4c59507"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
