)]}'
{"id":"openstack%2Fcliff~942241","triplet_id":"openstack%2Fcliff~master~If93c7c4439d2bb8c05dd18af75bb4fbd461ed802","project":"openstack/cliff","branch":"master","topic":"typing","hashtags":[],"change_id":"If93c7c4439d2bb8c05dd18af75bb4fbd461ed802","subject":"pre-commit: Enable mypy","status":"MERGED","created":"2025-02-19 20:12:57.000000000","updated":"2025-02-20 12:06:20.000000000","submitted":"2025-02-20 12:03:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"942241-typing","meta_rev_id":"621024105aa3a8715dad4113431d24f114b7d1d9","_number":942241,"virtual_id_number":942241,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-20 12:03:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2025-02-20 10:28:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2025-02-20 10:28:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-19 21:48:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"620e8c8d3fe3d7ca48099f6dcf64e1ca17d5a96c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-02-19 20:12:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e884b5f837c4fc259b6d673533d7968e39e03982","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-19 21:48:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b83018d2ca9548638e44cb087d49d910\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b3993a15ac2d4ff1ae286e607e9cd52e : SUCCESS in 1h 31m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0963501bf85b42379a93b05875a42a37 : SUCCESS in 3m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/995f60d676d14098ac9e3befa027bc84 : SUCCESS in 3m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/72d33c567d794403bf9c08c63fb38359 : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9dfea813d059430293aaefb8285d969c : SUCCESS in 3m 30s\n- cliff-tox-py312-neutronclient-tip https://zuul.opendev.org/t/openstack/build/c8c365183ac242a68b7f4f231aabdb10 : SUCCESS in 2m 40s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/636ef9fed69343d7a382e42b9469d28f : SUCCESS in 6m 26s","accounts_in_message":[],"_revision_number":1},{"id":"baf3d2ba7f7d06ba929b7a5957efa7148441eb37","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-02-20 10:28:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"670c7316021638292353904350f0e677f616cb40","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 10:28:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a3b42b3a321f56af091731bf40db2f2d9751e909","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 12:03:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20b76fb3a2e24a27a9b6602cc877fb42\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0edaea58f35f4ed3b48ef1fabcc4f2bd : SUCCESS in 1h 30m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/160b47bcc9384320973c5ba1c1082ba8 : SUCCESS in 2m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/449e95dea07f425e92cff9d01f9de47c : SUCCESS in 4m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b337b6bc5bf3484583b3e09953dd0ad9 : SUCCESS in 3m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fdd38319fff434c8631048384b39e6e : SUCCESS in 2m 59s\n- cliff-tox-py312-neutronclient-tip https://zuul.opendev.org/t/openstack/build/98b8aa05c2f346a4975af2d332a33198 : SUCCESS in 3m 46s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/5a5c029f2642482fb4078f8c74e614ba : SUCCESS in 5m 21s","accounts_in_message":[],"_revision_number":1},{"id":"00042ad1fc65b5315093df145d183abf8c31dfd1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 12:03:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"621024105aa3a8715dad4113431d24f114b7d1d9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 12:06:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/764737279055499691c81e1783ca3aa7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2dfa3c7f35134b1fb83fd8c200db7c83 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"27e08688262de529cc3dc551ba9a65447357421c","revisions":{"27e08688262de529cc3dc551ba9a65447357421c":{"kind":"REWORK","_number":1,"created":"2025-02-19 20:12:57.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/41/942241/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/41/942241/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/41/942241/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/41/942241/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/41/942241/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/41/942241/1"}}},"commit":{"parents":[{"commit":"9bb5c6ec2de1f699e9e22fcdcd4fc520db0784da","subject":"typing: Fix initial typing issues","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/9bb5c6ec2de1f699e9e22fcdcd4fc520db0784da"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-02-19 20:08:23.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-02-19 20:08:23.000000000","tz":0},"subject":"pre-commit: Enable mypy","message":"pre-commit: Enable mypy\n\nChange-Id: If93c7c4439d2bb8c05dd18af75bb4fbd461ed802\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/27e08688262de529cc3dc551ba9a65447357421c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/27e08688262de529cc3dc551ba9a65447357421c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
