)]}'
{"id":"openstack%2Fcliff~942247","triplet_id":"openstack%2Fcliff~master~I7b889a775f5b61b28b4b22c866b7301393206e7d","project":"openstack/cliff","branch":"master","topic":"typing","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-02-20 16:19:51.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"hashtags":[],"change_id":"I7b889a775f5b61b28b4b22c866b7301393206e7d","subject":"typing: Add typing to cliff.command","status":"MERGED","created":"2025-02-19 20:12:57.000000000","updated":"2025-02-20 19:16:22.000000000","submitted":"2025-02-20 19:15:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"942247-typing","meta_rev_id":"a8eaf08fb6cf9a98b34c70d49307c03510ba0585","_number":942247,"virtual_id_number":942247,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-20 19:15:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2025-02-20 10:28:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2025-02-20 10:28:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-19 22:10:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"babc8b8d649564da82ace2d636ff5660bf9e049b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-02-19 20:12:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d74fe4f930a55432fed8f42021620bf4b573f10f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-19 22:10:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00826cd1472848dbb25c02fcd744613a\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2937d2a15b444854a8c2296c04ad52ed : SUCCESS in 1h 53m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b65e5d64ccb5414f8a68bebfa28204c5 : SUCCESS in 3m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f0421d4814a84e35b78d8f6f9fa19acf : SUCCESS in 3m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c4a71cb996d44d7dad310a15e78ceb16 : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/884ed9ca269e49fb96fd6a2beea799d5 : SUCCESS in 3m 16s\n- cliff-tox-py312-neutronclient-tip https://zuul.opendev.org/t/openstack/build/c29f9f9d5e26451890906241956a8592 : SUCCESS in 3m 15s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/91640cd72ed84b91a9c2044ef2abd20e : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":1},{"id":"9322a274b2728f8aa845277b09c8c4673f0e7118","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-02-20 10:28:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"daabdb28c3284bc6c3aa8f134e9cbf799036c5b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 10:29:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5bb6c0e5f1002e55b2f6b9065dc3c061e9150a9b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 13:03:49.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c39969494221473b97831344fba6b582\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/052ddbd09deb42d3be3fe88216293433 : FAILURE in 1h 04m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92c90582a5a84d938f92ee48363e483a : SUCCESS in 4m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3db161b9e47f4e6097ac59ea7d4305e4 : SUCCESS in 4m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e5fe356d7d3444aa8b5c25897f4d0e20 : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2b10cec2fad47dbafad0adea518ef5e : SUCCESS in 3m 30s\n- cliff-tox-py312-neutronclient-tip https://zuul.opendev.org/t/openstack/build/da4d462b24f346b2b5ec996b35b13edd : SUCCESS in 5m 06s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/4a4e6798cc4443fcaaf7b6e0b1d46b0e : SUCCESS in 7m 57s","accounts_in_message":[],"_revision_number":1},{"id":"1a6ebebcaee51a0d5f14bfbf6c2fe300589ca539","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-02-20 16:19:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ea51ab543a9b8f9b8a531a448e11c333c3d8a194","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 18:09:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/59dbe9eade41416299ed4f36a1fee39a\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/931e6d8052c849b0a844d1c2499536c9 : SUCCESS in 1h 47m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1c18ece7d0542cfa5bdcfff0a329210 : SUCCESS in 3m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0c6e13773b724d3183cef2440b4c5c86 : SUCCESS in 4m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0fd56ce392c146fbacf89e91b5df82f9 : SUCCESS in 3m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3763f4b133d040a9bab08f917181b9d6 : SUCCESS in 3m 09s\n- cliff-tox-py312-neutronclient-tip https://zuul.opendev.org/t/openstack/build/3c563e6b6be34b788ebd309248a88ffd : SUCCESS in 3m 26s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/89b5d506fc0e419487b372ce98a6e84c : SUCCESS in 5m 16s","accounts_in_message":[],"_revision_number":1},{"id":"cd27ac129aa261aa966407ad821ea339b64faa21","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 18:10:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"768823f52bc14e5eb4ecb623dc199f681c1f152a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 19:15:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a54b7c349684b38b8843acde2de32b8\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d3c2461260014a4cb098bd6bf3d24b8d : SUCCESS in 1h 04m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/923806ded2fe42d386d61e3bb8099025 : SUCCESS in 3m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/135ace9102874f578f7dc4fd62757f0a : SUCCESS in 5m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/88ff29801b5b4c9490380d85e2b17c65 : SUCCESS in 3m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a483400d1e20417f8c995ee22e4439ed : SUCCESS in 2m 40s\n- cliff-tox-py312-neutronclient-tip https://zuul.opendev.org/t/openstack/build/817879c9d389403ca923882b12f25199 : SUCCESS in 3m 19s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/8b9769ff80f248ee9322a45bfeef0930 : SUCCESS in 7m 13s","accounts_in_message":[],"_revision_number":1},{"id":"6a347adc4fb18ce054db1b79a5378918d33f0b49","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 19:15:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a8eaf08fb6cf9a98b34c70d49307c03510ba0585","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 19:16:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45bf8a11a43f483c93929aa7ed750b9f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f152882b7e34612801e6f90cd32d918 : SUCCESS in 34s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8fe0f2ced5ebd06f1bed1bb1facb36971b6db550","revisions":{"8fe0f2ced5ebd06f1bed1bb1facb36971b6db550":{"kind":"REWORK","_number":1,"created":"2025-02-19 20:12:57.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/47/942247/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/47/942247/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/47/942247/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/47/942247/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/47/942247/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/47/942247/1"}}},"commit":{"parents":[{"commit":"3b562b294ecf5addd9dccd7bd17fb8521ade3491","subject":"typing: Add typing to cliff.commandmanager","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/3b562b294ecf5addd9dccd7bd17fb8521ade3491"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-02-19 20:08:23.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-02-19 20:08:23.000000000","tz":0},"subject":"typing: Add typing to cliff.command","message":"typing: Add typing to cliff.command\n\nWe also type all the in-tree commands, since it tests our signatures for\nthe base classes.\n\nThe only potentially controversial change we have here is that we no\nlonger set \u0027self.app_args.cmd\u0027 in \u0027HelpCommand\u0027. I\u0027m not entirely sure\nwhy we were doing this: it seems to have been introduced as part of the\nfix for bug #1421585 [1], but removing it causes no harm and I don\u0027t see\nit doing anything useful.\n\n[1] https://bugs.launchpad.net/python-cliff/+bug/1421585\n\nChange-Id: I7b889a775f5b61b28b4b22c866b7301393206e7d\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/8fe0f2ced5ebd06f1bed1bb1facb36971b6db550"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/8fe0f2ced5ebd06f1bed1bb1facb36971b6db550"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
