)]}'
{"id":"openstack%2Fcliff~942248","triplet_id":"openstack%2Fcliff~master~I3cf1170a44b614f5bd36c607bf121ac5378bb462","project":"openstack/cliff","branch":"master","topic":"typing","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-02-20 16:19:55.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"hashtags":[],"change_id":"I3cf1170a44b614f5bd36c607bf121ac5378bb462","subject":"typing: Add typing to cliff.complete","status":"MERGED","created":"2025-02-19 20:12:57.000000000","updated":"2025-02-20 21:11:59.000000000","submitted":"2025-02-20 21:10:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"942248-typing","meta_rev_id":"7e3551835895b79c957b0f17bfb2f45985b37cce","_number":942248,"virtual_id_number":942248,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-20 21:10:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2025-02-20 10:28:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2025-02-20 10:28:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-19 22:00:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0063d41a477ae132adba509887287706c40c1db5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-02-19 20:12:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"358288d9934af55a1ae29c279e1d20b6096ce637","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-19 22:00:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1d442e377ae4239964d3747982a6dc6\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f42fa1520bac44aba45e9cf0804bb33b : SUCCESS in 1h 40m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00d1e1762afc44b3be2bdd81da1515c5 : SUCCESS in 4m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cc215f85c26e4e26b9b97bc2be0996f7 : SUCCESS in 5m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1f87940ea8a747c1acecdc6b4bbd37aa : SUCCESS in 3m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63d493d9821e4e1d839de49678c9ab81 : SUCCESS in 3m 59s\n- cliff-tox-py312-neutronclient-tip https://zuul.opendev.org/t/openstack/build/af0c100a274c40ad8c081f6175b5be7d : SUCCESS in 4m 08s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/943c91cb95504a0da83b5240e337b23c : SUCCESS in 5m 25s","accounts_in_message":[],"_revision_number":1},{"id":"9f2fcf7e8f68b6bae1cfd55a023dd6f2aa50db8b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-02-20 10:28:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5a4b5ed3741008b2477b344cf82a0f406491ad3c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 10:29:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9acc0730c3e1545b9581988fa8eb1e9d6b08a093","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 13:03:50.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/942247 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"b5ce60ce296c9ad8bb298a059f66ee274851ed5d","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-02-20 16:19:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3830bb50c9e38a11f140b2df3c13b0d57af0da75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 18:28:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a9508b7d37e410eb88bff3bcc6f4399\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a2af12ff5dde4b58ab884770f8df1c5e : SUCCESS in 1h 52m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/666a612562a2409e89cf157304fd27d6 : SUCCESS in 2m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9919ff835230481dbbed1fd51d1eed11 : SUCCESS in 3m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/97fc64d94a1b430d8e71829df46400fe : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b497337097f45e79332a7b700a79668 : SUCCESS in 3m 03s\n- cliff-tox-py312-neutronclient-tip https://zuul.opendev.org/t/openstack/build/4622accff5a04407b8ab0832acb07abc : SUCCESS in 3m 21s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/0badac837bab4753971d77a5ed51d155 : SUCCESS in 4m 55s","accounts_in_message":[],"_revision_number":1},{"id":"be8fdc0e2d37acbc8ec9f0129e139f293e6469f6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 18:28:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ae42ce8124d7c3d375fed40b78a257ff85c456d9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 21:10:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/624663c50c124d6d99cbcc89560ddd91\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/72da331b4a6c409595e244ff4535e17b : SUCCESS in 2h 14m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb68c66302bb4bbda0abea384525ebc0 : SUCCESS in 3m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ee495b21e9cc432db0f54278cef8d277 : SUCCESS in 3m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f98017d07bff4224a2937bf986a8e9d1 : SUCCESS in 3m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b009e7cd3161486fa9580081ca955fe8 : SUCCESS in 2m 51s\n- cliff-tox-py312-neutronclient-tip https://zuul.opendev.org/t/openstack/build/30aa42ddd3be482ca5c7201213f75cfb : SUCCESS in 4m 32s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/6f9ba831c9964c45bad91c37b45a0c26 : SUCCESS in 4m 55s","accounts_in_message":[],"_revision_number":1},{"id":"72bafb4d419239c3456a7b2cff85d1e884c80d98","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 21:10:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7e3551835895b79c957b0f17bfb2f45985b37cce","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 21:11:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/157d586955244899990fcf5da5cca3e9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/560752026e9d4682b66aa7602690a40a : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8ade2ee41f0686cfdc0d4ed2c218fee7df0494b1","revisions":{"8ade2ee41f0686cfdc0d4ed2c218fee7df0494b1":{"kind":"REWORK","_number":1,"created":"2025-02-19 20:12:57.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/48/942248/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/48/942248/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/48/942248/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/48/942248/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/48/942248/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/48/942248/1"}}},"commit":{"parents":[{"commit":"8fe0f2ced5ebd06f1bed1bb1facb36971b6db550","subject":"typing: Add typing to cliff.command","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/8fe0f2ced5ebd06f1bed1bb1facb36971b6db550"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-02-19 20:08:23.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-02-19 20:08:23.000000000","tz":0},"subject":"typing: Add typing to cliff.complete","message":"typing: Add typing to cliff.complete\n\nOr at least, what is left of it.\n\nChange-Id: I3cf1170a44b614f5bd36c607bf121ac5378bb462\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/8ade2ee41f0686cfdc0d4ed2c218fee7df0494b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/8ade2ee41f0686cfdc0d4ed2c218fee7df0494b1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
