)]}'
{"id":"openstack%2Fcliff~942249","triplet_id":"openstack%2Fcliff~master~I9ccfcd8fde009e930d97f9c7ae7dde5067e93fe2","project":"openstack/cliff","branch":"master","topic":"typing","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-02-20 16:19:59.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"hashtags":[],"change_id":"I9ccfcd8fde009e930d97f9c7ae7dde5067e93fe2","subject":"typing: Add typing to cliff.help","status":"MERGED","created":"2025-02-19 20:12:57.000000000","updated":"2025-02-20 21:10:53.000000000","submitted":"2025-02-20 21:10:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"942249-typing","meta_rev_id":"66c5c782f388fb581e0f7cfa2cbb94d1d0368895","_number":942249,"virtual_id_number":942249,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-20 21:10:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2025-02-20 10:28:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2025-02-20 10:28:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-19 21:58:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cca6ceee428d44551474042be0c81b4de430e88a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-02-19 20:12:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ce5af94823ed30cb11491a32fa500246c48aa653","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-19 21:58:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/185afd27948648e8be02b3eea7eb04bc\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d3bd03c5f7ef43f692bc9601f93d3ccb : SUCCESS in 1h 38m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c69a092dedb04033b908a13bfbef6186 : SUCCESS in 4m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d89d2dfe6f7546de92e10864dbdfbba6 : SUCCESS in 3m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/27bfe9078e824d73957dde2e1d6d6c3a : SUCCESS in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0ea8bb1ae534794ba54557ddb32a221 : SUCCESS in 3m 05s\n- cliff-tox-py312-neutronclient-tip https://zuul.opendev.org/t/openstack/build/409bbff23475413d83598b67084a419e : SUCCESS in 6m 17s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/cf04486d841a49db8a6b74fd3db51031 : SUCCESS in 7m 10s","accounts_in_message":[],"_revision_number":1},{"id":"a3ad3708ab0a8d6c764fd94156fc5bbe93c648d7","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-02-20 10:28:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0368c942c8157805c8269f34610a44b6916cc755","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 10:29:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d1a7b2ba24f4211ddda9f48597861f7fb2fdfa40","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 13:03:51.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/942248 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"f3ebd029965723360667d9d84cf6f4b56464952b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-02-20 16:19:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"29b556784d119535fdfef8a475b6c9b9905e500e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 17:56:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b1e89099e4e4cbab68455b7dffb688f\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d84bcb700caf4e89960570643abddd06 : SUCCESS in 1h 23m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2854744fa49943e495c11048b6b216f6 : SUCCESS in 3m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f83fa65a978e4b309fdcf5e80b0a54b8 : SUCCESS in 3m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/670b51cd5c5346548c8f1cc1a52d397f : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/853bac9574714022bdd5450179f7ffe5 : SUCCESS in 3m 25s\n- cliff-tox-py312-neutronclient-tip https://zuul.opendev.org/t/openstack/build/6650448511134bc090f884fff13d0224 : SUCCESS in 3m 11s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/3fc5c2e057ca43e481506b7d2f6be3b7 : SUCCESS in 7m 03s","accounts_in_message":[],"_revision_number":1},{"id":"fa645cef4b1352e3f197624f4d986cb37b1c68a9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 18:28:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7b438d7ac59d589adb1637413f19407ace43b7ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 21:10:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e908680bde9c4de984cdc04359793a38\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e12e8ac0fcff4de79e7387c40045f5f3 : SUCCESS in 1h 45m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b0e64ad7c0b4de1942e803c3e8431b8 : SUCCESS in 3m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f1a82d8dc39b4bc1b635f6ab8a824bf6 : SUCCESS in 2m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/99737b6b0bfe4b458e8da22c1e3c471a : SUCCESS in 3m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47fac521d53446fe84f12f444a5052b6 : SUCCESS in 3m 07s\n- cliff-tox-py312-neutronclient-tip https://zuul.opendev.org/t/openstack/build/fe31af60224446188a5227aa2b9d68c3 : SUCCESS in 5m 10s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/a6df777c09034c249b3482b1ed318f87 : SUCCESS in 5m 30s","accounts_in_message":[],"_revision_number":1},{"id":"66c5c782f388fb581e0f7cfa2cbb94d1d0368895","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 21:10:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3a851069cc2db07cca75e94fc61263e4ddc8658f","revisions":{"3a851069cc2db07cca75e94fc61263e4ddc8658f":{"kind":"REWORK","_number":1,"created":"2025-02-19 20:12:57.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/49/942249/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/49/942249/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/49/942249/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/49/942249/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/49/942249/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/49/942249/1"}}},"commit":{"parents":[{"commit":"8ade2ee41f0686cfdc0d4ed2c218fee7df0494b1","subject":"typing: Add typing to cliff.complete","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/8ade2ee41f0686cfdc0d4ed2c218fee7df0494b1"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-02-19 20:08:23.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-02-19 20:08:23.000000000","tz":0},"subject":"typing: Add typing to cliff.help","message":"typing: Add typing to cliff.help\n\nAgain, what is left of it.\n\nChange-Id: I9ccfcd8fde009e930d97f9c7ae7dde5067e93fe2\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/3a851069cc2db07cca75e94fc61263e4ddc8658f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/3a851069cc2db07cca75e94fc61263e4ddc8658f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
