)]}'
{"id":"openstack%2Fcliff~942252","triplet_id":"openstack%2Fcliff~master~Idf39c04cabef65683bc621e445e902421e5d60e8","project":"openstack/cliff","branch":"master","topic":"typing","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-02-20 16:20:09.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"hashtags":[],"change_id":"Idf39c04cabef65683bc621e445e902421e5d60e8","subject":"typing: Add typing to cliff.formatters","status":"MERGED","created":"2025-02-19 20:12:57.000000000","updated":"2025-02-20 21:10:57.000000000","submitted":"2025-02-20 21:10:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"942252-typing","meta_rev_id":"23558a7342aefba01849c566c62ce90362606dab","_number":942252,"virtual_id_number":942252,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-20 21:10:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2025-02-20 10:29:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2025-02-20 10:29:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-19 22:42:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"699da5e9f5b50e1aef69490eab8c9f40f15efbd2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-02-19 20:12:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a855b3108535ba7f0084992747f9662050d85593","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-19 22:42:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ade31b4abfbf4c199aad2debc7f445c2\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/82184bd044e64d6396e99c0c03b27556 : SUCCESS in 2h 02m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99f9f59d2cb248858b4b1ebee59b4510 : SUCCESS in 5m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7d62edb779bd48bc870d0fa05e116e87 : SUCCESS in 3m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/01a9972a73d046c1830867b0aad574c4 : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c735b4e1ddb4fb5bdc9c43b68136db9 : SUCCESS in 5m 29s\n- cliff-tox-py312-neutronclient-tip https://zuul.opendev.org/t/openstack/build/c723f913989b40b28c5d0e2e65c325ed : SUCCESS in 2m 52s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/2a985a9e8b864b1e90885adb642f88ef : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":1},{"id":"13c0db9f0b66989be73b7b0453ab373015de3cff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-02-20 10:29:29.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"699314ed0d168d18165cff47eba126f13b258e80","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-02-20 10:29:40.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3e4735bba92f4f9abba7f1253a1e795f21ab1679","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 12:30:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b133cc667e240ec9dae180a766adff4\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6bb68bd48bce48e6ac7284bf9bf2ca14 : SUCCESS in 1h 44m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbc54def6f944cc688fdd7f59acc927d : SUCCESS in 2m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f31ad26fcc314002a7a0ac6afe542437 : SUCCESS in 3m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5923d1b096b542c6880f3a008ef21b3a : SUCCESS in 2m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f44529c212a2449a9a612df67b1c4d10 : SUCCESS in 3m 08s\n- cliff-tox-py312-neutronclient-tip https://zuul.opendev.org/t/openstack/build/b35eb86913604f489431b4dd2e1b5a46 : SUCCESS in 3m 37s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/79373a05647943b180b9bfeb73a0cb83 : SUCCESS in 6m 48s","accounts_in_message":[],"_revision_number":2},{"id":"43c5f56a117809696312bd70c98abe0914195127","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 13:03:52.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/942251 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"3234e6cf96edc0f9f56c53277602054d0fe6c584","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-02-20 16:20:09.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4309216ed52b3ac686005e13a358e8330ccef2f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 18:25:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/afb05887797a4f13be27ebce83361a0e\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1fc5d45deca44372a6b9947a7cca0257 : SUCCESS in 1h 49m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c50f4f57312429f81ab236260ee587f : SUCCESS in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d41f691b076a430eaa5cf7447f2d7c27 : SUCCESS in 4m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/93432d10d1014b60b42148621ff942be : SUCCESS in 3m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7823a5ddd34349a787fcecfeced12250 : SUCCESS in 2m 59s\n- cliff-tox-py312-neutronclient-tip https://zuul.opendev.org/t/openstack/build/208f4136b59444729aacd16bbc3232fa : SUCCESS in 5m 06s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/da3d900b470740779f45706c862fdcb1 : SUCCESS in 7m 24s","accounts_in_message":[],"_revision_number":2},{"id":"eb5b545532de5c41616aea44faffc2dc74e7878b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 18:35:29.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c559533df936e39d3a85dc5cad462eaf634789f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 21:10:57.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0df9cf9dbbd4aa48b087ce70b360c94\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5325c8bdb06a4c458ed66301084cf4a1 : SUCCESS in 1h 05m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a29ec639f9824dcd9550faa34e88c0e9 : SUCCESS in 3m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/273e546a5727456f87ca9c0081b5e387 : SUCCESS in 3m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b60ce71045f540a0a8952eb31c0ed655 : SUCCESS in 3m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c989d4625e04b8d8699687905c106ac : SUCCESS in 2m 59s\n- cliff-tox-py312-neutronclient-tip https://zuul.opendev.org/t/openstack/build/c4f82ec4d6a64b858e7e70dc92cf6959 : SUCCESS in 2m 38s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/15ba6d91945c4777b295c2e9b4610771 : SUCCESS in 5m 17s","accounts_in_message":[],"_revision_number":2},{"id":"23558a7342aefba01849c566c62ce90362606dab","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 21:10:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"348c774ca21f37c9bc7d61e184c53cf8a51eb8d5","revisions":{"852d7c8b425fbe9e5a2c839ef9e542206805a599":{"kind":"REWORK","_number":1,"created":"2025-02-19 20:12:57.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/52/942252/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/52/942252/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/52/942252/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/52/942252/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/52/942252/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/52/942252/1"}}},"commit":{"parents":[{"commit":"74179e610062b7f4d97db5238e4c5fa84416ebdb","subject":"typing: Add typing to cliff.sphinxext","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/74179e610062b7f4d97db5238e4c5fa84416ebdb"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-02-19 20:08:23.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-02-19 20:08:23.000000000","tz":0},"subject":"typing: Add typing to cliff.formatters","message":"typing: Add typing to cliff.formatters\n\nThis marks the completion of the typing effort so we can add a py.typed\nfile.\n\nChange-Id: Idf39c04cabef65683bc621e445e902421e5d60e8\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/852d7c8b425fbe9e5a2c839ef9e542206805a599"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/852d7c8b425fbe9e5a2c839ef9e542206805a599"}]},"branch":"refs/heads/master"},"348c774ca21f37c9bc7d61e184c53cf8a51eb8d5":{"kind":"REWORK","_number":2,"created":"2025-02-20 10:29:29.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/52/942252/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/52/942252/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/52/942252/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/52/942252/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/52/942252/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/52/942252/2"}}},"commit":{"parents":[{"commit":"74179e610062b7f4d97db5238e4c5fa84416ebdb","subject":"typing: Add typing to cliff.sphinxext","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/74179e610062b7f4d97db5238e4c5fa84416ebdb"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-02-19 20:08:23.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-02-20 10:29:21.000000000","tz":0},"subject":"typing: Add typing to cliff.formatters","message":"typing: Add typing to cliff.formatters\n\nThis marks the completion of the typing effort so we can add a py.typed\nfile.\n\nChange-Id: Idf39c04cabef65683bc621e445e902421e5d60e8\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/348c774ca21f37c9bc7d61e184c53cf8a51eb8d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/348c774ca21f37c9bc7d61e184c53cf8a51eb8d5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
