)]}'
{"id":"openstack%2Fcliff~942253","triplet_id":"openstack%2Fcliff~master~I7406be7cf5d958e86107a6928f8f30facb6efab8","project":"openstack/cliff","branch":"master","topic":"typing","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-02-20 16:20:12.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"hashtags":[],"change_id":"I7406be7cf5d958e86107a6928f8f30facb6efab8","subject":"typing: Enable basic typing for tests","status":"MERGED","created":"2025-02-19 20:12:57.000000000","updated":"2025-02-20 21:12:37.000000000","submitted":"2025-02-20 21:10:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"942253-typing","meta_rev_id":"a3f293fa269e7e5bee9be97f89d25b074a6f2033","_number":942253,"virtual_id_number":942253,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-20 21:10:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2025-02-20 11:26:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2025-02-20 11:26:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-19 22:15:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"759fcd3dca9ca5a1e74a045c10c94cc6ca92970f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-02-19 20:12:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c36786a29d86adaa43e00adba8ca10a14253a574","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-19 22:15:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7931c06272e349eaabc1cca2e7c57ab9\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5bfb4cc97df54ce58c8f0f5ff4b130bf : SUCCESS in 1h 53m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77e304e5e27c4689b760929d6d31eb37 : FAILURE in 3m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d7f311045e2746b49b552bc0fa4c9fb7 : SUCCESS in 3m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0447137e484b41b59abd16d9264badc1 : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31da0281e1c549a9ae9c2d07efd70397 : SUCCESS in 3m 21s\n- cliff-tox-py312-neutronclient-tip https://zuul.opendev.org/t/openstack/build/c783555e36da4400921a36692deac4fa : SUCCESS in 3m 45s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/c7a844cda5b84e9b825522a835e1b713 : SUCCESS in 5m 55s\n\nWarning:\n  Comments left for invalid file cliff/tests/base.py\n  Comments left for invalid file cliff/tests/test_app.py\n  Comments left for invalid file cliff/tests/test_columns.py\n  Comments left for invalid file cliff/tests/test_command.py\n  Comments left for invalid file cliff/tests/test_command_hooks.py\n  Comments left for invalid file cliff/tests/test_commandmanager.py\n  Comments left for invalid file cliff/tests/test_complete.py\n  Comments left for invalid file cliff/tests/test_formatters_table.py\n  Comments left for invalid file cliff/tests/test_formatters_value.py\n  Comments left for invalid file cliff/tests/test_interactive.py\n  Comments left for invalid file cliff/tests/test_lister.py\n  Comments left for invalid file cliff/tests/test_show.py","accounts_in_message":[],"_revision_number":1},{"id":"681bc1d0a7b9022903835b3464049edb0450c9b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-02-20 10:29:29.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"7ee6180ebce28d638c766541cc8adced6ac0c373","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-02-20 11:26:35.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"aed9e3568f05f3164437894ab6d29b88d0b702b8","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-02-20 11:26:54.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"f07179fde435f3d1e9b4bb970ecf950a06c9a9ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 12:32:30.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85fe9e55ad45414aa096fa5efda20ae3\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/798cce2d3cd44e219960645b628a0b6b : SUCCESS in 1h 03m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2b68ed293cc4e3abd161879d1b732ef : SUCCESS in 2m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/067208a0bc524730a82aadcdddf9127b : SUCCESS in 3m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/190b24441f144b278b72f20468a7bb55 : SUCCESS in 3m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/695329859a734d2896a3effbeb962a8d : SUCCESS in 3m 42s\n- cliff-tox-py312-neutronclient-tip https://zuul.opendev.org/t/openstack/build/e260b8917f3146629595e86538295112 : SUCCESS in 4m 10s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/ee33bc311237489ab9193ea77aa9f459 : SUCCESS in 5m 27s","accounts_in_message":[],"_revision_number":3},{"id":"43c4b4487cbdbfbb5768af2414975667b7271b42","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 13:03:52.000000000","message":"Patch Set 3: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/942252 is needed.","accounts_in_message":[],"_revision_number":3},{"id":"2dbb305f99e9f523a204af9ffa7812ec71371b42","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-02-20 16:20:12.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"df88a605127e292776cff0ee5ffbdcf927cd754a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 18:25:33.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f819a37471047ce9c482fb8688a8193\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/485fd94cd3d148c290e8a52aaae735f5 : SUCCESS in 1h 48m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfd311d5bba64d7b974db7bd4f88af12 : SUCCESS in 3m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7407612beb0b4bcc91c79bfa6bd4a825 : SUCCESS in 4m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c0282a4008d642d483a935e5ba3c2fe9 : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e97ed9a82c9495091eb729c661ea04d : SUCCESS in 4m 34s\n- cliff-tox-py312-neutronclient-tip https://zuul.opendev.org/t/openstack/build/bb679d461cd447ef9f076da937abcff4 : SUCCESS in 3m 56s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/356210b8d0fd4fc496c14b885ce23966 : SUCCESS in 5m 38s","accounts_in_message":[],"_revision_number":3},{"id":"e7b9ad799fe2243ed4b1898b2ac722f7d9c3e62b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 18:35:29.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"2eb1792f8596d595ff08804d1f1a1d0656056cb8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 21:10:59.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c006cf294816485eb16b01b8239442f9\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c3aba9428f6342feadc3e099edef4fad : SUCCESS in 1h 43m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90f7038044524659a8b4fe7f7b7c79c5 : SUCCESS in 2m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/00a4291b2b0540128ec902e7ab9a4749 : SUCCESS in 3m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/067dcaa3bac54e799bfad5b8a44f1b9c : SUCCESS in 3m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2b21b66052a4c9e9e50f5718815c541 : SUCCESS in 2m 51s\n- cliff-tox-py312-neutronclient-tip https://zuul.opendev.org/t/openstack/build/15272dbcb4a9404eba30d05f0bad87ae : SUCCESS in 4m 54s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/d7dc17ddebc84474aca8d8cdaf057fb3 : SUCCESS in 6m 53s","accounts_in_message":[],"_revision_number":3},{"id":"9c92c559a1937235a91637c4bd8550b460569a83","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 21:10:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"a3f293fa269e7e5bee9be97f89d25b074a6f2033","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 21:12:37.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db17cd72cbfe41438a02ffeb648f2f0d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2c81e0af3ed4221a214b65f20f2f26d : SUCCESS in 33s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"71b1e9fa028f1df6b8282eb55646fa790176f2ec","revisions":{"62c1c1de132ab9f491f81c3af626925e01f2dc57":{"kind":"REWORK","_number":1,"created":"2025-02-19 20:12:57.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/53/942253/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/53/942253/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/53/942253/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/53/942253/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/53/942253/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/53/942253/1"}}},"commit":{"parents":[{"commit":"852d7c8b425fbe9e5a2c839ef9e542206805a599","subject":"typing: Add typing to cliff.formatters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/852d7c8b425fbe9e5a2c839ef9e542206805a599"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-02-19 20:08:23.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-02-19 20:08:23.000000000","tz":0},"subject":"WIP: typing: Enable basic typing for tests","message":"WIP: typing: Enable basic typing for tests\n\nMost of these are false positives but they\u0027re a hassle to fix.\n\nChange-Id: I7406be7cf5d958e86107a6928f8f30facb6efab8\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/62c1c1de132ab9f491f81c3af626925e01f2dc57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/62c1c1de132ab9f491f81c3af626925e01f2dc57"}]},"branch":"refs/heads/master"},"e706c5b51abe37cc53f3b8f6520a0d39e6db68cf":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-02-20 10:29:29.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/53/942253/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/53/942253/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/53/942253/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/53/942253/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/53/942253/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/53/942253/2"}}},"commit":{"parents":[{"commit":"348c774ca21f37c9bc7d61e184c53cf8a51eb8d5","subject":"typing: Add typing to cliff.formatters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/348c774ca21f37c9bc7d61e184c53cf8a51eb8d5"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-02-19 20:08:23.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-02-20 10:29:25.000000000","tz":0},"subject":"WIP: typing: Enable basic typing for tests","message":"WIP: typing: Enable basic typing for tests\n\nMost of these are false positives but they\u0027re a hassle to fix.\n\nChange-Id: I7406be7cf5d958e86107a6928f8f30facb6efab8\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/e706c5b51abe37cc53f3b8f6520a0d39e6db68cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/e706c5b51abe37cc53f3b8f6520a0d39e6db68cf"}]},"branch":"refs/heads/master"},"71b1e9fa028f1df6b8282eb55646fa790176f2ec":{"kind":"REWORK","_number":3,"created":"2025-02-20 11:26:35.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/53/942253/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/53/942253/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/53/942253/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/53/942253/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/53/942253/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/53/942253/3"}}},"commit":{"parents":[{"commit":"348c774ca21f37c9bc7d61e184c53cf8a51eb8d5","subject":"typing: Add typing to cliff.formatters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/348c774ca21f37c9bc7d61e184c53cf8a51eb8d5"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-02-19 20:08:23.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-02-20 11:26:05.000000000","tz":0},"subject":"typing: Enable basic typing for tests","message":"typing: Enable basic typing for tests\n\nWe don\u0027t insist on fully typed signatures cos who\u0027s got time for that.\n\nChange-Id: I7406be7cf5d958e86107a6928f8f30facb6efab8\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/71b1e9fa028f1df6b8282eb55646fa790176f2ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/71b1e9fa028f1df6b8282eb55646fa790176f2ec"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
