)]}'
{"id":"openstack%2Fcliff~953543","triplet_id":"openstack%2Fcliff~master~I4153d81027c01a692997bd339c956b6854c42c92","project":"openstack/cliff","branch":"master","topic":"pyproject","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-06-27 17:11:10.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"hashtags":[],"change_id":"I4153d81027c01a692997bd339c956b6854c42c92","subject":"Migrate setup configuration to pyproject.toml","status":"MERGED","created":"2025-06-27 08:26:55.000000000","updated":"2025-06-27 21:28:37.000000000","submitted":"2025-06-27 20:03:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"953543-pyproject","meta_rev_id":"7d9b3ffc8693f05ec828f1c0125d4aa31840fb4a","_number":953543,"virtual_id_number":953543,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-27 20:03:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2025-06-27 17:11:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2025-06-27 21:28:37.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2025-06-27 17:11:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-27 09:52:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-27 21:28:37.000000000","updated_by":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"reviewer":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"state":"REVIEWER"}],"messages":[{"id":"57ad1bd91e128dc27b0721160bd5badeb8bee1d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-06-27 08:26:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b1ed0d2beaaedbbee68ce704ebeb8941b1e31164","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 09:52:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f8379f66764b44d0aa8fd7924da9494b\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3243051e4f4142c3af8992f4724c1bcb : SUCCESS in 2m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d39b6879f4d44719b0f05f2d15c59dab : SUCCESS in 1h 24m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2fcca3f3349429992c568ad8bce4d59 : SUCCESS in 4m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/02d7c4ac2d0b4857a37c087b0e77c8d8 : SUCCESS in 1m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0641fe3eaa8a4be6849be2353fed3faa : SUCCESS in 3m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/01c84a54bc8043b181f3d048fac8288b : SUCCESS in 4m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9fd6d07640f4f0da4287c621260abd4 : FAILURE in 3m 12s\n- cliff-tox-py312-neutronclient-tip https://zuul.opendev.org/t/openstack/build/edcbe96db78e46ea951ce950be0a2e92 : SUCCESS in 3m 52s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/e7b8b2ccc12a41da88001c25e806d841 : SUCCESS in 2m 23s","accounts_in_message":[],"_revision_number":1},{"id":"79fcf00842846b12b8a913f6337f4b1a2cb67b26","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-06-27 17:05:40.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"02ef5c82e48ff4dc5f0d729a594f44e24a783101","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-06-27 17:11:10.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c8818b6dec093360d850f1190b251fa570fdb56d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 18:50:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/861c338fd6684306891ae8c511678e45\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/499e1a4a3df241e78409fdeb59433954 : SUCCESS in 3m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3dd6906cca7e4178a69224c9e37b7212 : SUCCESS in 1h 37m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd78380c4e774d1384d40f471810e5ab : SUCCESS in 2m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dab0a375c9674dc8ad2df4d1dea41b80 : SUCCESS in 4m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ca9cfaa3224a4b19a93732331cbbf8b3 : SUCCESS in 2m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/917561086e2e485dbda328818bf5151f : SUCCESS in 3m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/099931a0a00f4aaa9b84c49517decb19 : SUCCESS in 2m 20s\n- cliff-tox-py312-neutronclient-tip https://zuul.opendev.org/t/openstack/build/692c50845b594494adee40fdba436c41 : SUCCESS in 2m 21s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/aa883d38acdf4b988bf65f5f30f1fcf7 : SUCCESS in 5m 39s","accounts_in_message":[],"_revision_number":2},{"id":"00a0c53f6371cd973ac40aa940b8c9f4337b9a88","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 18:50:41.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f603140280eae0cd7e46e6834cf846cddf9ce7d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 20:03:02.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a7b7c49185e349cb98ccab1029d8e771\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6a1c9931610a4427ab374f210c96d53b : SUCCESS in 2m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/35b9a31fe7b34e9d979d0b8d8b973be2 : SUCCESS in 1h 11m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23a9ac19291643619fd36d31d4b0f940 : SUCCESS in 3m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/20cbf70fed2d47848c6e4667500f5b93 : SUCCESS in 2m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/43b6db3d2aa6439f91303d260a0ef0da : SUCCESS in 2m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ac375e4d0654ffb925f001107fe136f : SUCCESS in 2m 43s\n- cliff-tox-py312-neutronclient-tip https://zuul.opendev.org/t/openstack/build/70e54f83d56545fdae81b0448f86fa2c : SUCCESS in 3m 01s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/35919e808472452f9d110f4eda81b5a6 : SUCCESS in 4m 20s","accounts_in_message":[],"_revision_number":2},{"id":"15fb7d30bd92959993fa9bf98238972d417a3c89","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 20:03:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"d65953db56eb2f7b11e90f99e60c23c2c263813c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 20:03:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba49ee3fa31044c1a96f6860802ea963\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21c12b29a2b84493a10f58e45700cc96 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":2},{"id":"7d9b3ffc8693f05ec828f1c0125d4aa31840fb4a","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2025-06-27 21:28:37.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"04285012e5d5ef0fee109cdb017856ccae04d0f2","revisions":{"2040eb667e97041b9c4a473fef1395dadbc0ae9b":{"kind":"REWORK","_number":1,"created":"2025-06-27 08:26:55.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/43/953543/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/43/953543/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/43/953543/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/43/953543/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/43/953543/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/43/953543/1"}}},"commit":{"parents":[{"commit":"af17b570351c6d1d9ce2a2a8879ba0d93d07ca7d","subject":"Bump Python version used for linters to 3.10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/af17b570351c6d1d9ce2a2a8879ba0d93d07ca7d"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-06-27 08:26:10.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-06-27 08:26:10.000000000","tz":60},"subject":"Migrate setup configuration to pyproject.toml","message":"Migrate setup configuration to pyproject.toml\n\nChange-Id: I4153d81027c01a692997bd339c956b6854c42c92\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/2040eb667e97041b9c4a473fef1395dadbc0ae9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/2040eb667e97041b9c4a473fef1395dadbc0ae9b"}]},"branch":"refs/heads/master"},"04285012e5d5ef0fee109cdb017856ccae04d0f2":{"kind":"REWORK","_number":2,"created":"2025-06-27 17:05:40.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/43/953543/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/43/953543/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/43/953543/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/43/953543/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/43/953543/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/43/953543/2"}}},"commit":{"parents":[{"commit":"af17b570351c6d1d9ce2a2a8879ba0d93d07ca7d","subject":"Bump Python version used for linters to 3.10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/af17b570351c6d1d9ce2a2a8879ba0d93d07ca7d"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-06-27 08:26:10.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-06-27 17:05:40.000000000","tz":0},"subject":"Migrate setup configuration to pyproject.toml","message":"Migrate setup configuration to pyproject.toml\n\nChange-Id: I4153d81027c01a692997bd339c956b6854c42c92\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/04285012e5d5ef0fee109cdb017856ccae04d0f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/04285012e5d5ef0fee109cdb017856ccae04d0f2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
