)]}'
{"id":"openstack%2Fcliff~957350","triplet_id":"openstack%2Fcliff~master~I30117478b6e841f318b54b2e439bfd0135cd65d8","project":"openstack/cliff","branch":"master","topic":"release-health-check-flamingo","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2025-08-14 14:31:57.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I30117478b6e841f318b54b2e439bfd0135cd65d8","subject":"DNM: gate health test","status":"ABANDONED","created":"2025-08-14 09:39:21.000000000","updated":"2025-08-14 14:31:57.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ff2115b9e83ff026058fdb8686051a5386455179","_number":957350,"virtual_id_number":957350,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-08-14 12:35:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-14 12:35:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0c5df76538cdc3af03a4c9655c6be56f081ea95c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-08-14 09:39:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e42802b3a7881c003c7b706455c9a7324ccbf78f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-14 12:35:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/af3e06d7a5d0491aadce68c08f6ded1b\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0dc06dc1d8694c2da95a683253a7097c : SUCCESS in 1h 44m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f72a320bd2c43a8a0e42fdf16b03c1b : FAILURE in 1m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8f9ea72605434063a253b8552715ddeb : SUCCESS in 2m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d282b57cb4d34ba5aceb08895c445ee9 : SUCCESS in 3m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/84df747d06f04782be9f5fe3a1174491 : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/838f52b4a7b0462594f6146df6bf111a : SUCCESS in 2m 54s\n- cliff-tox-py312-neutronclient-tip https://zuul.opendev.org/t/openstack/build/bb67b15720234b14b431cf2c8427f730 : SUCCESS in 1m 57s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/19cf48ca20f44bf0897399d4ca1d15b6 : SUCCESS in 3m 55s","accounts_in_message":[],"_revision_number":1},{"id":"ff2115b9e83ff026058fdb8686051a5386455179","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-08-14 14:31:57.000000000","message":"Abandoned\n\nthe only job failure is pep8, due to the generated comment, otherwise the gate jobs are passing, so i think patch can be abandoned.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9fdd95a9c5cce0fd55302c99cd2e242260589b51","revisions":{"9fdd95a9c5cce0fd55302c99cd2e242260589b51":{"kind":"REWORK","_number":1,"created":"2025-08-14 09:39:21.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/50/957350/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/50/957350/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/50/957350/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/50/957350/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/50/957350/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/50/957350/1"}}},"commit":{"parents":[{"commit":"0008efa45b4ee16bd93aa1d38ca4dfe7cbcd92a0","subject":"mypy: Enable strict mode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/0008efa45b4ee16bd93aa1d38ca4dfe7cbcd92a0"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2025-08-14 09:39:11.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2025-08-14 09:39:19.000000000","tz":120},"subject":"DNM: gate health test","message":"DNM: gate health test\n\nChange-Id: I30117478b6e841f318b54b2e439bfd0135cd65d8\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/9fdd95a9c5cce0fd55302c99cd2e242260589b51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/9fdd95a9c5cce0fd55302c99cd2e242260589b51"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
