)]}'
{"id":"openstack%2Fcliff~966817","triplet_id":"openstack%2Fcliff~master~I8558b31f5d69f3cbf495e7d80f3877edee33fabc","project":"openstack/cliff","branch":"master","topic":"conflict-resolution","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-11-17 17:43:52.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"hashtags":[],"change_id":"I8558b31f5d69f3cbf495e7d80f3877edee33fabc","subject":"typing: Update for latest docutils type hints","status":"MERGED","created":"2025-11-12 13:25:58.000000000","updated":"2025-11-17 20:53:18.000000000","submitted":"2025-11-17 20:52:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"966817-conflict-resolution","meta_rev_id":"6995ce1d20c1f9d577506ba0adae342c9c61a313","_number":966817,"virtual_id_number":966817,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-17 20:52:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2025-11-17 17:43:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2025-11-17 17:43:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-12 15:28:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"548b7e1d519cf3b6f92d6fa9ad4b1f9f3155b981","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-11-12 13:25:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"090307b61dc7677fb3e9efffed47c543f04f9a72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-12 15:28:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/603e112e3e294e1ea49f4a45f7c73882\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8d2a286b767044148f620d6d4928b899 : SUCCESS in 1h 44m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0a8ccc5b83b4ba0b85d7a4dd3a0744d : SUCCESS in 2m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fc85298e070d4a948651510175d86107 : POST_FAILURE in 3m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cebf7423aaa6457b959b67c2f772d6f3 : POST_FAILURE in 3m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a3dc8d43ad13405bbed3ce425b34dfdd : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/123b4832a36c498abf1042328fe58b17 : SUCCESS in 3m 00s\n- cliff-tox-py312-neutronclient-tip https://zuul.opendev.org/t/openstack/build/359c76793c174ebfada786c839abb33b : SUCCESS in 3m 28s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/5b7cbc2b5e7d4c5bb11aeaba9d299689 : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":1},{"id":"6a1f53df3ffc20a469dba4481af931d13a487c9f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-11-12 18:41:07.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"24b90217fe744ae15d5782da652a50c16d36697f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-12 20:25:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99fadb4d06c2408fb0ed7ad6bb3e01a2\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f8ff818e7167456d997622efe88a3c86 : SUCCESS in 1h 42m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0208775b4e1e4445900899382663abb2 : SUCCESS in 2m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/09ca6144a0f342549905efbe5f9c1277 : SUCCESS in 4m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/175ac26a1ddc4252ac77e6a277958ce9 : SUCCESS in 2m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1b3b1391d4b64aa6855fd63e4450fd16 : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae3817ebe68a4aa99b4756d095b7f243 : SUCCESS in 2m 40s\n- cliff-tox-py312-neutronclient-tip https://zuul.opendev.org/t/openstack/build/8f240dcb4c4e42efa09b21229ec16438 : SUCCESS in 2m 57s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/8f1deaffed704c7ca34ccaaecda5fa1b : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":2},{"id":"70f46f519db5f6b2e72eb1aec2a49a9745b16677","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-11-17 17:43:44.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"5aedca8a344d8489b95bb759d57288d424bc0e08","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-11-17 17:43:52.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"090f262c3ba4fea689b2dd6c119e057396610aa3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 18:44:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c73edc61740f4d318f3e228cf5892560\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/386d50644ad84ca182e8377341afc938 : SUCCESS in 2m 18s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a668c437e0624679b02cb326118295aa : SUCCESS in 1m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0b1cb3264b1946429ba9e9ade584aa3c : SUCCESS in 59m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4f52c53ae3b485bad6fc9d7f3ed78dd : SUCCESS in 1m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fece66b1b832477fb214a956054da40d : SUCCESS in 3m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/41958e29519a4e25a0b10e98c1d46a87 : SUCCESS in 3m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/121966f137d54a82aa93df019efc58e6 : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df7dd2828148474b9273d62b43e62482 : SUCCESS in 3m 07s\n- cliff-tox-py312-neutronclient-tip https://zuul.opendev.org/t/openstack/build/24e884c78d274fb9bdf35217984c28aa : SUCCESS in 2m 23s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/0c1330d2b9b64462aa36f5a51c4cb7f3 : SUCCESS in 2m 46s","accounts_in_message":[],"_revision_number":3},{"id":"8fa599b98b5de050a03605c3368b59cb403f2b25","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 18:44:45.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"fe9e4d5a034e4093a543f7e55d614d42057022d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 20:52:32.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b6ad196b0c144c63a9471c1b38055e11\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0df75d4ab5044ef989e5b34b856e3aa2 : SUCCESS in 1m 56s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f8205ac60efc469198a4d3e76fba656b : SUCCESS in 1m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2a3a8cac34bc4d30b6ed5aac10cf7910 : SUCCESS in 2h 03m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a46407f1475444395df43487f878e5d : SUCCESS in 4m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ad72fc7dc91d455a9e0a4ae30d32723d : SUCCESS in 2m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8a384a10dab84b51ac11d470ed826b8e : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee3cf0ac798d43c198bb6f92b851f930 : SUCCESS in 1m 48s\n- cliff-tox-py312-neutronclient-tip https://zuul.opendev.org/t/openstack/build/af5b3d360db04882b5263b44c92a5385 : SUCCESS in 2m 55s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/ab54b7f885a4498f8c2980265e9ec266 : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":3},{"id":"6eb114921b52482beca4ed130bb705ec867e586d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 20:52:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"6995ce1d20c1f9d577506ba0adae342c9c61a313","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 20:53:18.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/061a146a75c04b46af2d1a31928a509a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eba7b4dbbc2c4acd8c3c43e89005159e : SUCCESS in 35s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"32837fdf550c621b38b71c1be95212c6fc035cdc","revisions":{"518930d009ff88fe70942a107a3ba314088bf5af":{"kind":"REWORK","_number":1,"created":"2025-11-12 13:25:58.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/17/966817/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/17/966817/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/17/966817/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/17/966817/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/17/966817/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/17/966817/1"}}},"commit":{"parents":[{"commit":"0008efa45b4ee16bd93aa1d38ca4dfe7cbcd92a0","subject":"mypy: Enable strict mode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/0008efa45b4ee16bd93aa1d38ca4dfe7cbcd92a0"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-11-07 15:39:10.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-11-12 13:24:51.000000000","tz":0},"subject":"sphinxext: Use StringList","message":"sphinxext: Use StringList\n\nThis is simpler and is what we should have been using all along.\n\nChange-Id: I8558b31f5d69f3cbf495e7d80f3877edee33fabc\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/518930d009ff88fe70942a107a3ba314088bf5af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/518930d009ff88fe70942a107a3ba314088bf5af"}]},"branch":"refs/heads/master"},"2d1e1d0fdf912ebf04ab62df4872b39c4a5f32b6":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":2,"created":"2025-11-12 18:41:07.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/17/966817/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/17/966817/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/17/966817/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/17/966817/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/17/966817/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/17/966817/2"}}},"commit":{"parents":[{"commit":"6c5062ca61d15e361ee0b01186138d8deb0117bb","subject":"Remove pkg_resources remnants","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/6c5062ca61d15e361ee0b01186138d8deb0117bb"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-11-07 15:39:10.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-11-12 18:04:29.000000000","tz":0},"subject":"typing: Update for latest docutils type hints","message":"typing: Update for latest docutils type hints\n\nChange-Id: I8558b31f5d69f3cbf495e7d80f3877edee33fabc\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/2d1e1d0fdf912ebf04ab62df4872b39c4a5f32b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/2d1e1d0fdf912ebf04ab62df4872b39c4a5f32b6"}]},"branch":"refs/heads/master"},"32837fdf550c621b38b71c1be95212c6fc035cdc":{"kind":"REWORK","_number":3,"created":"2025-11-17 17:43:44.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/17/966817/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/17/966817/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/17/966817/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/17/966817/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/17/966817/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/17/966817/3"}}},"commit":{"parents":[{"commit":"0008efa45b4ee16bd93aa1d38ca4dfe7cbcd92a0","subject":"mypy: Enable strict mode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/0008efa45b4ee16bd93aa1d38ca4dfe7cbcd92a0"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-11-07 15:39:10.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-11-17 17:40:11.000000000","tz":0},"subject":"typing: Update for latest docutils type hints","message":"typing: Update for latest docutils type hints\n\nWe also bump the pre-commit hooks while we\u0027re here.\n\nChange-Id: I8558b31f5d69f3cbf495e7d80f3877edee33fabc\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/32837fdf550c621b38b71c1be95212c6fc035cdc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/32837fdf550c621b38b71c1be95212c6fc035cdc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
