)]}'
{"id":"openstack%2Fcliff~966911","triplet_id":"openstack%2Fcliff~master~I09ae40afe9f26659d44e26a16d10f1ff4eb57d1d","project":"openstack/cliff","branch":"master","topic":"conflict-resolution","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-11-17 17:44:01.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"hashtags":[],"change_id":"I09ae40afe9f26659d44e26a16d10f1ff4eb57d1d","subject":"Remove pkg_resources remnants","status":"MERGED","created":"2025-11-12 18:41:07.000000000","updated":"2025-11-17 21:47:21.000000000","submitted":"2025-11-17 21:47:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"966911-conflict-resolution","meta_rev_id":"480fae1be76ac039eb6aa532ddf2555cf9dca191","_number":966911,"virtual_id_number":966911,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-17 21:47:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2025-11-17 17:44:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2025-11-17 17:44:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-12 20:33:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"988a6bea101d988bb84d5b24d3ef5818dd34b184","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-11-12 18:41:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cfaf690f9359e4048d77a9ef2368ce04d58d3073","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-12 20:33:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/500acdfea52d4d70bdb3000b60bb82f7\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ca4b793abaee4e5897b61d60a76082b0 : SUCCESS in 1h 46m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc4d047badcf493290c9c144021c6d5e : FAILURE in 2m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eb38354a3038497bb4424f2e176d8a0b : SUCCESS in 2m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ecfbceac318a496fb9f2c55ab789d120 : SUCCESS in 2m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f09edb370bda4b4fa3b24f7462583240 : SUCCESS in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88c7d6a2e8a0415fade3419ed8c9e9d9 : SUCCESS in 4m 07s\n- cliff-tox-py312-neutronclient-tip https://zuul.opendev.org/t/openstack/build/cd01b87962d0415a9463850642bb3e6d : SUCCESS in 3m 30s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/33b1d367f1de42528b63ff99f8ab18db : SUCCESS in 4m 18s\n\nWarning:\n  Comments left for invalid file cliff/sphinxext.py","accounts_in_message":[],"_revision_number":1},{"id":"9652b8c6c84a2682849b509cfe018f0d2a52ebb5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-11-17 17:43:44.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"07c8dd883ed05dbefad6be98b5b683ee64f2b707","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-11-17 17:44:01.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e46778c48f8c029b4b03e689e6ca615132eafdff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 19:32:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a81273145aed451f9babe0ab4141a516\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a1f2d7fcd2d54745945230593de4505b : SUCCESS in 1h 43m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef6041ccff5a4101b683ebd60bdd216b : SUCCESS in 2m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/94ec6299ae2c40308460d377581244e8 : SUCCESS in 3m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b8a2382c3b96466d9a9256f3726ee57e : SUCCESS in 3m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6ef43376571f41dbaa16e6507e90c2e9 : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40959c77b3e7444c91f8a5528527445f : SUCCESS in 2m 58s\n- cliff-tox-py312-neutronclient-tip https://zuul.opendev.org/t/openstack/build/544896052729442ab4fc84886ed181d1 : SUCCESS in 5m 03s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/060470e64a1b4c5aa1f57950c5b98426 : SUCCESS in 2m 50s","accounts_in_message":[],"_revision_number":2},{"id":"af4f50e10413e3d71a23b6c8347128661aa5a62d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 19:48:45.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6c7169aba34e95028df6419f1abc0a9b37ed4ed5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 21:47:21.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bde4fe7cb1c94c4baa18c98373bb8105\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0420244a92514c81ba4766df519063ee : SUCCESS in 1h 22m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dde909a193984b619ca771ad35324a42 : SUCCESS in 5m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1ada5f4495f0475f800d973b41271490 : SUCCESS in 3m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/97dae58d6c1846628c5dc9173339b8cf : SUCCESS in 3m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e661f90d86204ea09fd998ca073367c6 : SUCCESS in 1m 58s\n- cliff-tox-py312-neutronclient-tip https://zuul.opendev.org/t/openstack/build/9f2210dd83b04c1d98ea87b6406c55a8 : SUCCESS in 3m 32s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/9104889d7f704b0cb56099ba4fe3d90a : SUCCESS in 3m 41s","accounts_in_message":[],"_revision_number":2},{"id":"480fae1be76ac039eb6aa532ddf2555cf9dca191","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 21:47:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0833a4de945aab4238cca31f4a76dac0c897a543","revisions":{"6c5062ca61d15e361ee0b01186138d8deb0117bb":{"kind":"REWORK","_number":1,"created":"2025-11-12 18:41:07.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/11/966911/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/11/966911/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/11/966911/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/11/966911/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/11/966911/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/11/966911/1"}}},"commit":{"parents":[{"commit":"0008efa45b4ee16bd93aa1d38ca4dfe7cbcd92a0","subject":"mypy: Enable strict mode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/0008efa45b4ee16bd93aa1d38ca4dfe7cbcd92a0"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-11-12 14:19:32.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-11-12 18:04:29.000000000","tz":0},"subject":"Remove pkg_resources remnants","message":"Remove pkg_resources remnants\n\nstevedore switched its internal implementation to use importlib.metadata\ninstead of pkg_resources in v3.0.0 [1]. The EntryPoint class provided by\n\u0027importlib.metadata\u0027 only provides a \u0027load\u0027 method that takes no\narguments [2]. There is therefore no reason to check for a \u0027resolve\u0027\nmethod, nor to check for a \u0027require\u0027 argument that is not needed for\n\u0027importlib.metadata\u0027. Bump the stevedore minimum version to v3.0.0 and\ndrop this logic.\n\n[1] https://opendev.org/openstack/stevedore/commit/d5297167e08468c75d2477f15004df61cf98e57e\n[2] https://docs.python.org/3/library/importlib.metadata.html#importlib.metadata.EntryPoint\n\nChange-Id: I09ae40afe9f26659d44e26a16d10f1ff4eb57d1d\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/6c5062ca61d15e361ee0b01186138d8deb0117bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/6c5062ca61d15e361ee0b01186138d8deb0117bb"}]},"branch":"refs/heads/master"},"0833a4de945aab4238cca31f4a76dac0c897a543":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-11-17 17:43:44.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/11/966911/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/11/966911/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/11/966911/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/11/966911/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/11/966911/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/11/966911/2"}}},"commit":{"parents":[{"commit":"0d92e17bb3bd336965f4297b82b88c701bbc59b8","subject":"ruff: Enable E5 rules","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/0d92e17bb3bd336965f4297b82b88c701bbc59b8"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-11-12 14:19:32.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-11-17 17:43:34.000000000","tz":0},"subject":"Remove pkg_resources remnants","message":"Remove pkg_resources remnants\n\nstevedore switched its internal implementation to use importlib.metadata\ninstead of pkg_resources in v3.0.0 [1]. The EntryPoint class provided by\n\u0027importlib.metadata\u0027 only provides a \u0027load\u0027 method that takes no\narguments [2]. There is therefore no reason to check for a \u0027resolve\u0027\nmethod, nor to check for a \u0027require\u0027 argument that is not needed for\n\u0027importlib.metadata\u0027. Bump the stevedore minimum version to v3.0.0 and\ndrop this logic.\n\n[1] https://opendev.org/openstack/stevedore/commit/d5297167e08468c75d2477f15004df61cf98e57e\n[2] https://docs.python.org/3/library/importlib.metadata.html#importlib.metadata.EntryPoint\n\nChange-Id: I09ae40afe9f26659d44e26a16d10f1ff4eb57d1d\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/0833a4de945aab4238cca31f4a76dac0c897a543"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/0833a4de945aab4238cca31f4a76dac0c897a543"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
