)]}'
{"id":"openstack%2Fcliff~966912","triplet_id":"openstack%2Fcliff~master~Ib737311e80a7e2d408c59373f8c7171d35d46141","project":"openstack/cliff","branch":"master","topic":"conflict-resolution","hashtags":[],"change_id":"Ib737311e80a7e2d408c59373f8c7171d35d46141","subject":"typing: Clarify behavior of _run_after_hooks method","status":"MERGED","created":"2025-11-12 18:41:07.000000000","updated":"2025-11-17 21:48:46.000000000","submitted":"2025-11-17 21:47:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"966912-conflict-resolution","meta_rev_id":"4169abef6de54a5c6617eb0dbbe008b9efc4de72","_number":966912,"virtual_id_number":966912,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-17 21:47:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2025-11-17 17:44:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2025-11-17 17:44:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-12 20:25:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0911209e87fbc320a3071c657b9edf00a6330945","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-11-12 18:41:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9bea8ad4a44adbbf27979969eb583c1e7ff50a1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-12 20:25:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf0617ff8fe14e5cb2912e5e3558a167\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c2c6b82d909a44e2b8f0ccf364e25f13 : SUCCESS in 1h 39m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebe93da038484d01ad1e28fb1982a08b : SUCCESS in 3m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ee091034d353485097868d746f2970f1 : SUCCESS in 2m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d71315455e474af19a996d237229c8eb : SUCCESS in 4m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c095e88477ef4fa5a179b0de6ba987ee : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c35a3309a3934b90a1bebb98fe92f0e4 : SUCCESS in 3m 09s\n- cliff-tox-py312-neutronclient-tip https://zuul.opendev.org/t/openstack/build/fdaddf9a4dd5459e99fc97ddd4e2e39e : SUCCESS in 4m 22s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/db7c31a95db845ccae83f530e15fcbf4 : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":1},{"id":"d912a79c698ec33f811ae6030109908cf53522bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-11-17 17:43:44.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"abbab710a044e318510aab191136ba713d865555","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-11-17 17:44:05.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"bfc829ae95a20d4fa5d9ce8337cd31089304600b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 18:37:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f4c934be07a45a6a1241af949998ca4\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/65aefcc627e74543af39fa4352768d90 : SUCCESS in 52m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebef9f52309344cab12fd7e31ffe7155 : SUCCESS in 4m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1008d501ec4e4d299503f4eb807b6ce1 : SUCCESS in 1m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c2e525359b6b4f3d88ee3b0298855ce7 : SUCCESS in 1m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/547153d054094599ac9c60ed2786260c : SUCCESS in 4m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7c37e8ace2a4cebbec190bfaa358d35 : SUCCESS in 4m 25s\n- cliff-tox-py312-neutronclient-tip https://zuul.opendev.org/t/openstack/build/d0a678fd48f446adbf1399e7ef4688d4 : SUCCESS in 2m 08s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/a9e1d871e4644f398cbda8a4dbbdfeff : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":2},{"id":"2d1fe938d1824b0e5296af159a23694487925454","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 19:48:46.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"945e2032a564d50bbc45db68c1440b16ab7d027c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 21:47:22.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8fa43a19659b4472851bda2d6937b5aa\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dbfde22cdcea40279c86fd79343863ec : SUCCESS in 1h 00m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c222ec48295c4e2ba6d9201e36858330 : SUCCESS in 1m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e06282ff54454316b17255259941c1fe : SUCCESS in 2m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e8db54807bd54571b70f11cb9eaf0528 : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/301a1a076acf48fea08bddf83b9c5d6e : SUCCESS in 3m 15s\n- cliff-tox-py312-neutronclient-tip https://zuul.opendev.org/t/openstack/build/5abf82a52a40489092c55880c208276b : SUCCESS in 6m 05s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/96ef9b2000824f98bb02145cc93c187f : SUCCESS in 7m 16s","accounts_in_message":[],"_revision_number":2},{"id":"136536c6bb9f5fe0217029063af78f5f11295a48","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 21:47:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"4169abef6de54a5c6617eb0dbbe008b9efc4de72","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 21:48:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b61eac8d29e4fe9b62c12a1e9e143a3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d2a0df53a9247b58b228b8db22101de : SUCCESS in 33s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"12c21fa7ef2e67f273a7394e5e03fd94a6c79b92","revisions":{"9dca7a21d98de705ff0731399618a955f6192533":{"kind":"REWORK","_number":1,"created":"2025-11-12 18:41:07.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/12/966912/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/12/966912/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/12/966912/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/12/966912/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/12/966912/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/12/966912/1"}}},"commit":{"parents":[{"commit":"2d1e1d0fdf912ebf04ab62df4872b39c4a5f32b6","subject":"typing: Update for latest docutils type hints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/2d1e1d0fdf912ebf04ab62df4872b39c4a5f32b6"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-11-12 18:02:59.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-11-12 18:04:29.000000000","tz":0},"subject":"typing: Clarify behavior of _run_after_hooks method","message":"typing: Clarify behavior of _run_after_hooks method\n\nWe need to override this for displayable commands.\n\nChange-Id: Ib737311e80a7e2d408c59373f8c7171d35d46141\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/9dca7a21d98de705ff0731399618a955f6192533"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/9dca7a21d98de705ff0731399618a955f6192533"}]},"branch":"refs/heads/master"},"12c21fa7ef2e67f273a7394e5e03fd94a6c79b92":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-11-17 17:43:44.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/12/966912/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/12/966912/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/12/966912/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/12/966912/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/12/966912/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/12/966912/2"}}},"commit":{"parents":[{"commit":"0833a4de945aab4238cca31f4a76dac0c897a543","subject":"Remove pkg_resources remnants","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/0833a4de945aab4238cca31f4a76dac0c897a543"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-11-12 18:02:59.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-11-17 17:43:34.000000000","tz":0},"subject":"typing: Clarify behavior of _run_after_hooks method","message":"typing: Clarify behavior of _run_after_hooks method\n\nWe need to override this for displayable commands.\n\nChange-Id: Ib737311e80a7e2d408c59373f8c7171d35d46141\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/12c21fa7ef2e67f273a7394e5e03fd94a6c79b92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/12c21fa7ef2e67f273a7394e5e03fd94a6c79b92"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
